)]}'
{"id":"openstack%2Fkolla~904206","triplet_id":"openstack%2Fkolla~master~I6a56639d9b61944cf2ea5efd1063ebd3c8950113","project":"openstack/kolla","branch":"master","topic":"bump-gnocchi","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-12-27 16:33:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a56639d9b61944cf2ea5efd1063ebd3c8950113","subject":"Bump gnocchi version to 4.6.1","status":"MERGED","created":"2023-12-21 12:55:59.000000000","updated":"2023-12-27 16:34:23.000000000","submitted":"2023-12-27 16:33:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904206-bump-gnocchi","meta_rev_id":"ce60e02b97c6f268cceb2cd62f1837951e488d3c","_number":904206,"virtual_id_number":904206,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-27 16:33:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-12-21 14:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-12-21 13:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2023-12-27 14:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2023-12-27 14:51:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-12-21 15:48:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-21 13:16:30.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2023-12-21 14:16:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-21 14:35:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-12-21 14:44:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-27 14:51:15.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"6cc1b31b54040dd047d37b90952d0f76971a2d3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-21 12:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"331b294565b4b118178be4fc2e2e40dcbd8bbe40","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-12-21 13:16:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a8b47abf19315c87d6fbf10c744cadf42b64c06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 14:16:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f710828110a94a32a93bba33f5f298ce\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b0ab725c1cbc48c19d5645616f41730d : SUCCESS in 25m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5d950bbae2054da1b2e810da7d45b2b1 : SUCCESS in 19m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3dac89ab7ec64bdf85b0a557051a1bd2 : SUCCESS in 1h 19m 12s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a67d755b9beb49e198df8b9a3202cbce : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eaba662d929d47b8be5148f5d90da261 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f5dc89c72c25446784aae99c1b50d3b6 : SUCCESS in 4m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/77677c20bbf9486bbf02d44e44076e1f : SUCCESS in 25m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a1b6804f60104db7a62b6b07634a0de7 : SUCCESS in 54m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15ea7b99a2bd3eff2383bf0f3f49f90f53bdcaf7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-21 14:35:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"644ea7c1511ab2d5c93ed6995eb03368c082ba70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 14:44:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b38a37326ef042d895bad9653c4c5381\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/842e65bada7b4972880b7ab25bf070f2 : SUCCESS in 18m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b7f1dc0d405e41bea09be5f842797b43 : SUCCESS in 55m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e924222eba4a4d21ab24cff933056dcc : SUCCESS in 21m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dea806a352364f3db9ad856167d19c3c : SUCCESS in 36m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/514f8ca7a7584574bdd4b2dd09c1e15c : SUCCESS in 1h 10m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/648e33fb953a49fd8063ebc40c0a3914 : SUCCESS in 1h 41m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d633ec5b8c4f758157320beb32f74e : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39c5203ff00148ccbcf65ceee62367ed : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34f02f582c6f41af88d32a687420a7d6 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b1a801db49544938c3456077cace46e : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df448ec0ec00478aac231a7c7a69f975 : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cf933cd3b2f14367af46770a5d83ea64 : SUCCESS in 24m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ecda3005ef744ce29a1e23025f4a506a : SUCCESS in 24m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4abb2d2d2da6414e908eac5f23831d5f : SUCCESS in 1h 19m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fe2561ce75424bb98fd36216efe71e0e : SUCCESS in 1h 42m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/41a1e1d0ffdd40e797d22d4646305ef9 : SUCCESS in 22m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6c86113087e8460595784bdf9f78916c : SUCCESS in 1h 15m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fc2eaa022f0a4bdd85d5b3c52d5ea503 : SUCCESS in 52m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6b1b1557a3e248c680ff96384728fa68 : SUCCESS in 1h 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"276729b5ed336cff81340c8e68237e04c27e3d92","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-21 15:48:47.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9e48db5a9ddde5968dd1d72a9c0bb38b44ecca4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-12-27 14:51:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b227ecf29e2cae5f57ffa716c5e08af9dd906b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 14:51:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d7fdbdb46d0ee168e29ad7e37af9198a1db2a74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 16:33:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e78b4d0226442bb45d0a3cfc02e111\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/29ff2ee0b45b4fe38a5e7a9ecf587b69 : SUCCESS in 15m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5063dd1ce7db44c4b5f51eeef211e13d : SUCCESS in 29m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/32ffd2ae67d6489fb8b8adcf04cb1945 : SUCCESS in 1h 02m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b6f37bf4d64e41f8a1fcdb9c001a024b : SUCCESS in 1h 36m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ec3ab518d045cf8ea8531b0531063a : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/799cb43238174282b60028de9e3f1fce : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09034db6acab464a9257e240eac16e6b : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32dc86987f154628b6a5c33a6d449bb8 : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6cafa31dbbb24f63a5a817e6df925345 : SUCCESS in 19m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8261436b0b0a4004ab975a70998a459c : SUCCESS in 27m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9f993ae139f6484aa3db6228c5670d33 : SUCCESS in 1h 14m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ba0c588c543147b299e0369d03d660e3 : SUCCESS in 1h 37m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5886a59a145940a6841921658f44f6f7 : SUCCESS in 20m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cdfcfeba859d4c6ab858dae34c3d8a10 : SUCCESS in 1h 04m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e98ad5cb35af4005a98d6dfdb74360a9 : SUCCESS in 53m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e6711bb403174fe5946d6416982939a7 : SUCCESS in 1h 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e743fe9d847e21aba73aeb4556c3dee9eeeb9f14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 16:33:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ce60e02b97c6f268cceb2cd62f1837951e488d3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 16:34:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e723f5c3f61e40d49033e06a682fa42b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf8ae07f2aa4561b96422c2ceb2e3c3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7c6ed79bfe874b8f9b7b70f3c3c3a6942d2bb79","revisions":{"b7c6ed79bfe874b8f9b7b70f3c3c3a6942d2bb79":{"kind":"REWORK","_number":1,"created":"2023-12-21 12:55:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/06/904206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/904206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/904206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/904206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/904206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/904206/1"}}},"commit":{"parents":[{"commit":"316bafaf075ee85f555fce5e5c4df72f88e52450","subject":"CI: Use newer podman/buildah on Ubuntu Jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/316bafaf075ee85f555fce5e5c4df72f88e52450"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-12-21 12:54:50.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-12-21 12:54:50.000000000","tz":60},"subject":"Bump gnocchi version to 4.6.1","message":"Bump gnocchi version to 4.6.1\n\nGnnochi has made a new patch release, let\u0027s use it.\n\nChange-Id: I6a56639d9b61944cf2ea5efd1063ebd3c8950113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7c6ed79bfe874b8f9b7b70f3c3c3a6942d2bb79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7c6ed79bfe874b8f9b7b70f3c3c3a6942d2bb79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
