)]}'
{"id":"openstack%2Fkolla~904264","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ic64eb51325b3503a14ebab9b9ff2f4d9caec734a","project":"openstack/kolla","branch":"stable/yoga","topic":"bug/2045558-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-12-27 16:50:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic64eb51325b3503a14ebab9b9ff2f4d9caec734a","subject":"Sync only local cell in nova bootstrap \u0026 upgrade","status":"MERGED","created":"2023-12-22 07:22:24.000000000","updated":"2023-12-27 16:51:28.000000000","submitted":"2023-12-27 16:50:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904264-bug/2045558-stable/yoga","cherry_pick_of_change":902057,"cherry_pick_of_patch_set":3,"meta_rev_id":"2aa463d3bd98c906fcbe717e6fcdfa2bfd5ffb28","_number":904264,"virtual_id_number":904264,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-27 16:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-12-27 14:49:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-12-22 18:48:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-12-25 01:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-12-27 14:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-12-27 14:52:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-22 07:22:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2023-12-22 09:03:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-22 18:48:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-12-25 01:14:42.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-12-27 14:52:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f30c5a6845de65ee6ff0fdec72cefda7b95f4af3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-22 07:22:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"91617246aca8ecd98f6470819176a82ceb560701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 09:03:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8359e0b76aad4ed2bed7f19f45040c5c\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/db75ffae1524489a9d463261db235d1b : SUCCESS in 40m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/23d0dce5e1194c6fa92414b73785980d : SUCCESS in 43m 34s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/497704317fe341acbf88f97eb5a16f6f : SUCCESS in 34m 02s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/21a1635fb8f84dcea310fbac8d741c0f : SUCCESS in 1h 09m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/210deea6516e46e0b32fc47396e69d76 : SUCCESS in 1h 06m 41s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/81efdd633d54410f84669589de351a73 : SUCCESS in 53m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/23a6e855c7c040af8b6f4d854e956431 : SUCCESS in 21m 02s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/719c5356dabd45738ac9951fc7d77d65 : SUCCESS in 21m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/143009b50da4429bb52e483d67f1b3f9 : SUCCESS in 1h 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf4b202db6b4ccca0cf6aab762af9f5 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02cdc22707714cb89bf8f2cfc28d070f : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c7b2fc3a83e4b87be1edb620d4c6ac2 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17b5ae71fba48f5b0825ff198745187 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9d8ec3f5628429e98f098911b16f8cf : SUCCESS in 3m 31s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/60aaa54738544d7aa6aae291ed02a188 : SUCCESS in 25m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/7213bae8f8aa4abaaf020b8545fe7c48 : SUCCESS in 1h 05m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cfc3f38baab7435f91ff6ee7f7b95fc6 : SUCCESS in 22m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3460e9e8e7d4f4fbb33aa76482757f4 : SUCCESS in 24m 29s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/ae099c36340b4357ae8351e906116203 : SUCCESS in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"580a25f1683bc0ad40bc61bcf79df68c2f32736b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 09:46:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6f1f85c3a864b138636b767d9a4c111\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2fac74a8b883466193020c2981285e98 : SUCCESS in 2h 21m 31s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4616477ec8244f2ba1963cbe7701a6c : SUCCESS in 1h 38m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8da611f9945f44f98bae6ad61c80b3b9 : SUCCESS in 1h 46m 26s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a017ffcc52b140d38cd2d3f2af3158bf : SUCCESS in 1h 18m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/898eaf7245a5432ba74ecd88c5eb13a5 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78668addec7a4e3b8da250ca682795b1 : SUCCESS in 4m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/802596a080f1436787965368137d1b25 : FAILURE in 25m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f399fedcedf00bd6616400e6dc32ef3ff63a90ec","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-12-22 18:48:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"74f7eb2bff06c1dc6fc4b51247de67ea719eef37","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-12-25 01:14:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bd36d79d3e5d22a277016d3a497eb3e7639b871f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-12-27 14:49:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"948116fb7907b03c1402081d1f498bcee5e7a822","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-12-27 14:52:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a6a282ac320144475ec96b3c426cc0dad618337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 14:52:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32e38ec91c143c14f9741bd0c1addd5a5c8ad070","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 16:50:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a2dfd4d0fd54f5d93f347b4f9562e23\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8c302ed979984dda988562e8eeae7488 : SUCCESS in 36m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/79440926c64d4cc88021277b46a245a5 : SUCCESS in 1h 10m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/82306b19470a4147a50d8ee76cb03865 : SUCCESS in 26m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/250a76a1af944d0eaaec96a5a8de0ab1 : SUCCESS in 1h 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef435db237dd437db06189a1915b6819 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/799316de1e71437d87a49edb327d8951 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7669bce2bbde4bea83320d0dbc09b535 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8ad9c042624e2f95fbb4aca2b4f64b : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ec7ca6ab76a414aa66715290b6a47e2 : SUCCESS in 3m 47s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/be3539789d3746ada68b39172db4fd8c : SUCCESS in 1h 12m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c13f385ef244ecaba54dc0875b4ec3d : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1b28ac6c9d4d7df5aa18d6b7983cb1debc66f083","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 16:50:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2aa463d3bd98c906fcbe717e6fcdfa2bfd5ffb28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 16:51:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30ffdd8e02504ea2bd6a01cfd01745f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a18e7cbc104bedbaeb83b9f4b847ae : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a68f5aacd8a48ee8eed2b375d592b5d : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82fe25505592ea121e91a33f467bdb059da01a6a","revisions":{"82fe25505592ea121e91a33f467bdb059da01a6a":{"kind":"REWORK","_number":1,"created":"2023-12-22 07:22:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/904264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/904264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/904264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/904264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/904264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/904264/1"}}},"commit":{"parents":[{"commit":"f178ec3fc08b0f9a8fe76f18d931a08cc0c670a5","subject":"Merge \"Add rsync to nova-compute\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f178ec3fc08b0f9a8fe76f18d931a08cc0c670a5"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-11-28 13:08:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-12-22 07:22:24.000000000","tz":0},"subject":"Sync only local cell in nova bootstrap \u0026 upgrade","message":"Sync only local cell in nova bootstrap \u0026 upgrade\n\nAdded the --local_cell argument to nova db sync commands during\nbootstrap and upgrade.\n\nThis was previously thought to have no effect [1], but has since been\ndiscovered to fail when rotating the nova database password.\n\n[1] https://opendev.org/openstack/kolla-ansible/src/branch/master/ansible/roles/nova/tasks/bootstrap_service.yml#L2-L3\n\nCloses-Bug: #2045558\nChange-Id: Ic64eb51325b3503a14ebab9b9ff2f4d9caec734a\n(cherry picked from commit de1487f05187387aea3f113dcbb6926734b1ce6b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82fe25505592ea121e91a33f467bdb059da01a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82fe25505592ea121e91a33f467bdb059da01a6a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
