)]}'
{"id":"openstack%2Fkolla~904722","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Id1a87995a36e202fc0aaf82daf022ec10d78e37d","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-01-12 12:07:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1a87995a36e202fc0aaf82daf022ec10d78e37d","subject":"rabbitmq: Use timeout in healthcheck script","status":"MERGED","created":"2024-01-04 09:38:58.000000000","updated":"2024-01-12 12:09:01.000000000","submitted":"2024-01-12 12:07:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904722","cherry_pick_of_change":886449,"cherry_pick_of_patch_set":2,"meta_rev_id":"6408128458c06f91414e6e77e694bc4acfe3346b","_number":904722,"virtual_id_number":904722,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 12:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-01-04 12:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-01-12 10:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-01-12 10:17:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 09:38:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-01-04 11:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-04 12:16:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-01-12 10:17:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"460e6b8e704bf6bf8dd7adf954e62a3d16755eb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-01-04 09:38:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c66446747c7e1104ec869930df65c8dd95e56e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 11:06:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7d7c0152ebd406aa6ca26c793b4c90a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c5ab0df094d94013aca82351f37d9ba1 : SUCCESS in 39m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dba3c2235f6a4b05a93f5a72ede5fd2b : SUCCESS in 41m 10s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/ac4d66afa3e6429a8a3b9212d7bd9950 : SUCCESS in 27m 50s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/22c8d042d1ad40d6b9b2844ef8506b13 : SUCCESS in 1h 22m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1ee28c5789714159aa33e3f8987d0f29 : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/80b4d851cc484cb4ba0cbf930b3e7a60 : SUCCESS in 1h 07m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7bbe40d7e4fe48c6b69239b5e60eb9d0 : SUCCESS in 18m 51s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6bf4b5eb6a0b4a37889336ac76d8dc19 : SUCCESS in 22m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1ba48978bed84454b11eeb0b0cd4306b : SUCCESS in 1h 09m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b558c145a0417d9caa9916eeb445d8 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1849bd8f447450c8aa62afb8cfd0116 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4b777fad0244d8f85075aa28b1ec364 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88179fe21a214a868590358f8d670601 : SUCCESS in 6m 11s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/f43dd8ce05c543d6b02e3092243a74cb : SUCCESS in 21m 03s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/edb285e7ec754d26985493f985c03411 : SUCCESS in 59m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5f55e6fdc5b3481eabde2706cd100761 : SUCCESS in 20m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/906b05787ea04ef9a9f68c99fed63ff3 : SUCCESS in 25m 06s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/f776b548b58a45dfb77ec6a575fe287a : SUCCESS in 52m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6960ab210df9dd2b7b4c059f4a648d0bf6500e19","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-04 12:16:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fbd1caedb98d2046dd3b8cc0053051aa18af3684","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 12:35:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91fff24fb54743fa88414bbecbc02c7b\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a53c415d84042ca92a6f321236f4744 : SUCCESS in 2h 53m 33s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/67d6a1c7c939430297fc6991bca464da : SUCCESS in 2h 14m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e8edc32347648beba6351f634b0d147 : SUCCESS in 40m 33s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/11089713d85b433391b4971dd94d1a88 : FAILURE in 47m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91d293bce739435a8337a59b3413f52d : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74a01a237c7b43c396b2730872832cce : SUCCESS in 5m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/71575594b0c34292b21b8f48b4090e68 : FAILURE in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38eb1815b0edecf6710dbada9126b5d2d1bdf6ab","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-12 10:17:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ba66b0c755e73a7e8980eb48e32058fba069ea0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 10:17:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d127e55628b560c1e2af75b720b615e37baea07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 12:07:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/319b0d4012124487a63818b4124aa9dd\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/79b7b417c7064850a21e919b47b91310 : SUCCESS in 39m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d0aaa2fd163a44069f0ff9cc8ada9ea7 : SUCCESS in 1h 41m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3a933ce508a549ee9a4ceb853d1ca06f : SUCCESS in 25m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2395499f873843a09747712201cf5f49 : SUCCESS in 1h 04m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/678142a104c843bab9fd35cc5faaa5cf : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c67e95de6d140c6b9c8b82cd8bfc6a3 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3181c726bac488f8a3c07807d3e1611 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f522309a077e4db0bfab82ede6563dc1 : SUCCESS in 5m 27s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a6a98cff4a7c472bbacbcabbc7a54d2f : SUCCESS in 1h 10m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10ae19e2890b4a0785968a10de87648a : SUCCESS in 23m 36s","accounts_in_message":[],"_revision_number":1},{"id":"088b0cb350aed612a23615435bf51b493a4d6218","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 12:07:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6408128458c06f91414e6e77e694bc4acfe3346b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 12:09:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec1157dceeb944aabb14a00ed408e92b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087da52f2fac49b9855edfa801a21f47 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1eb0f2bf5b685b0c6a7c98e4b12ce6f86d25b298","revisions":{"1eb0f2bf5b685b0c6a7c98e4b12ce6f86d25b298":{"kind":"REWORK","_number":1,"created":"2024-01-04 09:38:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/904722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/904722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/904722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/904722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/904722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/904722/1"}}},"commit":{"parents":[{"commit":"82fe25505592ea121e91a33f467bdb059da01a6a","subject":"Sync only local cell in nova bootstrap \u0026 upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82fe25505592ea121e91a33f467bdb059da01a6a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 07:17:21.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-01-04 09:38:58.000000000","tz":0},"subject":"rabbitmq: Use timeout in healthcheck script","message":"rabbitmq: Use timeout in healthcheck script\n\nCurrently we don\u0027t set timeout and the default is infinity.\n\nChange-Id: Id1a87995a36e202fc0aaf82daf022ec10d78e37d\n(cherry picked from commit 36df37be3136d1aa14b23f87632fa461363509b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1eb0f2bf5b685b0c6a7c98e4b12ce6f86d25b298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1eb0f2bf5b685b0c6a7c98e4b12ce6f86d25b298"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
