)]}'
{"id":"openstack%2Fkolla~904731","triplet_id":"openstack%2Fkolla~stable%2F2023.2~Ifce506b49571950ee16e26aae06f20aa94567722","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-01-12 07:04:57.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ifce506b49571950ee16e26aae06f20aa94567722","subject":"CI: Use newer podman/buildah on Ubuntu Jammy","status":"MERGED","created":"2024-01-04 13:29:45.000000000","updated":"2024-01-12 09:08:07.000000000","submitted":"2024-01-12 09:07:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904731","cherry_pick_of_change":903161,"cherry_pick_of_patch_set":3,"meta_rev_id":"3dc271012bcadf0aa1c109b502f75560970f983c","_number":904731,"virtual_id_number":904731,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-12 09:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-01-12 07:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-01-09 11:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-01-12 07:04:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 14:42:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-04 16:32:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-09 11:33:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"b255b70a23065c4eb8c2c088eeb3fb0021975990","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-04 13:29:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02ec231d4b6c73c1dcf183e1d216576e67a087b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 14:42:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d509a797ca2f40a0965d4bcf78db8b9a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b04e5c76671d47f4ac1168daefea25f8 : SUCCESS in 19m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8e1c531c6ce463bba15d43928944e91 : SUCCESS in 16m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/554efd7321a24f58b803daa6aac83c4b : SUCCESS in 1h 11m 06s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/497d36feccfd4359ad3080e0920893fa : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7de0830607a48d393d0ad19b5ad9912 : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1ab5e435d594b9fa0c5ff9c2fb12145 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ef9d432ec264176beb4233f7ae1f438 : SUCCESS in 2m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6020430ec0d5430687ba29fa71016774 : SUCCESS in 17m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/78f446d006db4c24a0b552da89759e11 : SUCCESS in 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2815388e9bf26633a4d9c13830045c494245cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 16:32:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95b3468fafe04adb8cc4435b87ee49ec\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8fcc79af035e48d4a733d4aa75cff05f : SUCCESS in 18m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bba5e2efb7c341ceac9e8ffe4dd3e174 : SUCCESS in 58m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d109c624233f4d8a96515fbf62b97f87 : SUCCESS in 20m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a3c5006fb2a04ea88a5a75b94943ee15 : SUCCESS in 28m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7029a238f19a4b8b82c0f7e2a8414ee3 : SUCCESS in 1h 03m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/91f52c6e9ffb4586990143d2e45675b4 : SUCCESS in 1h 23m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec179a032124338b76da7201e66eda5 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34d8c92e32084ed2b92d832735652548 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/458180406cff4991a2e1534c4296d5b0 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6be62bf0b8c84cdeaecb6d4c358857ab : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69bdd00f39aa435397ffdfc5f1f0d4a4 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/377825e3c577462ea631a4416aa84e6f : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89fabe4eb07340edbf7256b9a3a67cba : SUCCESS in 4m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/abfca45ac55741a782a4baf5c32df397 : SUCCESS in 21m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7ad6341784a043b699be27337aebd1e0 : SUCCESS in 30m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0d11d559182b44d4b14e0db9fb112684 : SUCCESS in 1h 00m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0aa661486d2642639371f997e1edd417 : SUCCESS in 1h 22m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/74120da0e68e4336a3ee022631c05dcb : TIMED_OUT in 3h 02m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9cd52456c2f34d23976f810d260b2a60 : SUCCESS in 48m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b3dee1358618423f9c8924ca33209eec : SUCCESS in 44m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fb98ccb016f8438da1b0d88316445d4d : SUCCESS in 1h 20m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2c66657f63196b095d840309c3aa277e80dd6cfb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-09 09:15:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1daef76e06ed99a2b7fd1af038a3e2982c21975","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 10:49:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66fdd15903b24cf68aaa44ec9297b879\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8be5b8dedf794881aa5602fcc4a9b226 : SUCCESS in 1h 21m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f35f0e9a2b3d43a29b4d6a92e1c2dc2d : SUCCESS in 1h 31m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4807a06a81fc4e248f33d1f2721db788 : SUCCESS in 1h 14m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d35df967ff49411b91764080560d9e58 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80075add5d454954a642d6c6304c5a3b : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aba5e8e9d03b4b38963180ba99363b6a : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fed1369376de464aa44409b3346aebf9 : SUCCESS in 4m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ce99cc3f89124840a332e4a72d73ecdf : SUCCESS in 26m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6668b8426da24304adafdc65ae6b1667 : SUCCESS in 24m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e605f8c0cef9d89d77dd301a3598cfe00b1527f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 11:09:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9ea77272887463097f8562e5b43a091\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/25f7eca2152944f1a98e0421cbbc5e8b : SUCCESS in 20m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d9f5c047fef34b68be43a358a6770e05 : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/984cbc353fdf4073ae7d09cdafa1b610 : SUCCESS in 19m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6964612adeb84b67a7f42f2fdadfe5e7 : SUCCESS in 28m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6ae3fe63cc394272ade3c4250f403d48 : SUCCESS in 58m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dec37ac865f14e63abdcbae29aae1f86 : SUCCESS in 1h 33m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feeef5680fff4a17ae8d1091ea6912bd : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ff74eb406f4c64aa07ae34adf931e4 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f451bb3cc4c459eb8f1d7bb02ba7118 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3894b12c04f4856b82e85724894053c : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6179c89ba7774b9cbd73bc74145d2f16 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e68078f15b64458a9f822b4b1e33c6d8 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7c560696604154b26b43f328bc9366 : SUCCESS in 6m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2c0a7c77cdff4912b28c3cd13ce8cb31 : SUCCESS in 19m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e65bf36c702e4234b4a4196057f3f223 : SUCCESS in 28m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d9122b6e953c41c09230615b9a661e2b : SUCCESS in 1h 03m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4c0ef03e6dd7459d9a10cb703913feff : SUCCESS in 1h 31m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d5cc20c8ee2d4d2290d2294859f3b282 : SUCCESS in 16m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/45679736b1f049adb5212fee31fc7459 : SUCCESS in 1h 03m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b35a2e4a8db4467c8abe0d8c03678d86 : SUCCESS in 59m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3638407d403a42789f17309576a02528 : SUCCESS in 1h 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ea22cfbbbf9da01cea6abcc521e95934d3ac1104","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-09 11:33:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e84d648f7e1a529a9e4e2cc03da7b4b3681eef4b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-12 07:04:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21a7849cbc77caaf7f5ee4ba9e4a45507c24b954","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 07:05:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ec44dd7c930c80a33dc00cf3d6234433033fe08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 09:07:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95070f4739fb40f68d0185054adb40da\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/df1f080261c849669955bf772ad4a9c4 : SUCCESS in 20m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7da621d9d6b54d37b49b064b3d33d97f : SUCCESS in 26m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9657a3a93b5a4234a42c6e0cbc4dfaa3 : SUCCESS in 1h 03m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b123cc97fc1045f6a28181dd8c4a9fb2 : SUCCESS in 1h 46m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ca4a278273462e9240df34eccb7d04 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2abe43f2f40e46398278700678ce913a : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c78a8fdbf0045a799faa9ac874dc564 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/684be7602b714b7ab347181414b07886 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14da9f230ec49aa85507fb598083978 : SUCCESS in 4m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/712ca6839e86409e8856df1c1d42265c : SUCCESS in 18m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4d8fd99c5d9f49cfb454753a48edb594 : SUCCESS in 28m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f510807c8b114f58ba144df65c2f7189 : SUCCESS in 1h 11m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5aad234a61c04ef1af01f7cc846c2a59 : SUCCESS in 1h 46m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/644a3d5a488e444a8e7b3e00f6a664c2 : SUCCESS in 19m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c72b3df8960e4b789e4f626e8b0a37c3 : SUCCESS in 1h 17m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/653bacd7b3264dcca834a63126fd8372 : SUCCESS in 44m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f21e756418204f438e861099128399b6 : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2cc077b77f602bebfa833a2a17e00711e414496d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 09:07:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3dc271012bcadf0aa1c109b502f75560970f983c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 09:08:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9af393cf790b40a5beef014907ab331b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1258ac6928942af821cf99d4856d64f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efb32d62ffecc730944306d5eb54b5707da60ea0","revisions":{"efb32d62ffecc730944306d5eb54b5707da60ea0":{"kind":"REWORK","_number":1,"created":"2024-01-04 13:29:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/31/904731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/904731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/904731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/904731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/904731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/904731/1"}}},"commit":{"parents":[{"commit":"8412887491cada67fa2e6c4ae146cc5bd75973fb","subject":"Sync only local cell in nova bootstrap \u0026 upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8412887491cada67fa2e6c4ae146cc5bd75973fb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-12-08 09:55:33.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-04 13:29:45.000000000","tz":0},"subject":"CI: Use newer podman/buildah on Ubuntu Jammy","message":"CI: Use newer podman/buildah on Ubuntu Jammy\n\nChange-Id: Ifce506b49571950ee16e26aae06f20aa94567722\n(cherry picked from commit 316bafaf075ee85f555fce5e5c4df72f88e52450)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efb32d62ffecc730944306d5eb54b5707da60ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efb32d62ffecc730944306d5eb54b5707da60ea0"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
