)]}'
{"id":"openstack%2Fkolla~904781","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Iddd1f1ee6d4cd6b9a15ead323178052fe6b6e4fb","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-05 10:29:44.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-01-05 10:29:44.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-01-05 08:56:37.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-01-05 09:23:31.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Iddd1f1ee6d4cd6b9a15ead323178052fe6b6e4fb","subject":"[yoga-only]: Use pyroute2 from distro vendors","status":"MERGED","created":"2024-01-04 18:10:07.000000000","updated":"2024-03-27 09:53:23.000000000","submitted":"2024-01-05 10:29:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904781","meta_rev_id":"c7c8aba9686320f952b8c13337cf0874934dac91","_number":904781,"virtual_id_number":904781,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 10:29:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-27 09:53:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-01-05 08:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-05 08:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2024-01-05 09:29:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-01-05 08:56:37.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-01-05 05:54:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 19:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-04 22:42:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-01-05 08:38:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-01-05 08:53:48.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-01-05 08:54:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-05 08:54:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-01-05 09:29:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-03-18 15:58:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2024-03-27 09:53:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"5696c33a252fe3aef1003280d1ca10af9c5fd346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-04 18:10:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fba1c90d50d7901d7ccd090a81345962843edfd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-04 18:10:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7c6287d615ccc16fecfd3d2aa2668a502e31480b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 19:56:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c1ad64531444b4b2ae5776ed2fbb24\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2bd1bf3fd2fd4658a3fbb4f8e857ee76 : SUCCESS in 38m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6539ae4683df4309aa1559a12ea9a118 : SUCCESS in 35m 45s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/24c2f3f8eb1749399bad9bbb23d397f3 : SUCCESS in 22m 08s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/23960ec48cbd46e0b290d4c38aa2401b : SUCCESS in 1h 07m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/446069a03c924277a3fce5ad7d95d72e : SUCCESS in 1h 11m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4587ccc1b21a4c5c902d99473e1b366b : SUCCESS in 1h 08m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c6b889cef274faab8ff7b086eea79de : SUCCESS in 22m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0bc9cae813084d3abfb08037d62cee14 : SUCCESS in 15m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a45b3963114643db9610f2aae291b0b8 : SUCCESS in 1h 07m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/258ca5d3db3941c1a5a76aa34b643571 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/866794aa3a5a41f1b9cf8ea037d1c21e : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6d02f63fe5e4ca3ad0fe2648cfe3a05 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad198c356ee4af7be3ccb6f948dd441 : SUCCESS in 5m 47s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/8c6d5a0959784c67b564211591fb0440 : SUCCESS in 26m 04s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/678f27370d9d4ced9f5655853bcd4a56 : SUCCESS in 1h 14m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/202cbe158e9047e7a73f2d26cfd18998 : SUCCESS in 21m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ece6dace0e3b46bc8f14e52ec48769c7 : SUCCESS in 23m 40s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/1e0b4e72b90947a6b3a48568c4d5b999 : SUCCESS in 48m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b4cf0d9cc9d0ea5105252a106b690ca0a0c71123","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 20:23:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a802160714bb4fb9899b5a3aa76848f0\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a03f519a98a4d0e81b2f2bc715a2584 : SUCCESS in 2h 09m 46s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/80038cfa6ae44a3da137e9f23af60ef1 : SUCCESS in 19m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e6804eebc5f4db68d831b0076324cf5 : SUCCESS in 1h 40m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/72d94b3006e74803bf8e870f9d349a47 : SUCCESS in 1h 13m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d7f6cc1fc834ab39dd306c815c1bb07 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a03277bd462489196fea8f3766fc2f9 : SUCCESS in 3m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1f1b1da63dc47bc8930f31c9be4f098 : FAILURE in 1h 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5be028df1a43aa9bd60249fba36bee1e2211574b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-01-04 22:42:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e0e7b6370772bee4ab1fbde87853e754457c3b0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-05 05:53:28.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5f66ca960168874aac1b1ae1fad6d98b66f56ed5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-05 05:54:11.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3727f3294da53df8d50e60b3fff0ec0399e3eabf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-05 06:06:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c48e5a5373adf587c3e32146eee8f6b0226007ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 07:03:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb590cbacd74b47ae7a5b109354579e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0afdd6988a8e4b98a08dd4688729dd2f : SUCCESS in 25m 57s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef37410c4cc345489bd5829c2903537d : SUCCESS in 22m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6cd69a2786db4beca1d181fcec8f296e : SUCCESS in 1h 00m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/98ba6c648a0a4ad78a4b90b83f955950 : SUCCESS in 1h 09m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f591ef89ed3e439081b58e8728145a79 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d671bf5bf8c2433785cef8400eb0ccd7 : SUCCESS in 3m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b72b66283de4c1c91faa9dcf9ced14d : FAILURE in 55m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92ec3cc0cb7cd50dbc88d43eb6bd0be0972325db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 07:14:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6926911dc9ab4e1780489dcde28a3711\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d618be896312417dbadf171e1e64e86c : SUCCESS in 35m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a1c8780f47944a1fbc5791f4bfcd0571 : SUCCESS in 34m 21s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/aa838b47b93048b4bf6fda8d75e789a0 : SUCCESS in 21m 11s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dc21f63e0f574270a42260d33a4bcf59 : SUCCESS in 1h 20m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7a3bbe600ac647e3b6b0c2c3af89ba00 : SUCCESS in 1h 14m 20s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/101735846d444777ad640a5c9bbfa5e6 : SUCCESS in 1h 03m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/604c647113f14119926b188bbd73cd33 : SUCCESS in 17m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/18d365b67a8d444d87a72f9d95989f20 : SUCCESS in 14m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/161479e901b74f6f8d5297edae4dac6a : SUCCESS in 1h 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befb23232db04262ab44ca894eef4194 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9089c101c7c04935b1fac68f892fde9b : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e35cdab9bef45d582c6507c52e75980 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc42ca199cd44cb7acb02032b5d9d619 : SUCCESS in 4m 52s\n- kolla-build-rocky9-source https://zuul.opendev.org/t/openstack/build/de99635914c746b58427e0ea9ee273ae : SUCCESS in 24m 32s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f3ce494511194176a837408b6cf40e33 : SUCCESS in 1h 07m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fdee36753a664b7dbc0b5cb35fcdd5b2 : SUCCESS in 21m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e94adebe46e34a31ac80213f94a2db5a : SUCCESS in 26m 02s\n- kolla-ansible-ubuntu-binary-focal https://zuul.opendev.org/t/openstack/build/681ef9548433414bb8888355f8556019 : SUCCESS in 54m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1424babc9eb84dfa46a8f556ec88af14796078e4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-05 08:53:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f7c85520371fec4aae21a2e51a7eb2477cddf5c1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-05 08:56:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9bcab42631e771d7396696ece1268bf8d71a5259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 08:56:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5600c9b5222a9d26edbef72a995a02819b7240bc","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-05 08:57:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"52746509dd68f8ac9e263dc798fdc6a869f98a79","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-01-05 09:23:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8b30b3f30601e808f72fd93cc8ba19be6d4d96f7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-01-05 09:25:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca35b33c498543bf24a30c45e769e4ea8c380479","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-01-05 09:29:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"35ec427e38acfdf69937b483d87017a5b4368610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 10:29:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a95986d99d4622a557898b1a6275cd\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a8bd93c1d3aa40d8bb07c912ba2e87ed : SUCCESS in 38m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b8231389682347098b47eb1dbff4ef05 : SUCCESS in 1h 26m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c44abdfcfb944eeb9d529bdb809c900c : SUCCESS in 26m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/05c1b6449d5d45a78aa05d9c90e67ae1 : SUCCESS in 1h 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b152e275f214192b4d0544f76e17680 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2749579107149d6b39e28d7d54eeca2 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50e9e102ed774b298b7fddcf7640dda9 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99e22d9a9d644110b55e1fc74751849b : SUCCESS in 5m 53s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b29afa17009346518bd63a6ae5d26535 : SUCCESS in 1h 10m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/61dc7c70417941e48632b9b176f1c500 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":3},{"id":"3503bcd3151d4429f7835e0a3f03540adb2b1255","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 10:29:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d7b14ef6b0e127e1ca4e02507636eefce48d0626","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 10:30:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89601cf844c64802a095363cedd628ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c45d7b6e7b4d5087625ad631618f55 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3},{"id":"0123b7d3b422804689131b6859358e388b7fd377","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-18 15:58:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a76c35355a41027efc142012414af9d777b11684","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-18 16:14:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c7c8aba9686320f952b8c13337cf0874934dac91","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-27 09:53:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63031520e51b160ba4c5429fb811dd27e10e05a4","revisions":{"42a17404b53d16a6092b756e40a323da7c6c78e8":{"kind":"REWORK","_number":1,"created":"2024-01-04 18:10:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/904781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/904781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/904781/1"}}},"commit":{"parents":[{"commit":"82fe25505592ea121e91a33f467bdb059da01a6a","subject":"Sync only local cell in nova bootstrap \u0026 upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82fe25505592ea121e91a33f467bdb059da01a6a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-04 18:08:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-04 18:10:05.000000000","tz":60},"subject":"[yoga-only]: Use pyroute2 from distro vendors","message":"[yoga-only]: Use pyroute2 from distro vendors\n\npyroute2 0.6.6 derived from upper-constraints has a bug which manifests\nin Operation not permitted problems on newer kernels.\n\nDistro vendors (RDO, UCA and Debian OpenStack) have already backported the\nfix to their python3-pyroute2 package - let\u0027s use it.\n\nCloses-Bug: #2042954\nChange-Id: Iddd1f1ee6d4cd6b9a15ead323178052fe6b6e4fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42a17404b53d16a6092b756e40a323da7c6c78e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42a17404b53d16a6092b756e40a323da7c6c78e8"}]},"branch":"refs/heads/stable/yoga"},"c1bbe8d4601a74f38514733a106fc48c1878d7d3":{"kind":"REWORK","_number":2,"created":"2024-01-04 18:10:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/904781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/904781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/904781/2"}}},"commit":{"parents":[{"commit":"82fe25505592ea121e91a33f467bdb059da01a6a","subject":"Sync only local cell in nova bootstrap \u0026 upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82fe25505592ea121e91a33f467bdb059da01a6a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-04 18:08:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-04 18:10:45.000000000","tz":0},"subject":"[yoga-only]: Use pyroute2 from distro vendors","message":"[yoga-only]: Use pyroute2 from distro vendors\n\npyroute2 0.6.6 derived from upper-constraints has a bug which manifests\nin Operation not permitted problems on newer kernels.\n\nDistro vendors (RDO, UCA and Debian OpenStack) have already backported the\nfix to their python3-pyroute2 package - let\u0027s use it.\n\nCloses-Bug: #2042954\nChange-Id: Iddd1f1ee6d4cd6b9a15ead323178052fe6b6e4fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1bbe8d4601a74f38514733a106fc48c1878d7d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1bbe8d4601a74f38514733a106fc48c1878d7d3"}]},"branch":"refs/heads/stable/yoga"},"63031520e51b160ba4c5429fb811dd27e10e05a4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-01-05 05:53:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/904781/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/904781/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/904781/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/904781/3"}}},"commit":{"parents":[{"commit":"82fe25505592ea121e91a33f467bdb059da01a6a","subject":"Sync only local cell in nova bootstrap \u0026 upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82fe25505592ea121e91a33f467bdb059da01a6a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-04 18:08:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-05 05:53:27.000000000","tz":0},"subject":"[yoga-only]: Use pyroute2 from distro vendors","message":"[yoga-only]: Use pyroute2 from distro vendors\n\npyroute2 0.6.6 derived from upper-constraints has a bug which manifests\nin Operation not permitted problems on newer kernels.\n\nDistro vendors (RDO, UCA and Debian OpenStack) have already backported\nthe fix to their python3-pyroute2 package - let\u0027s use it.\n\nCloses-Bug: #2042954\nChange-Id: Iddd1f1ee6d4cd6b9a15ead323178052fe6b6e4fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63031520e51b160ba4c5429fb811dd27e10e05a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63031520e51b160ba4c5429fb811dd27e10e05a4"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
