)]}'
{"id":"openstack%2Fkolla~905116","triplet_id":"openstack%2Fkolla~master~I2f6c9045aa990b36716bb18d9f2eb699a35ee51a","project":"openstack/kolla","branch":"master","topic":"kolla_sigterm","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-26 10:36:53.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-03-26 10:36:53.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-23 19:06:58.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2024-03-26 10:36:53.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-03-25 10:36:09.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-03-26 08:20:01.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-03-26 08:52:32.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I2f6c9045aa990b36716bb18d9f2eb699a35ee51a","subject":"openvswitch: Move db conversion from kolla-ansible","status":"MERGED","created":"2024-01-09 13:12:57.000000000","updated":"2024-03-26 10:37:53.000000000","submitted":"2024-03-26 10:36:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905116-kolla_sigterm","meta_rev_id":"eb9c7ca952ff0525fcb5c7417eebd38d2f6d3df0","_number":905116,"virtual_id_number":905116,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 10:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-23 19:06:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-25 10:36:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2024-03-26 08:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2024-03-26 08:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2024-03-26 08:52:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-03-23 14:41:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-09 14:44:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-09 14:57:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 11:26:23.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-01-12 11:34:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-02 07:22:21.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-02-12 13:37:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-03-14 18:29:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-03-23 12:56:02.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"9eda8f68621f3bdf9cafd4a596c9efbc75eef0c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-09 13:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aec32ba28beaf8d11a4a410a89ab6d98d1011e48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 14:44:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cadea78e6095432084116b8cb147a191\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/75d2afef174c4ec9a7f8684dc03c53c2 : SUCCESS in 32m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d06d40e68cc84178b307a192754a3b61 : SUCCESS in 30m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6147cbb173094652aa15f3ee1b44216e : SUCCESS in 1h 29m 17s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6e7a834924a424c860bdba45f37c142 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f04c2b3884c543ea9ea3ab5df6f155c6 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d7c4882096642c6846bb22f1c5eac36 : SUCCESS in 4m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2b263ca8ba8c46efb5a0fabdffc77b4e : SUCCESS in 33m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d220aef1f0ce478e9a8554658228e756 : SUCCESS in 30m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e8bfe47c9da53e33f459127f8ee88b6e983b4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 14:57:27.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06c6df648cd34f559c0133c5eca0a6fa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ea7c59b65fa84e2e9c877f6d33cfc118 : FAILURE in 28m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c8fd42ec80cc4846ba48d4a3c659e684 : SUCCESS in 1h 26m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1e91dbc07979477f965cc1cada72d48d : SUCCESS in 22m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6d05745596674447aa72b458e032f9f5 : SUCCESS in 31m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9d4b1f85d1b641788c3335afa4cd2078 : SUCCESS in 54m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fb5125586c7d40379cc9aebd0614927f : SUCCESS in 1h 26m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa5c445ca3540b887d0f29c8da48dd3 : FAILURE in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28389a763b864a06ab4ff7f2d4be0112 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79feb4d94f2c47e4b12f08e2ce15a9a7 : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31a6a396ce8f4ee7ac75025627db44c6 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c2a389d9814e1793bd0be952a14295 : SUCCESS in 5m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bb7ab2ebacd541de9b90f88988433778 : SUCCESS in 22m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4e6a02f85be9421aafc3b6828e9f1ab0 : SUCCESS in 32m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4bcb09db2dba48e6840d5cc51a780ea8 : SUCCESS in 1h 10m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ed1255cb64344077950cddd5662a06f5 : SUCCESS in 1h 17m 53s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d545b3e6f64643cc9a16ff4a3615364a : SUCCESS in 1h 14m 59s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ab96d1d66de24c5ca4330cad62dd029d : SUCCESS in 1h 37m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/db82b8e62a5941238bdd366f3b9bb7b6 : SUCCESS in 18m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dd596c2e7de74a7eb8a26015c55a2af0 : SUCCESS in 45m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b65a3d5d894a43fa9456054f9e0352f1 : SUCCESS in 1h 09m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/331ce16d5bf64d4392f38cc4736ef1ff : SUCCESS in 1h 10m 09s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/322bd82d7cd547c58a7d142b6e74a548 : SUCCESS in 59m 23s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/60ea0007c06945d0b5e31c3e7ec44567 : SUCCESS in 1h 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fd90e1df696cd1590f89d8bf48e5dce96261263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-01-09 14:59:44.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d34119d9fab02876929773e8b5f2042cfd4f4b2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 16:48:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/364af2b6f2b5419fbd6319b324131ea5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1eab467111c14d08bb1f16248fdb7af6 : SUCCESS in 26m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b252564fa2464b3bb8d60148b3d15cf6 : SUCCESS in 1h 07m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8dbbdc9b902a49a2a8439a5ec68650a8 : SUCCESS in 19m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ba86fa538b2e4827a10da92dcad33317 : SUCCESS in 35m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/359955f320974382b95f40002ed907d3 : SUCCESS in 1h 06m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c352af721b184bb98b8e3a9aa6a78609 : SUCCESS in 1h 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/961595cae74f43d3a7885a9670ead687 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a99cce29fee047d0a02d4ae14e3756ea : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f2a6db44d30404e97ad33852f3ae819 : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b8bce5df892405490b44ca4411070d3 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be26b48946c4fb8a14f2dc5663b3e5c : SUCCESS in 5m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/53ee984dbdfe41008e2b037e88005b66 : SUCCESS in 23m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c15f9f5069c143cab7f7a80b9924814d : SUCCESS in 28m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/87de0d90d85140b6baf712a7c89d6328 : SUCCESS in 1h 10m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/58f0bf727f754413a60b260b4a93548e : SUCCESS in 1h 41m 47s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5feef64ade054d8a9ee66b8ded2729f1 : SUCCESS in 1h 17m 03s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/be5dbb15ebfe48a6bb5249e7c46f8a61 : SUCCESS in 1h 30m 29s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/529b8b9caf9f4f7aa6b1b05898e0fa35 : SUCCESS in 19m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2e7d7fc9265f457db50a6a6c2d9376ec : SUCCESS in 1h 03m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2e051d29075a44c9a044166e0b73a649 : SUCCESS in 47m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/45786898c75346edbc427052cf98cfe9 : SUCCESS in 1h 23m 50s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/bf26a6de4a3247e0b1dff28f588d7153 : SUCCESS in 1h 09m 30s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/22a96b5a48a94d15a0794e8c1194474e : SUCCESS in 1h 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf476785fdf9ee39abefcd2e18fdb0417dec7deb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 16:49:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/922fa48a3acb4686ab673130fa3a0914\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/698ea067b3d140c7a9f7b9d0691820fc : SUCCESS in 1h 47m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/57ff8d73dd2244979de58b1466ea34cb : SUCCESS in 20m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/64ed22234004483ca23b5827044b2a18 : SUCCESS in 1h 41m 27s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e12372347cd14218948d4dca37d4e02d : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d25bf593c99b46799007acfc93c11011 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1f9ef9326ba42e9922c2b75e6d58204 : SUCCESS in 3m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/db9eac2ebd734e7e92d1150817d2a7a5 : SUCCESS in 21m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2adad73571974df687e439daae4097a0 : SUCCESS in 20m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"331705e85a1749efeb79f8723f662d1a773a6cc4","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-01-12 11:26:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a1ab0a8d9ff19ffa996c6b3fd8f3e66e256b86e1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-12 11:34:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e61717b5ea7d405857ef7f379c4e0228fa33fae8","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-02-02 07:22:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a5d49aa0515adb70fd24cc307709764effc78b5d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-02 13:42:25.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"dad58bd47effb989045cf6932b09f4aebf4d995f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-12 13:37:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6e0b432261861dfdfc0a3648e4ad5fbe1ab11571","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-14 18:29:55.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7788f9fb3a1a7b3cf81d14b5643413cca1d3f2b1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-23 12:56:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7655c2681f31cbf1497e8b1629cf172b269ffbb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 12:56:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"afc2cce521e2f12b252848421e09cf5d71e70ad2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-23 12:56:30.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"b17553190e397d34a77865f83381d700c1d3b1f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 12:56:38.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"10480ed9a7facae3be4759b615558b2ae9d829b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-23 14:40:55.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n","accounts_in_message":[],"_revision_number":3},{"id":"74d1f3f661c1907748799d8fe0976540b6ae797b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-23 14:41:05.000000000","message":"Patch Set 3: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4cc359124cc4312cbfd63a3b4b1e415831caa12b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 16:16:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b034859c9d4b509b39dd9477fb4023\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/64944ee7cde4408a89eed81f689d0e89 : SUCCESS in 20m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/77413dc784164d43b8882065427dbdd7 : SUCCESS in 19m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/220799a199e744d3922c8a6ba23a89d0 : SUCCESS in 1h 33m 15s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15a494aed19049f98e24207a8a740b37 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ef41035a3e146dc9f360a405002ee58 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/05a7af62eb534bcfac3dd6fe87599c4b : SUCCESS in 2m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7c7c178e59fa4a85ba1e9c208de8cd82 : SUCCESS in 1h 02m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7b6d472537724835b76081779cc663c3 : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a4175dd9ad5e6015a25e102cdc1d34286084067c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 16:23:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c33a659e6c99492c90d5c32ff44f2704\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e4af5cc15e0b479289bd1b1f17556182 : SUCCESS in 19m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6908647eb8e54383b533e24d58ec597e : SUCCESS in 54m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/42a74126e4784b5f9b6134cb5d15ffef : SUCCESS in 13m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/392eb36fd5d848819d504d7b826606de : SUCCESS in 19m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/502b50fd98d9488eb6843aa5f63ff338 : SUCCESS in 48m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/93fbdf483c044852ba6e417e98ab5fa7 : SUCCESS in 1h 13m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1125a43466c643e39475f4da4bf2cdda : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1005a9b264b849bdb4a9e4587a6fb866 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22b23a6dc296457fb884ca13365d43ef : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae567d2a4edf4bb59400197dd1dc1d17 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9130d9d638040258456078ff8ce45d7 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26fcd837a2bd4d24b656e020b2644282 : SUCCESS in 4m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/11d374b694744177a9220349d20cee7a : SUCCESS in 25m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1b85bf6b084a46f78a067f1b42cd508b : SUCCESS in 36m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9490cc785c484e20862418a0fff92ee2 : SUCCESS in 1h 11m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a09f226e494e42bd8b3f4bebce9e1c11 : SUCCESS in 1h 37m 50s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/85128d8fd1334a31b502cb9f936987b2 : FAILURE in 1h 06m 10s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/cccb0bb8236d444baec9942fefc06d4b : SUCCESS in 1h 31m 38s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c600f72ceecf4b68b7be27ce40cb100e : SUCCESS in 17m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fe48d59a85c845c79ad3391baef399d0 : SUCCESS in 46m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a880188155e14e6293f87ca022b5241d : SUCCESS in 43m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/79ccd86005e94024b9c33d0355422014 : SUCCESS in 1h 06m 54s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b738ba18dbf84cd389ab125e2ff0a70e : FAILURE in 43m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1f6912a3f68e4f6f903c1b90bca3bd13 : SUCCESS in 55m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02f9f5126dfc69018d954b45a4dc85c0c5f10f78","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-23 19:06:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"53220bfbebd785b0978b5ed0dce4f795dee0e3fe","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-25 10:36:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"adcbf7d6458c4b9fb6b61dbd87090b7b3bcef624","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-26 08:20:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ca5e7844e3b84b4088b6e0ee174944a891986c6b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-26 08:20:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4f0607cf321529ebe21c00c5bd58c767e47fd96","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-26 08:52:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"189e3337f4927b7fbcb241a284074fc536009bce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 08:52:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2e4ab33b92d71b6d0fc30812b17d5dba32e0eef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 10:36:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05fb5a29724b4253840e689f8992d51c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/07b861bfbb41476f9253e0e374126d32 : SUCCESS in 18m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/07be01d7190e47949bbd71254130fe1c : SUCCESS in 29m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bb3934b2806b4313821e6321cfbede64 : SUCCESS in 1h 13m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3ce2405e694d4f9bbd63b4307a83855e : SUCCESS in 1h 39m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1989127541174dbbac84ec8a8d61fce8 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dafaedea73fd4a6499a5256579dfe2bd : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ea85c95e1f24306b86820b9cb14c442 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e13af72969b4ba28defca231c8e8672 : SUCCESS in 4m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/efdcce3a2b7a46ab92076ff24e73ad05 : SUCCESS in 22m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c4715336a88b4538b8df811d4c963cf6 : SUCCESS in 28m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/50e2ce6b8e9e4f74bf87d33d83591f7a : SUCCESS in 1h 00m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/47c4d8440f61472cab67ce619c1ec5db : SUCCESS in 1h 25m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/99dc1c2de37d4ffe8ba9b09d8d83578c : SUCCESS in 14m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/29fe4999ff3b41a49bc31546c470eeeb : SUCCESS in 56m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ed0d1903116942a99334d12b82ae84d0 : SUCCESS in 50m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/17471effed7145e4b560e1e8807937bb : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":3},{"id":"e5e3417a483b774019ca61c15884c4faa8ef14c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 10:36:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"eb9c7ca952ff0525fcb5c7417eebd38d2f6d3df0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 10:37:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f70ba6821244afe8f536b24c7de6954\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dee80d0c3e84241ac0becf594ce1f52 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e8a6a457211b58ed966ed2faab0266bb8e3a7595","revisions":{"ceabb37b5ab887937ba5ef5f310345255cdc66c6":{"kind":"REWORK","_number":1,"created":"2024-01-09 13:12:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/905116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/905116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/905116/1"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-09 13:12:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-09 13:12:54.000000000","tz":60},"subject":"openvswitch: Move db conversion from kolla-ansible","message":"openvswitch: Move db conversion from kolla-ansible\n\nPartial-Bug: #2048130\nChange-Id: I2f6c9045aa990b36716bb18d9f2eb699a35ee51a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ceabb37b5ab887937ba5ef5f310345255cdc66c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ceabb37b5ab887937ba5ef5f310345255cdc66c6"}]},"branch":"refs/heads/master"},"afeee6f044176a5484c0e72f7b7ccb70c2beeda9":{"kind":"REWORK","_number":2,"created":"2024-01-09 14:59:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/905116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/905116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/905116/2"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-09 13:12:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-09 14:59:44.000000000","tz":0},"subject":"openvswitch: Move db conversion from kolla-ansible","message":"openvswitch: Move db conversion from kolla-ansible\n\nPartial-Bug: #2048130\nChange-Id: I2f6c9045aa990b36716bb18d9f2eb699a35ee51a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afeee6f044176a5484c0e72f7b7ccb70c2beeda9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afeee6f044176a5484c0e72f7b7ccb70c2beeda9"}]},"branch":"refs/heads/master"},"e8a6a457211b58ed966ed2faab0266bb8e3a7595":{"kind":"REWORK","_number":3,"created":"2024-03-23 14:40:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/905116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/905116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/905116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/905116/3"}}},"commit":{"parents":[{"commit":"0eddd35da52c3e74a4c372796bf65247b261f7f0","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0eddd35da52c3e74a4c372796bf65247b261f7f0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-01-09 13:12:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-23 14:40:55.000000000","tz":0},"subject":"openvswitch: Move db conversion from kolla-ansible","message":"openvswitch: Move db conversion from kolla-ansible\n\nPartial-Bug: #2048130\nChange-Id: I2f6c9045aa990b36716bb18d9f2eb699a35ee51a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8a6a457211b58ed966ed2faab0266bb8e3a7595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8a6a457211b58ed966ed2faab0266bb8e3a7595"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
