)]}'
{"id":"openstack%2Fkolla~905580","triplet_id":"openstack%2Fkolla~master~If10006bf7f037a44b9b664b70ad423f7c6c08f45","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-09-09 17:11:24.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-01-16 11:20:34.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-06-06 15:55:54.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"If10006bf7f037a44b9b664b70ad423f7c6c08f45","subject":"Support collecting libvirt qemu domain logs","status":"ABANDONED","created":"2024-01-15 11:52:08.000000000","updated":"2025-09-09 17:11:24.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6ac26cfac525b2b1107bd6f2e78aab4a79e01e20","_number":905580,"virtual_id_number":905580,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-04 10:49:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":1,"date":"2024-06-04 09:00:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-06-06 15:55:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2024-06-06 15:55:54.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-15 12:42:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2024-01-15 13:32:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-16 11:20:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-06-06 15:37:51.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"b31820a1b89a77d509a7711a4fb46a5e5b851d29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-01-15 11:52:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db09c79fabc4aea19ed56c053c258d13bcaf1beb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-15 12:42:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e7cf240a4e2b94b41206ad0bd4a88cce9cf2e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:32:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b154831cecb14379a9163e3874051318\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1eda8ecd99d248c7b96545b3f14a34ae : SUCCESS in 23m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5e59077406774bd1947e5591cb03c7cc : FAILURE in 30m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f802c0e98027437ab6985c81d64895f4 : SUCCESS in 18m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1a6dd5842a9048c4a6e4c4b998686045 : SUCCESS in 30m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/745c885cf8b94bc9aa1b61f76dc8641e : FAILURE in 51m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3bf706ba1a734b03bac8583534afe99c : FAILURE in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b8e56e217e4dc89d19324c789e7e02 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac5e703d2df74783b315b654d6d26b6b : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0453447f0564eb3ab8f36a249daf18d : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a7f37911b5d4895a80e0e5b17329511 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d5530b1656434bb45c56ebf6ba04d8 : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fe3182c5b0314ac6b7edddedab96e28e : SUCCESS in 25m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/06ba297dcc3c4a0ca206c0ffdb273477 : SUCCESS in 30m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9fa8b6bed2e4a46bd75b317afc57fe6 : FAILURE in 27m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fdb14bd572cb43a69ea2301739c1fc00 : FAILURE in 14m 23s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f6f1e7c01d334c0aad4db02065b0bca0 : FAILURE in 24m 35s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/5d7317de4b23468f893ce43e524e1e02 : FAILURE in 28m 04s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e4d8ef80a3c24ddea4bb9cf4c8a55606 : SUCCESS in 24m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b50d574f9acb4e828178b2e993c46889 : SUCCESS in 1h 17m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1d3137c244e44f0e849ef60e899c5b06 : FAILURE in 40m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0d69a861105744548e4ec06b0d8d5c60 : FAILURE in 12m 04s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ec814a540e704c4497573f9358266dba : FAILURE in 26m 05s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/cfde26090a0b4774a553dab8a456ca90 : FAILURE in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae23af957cffe30a99b678cace6a1cff2ddd8ccb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:41:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ef361c00ce4fe5a2391b6e4840ea8d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7667b7f3f446454fb81391000d928746 : SUCCESS in 1h 33m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9af01a9ff92b4853abcce7d232244fdf : SUCCESS in 1h 47m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0d1a025abd5447fb89fa12541c188ca1 : FAILURE in 1h 40m 07s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3ff58d132cd4034b62f13e86608c3aa : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c82ee91d34814084a2547c387b23edff : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9712c3ffc02e4600b5be4daef3ef80bb : SUCCESS in 4m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1a0fab8904af4a7d80701bb7bb1acd61 : SUCCESS in 1h 42m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fd9de27015c84cf296f720b3ca897f50 : SUCCESS in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75ec12e7d2f1fc0967a51dd77362dc383111a4ac","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-01-16 09:30:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f9fdd530d68b4aa5cffd6363afa284524d8d2d5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-16 11:20:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d56455e11b5ed37d18b77200531387e67429c929","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 14:05:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98a9a0741ce3738a294046858c4ce94534ceefc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-04 09:00:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e890f60467472e8643166c54e21a82110e4b52a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 10:49:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b06b4509bedc407581331a5c0b937d43\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4357912e224a426b84f9c336ec6570d2 : SUCCESS in 23m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2400becc396645e0b0b9aa97472fde57 : SUCCESS in 1h 14m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1aaf7992ac524213aa0d47acabf636ed : SUCCESS in 20m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2cbd3d1b760b4ad0b38ae1ea674b4126 : SUCCESS in 30m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/35ae8f7b298d48f681fb3fc88d687fa3 : SUCCESS in 1h 07m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a1aa4cfd1fc54d7dabdc306a5d190a9f : SUCCESS in 1h 39m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93741af77514b06bc7c85e7ed378344 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d535b1873c24cdda160f8a70cf8aed4 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aee4aa5b2ed747dc932f8fdfcaf6ef7a : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bfa71e24e0b4eef85935cfbc1dbe7fb : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/188a6f2ba3624c1f8145d3dabd336f37 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e6b2770c7248178df1cdc6df5f437d : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c6e3ab62a1a3451d8c04dca8e43a94ff : SUCCESS in 23m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4ce39ad0350a4070ab7ceabc08370ab7 : SUCCESS in 29m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b4f913f980ef4a56a49007ed1ecfffa8 : SUCCESS in 1h 01m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b25a48a9ac1b4b45b182e10fa168d244 : SUCCESS in 1h 23m 49s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8e3d9a8f71034a00911c4c4dd060505d : SUCCESS in 29m 00s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/f45e7553e05b45b1aedaae7fb3d927eb : SUCCESS in 53m 46s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8c9249c8ef904c118d1db00426344934 : SUCCESS in 16m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/668076448bf84630969f83c801d7f14d : SUCCESS in 50m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1c854d29982041e6bd1683a137b082c1 : SUCCESS in 41m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/48a6a152271c4e62b4fa6361b1f818d5 : SUCCESS in 1h 18m 29s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5e6b0a1cc5094628a7fd8b77bea87361 : SUCCESS in 28m 00s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/4796c0f2fae94ed995d1fc66c3f55be0 : SUCCESS in 36m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea55c5849563791894d5d46acab54e67a7e15444","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 12:11:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cffeef8c3c545099d479b0abc77268e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e28147f1277943e285bd26cde22f6fae : SUCCESS in 25m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f3a57eb43a994d6097f7581b31eb01c1 : SUCCESS in 2h 32m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d48e5e0787ad4bd88ce5e02821855352 : SUCCESS in 1h 00m 14s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ba48caa319549f6ad1e08e5b7b1452f : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15d4280dc8ee4459bc3d67a7b6fa93a5 : SUCCESS in 2m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ee54c92628da4d76bf5191c08ab4dc4c : TIMED_OUT in 3h 07m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9c38509332a344b7b60cac449d76a4d8 : SUCCESS in 24m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c62d2ba4b84a13cb7f80d5cab3527b99a21d563","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-06 15:55:54.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6ac26cfac525b2b1107bd6f2e78aab4a79e01e20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 17:11:24.000000000","message":"Abandoned\n\nno updates since 2024","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ed4180e9401f324a84e2c32aec4e5bcf4f1ddc7c","revisions":{"0f6a1aeed7c5fb32555c61293769797c4c60c88c":{"kind":"REWORK","_number":1,"created":"2024-01-15 11:52:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/905580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/905580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/905580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/905580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/905580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/905580/1"}}},"commit":{"parents":[{"commit":"a1bc328d40325a4bf8eb0804d0341b1a1b9f932c","subject":"Support CAP_DAC_READ_SEARCH capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1bc328d40325a4bf8eb0804d0341b1a1b9f932c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-01-15 11:36:44.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-01-15 11:50:44.000000000","tz":0},"subject":"Support collecting libvirt qemu domain logs","message":"Support collecting libvirt qemu domain logs\n\nThis change causes the logs to be collected in /var/log/kolla.\nUnfortuantly, it was not possible to change the log directory via a\nconfig option, so we need to symlink the hardcoded path to a directory\nunder /var/log/kolla. The consequence of this is that the logs are\ncollected in the kolla logs volume where other programs, such as\nfluentd, can read them.\n\nChange-Id: If10006bf7f037a44b9b664b70ad423f7c6c08f45\nPartial-Bug: #2049389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f6a1aeed7c5fb32555c61293769797c4c60c88c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f6a1aeed7c5fb32555c61293769797c4c60c88c"}]},"branch":"refs/heads/master"},"ed4180e9401f324a84e2c32aec4e5bcf4f1ddc7c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-04 09:00:49.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/905580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/905580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/905580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/905580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/905580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/905580/2"}}},"commit":{"parents":[{"commit":"a2034e299be947540b41c73ae5c967b2bd2941c4","subject":"Support CAP_DAC_READ_SEARCH capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2034e299be947540b41c73ae5c967b2bd2941c4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-01-15 11:36:44.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-04 09:00:49.000000000","tz":0},"subject":"Support collecting libvirt qemu domain logs","message":"Support collecting libvirt qemu domain logs\n\nThis change causes the logs to be collected in /var/log/kolla.\nUnfortuantly, it was not possible to change the log directory via a\nconfig option, so we need to symlink the hardcoded path to a directory\nunder /var/log/kolla. The consequence of this is that the logs are\ncollected in the kolla logs volume where other programs, such as\nfluentd, can read them.\n\nChange-Id: If10006bf7f037a44b9b664b70ad423f7c6c08f45\nPartial-Bug: #2049389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed4180e9401f324a84e2c32aec4e5bcf4f1ddc7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed4180e9401f324a84e2c32aec4e5bcf4f1ddc7c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
