)]}'
{"id":"openstack%2Fkolla~906199","triplet_id":"openstack%2Fkolla~master~Icb158d296989ba7c99ba108e7697ca71176f07db","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2024-01-25 11:41:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb158d296989ba7c99ba108e7697ca71176f07db","subject":"trivial: fix typos in let\u0027s encrypt status logs","status":"MERGED","created":"2024-01-20 19:53:50.000000000","updated":"2024-01-25 11:42:24.000000000","submitted":"2024-01-25 11:41:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906199","meta_rev_id":"1322cccdb994e60f4830ecde613f17322574bf7f","_number":906199,"virtual_id_number":906199,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 11:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2024-01-25 09:18:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2024-01-25 09:18:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-20 20:57:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-20 23:01:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-24 21:49:33.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2024-01-25 09:18:26.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"0dfaecd227cb7fc998b38c2cfc8547e4ccd5de89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2024-01-20 19:53:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02bbf5d9699561377e1dacb88111c4ab1e3b1305","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-20 20:57:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56441227eb8444e387594803932942e0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7ab5ac85c6594d5185cd8c8eba0cd4bb : FAILURE in 20m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5fa463b6806748889cd03adc1ec12185 : FAILURE in 16m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/886431213c194d2b9d6fba8d1280e029 : FAILURE in 58m 57s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9c2190b82624baeb162748abd05e355 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/466c9c158a2f447ea872c521b9abdae0 : SUCCESS in 2m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09952912db50420e81907cae75f2711a : SUCCESS in 4m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/53e684d3c69b47ada5b386b155de376d : FAILURE in 27m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f263f5685e104f92a96c686783641a7e : FAILURE in 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4d25d556f20d0aead3df1a3267806d96ac8cf98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-20 23:01:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7fd90a83f23423bbc5de175b661173e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c6a1a18c12834f26b00677280cb1dc87 : FAILURE in 19m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/98de033d0fe546219623f48e6f042007 : FAILURE in 43m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/92861036bc9f439889f94c65a2b62537 : FAILURE in 16m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/15a0c90ded114583a69feaf780440711 : FAILURE in 21m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c160aa8fe8934469b7b1a4c9ac67b613 : FAILURE in 39m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4710c73d86064dec8864dadf728739cc : FAILURE in 1h 34m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3bc4208f9343d7a1fc30de2180b650 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb826527cc5f43fbb1b6391df304834b : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e68db52157cc422a903444ec643f7949 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a35accc101b4209b886f03010e44900 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e12e04cf514717b4d2096c0b1558a6 : SUCCESS in 5m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2cc9e9deb81641078b9d6f9a91a41ef1 : FAILURE in 28m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/439b99b61c1b4a96a0f1751d6c121e54 : FAILURE in 26m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a5aa4bf358494288a6ee2331c0556997 : FAILURE in 49m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8aff1829bef84c3b90d9a6de0e993f9d : FAILURE in 1h 34m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e22e7331901418aac5fa3dbbbab20d0 : FAILURE in 18m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/afc24b4ccf4a4745a4567bab428b73ed : TIMED_OUT in 3h 01m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d34296a6b6bf40639b2a99d9d4c4d60d : FAILURE in 35m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/790e5ec85dc24a89bc102589b2fc6214 : FAILURE in 1h 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7889daa8c80a16aaa5d474471867c45a567c36a2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-24 21:49:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e66ea19c6b41a95af543367f62f501bfc5847985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 23:28:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/245e1b632b7942b3a1d72b356b6724c2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3a865122689444b79b87360f2d3611f6 : SUCCESS in 24m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6c1e0f4c02a24dae8eb5c134100f4299 : SUCCESS in 55m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/69ca9fc377b5476d86e4a114ce4979a2 : SUCCESS in 23m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7a0118aa5ede40aea892d12287124673 : SUCCESS in 26m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/47657ee3b05b41fa98258f852d430f45 : SUCCESS in 1h 07m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/abe997c3b41244d6975360b1df04377e : SUCCESS in 1h 31m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17beb80fb852482585572b3fe46fe563 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10a36d96e457473bb86cfb49ab3f07f6 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/862cbb46e5414535b14da2c764589d0b : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/758958210a254d248c8386d576ceaf8b : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6946358b9984559a00b2fc3c70198a8 : SUCCESS in 6m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9c61f85cedbe4f3a89d0358c3fa2f784 : SUCCESS in 22m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/00cbad6e5f7d45deb9b78b1617de650c : SUCCESS in 31m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/25d0c1c73ce64bdd94f1e2d8d9bf5268 : SUCCESS in 1h 01m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/275b1c3379314710bdc35a63b4a2292f : SUCCESS in 1h 36m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5fcb23eece564c28beb3f1fb86731cfe : SUCCESS in 21m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/54e119b32d014633912053e20a8e667f : SUCCESS in 1h 12m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/089ef3a2abae4afb926a1e3d09c636ea : SUCCESS in 58m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/00c7ea18353b41418adf1236e4ae11b4 : SUCCESS in 1h 23m 26s","accounts_in_message":[],"_revision_number":1},{"id":"42e8dc564739384bd4b38e0fbbb672c16585e3dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 23:55:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b504b28e94e44df589baf4854525b30a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b65d2220b519435f821f27224ba425d6 : SUCCESS in 33m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9007f483960c444b8e64a0d3cda163b9 : SUCCESS in 1h 36m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0df8c10aaf7c487c8f4f12601173895f : SUCCESS in 1h 08m 58s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca6b9725183a4948b00cc668af426314 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7304a13afc84492db4a4b56619e1210a : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d3e0516a56234ec3aebdcaa9aea0cc77 : SUCCESS in 4m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3795c2395e364965bb694985f882bad3 : SUCCESS in 26m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ae8761fed30418dbdcb6cb6d480205a : SUCCESS in 1h 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0461dfd1848fd972089ff8d7552a2f3a3912fe1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-25 09:18:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c89729ffd2cef409c1032aff8c010a5829a6deec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 09:18:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4ab6137fa84644d1e2ae29b1d0fd993eab74d86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 11:41:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a19a2685e0894777a4d1de782d3bb7b2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/649145757c3242aeaebed0ceb490af9f : SUCCESS in 20m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5dc49cb58991427fbe0a673656da23e4 : SUCCESS in 29m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1b83e309399944b0acab063da9de3cef : SUCCESS in 56m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0e94aa25ab31447abfc37df54b2f9e57 : SUCCESS in 1h 42m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457d02ccb0744ef69197afa61653f49c : SUCCESS in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27292d84e0fd4991b780fc27e6a6ce3f : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5c8503c7370479889abc07ef1e63000 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6289751ba2d840f5b3bd74844f94d344 : SUCCESS in 5m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/02fe0543b7f84d2c86266f55ffb55118 : SUCCESS in 22m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0fc2c8864b034cc3918f86b351897e22 : SUCCESS in 30m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b25a0f2b397b4a19af88a09432d70acb : SUCCESS in 1h 21m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/320593e1ad564474b30ddf6cd0e1c397 : SUCCESS in 1h 48m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/317329a824d247c785ff2a3bcc9b03d4 : SUCCESS in 16m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e2b0dd5c300b4550aec3271655e71353 : SUCCESS in 58m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/46b3588c3dcf4084af76b9395115cba7 : SUCCESS in 49m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/355b66e8f8a641a9a6ac83c4cc88e157 : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":1},{"id":"00fb4476cffdd59aaaf59841b19f0a7c7a7c831a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 11:41:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1322cccdb994e60f4830ecde613f17322574bf7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 11:42:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eae1dfc8c74437fb85a57f55ea0deb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/251f750c5115488594acee2fe4b9441e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc70184a0cf5abbdef078f7ee2994fb0a45417c8","revisions":{"fc70184a0cf5abbdef078f7ee2994fb0a45417c8":{"kind":"REWORK","_number":1,"created":"2024-01-20 19:53:50.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/99/906199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/906199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/906199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/906199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/906199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/906199/1"}}},"commit":{"parents":[{"commit":"1580d778219b2d4e40c53ca9d5a437fe2a82e524","subject":"Merge \"Stop removing missing /run/nologin in RPM images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1580d778219b2d4e40c53ca9d5a437fe2a82e524"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2024-01-20 19:53:09.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2024-01-20 19:53:09.000000000","tz":60},"subject":"trivial: fix typos in let\u0027s encrypt status logs","message":"trivial: fix typos in let\u0027s encrypt status logs\n\nChange-Id: Icb158d296989ba7c99ba108e7697ca71176f07db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc70184a0cf5abbdef078f7ee2994fb0a45417c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc70184a0cf5abbdef078f7ee2994fb0a45417c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
