)]}'
{"id":"openstack%2Fkolla~906349","triplet_id":"openstack%2Fkolla~master~I09cc2635cd422a859ccded887affb58dfbdc60ad","project":"openstack/kolla","branch":"master","topic":"requirements-change","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-23 12:01:19.000000000","reason":"removed on reply"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-01-23 13:51:11.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-01-23 17:25:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09cc2635cd422a859ccded887affb58dfbdc60ad","subject":"Fix gnocchi and skyline after requirements change","status":"MERGED","created":"2024-01-23 07:45:08.000000000","updated":"2024-01-23 17:26:17.000000000","submitted":"2024-01-23 17:25:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906349-requirements-change","meta_rev_id":"345dba016ed69ec774ba86d9c48cb04741615d48","_number":906349,"virtual_id_number":906349,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 17:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-01-23 12:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-01-23 14:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-01-23 12:01:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-01-23 12:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 08:03:48.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-23 08:03:48.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-01-23 09:33:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87332ff45a7203203772475b65f283958017292d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-23 07:45:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb606ff269a3e5c4d03df12dad2d4e7debe1d2ae","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-23 08:01:47.000000000","message":"Topic horizon-local-settings-d removed","accounts_in_message":[],"_revision_number":1},{"id":"61028bb156eb8b7b38d8df9c350e83834ef1bc5c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-23 08:01:53.000000000","message":"Topic set to requirements-change","accounts_in_message":[],"_revision_number":1},{"id":"fae754186d9231537c49a41105746a1282311e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 09:33:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3131c9c064d46b38bbae8545f76d000\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1e67e568e0924186a5ce5ec21c771fc6 : SUCCESS in 30m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c588136c34f24ea19f43c28b8609dd22 : SUCCESS in 1h 12m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5d6daf4ef5334276a051472fb8af8772 : SUCCESS in 18m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f64e05e229af4a129cde72c8a1db89bc : SUCCESS in 28m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8f0fc377ad2343549034404818e2ef70 : SUCCESS in 1h 03m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5d619d36963b471997d8a04029ce5a4f : SUCCESS in 1h 37m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae737fc2fc246f8a8450d218ca0eabb : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/943f345b07a045499f4d8c1af48dce0c : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3531faa49bf34b08af8d9402775f106a : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0126fc0ea8c4db1a97656ca3f699dd0 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d53e69e80e54989a2c7dbf3e1830bab : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/978d6de72072461784ff3d82acccd3af : SUCCESS in 23m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/22e0181e322e4e948f96b11a4c6778bc : SUCCESS in 38m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ecfd942fa95749bb970ee32456e880d3 : SUCCESS in 1h 07m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ebce27b25f6a47f29d78934f395be41c : SUCCESS in 1h 29m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9f444acf75284d24ba30a0667035cd99 : SUCCESS in 17m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/27d519c6f67c41f49c859be412ca8b0e : SUCCESS in 54m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/635d9ced0b094d5b90b129df3751bd6a : SUCCESS in 55m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/879bb8a29d864014afae4270772e3509 : SUCCESS in 1h 33m 44s","accounts_in_message":[],"_revision_number":1},{"id":"023ac46f8028b1774e0d8fa2a87b97b11d6e3144","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 09:33:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c87ba45ee84b8fa4e256c018734e6c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3b20a4db1eb6473faaf04d1ebd6cb885 : SUCCESS in 32m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e260159840ed4d9c87cfffa7c5b7d988 : SUCCESS in 18m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/20dc5dc698f4409c98e9119d1883acfe : SUCCESS in 1h 45m 59s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08420630af8b464aaa901a9a7f914dd4 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84b86e189398494ca681d0c8aba7b466 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e36c1fccc333458ca85ca5dda99de4bc : SUCCESS in 3m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d08a00e4c30c439ba50349e13afa5dfd : SUCCESS in 1h 37m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/87291b610fcd4b93bc74f120a0196747 : SUCCESS in 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4491138d236512bffbb92d38da6784df0fd6d88","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-23 12:01:19.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2 Workflow+1\n\nToo bad that skyline isn\u0027t properly integrating with upper-constraints, nothing new for gnocchi I guess. single-approving to unblock gate","accounts_in_message":[],"_revision_number":1},{"id":"4888f40b4a931166a10ea6a90051d979d1e5447e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 12:01:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed45d69d7d78c01d2894afdbba850b360eae4012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:45:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15835ef2ae6d445290b0b4e42882330e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/31a74194ea4340cdadff2fc9468c93da : SUCCESS in 21m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9efb2d17c4f048f4adecfae64f177800 : SUCCESS in 35m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f4ec394dc544428ea3b111758968113f : SUCCESS in 1h 06m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7b33eea389304050a011169e8bd5469c : FAILURE in 1h 16m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619e494a13ae44d69199b53b38281195 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96dd0025da14429f8269930ba910d3b2 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/934f50ccf09445ba875bd63a838eb6e9 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8fc3fd7ef74901897378b066191bbe : SUCCESS in 4m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d23c2cd03b894f56be26a2cab548ded3 : SUCCESS in 24m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b24965d666dc40ff9e578430fb016df0 : SUCCESS in 32m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c187f72d396e494e8c1bb9e544839dc2 : SUCCESS in 1h 07m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/27ece2760de8494ea6b9d1a4f9f54e4c : SUCCESS in 1h 39m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3cf8c14e4709470abec262f5ebd3b82a : SUCCESS in 16m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0568ad51215844f8a68ae4e7212d92f9 : SUCCESS in 57m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/569575307b414b119a18c20c21fa0f62 : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/00bb32dcd49348bba570281e2e0afdeb : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"13f4be2b32e90373191878c196dad9aaf30e56c7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-23 13:51:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c0a776f75eec10d64fe3d3bfe57fc448f9021ae","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-23 14:06:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b34bd6126c32fddb68a7d00c6e5c3e6e35cd1dc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 14:54:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/625398ed04494382aabd6deac5896d2a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ab946b3c587f4303b5a58e0551ff1253 : SUCCESS in 28m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb6393ca45b94f11bbd9678ad3720cbd : SUCCESS in 1h 01m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b30aec233d074ed58f0691ab919027e4 : SUCCESS in 1h 01m 26s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f860eb55b2b5466c91866ed800e81682 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7ec914f25f040b6aba2ae70f341a6ea : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2045a23314ba4a4295cc2ca27944597f : SUCCESS in 5m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0b85548770244f5e8104dfbafa3acd00 : SUCCESS in 22m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/925bd1c297634ce8b48cf4899702e196 : SUCCESS in 59m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"286289724aed1c8515d91250522b9f7a8c89156a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 15:42:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b529a4e746f7499083e550f7641edcd4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9d7da588e27b4e0689b36622a9dbd483 : SUCCESS in 27m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5f4a1d36c3aa4a21af5d4d2cd9f036c0 : SUCCESS in 1h 09m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1d51a64c8a5548ef9bc1b227cd3d5823 : SUCCESS in 17m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dd8278ec247344859e7a2eaec940c634 : SUCCESS in 31m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8e7b673d39e245828b476ec4322e4694 : SUCCESS in 1h 08m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8f16b48d36d64a2b90ddb6582db8443f : SUCCESS in 1h 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e007c1c8fdf64adaa37f450df76c69bc : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5817ebe476184113bb08e62c2632a802 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c5db9151417454485a7db771695e299 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29e539d60bfd44cd80de4bf542e38203 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baab0e5ae2424efc807d4a4f31f8dca3 : SUCCESS in 4m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/86098677d87041b29a41a6a221a870b6 : SUCCESS in 19m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d16f66a79ec34640acfd8244c3a9f17f : SUCCESS in 27m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/59b8d1c8904840ddb65d6ca4fe1fd8bd : SUCCESS in 1h 01m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8b5d7ac0269045348475057905b32ef2 : SUCCESS in 1h 42m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ebc4d5f9762b4989bb82a50a6e0e4452 : SUCCESS in 24m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6da4a147881548aa90dca2efdac2feec : SUCCESS in 1h 06m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b600f0c1b7b949f2b77ad4cacae48874 : SUCCESS in 1h 04m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a542bad579cf42ff908283f8b57507cc : SUCCESS in 1h 22m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9ac93615d994d8f7f4fd16f91ec9f6e71a375b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 15:42:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea2da998cfdb0ed92e63059760a0136fa91fc9a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 17:25:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/939b3360662a43a0b3a47c95f1d59213\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d86b44298fcd49b1830ccb28f3632cbf : SUCCESS in 20m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/08d18498762943d7937975d3eab5e32e : SUCCESS in 32m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cfedc56061ba4018b9c8ce003f47048e : SUCCESS in 1h 03m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/794e563ca8344077af9a2bb9b29939c4 : SUCCESS in 1h 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28288d8399364717b78209d3046a8f7c : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/790ec70706274ae5878c24825ce23083 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4b4b0e70fae485d98d967d02ddf93fc : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4844709d91fd48c28bd0bbc8676fa89f : SUCCESS in 5m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/de47c419245c4063a634ac0afce02f5d : SUCCESS in 27m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c82bfc245fdf412f8f1a272921a93bac : SUCCESS in 30m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e6c66c4d871c4c03810b37000de05eaa : SUCCESS in 1h 19m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0d57d9b02f4c4b388486b08f5361481c : SUCCESS in 1h 36m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/872836981dc24472982136b01f79d5c6 : SUCCESS in 17m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c3d74fc47aa5400a827a7e86afcdf5ef : SUCCESS in 1h 17m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/39e7785ef2664279a4887ca5570c7fa3 : SUCCESS in 53m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9779bcd058114742a6572b74171c5435 : SUCCESS in 1h 25m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4a81358aa3cb3663f532a11adf71e32a6838951e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 17:25:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"345dba016ed69ec774ba86d9c48cb04741615d48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 17:26:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a589201a28b240b08e29ba24956d355b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19afd92402c4ca796965d8880619ec4 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76d78b50c0208795c4bc93b932e7fb0e79a036fc","revisions":{"76d78b50c0208795c4bc93b932e7fb0e79a036fc":{"kind":"REWORK","_number":1,"created":"2024-01-23 07:45:08.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/49/906349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/906349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/906349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/906349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/906349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/906349/1"}}},"commit":{"parents":[{"commit":"1580d778219b2d4e40c53ca9d5a437fe2a82e524","subject":"Merge \"Stop removing missing /run/nologin in RPM images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1580d778219b2d4e40c53ca9d5a437fe2a82e524"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-23 07:42:25.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-23 07:44:53.000000000","tz":60},"subject":"Fix gnocchi and skyline after requirements change","message":"Fix gnocchi and skyline after requirements change\n\nGlobal-requirements changed upper-constraints in [1].\nBecause of that change skyline and gnocchi build is now failing.\n\nThis patch fixes upper-constraints for those projects.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/900435\n\nChange-Id: I09cc2635cd422a859ccded887affb58dfbdc60ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76d78b50c0208795c4bc93b932e7fb0e79a036fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76d78b50c0208795c4bc93b932e7fb0e79a036fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
