)]}'
{"id":"openstack%2Fkolla~906516","triplet_id":"openstack%2Fkolla~master~I2a4df6121236379d88d0d7f2f7a888feab273003","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2024-01-31 13:27:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2a4df6121236379d88d0d7f2f7a888feab273003","subject":"Use full binary path when invoking ip","status":"MERGED","created":"2024-01-24 12:24:00.000000000","updated":"2024-01-31 13:28:21.000000000","submitted":"2024-01-31 13:27:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906516","meta_rev_id":"1c284a04c667b93b202ae52604a061e320fd26f4","_number":906516,"virtual_id_number":906516,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-31 13:27:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":1,"date":"2024-01-26 12:34:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-01-31 10:49:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-01-31 11:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2024-01-25 09:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-01-31 11:43:53.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2024-01-26 12:34:30.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 13:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-24 14:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 09:09:34.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-01-26 12:34:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-01-31 10:49:28.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-01-31 11:43:53.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"7214cf66896ac829b7230cc7739b50fbbba102a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2024-01-24 12:24:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a62f4a56b63d56fd0015061f61d203c74007556f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 13:57:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b446f33f60bb41bf9af56bdec515db2c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e67c9a5c0e0f48959e68b90712ab1010 : SUCCESS in 1h 24m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0efda543b32d4f5cad4ca6409ff11b3c : SUCCESS in 31m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/889bbcd66d764748a22eef27dfcb0e78 : SUCCESS in 1h 15m 18s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6afbe08d0374d66afcb6a86985f933b : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4740ea282c374d1ea3cb451bc9ee2a58 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/78533461f49c417b80717807d95f9293 : SUCCESS in 4m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4ee2c715771b4925af4b82c93649f01c : SUCCESS in 34m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/222720aa3c804859b2f8974aecfc889e : SUCCESS in 1h 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d9ee04b9273f521c0f94ca9d2ef61df506db8db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 14:16:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/377dda1cfb5c4821af3ebd8037e98989\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/49a529940280430b91ef213d87c8bf02 : SUCCESS in 25m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fb30900db8f943f48af3122eee21ec2d : SUCCESS in 1h 04m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/935b3f941f7f4c97b5014882d6315ab1 : SUCCESS in 22m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0afdf40017774f5783c056eb6d1461ee : SUCCESS in 30m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f465d39dfe144229b4db51a1a8477e1d : SUCCESS in 1h 06m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/679ce064ced74bbb958cf9743fe513e0 : SUCCESS in 1h 40m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d24a7aaab04b0d9614f0e2efffcfe9 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75815cb6e8634d64940b87765daef28f : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d68d9137733a4f0c8fa7fe093c09fff8 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/849f7c1c48b1489e81696e896eebbff6 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c7b5d9bda7346da8086ec099bbceb2a : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58c841a289164ad1a1d301da621680b9 : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1acdc7310457457384809abbe9b0072b : SUCCESS in 23m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f68c50008191462283de4022e030b639 : SUCCESS in 32m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ea32a615a5874760bc5b319bd106e8b1 : SUCCESS in 1h 09m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7ab84a2d10d0454b82b0ab5dcdb2e643 : SUCCESS in 1h 46m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/66ecb74f99804c2f8224be1001d964b0 : SUCCESS in 18m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b790f56c29d642dc9e1fb79fcabc0b63 : SUCCESS in 1h 16m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2b15dcbff2b94a39835bf48131bee5fa : SUCCESS in 50m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6f1d70cef76144c5a76b885490e03416 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0b51efc55782d5db968ecdc61f40e90d4e1d1983","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-25 09:09:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5e2f713b5cd51d671996350b88cad0afa164e1f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-01-25 09:14:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c814ddf7c2d3fafd1c05d0955265b096824e3ae9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 10:49:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b533f33d378a4722a0edd68b41351bcd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/69bbdbf87c4546828cc24cfecdbb4351 : SUCCESS in 1h 32m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/23a14a533b414f7e811fa7b1ee8992f0 : SUCCESS in 1h 27m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e1b6592b57bd451a9dfa37e12bb3d933 : SUCCESS in 1h 22m 24s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f52b54d8b6f464789335f2de4608fc4 : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6276afd25074404ebfb3fcad9b8146b9 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6cb3a9c0d6fa43e49db069d9282cf60d : SUCCESS in 3m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/79a811e43c1e4411a2413a1f3f29d504 : SUCCESS in 38m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f357315e1caf4da1a22e5209b96d5405 : SUCCESS in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da35ce594e300184c89b1b689d0b7254c74e4ba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 11:10:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32548103ce9247808bcf75044f548a67\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d0cd769cff1344b6871dbadd7cf8b6f4 : SUCCESS in 18m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/64e18d865a394e7c9c6a301c4b6115fe : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/41ee80925e5c4dad91757193ea7b20b0 : SUCCESS in 21m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d546bc35b6ff4236a6b94a806dd56c77 : SUCCESS in 32m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2580dea5593a43179e8272d9cbb4b4b8 : SUCCESS in 1h 06m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0a266bdbd2be481592ae0ea5c52c1be8 : SUCCESS in 1h 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa658392ce854f0fab7b727e5dc5612d : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79a7fbb0a634bccb2354f560f18b91e : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e15a16d3254841b38afa95fdab530dca : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2785280f521849f38380c4af57f2f9a5 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a303e3121cdc4186a2151491d9a9bcfc : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f8c0625e27b430ea048d38fb9a1a913 : SUCCESS in 3m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/88703c6857b34d6a96003aa834fa6ac5 : SUCCESS in 24m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/47341fef642c4280b4eabbbd6e4fe9a8 : SUCCESS in 32m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/063ab9ea37404afbbf121e99066f35d1 : SUCCESS in 1h 15m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5f928ce8ba7d4e27bf3ecba6653740c7 : SUCCESS in 1h 37m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1a8327ba829a4ae79c58cb3fe3c3b18e : SUCCESS in 17m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/21184243efd5401aae4555ccd8b28618 : SUCCESS in 52m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/801aa559a3ab40c4a67beee0778d9b29 : SUCCESS in 51m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/da16cdcbc80a4971bc97f41619da6200 : SUCCESS in 1h 23m 02s","accounts_in_message":[],"_revision_number":1},{"id":"42e3a886f730863a5987c16e751293f3062a7707","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-01-26 12:34:30.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7f32f03dc1aac5b95635f9c2f13190d2987b02c4","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-01-31 10:49:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02be7c12bc4a471bca89352a9eba25dd910a3edc","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-31 11:43:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68b2d99efc4d1fa269d96ea71832c46df315b9a9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-31 11:44:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23b0f66df733f1621820d5f8a8f6b4d78177307f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 11:44:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7df577488811bc73ecc6e9a497de8ffa695faa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:27:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc8cb0fd9af842e49b1269d037358b3b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a4da4a50bf1d4176a8bd79e548cf635b : SUCCESS in 19m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/74003f3ccc754a3c8a34aae36b711814 : SUCCESS in 22m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1ab8250781b14e8e9558bbf40f7bf787 : SUCCESS in 1h 06m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fb3e07b440124e61b0e5f9812dcb2ed3 : SUCCESS in 1h 33m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690fdc6fd5f244438c2c2ce1581a49c1 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/904b43655f16473fa903138a1f7f552c : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c955fa79c9094e5ab6547cce17d8de62 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0249f9d06d87468ab9df3b869511bf28 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92122b4554e9484ca517c7edff73864b : SUCCESS in 3m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7643ab93ff69461aafad3984254541c3 : SUCCESS in 21m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/325b7181cbf143c6b1fb005174990d21 : SUCCESS in 28m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/44e454a918514a0f988a62e2c4a6737e : SUCCESS in 1h 10m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8b82e3557a14447690f06265c46c36ac : SUCCESS in 1h 36m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4beeb3c5cf434102bda2df75faaee9c1 : SUCCESS in 19m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/116afb6c0a0c44c5a751cd20291d2ea5 : SUCCESS in 55m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c1eb538771c84f1997ba396768216a9c : SUCCESS in 1h 17m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/246d33a701f148fbac074508c5da6dcd : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fadd8ea31b00399baf185c3972aa4013817cdb82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:27:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1c284a04c667b93b202ae52604a061e320fd26f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:28:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a52cb10802047b58dd409fa235232ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/707b23d8a6d44383bc1a24e00de8982d : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99cde9f536c346d3920af3b8f87dc37b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3eb169e6c62747cfbc5ddc8ba951f7f5dda6382","revisions":{"c3eb169e6c62747cfbc5ddc8ba951f7f5dda6382":{"kind":"REWORK","_number":1,"created":"2024-01-24 12:24:00.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/16/906516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/906516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/906516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/906516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/906516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/906516/1"}}},"commit":{"parents":[{"commit":"76d78b50c0208795c4bc93b932e7fb0e79a036fc","subject":"Fix gnocchi and skyline after requirements change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76d78b50c0208795c4bc93b932e7fb0e79a036fc"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2024-01-24 12:23:52.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2024-01-24 12:23:52.000000000","tz":60},"subject":"Use full binary path when invoking ip","message":"Use full binary path when invoking ip\n\nCloses-bug: #2051111\nChange-Id: I2a4df6121236379d88d0d7f2f7a888feab273003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3eb169e6c62747cfbc5ddc8ba951f7f5dda6382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3eb169e6c62747cfbc5ddc8ba951f7f5dda6382"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
