)]}'
{"id":"openstack%2Fkolla~907901","triplet_id":"openstack%2Fkolla~master~I9c723b666897883bf5b3343fe92e06aaa0c89d81","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-13 06:55:31.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-02-06 09:16:38.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"I9c723b666897883bf5b3343fe92e06aaa0c89d81","subject":"openvswitch: Drop ovs_ensure_configured.sh","status":"MERGED","created":"2024-02-05 18:43:39.000000000","updated":"2024-09-13 08:34:49.000000000","submitted":"2024-09-13 08:33:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907901","meta_rev_id":"87e4b6337a0e9cce345621be7b4005dff2a74c8d","_number":907901,"virtual_id_number":907901,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-13 08:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-13 06:55:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-05-15 06:34:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-05-27 12:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-29 12:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-09-13 06:55:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2024-05-27 12:37:54.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 20:15:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-05 20:29:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 08:45:18.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-02-06 09:15:30.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-16 18:47:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-05-29 12:06:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"29043b3fab0dc08c117ea58d15ead455cc2ac329","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-05 18:43:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abe501261d14acc824eb92198121c45cf720c4bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:15:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79d05113c312438b9aeffe319a518e45\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/554796250763472297a91d4fd58c279d : SUCCESS in 1h 30m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d89d6576388841989b401ef81b116b0e : SUCCESS in 1h 29m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c87457d574474658870239935548dec3 : FAILURE in 16m 08s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a8cc731edd043dd9efc066c2aabd13c : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca3529ff84a9492f97bf6212df385353 : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b9b147e2a204cb2a7700652162b83dc : SUCCESS in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b85f7454282242b083c014a2231e5bd5 : SUCCESS in 25m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0633f080e6144d3f9130e2b98d1a9b77 : SUCCESS in 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de3bcd43e319cda4137567800d89401b584cad2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:29:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9579ad93b35a4d77921f6cb304fe787f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2d7c38e9f9e143419461e15b7dc43f63 : SUCCESS in 28m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ecfc792b0a5c49bb8d60c06d49fc9a1d : FAILURE in 49m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3afe593e7dcd4b5baa33d28674929af2 : FAILURE in 5m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/896d69f30e0b4beabf0aa38de134d94c : SUCCESS in 28m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dd6f3b0bb5b34b499cb0d840346b5b55 : FAILURE in 8m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/61723ba5d9ca409482e8e1d45f0a3546 : FAILURE in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd943018f614950ad5bbe42f6819937 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1e5712ae5e24bd9adce318ce08a8cfb : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03538b04c86b49b08a58809b2ca2d79b : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e52345d1c6ed4168ac7757d64ea4819b : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da7a92695704c1b9756a3e2f41115ee : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f3dc6212b8584fc396e748106ed1e797 : SUCCESS in 22m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3c9cdf4b92ec4881a758dad667a91bba : SUCCESS in 34m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ef8151ac972e41d09332548ef318ccf6 : SUCCESS in 1h 00m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/28abde11441a4bfb8554f4810ce69110 : SUCCESS in 1h 40m 47s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c840bf86923d4b2998850ae487e53b9a : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/729a73178b4a40639a2e52a932642dbc : FAILURE in 52m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/36d6f614b36843b296446dbd95d77a53 : FAILURE in 5m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/90948d9e8fb14ada9260021b494e5ed1 : FAILURE in 5m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cb192bc42d444b28bba8e1d3c981ccc3 : SUCCESS in 50m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/69d179a85841484aa1a8d62ac2b69cbd : SUCCESS in 1h 12m 22s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b9e35caf907245bf9e73275f40d9a920 : FAILURE in 7m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/754a8600f5b64d12b2bde93959f7da24 : FAILURE in 47m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c6c65cabbe33d39c505ea5772d4a2ea8557fd97","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-06 08:40:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48b9de1ef7f0ce12a7c8339c422898119797a018","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-06 09:16:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9287e3e694dd17ace557ac322e2522c37101de90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 09:54:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1c4d0e987b94e7c8d7b70639d43795d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b0b68028dc97482dbcdfc43adbfc94c7 : SUCCESS in 32m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/679b239deed74472ab2348cd9d285d11 : SUCCESS in 26m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4a310951c36e407495ecd954f19d748a : SUCCESS in 1h 09m 00s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fabe4840fac541e39b9e0872f59c791b : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce0883e937d34ba289a7136f3484a829 : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/97cc79bcd7ec423e8a1249769a02a00f : SUCCESS in 4m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/eebffc34530a438b98c9bfe9cb19dc73 : SUCCESS in 26m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/14a9ca82c9414c08b546e6caf464ddf8 : SUCCESS in 42m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17483a0d43fe099aa7c1dce473e1df5dc1f02642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 11:16:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8275338a1311457698f1bba920e67df8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9f847dfa5d7b44f0805a6d3e3581e7ff : SUCCESS in 27m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/05ffad8930084f4fb8b9e0d4825531c9 : FAILURE in 49m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1522ce231d4746c1b6d3d3020ca33929 : SUCCESS in 21m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/94ca360e024a4950949724e266354a39 : SUCCESS in 32m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/172d05507df44383bdc482e1ac673267 : SUCCESS in 59m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f792c731d4a74698b6d8ea3642b70f19 : SUCCESS in 1h 32m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de2ea248a95401eac709e65dafb16da : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ce0de10e3234bbb8eb8ccd5436ae838 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb70fa829ca2495893c99989ac8dcb1d : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/157ce4b3f8334202a561fe40c9ea5687 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4dfe8463b94476a7b00a77f343f70a : SUCCESS in 4m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2f02c68f5b0d486f8911cd51626f4806 : SUCCESS in 22m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d021c5b678854d06acf2e17f4d8396fe : SUCCESS in 41m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8d118ac945594a9b9c33d6263ebb6606 : SUCCESS in 1h 09m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e200d5109b9c41aa81d6226b3320f6bd : SUCCESS in 1h 44m 16s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/dc2c470f83ac464b9f830def27972889 : SUCCESS in 1h 12m 39s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0c9773a4cc33498684bb8d35a3df9c1f : FAILURE in 1h 01m 49s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/38f97af611bd401eace66b05a5717cef : SUCCESS in 21m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d5cb795094e0402ab6938b8d17ad085f : SUCCESS in 1h 18m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/19cd0c072eab467c8ec9e55b2dd1897b : SUCCESS in 56m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1d53f52671be428eb3e3b424f7d3fe81 : SUCCESS in 1h 19m 13s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e75a9192cfc048e69abd80f3bdae6364 : SUCCESS in 1h 00m 38s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c8c3c211b26c49bda30cb21d247fe2d8 : FAILURE in 53m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9c340a59982244d04f64f90311e52b3a0d2649f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 18:47:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d932b136ad2c811cfc3e9b367cd85bb86b2c457","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-15 06:34:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5c6676e774632001b3803f58e16dfd94a8ef576c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 08:34:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d8385651bd4cb38217429bd3559e82\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/46fe463c82df4ffdbc4e5e4da13e8ca2 : SUCCESS in 35m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c4a7ca8c5cd144d3a8bd038c5127ccc9 : SUCCESS in 1h 02m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1437440fe9de43ce9b1a378fa5d3da37 : SUCCESS in 15m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f078ffc882e94f30ba6833ad8ddc133b : SUCCESS in 30m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9ed10cafdc8f445f9b5f701fb688b548 : SUCCESS in 1h 11m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b21e8206264c4af3803bcf567ad46025 : SUCCESS in 1h 38m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b59489d6c4b3445dbd5a9e0154c71b7c : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d6476ddce2447848d8525e8863a17c8 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/935a23e2378641f182ada45dfa9359c1 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a65369c30c8479b872ec75a76664ac8 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d61aa9e6a4c4f57909538675b0d7255 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d736b5f9e4094045bd7b1396af5c352d : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ba4f4fabb67b4d81930a37b7b013a525 : SUCCESS in 22m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1f282ebd3762402991d8f39c3e4a4522 : SUCCESS in 32m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4c5077939b624441bdb474c1fb39d47a : SUCCESS in 1h 13m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/49d127452f3f472d887c49e600958015 : SUCCESS in 1h 43m 06s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ecc2da800fb5484ca404c040eb958dc7 : SUCCESS in 1h 12m 45s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/73187ade6963477fa27a232964c722ad : SUCCESS in 1h 14m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2d70ef18cb3f47ceb3d695ea8c90abf2 : SUCCESS in 26m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/24ee5cdde8e5431dbd44c763d4e7b573 : SUCCESS in 48m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bb57fdd9f2c84b51824f7c3ea5a847ba : SUCCESS in 46m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/460f9dc03d6845a7ba948f52b34be6a6 : SUCCESS in 1h 12m 19s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9243e952bd334f01b09ca8cd7168c009 : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c2d4a79c3c524a3799ab752c3a46c5dd : SUCCESS in 1h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"767c80d85a0adcbbcf1c5975c9c0d3eea876cc49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 09:07:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/470128f09aab4fdca7f049dc3a2ebdaf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/565e76e2f0b3438ebe2068d38cb9548e : SUCCESS in 21m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c50c8f81a55a445e8092098b7c3b103f : SUCCESS in 17m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d8d52bb99bbc4f39bac89e00ca106f0a : SUCCESS in 1h 48m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c32ca887c994c1584bfda97919d5ff0 : SUCCESS in 2m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6fefb4822025461599569b6023916620 : SUCCESS in 6m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e030246b49854e7cbd01f8094efb643b : SUCCESS in 47m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5db6b167d3344cb096abcf75e8acc57d : SUCCESS in 2h 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97b57695d868e70babba7ad6d4af530b39ab14c5","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-27 12:37:54.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8efdb42fbbd81944138365413ab4fa76f8d7eb34","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-05-29 12:06:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"decc7d2e6b01febd1d35a968e9db8087ebe937cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-13 06:55:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed47e48627c0a510de3575dbb37de44c8bb387ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 06:56:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"498f67890ab57c17c77c522149863dc19055321f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 08:33:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89a7db24329248eda366de2273205c0b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5a123a4a77214c07950ce1f48ff80bb9 : SUCCESS in 21m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bb690c8965fb41e3b9886a0ff0756fa8 : SUCCESS in 31m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5793681de61f48b880d5e234ba5126c4 : SUCCESS in 1h 02m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0029e4475f3545ac876859f4080aa4a4 : SUCCESS in 1h 37m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9067baaedbce428fb73e22b78843eebc : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e8ebfe6a1cd4fd49b8bbddb28c5a57a : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb4a309785444d729052a06ca9ec67a7 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/327960df39894df2b6d7e7fd243ebec3 : SUCCESS in 7m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/26e0d57e2bed4a79af4ab35d02c8378f : SUCCESS in 26m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7aa5b702975a4ac2bcc9602d03a27167 : SUCCESS in 36m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ec256386749c47a8b81fcffa93e99759 : SUCCESS in 1h 06m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/88828a610c2746239038962ac2b92e47 : SUCCESS in 1h 21m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7737b0f733a24edbb196b4d54a09ed98 : SUCCESS in 17m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/21ac62949fc44eacaa034cdfdfb73858 : SUCCESS in 1h 07m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/945e51ac9b284ab09ed3f7e04dadcfc7 : SUCCESS in 47m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d11d63cc08d54da58314ff2de5902aed : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5661bd2e6e53c331a60ffebe71aa1c5d99b855c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 08:33:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"87e4b6337a0e9cce345621be7b4005dff2a74c8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 08:34:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/435573f3208e47df8566cfdfccd0039b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8be927e2e84e4c85388099398dfe1d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c66a2cd63bd87b7052e4b76595e5e0a5f807eabd","revisions":{"42657631412cf6815474db35d086c50f1e0facbe":{"kind":"REWORK","_number":1,"created":"2024-02-05 18:43:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/907901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/907901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/907901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/907901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/907901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/907901/1"}}},"commit":{"parents":[{"commit":"410fafa8f0b3487d0bb1737d0f6965bc9cbcacb9","subject":"reno: Update master for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/410fafa8f0b3487d0bb1737d0f6965bc9cbcacb9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-05 18:40:49.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-05 18:43:36.000000000","tz":60},"subject":"openvswitch: Drop ovs_ensure_configured.sh","message":"openvswitch: Drop ovs_ensure_configured.sh\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\nChange-Id: I9c723b666897883bf5b3343fe92e06aaa0c89d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42657631412cf6815474db35d086c50f1e0facbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42657631412cf6815474db35d086c50f1e0facbe"}]},"branch":"refs/heads/master"},"c66a2cd63bd87b7052e4b76595e5e0a5f807eabd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-15 06:34:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/907901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/907901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/907901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/907901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/907901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/907901/2"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-05 18:40:49.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-15 06:34:00.000000000","tz":0},"subject":"openvswitch: Drop ovs_ensure_configured.sh","message":"openvswitch: Drop ovs_ensure_configured.sh\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901695\nChange-Id: I9c723b666897883bf5b3343fe92e06aaa0c89d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c66a2cd63bd87b7052e4b76595e5e0a5f807eabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c66a2cd63bd87b7052e4b76595e5e0a5f807eabd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
