)]}'
{"id":"openstack%2Fkolla~909591","triplet_id":"openstack%2Fkolla~master~I5570e405815d5f1fd0eeb017875cd2fa2ffed6bc","project":"openstack/kolla","branch":"master","topic":"openstack-exporter-1.7.0","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-22 19:51:33.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2024-02-29 13:24:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5570e405815d5f1fd0eeb017875cd2fa2ffed6bc","subject":"Use openstack-exporter 1.7.0","status":"MERGED","created":"2024-02-20 15:17:57.000000000","updated":"2024-02-29 13:25:59.000000000","submitted":"2024-02-29 13:24:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909591-openstack-exporter-1.7.0","meta_rev_id":"97863321a061b4ec7c8c3228b1dc0a7ecdc36d27","_number":909591,"virtual_id_number":909591,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-29 13:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-22 19:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-02-29 11:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-02-29 12:34:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-29 11:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-20 16:55:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 10:24:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-29 11:48:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-02-29 12:34:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"b326617de39314b80248fe4f9ef640e3e349fee4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2024-02-20 15:17:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f44b8010922c9bcdcc8e634f8eb0e926c943f7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 16:55:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e98b614425141689682d07da6104f46\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7ab58983b6264844bd18dadd3809c2d1 : SUCCESS in 26m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/55155cd96817461baffaf1c467bb527f : SUCCESS in 1h 11m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/89b29fc1f1934850b862c3caadbbcf5b : SUCCESS in 19m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7c42b6e3e4ad463b8aa4bb2a5a16d911 : SUCCESS in 27m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ec23c456fb5a4660b263a2baaba5af5a : SUCCESS in 1h 10m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/486be33655f648238ec3b60b7b61f4c2 : SUCCESS in 1h 33m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6268f4b199ec49e2b38155745ff64b16 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81a5b9f7950245ce82b03eb1d43849d6 : SUCCESS in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3e4b018423a47f1be113adc505e1921 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a19ff5df73ca44f2aa43b1493c6ce699 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1b37fa0b8a4d2da4cc190b7ec0e195 : SUCCESS in 4m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5ac0795a8e3b4c058191b0e121300895 : SUCCESS in 23m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/07e32f5348c842e7a2d26e11dbe8bc17 : SUCCESS in 38m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0950b382623247dc93f49cb6b724abd5 : SUCCESS in 1h 13m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c05fe51723ed431aabb40401cc6ae174 : SUCCESS in 1h 33m 01s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7000dd3f3a0041ab87e1c64996e0c03e : SUCCESS in 35m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f6481ecb09cb4939b47c328b8fd097f8 : SUCCESS in 21m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4397b174c7224cc5ac0163a0d662d6b0 : SUCCESS in 50m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7b5d54c0420641259e90c370ad94e914 : SUCCESS in 53m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ccaa2d99992d4599973bf2211002bd76 : SUCCESS in 1h 17m 07s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c4138cc995fc49a0a0f615be87bb9e2d : SUCCESS in 27m 08s","accounts_in_message":[],"_revision_number":1},{"id":"02c7f41505dfab0f1bfa4c0481465270366ea0bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 17:18:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/230f81cf0479499285a06165d02c3514\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/599743ca35384431bc0b680e802f8994 : SUCCESS in 28m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/38f775c58b5d4a489ee1814c40a9b43d : SUCCESS in 26m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e8339ee43c8c44b49e7645f6ff8d4918 : SUCCESS in 1h 58m 17s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/983aaa15708c448583a023e88e56c24a : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fbfe17106ad145a99516fa55b4115fcb : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26cd1f5caacc4292a13018a71e6bbbfa : SUCCESS in 3m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8244fd86f8ec45d69dc4b8fa58c4cbcd : SUCCESS in 1h 22m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e75322c3b0f441f085e381bf3ee13e0b : SUCCESS in 22m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c126b2eb378dbb297a17c6d417cf904b56fe9662","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-22 10:24:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"215304df7cf7a6e1dc66f9790bd60a895f245cf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2024-02-22 19:42:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"411db3c67aaee9c846038a63180b519162ac9f98","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-22 19:51:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c89a7241ef38c82ef79060553c19bcacb386065f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 20:50:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63772d486aff45098754fa53405fb336\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8ea8907666174a98a2a88c81e3988bbb : FAILURE in 26m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee7b38b4acd04c5390002ffbc0126b58 : FAILURE in 19m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fc95a7575c2043f49be21048186ba34c : SUCCESS in 1h 06m 24s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9163ffaee3042b79a42e9ae462bc046 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2dde2052ec4476bb60f6b3dbc9ada6c : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2bfb87a6ce99498c938ec326224bf27a : SUCCESS in 4m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ff986c74c9be48708e79074257619d15 : FAILURE in 23m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2a253bcd927148f0b8e37cb9f9f40853 : FAILURE in 46m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b546ca975a82c2ff0193ea24a3c0720960d1275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 21:24:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d99d07238b0f48a6afa8f7e01296f3c8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a80429ded19e4a7abd1bd7dcfc6b44df : SUCCESS in 22m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3ffacfc291c444ec959ec95650036ce7 : SUCCESS in 1h 00m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/04bc7de41117479f81ad724e29107c75 : SUCCESS in 15m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fee837d4676049a4a9af1b38fad586af : SUCCESS in 31m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/382263eb6ced447e96d022a42813f6ac : SUCCESS in 1h 05m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3b67af599b7145579c2f1e0ae3486e0f : SUCCESS in 1h 33m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987236ff454e4f84a059f78417b48654 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5642ef4876945be959d190b30dd3ea2 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b37fab528c6c4c21984a77f5b71dbb5f : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee202a75346c4af6b3ed23036ef1002b : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d63bd494b04779a7dc67c179d6b702 : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4b14f845b47a4ecc89e849529d25b0a4 : SUCCESS in 23m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a47b31f66e434589a3db94c19a73a0bd : SUCCESS in 30m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/78e3c732b5c146e5aa47a5654810d02d : SUCCESS in 1h 10m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e5b45cda0e3744b99324dbb42e7527e3 : SUCCESS in 1h 22m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b83a6501b3cf4c339387f11cbbbd94a4 : SUCCESS in 25m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/56d64895ac7b487eb23d5313ff475ccb : SUCCESS in 57m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ccfe004c02424ede8cfa1648ac59efb6 : SUCCESS in 49m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8a89f1e91f6b4395a395da12f85e2fb0 : SUCCESS in 1h 19m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b8294e59b361fedaacb6dbb54f250409d202297a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-29 11:48:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb9418b1e183ea75d22820d5078bd12b8f51f286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 11:48:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a4ce221de64e6125f35332197ea49850379de80","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-29 12:34:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2676c3d5a3b7b119e9a66ce16e61592e749c4a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:24:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/541e600430f8466cbaa20e0a05e38939\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/34e65005c7934e1ba65eabd24e2a4e0f : SUCCESS in 19m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9e0d250b92c348b491bfd83663155efd : SUCCESS in 36m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/20763421b85b45a29e2460c9690153e7 : SUCCESS in 1h 08m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aca35867224c4d909cbe9bbe0763197f : SUCCESS in 1h 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65d44a9c18214599802e03b6ccc526cb : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdaed505d0d84f91a402ac4b0dbb2284 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67d4bc4b6b2945a4a180c13713e3cda6 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/607eb069932c4dba8e0d710cd4b8b31d : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7c9b216da1d8470c822152e13ec8ac8e : SUCCESS in 26m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6f5d09f464cc4f8ba68f2f7a1948e03f : SUCCESS in 26m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d7bdcbca77a44d6bb0a733ef79f26ac5 : SUCCESS in 54m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/642e08fba43c4fc5ac7ab31202ffcae1 : SUCCESS in 1h 11m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3de00c0e1d2a430a84d37c095be331ba : SUCCESS in 17m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f0d8bf6e5ca94b9dbea175640557a447 : SUCCESS in 1h 02m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/18e1f44efab54724bc248dceb257cb19 : SUCCESS in 55m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c5c86c4e4f044f8ca0b6ba404634408e : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":2},{"id":"292c3b9cb660b582b487be357a7afb5d0f787836","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:24:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"97863321a061b4ec7c8c3228b1dc0a7ecdc36d27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:25:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9236d36434614d3097238995a0f64897\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e5ffda520834c189d3212563bea4511 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4196f8d9fba874b88835a523f04649f90b608477","revisions":{"88c582ceeabf354fd7baab37e94c43fdd0d9e728":{"kind":"REWORK","_number":1,"created":"2024-02-20 15:17:57.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/91/909591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/909591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/909591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/909591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/909591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/909591/1"}}},"commit":{"parents":[{"commit":"b276a9cdb52502a9abe1b0f3cf866d952f201b5d","subject":"Remove the Prometheus HAProxy exporter image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b276a9cdb52502a9abe1b0f3cf866d952f201b5d"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2024-02-20 15:17:18.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2024-02-20 15:17:18.000000000","tz":60},"subject":"Use openstack-exporter 1.7.0","message":"Use openstack-exporter 1.7.0\n\nChange-Id: I5570e405815d5f1fd0eeb017875cd2fa2ffed6bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88c582ceeabf354fd7baab37e94c43fdd0d9e728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88c582ceeabf354fd7baab37e94c43fdd0d9e728"}]},"branch":"refs/heads/master"},"4196f8d9fba874b88835a523f04649f90b608477":{"kind":"REWORK","_number":2,"created":"2024-02-22 19:42:24.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/91/909591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/909591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/909591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/909591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/909591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/909591/2"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2024-02-20 15:17:18.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2024-02-22 19:42:01.000000000","tz":60},"subject":"Use openstack-exporter 1.7.0","message":"Use openstack-exporter 1.7.0\n\nChange-Id: I5570e405815d5f1fd0eeb017875cd2fa2ffed6bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4196f8d9fba874b88835a523f04649f90b608477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4196f8d9fba874b88835a523f04649f90b608477"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
