)]}'
{"id":"openstack%2Fkolla~909644","triplet_id":"openstack%2Fkolla~master~I6b6ef656bc2009a2dc3cdfcb01956d50fba4b7b9","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-21 16:10:54.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-02-21 22:41:44.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-02-21 16:06:10.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-02-21 22:41:44.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-02-21 16:58:34.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I6b6ef656bc2009a2dc3cdfcb01956d50fba4b7b9","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","status":"MERGED","created":"2024-02-21 10:10:46.000000000","updated":"2024-02-21 22:42:45.000000000","submitted":"2024-02-21 22:41:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909644","meta_rev_id":"a8ad96dbdee7bc865b8b41671f37ab2d7b1cdc78","_number":909644,"virtual_id_number":909644,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 22:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-21 16:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-02-21 19:42:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-02-21 16:06:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-02-21 16:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-02-21 16:58:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-02-21 15:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 11:43:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 12:26:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-21 14:08:01.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-21 15:34:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-21 15:34:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-02-21 15:34:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-02-21 19:42:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"fae333964b8168673edb0f12a47a0f64ff5804d8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 10:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc747cc7fbfea2c37746397cb3210e51c0589a6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 10:38:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7ea4962325422c875d77a85e90829b9ce3af43a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 11:43:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/005dae6822c843f58fdf37007c1ddd7c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ba136b1d539f46b58a625daa81b1ee2e : SUCCESS in 21m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b5e72e52a66f4196b18f59a8fa9bfd2c : FAILURE in 16m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/49b663332f87404aadb8a010a3ba3d20 : SUCCESS in 1h 03m 34s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/159a1305b9b44ac9944833168b5c6050 : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32f7e7aed72c4b37b7c18fc440bb6aa4 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7619a8f9f8741fbab6ab551e781b726 : SUCCESS in 2m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9e13884a0d8c474cb730cbca3a81c004 : SUCCESS in 1h 02m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5692834553284d48875f95dca9354069 : FAILURE in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"640f63296a0e4c9b1ba99d4ae2b19a0bb24355d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 12:26:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13ccba792c7841b183efcd8da54b08ef\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3178c74815e241f5aaca0dd1948e6c19 : SUCCESS in 25m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ef4d78f1593a4bb9b36f769cb9885212 : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ecc0b98b9fde409f9ab4797c71700ec1 : FAILURE in 17m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a563f2e51feb49e28ff354efdab3f3d4 : FAILURE in 25m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2febeabcee394e4085b3a002f53f192d : SUCCESS in 1h 06m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/397bba5d348e46bcb4f3f33ce5407571 : SUCCESS in 1h 37m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24f0b2be2692428989c4ef5e123b7033 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/323fc37ca6f642809b35abdb70630958 : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dba9123cde4a42cab61d201556e8bccf : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42f0230818a04b4ba5c247c40992791b : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723bd9317d0f467dbba7d07cf0793297 : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/67d164d7990b4d4bae755a820a14625f : SUCCESS in 24m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cf53163a259542cd8bb8e6ec9a474a7d : SUCCESS in 29m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dc8c2d60752740cd84dc25d65117a8a9 : SUCCESS in 1h 14m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/22d30dee0e744cbdb8cb2497c3189ed7 : SUCCESS in 1h 42m 23s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/02bd76f3cb7d421e9121d4cbe9c07f4e : SUCCESS in 33m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/54288541dab44ad5b3843a8bb5dbd791 : FAILURE in 22m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e25c337ab5c04dc1ac27961577a32cf6 : FAILURE in 50m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/307a7cd492324024a6c691875c68c67e : SUCCESS in 52m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/058c3d8dd48e452a9a62ac1b40ecaac1 : FAILURE in 1h 16m 14s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/62b7c73a29d44225b85a86e2a17fc0d6 : FAILURE in 16m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d15170bd7c4eab1246b8f5cf5362cd13f979bb30","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 14:02:00.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a9ffd519f8f8c62dc7b3f84b8491213d35a89827","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 14:41:51.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d9b93e84bd9df36e34f38c260430ca4f65bd258f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 14:41:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"b88ef2983bdc7c02abb3818e1a45789585e5503f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 15:28:37.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"05f9b87e993e232652bbf24ae6359b630f90643d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 15:37:07.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"15c795241879804bf4356ece118ad497555a8579","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-21 15:37:31.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c422d8e78ba47b64a436151dac3e7bf2ab0053b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 15:42:10.000000000","message":"Patch Set 5: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c8eeab5e3b98d7cd93cd9e8c3af762f9b422efcc","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-21 15:50:10.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb6fb2eeae5ca3707b73b09464ac6f718868ac35","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 15:53:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4208d19830089d2fa23c508916897c2b7d35715f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-21 16:06:10.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f6365696369a5047ac7ae354932acdc79635b920","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-21 16:10:54.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a4d7ad7c641c98491d93d8c4fa27ef179eac60c5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-21 16:58:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"43eb2416b463e4fe69770e687f42dd15a4c658f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 17:06:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9748037f734d2592f1fb77b270af62\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ea37f0c7cf1a4c4d83d06375af9dfdc5 : SUCCESS in 1h 17m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2d68c863783a46eea8024b637669116c : SUCCESS in 20m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b708f56c29b04367a5c8bb1af007097b : SUCCESS in 1h 27m 16s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d436c25651874d42821cdfb804cee7ec : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4fe19462a219498583738d4fba822373 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18cdd35688294780afac2ecb69580163 : SUCCESS in 3m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bc3681e0cdd54844887c886638310ef6 : SUCCESS in 1h 21m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a93811ee450f48369dfac1c67170ff6a : SUCCESS in 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a5e5e09875f1a3433e0aa23a3c2e1d394748f43a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 17:17:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d95d7c9245294a1f80aa3f334a480338\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ab7236d9139c40848af060b4199ca5fe : SUCCESS in 26m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c7e19a353f6e4e97b6c7f0c665dc2bb7 : SUCCESS in 1h 10m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d7160946f5194344aeea11a589c1fe25 : SUCCESS in 20m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4912167a35b84bbdb4618ba495d74a6f : SUCCESS in 31m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/22ea1ec59c554cd993b9afe01bc7618f : SUCCESS in 1h 05m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/20627589b6c94cf394caf4f9e5e891a5 : SUCCESS in 1h 32m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45807161434c4b52adcbf97e19a9efe7 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13e0754329f6490cbba5c4c4c350634b : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85da2251f27d49849bf3d8be99ed133f : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/809aef7f2cc742c0b38040288720a1a5 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/048c10ee85fb41de9b1452b2d359cb8b : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5702aaf8776645b3b45f65e8f903be6f : SUCCESS in 24m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c3c22de0c0eb4b87abd388a4f406824e : SUCCESS in 29m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/730be0169ee8478cbe056f608e9e8d92 : SUCCESS in 1h 06m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4ecdba9406da44a8b4bf2835ac5e3f5a : SUCCESS in 1h 33m 18s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/53d7432caa8140068f9755bc49e1ec1f : SUCCESS in 31m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/34b0d4d544904079ba507cbc13a47aa6 : SUCCESS in 23m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b661255ccf4e4dbe84ac8e7f29a5a41f : SUCCESS in 1h 03m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fbd748ad2e154e14a40ef65cd185a6a6 : SUCCESS in 52m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7cd93fcaaaa046db872dfa6e443ac17a : SUCCESS in 1h 35m 48s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/863822badfc240eda23eac4b8c6cedd0 : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":5},{"id":"b23d16e8a95e3fd812218fba3ec995e4f135e071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 17:17:49.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"52075719b3a09a0d094a6fc6a61a18c9c65b9ae3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:56:20.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ebef95e80a641959c416111e4e7b061\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e2445dd07456476f866f7bb266703fba : SUCCESS in 16m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7e22aeb8d0a248cc800b1f0485451f00 : SUCCESS in 27m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6f7a5a7cc8ea4ecfb4199873a9ca921b : SUCCESS in 50m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/633ed94f1d7445489e58fe7c7daa59e1 : SUCCESS in 1h 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c4fa928d02417295a983c0e2688e48 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0eb97172ad74c63963b9596c39be8a8 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1d970e2b60341438e9222de7348d3f5 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6690db7beea74191a38892554339701d : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8dae512d26c943df8af174f5db90556f : SUCCESS in 23m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b45f00b89bb84179bc611b96ab6e73e5 : SUCCESS in 27m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ef4ee78015c44c39854644a83f43dbe1 : SUCCESS in 1h 06m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/25ea66401fe24ea68c38f4d107ad99a1 : SUCCESS in 1h 33m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dc113a289a914baf803f2864197400d0 : SUCCESS in 19m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/db7cfb9468aa49bc965514fabf73847d : SUCCESS in 50m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5e184a2fc4b5404299e8680290bf4858 : FAILURE in 36m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/563b9720a5934807b5ba98d8d67dfed0 : SUCCESS in 1h 21m 37s","accounts_in_message":[],"_revision_number":5},{"id":"7e280043aee88c7c6f4d05de079a8d4c55e00524","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-21 19:09:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"908966d453bb0c81ac31c639d3cf6b0dc268556a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-21 19:42:21.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"580947b4ed789630862d17ac8794d22ed88e0398","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 20:22:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae84e37bd27c49ca80c556455448fbd1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f7a64c48ec624e6ca438ab247c12d0ba : SUCCESS in 21m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7c47ae356fd04b1986771953833da053 : SUCCESS in 58m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/58a6ec524d7c41a2aa2e291d7b43f371 : SUCCESS in 1h 09m 56s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a81f5b2b7ee14cc6ac9c1baa3a824aae : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/adbbb5b639d34b6499ab6c8bae11b23f : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ff6697d3adaa4a8eb6388aa1046b7faa : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/66b441e5e5c44920a69d90869e6f0eca : SUCCESS in 23m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/75fdda7749014a4994c76acc4030f26a : SUCCESS in 52m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fd4af5a1eb7746f3f4593e6a4f6b549bc06c317d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 20:44:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d915de8e0e264fe7a67b5cc2b40078c1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d75ee94d279047beba8fd9d198de01d4 : SUCCESS in 24m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/861e62500c7b453e8a8c96deae32960e : SUCCESS in 1h 00m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8fcf9ab6e71d496d8daf187ba1281c92 : SUCCESS in 16m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1a156e3905884e3589080bc3460fee5d : SUCCESS in 24m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5090143a91b24315bfc94e08a833f668 : SUCCESS in 1h 09m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/538329490ebe49019fba63882c4dfde5 : SUCCESS in 1h 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6437a2a0a026496aa789e560900b175a : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f53aa3b41b1f43d896e5898181afe42b : SUCCESS in 2m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b339a63086c34a9b86e609766932cd77 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98427b02dcea4e20986e33d1703007b9 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40df831f23b44309b38b376cfd99905a : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2acda73a5eeb4135a78c447badb4dcc2 : SUCCESS in 22m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/13ee2a769ffe4ffca7b499ae3ec1e1b1 : SUCCESS in 32m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5dbea669830349a5a48092b81f7252f2 : SUCCESS in 1h 10m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/40d31e5e3ee44909aaa5e3dcb0c44d8c : SUCCESS in 1h 28m 24s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8cd6eff10f7d4fc08325288f5171d036 : SUCCESS in 33m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/61d79a613ac14c04b246192c04227e1f : SUCCESS in 18m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/959a51f15f3a42e4ac963526c197e7c2 : SUCCESS in 47m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/38964eafd27042ee94a99764f49a83a0 : SUCCESS in 50m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ee8c2074fa4b44afa8f011562eecf85d : SUCCESS in 1h 08m 16s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/82c2d21be6ea4f0299cbe81a2c4b26df : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":5},{"id":"ce2d471ccb91a8c3411f86b263c486689717590f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 20:44:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c08ad695daa6a9cfbff0806f1d6aea15a2156511","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 22:41:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adfedad9c01a43af9e90edd5e0d4c0e2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a4a350209ced47c4ae470bcd8fda4f8b : SUCCESS in 13m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/754d0179a5b64bf4b2e0d70512b0043d : SUCCESS in 20m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/71f81676c7204cb59a67e70b405a633b : SUCCESS in 1h 06m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b30881c38f0b49a1aca9863518687849 : SUCCESS in 1h 22m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf139b087a0f41f380bdb07960a970c6 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43681f02c1564912a79bf803b8a9896f : SUCCESS in 2m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6490455fe90d4389bc55690dfc318d65 : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72806603f1d84b6d83a2a7fe13f48221 : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2d89b33ddaf64de7827c9165fcab4807 : SUCCESS in 25m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bbdaa1afe135434aa419802bad96858c : SUCCESS in 33m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f6ebcf0cfe3748a38ef3caa5300a3e30 : SUCCESS in 1h 11m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/db057c43db534e10b17348b668f08e28 : SUCCESS in 1h 49m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1fe8986a66794bec8e22a128fcad8dba : SUCCESS in 15m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5805ebb9618140c29c1d3787a4b2eead : SUCCESS in 44m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/526ded382dc844b9a9f512ab9151890b : SUCCESS in 46m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/23b538d9c2614c788009d40eb781d915 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":5},{"id":"301884ee3fe9faa6f8b820b45778369879dd64d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 22:41:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a8ad96dbdee7bc865b8b41671f37ab2d7b1cdc78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 22:42:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13483ad089f44485a8d40e4980ad076e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be76d642f8345ad967ce54e41d026cc : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"756ee984c3c761514115c342e7cba9a3cb533d02","revisions":{"c5604f59207725a8529e0e1404c878a140bf594f":{"kind":"REWORK","_number":1,"created":"2024-02-21 10:10:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/909644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/909644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/909644/1"}}},"commit":{"parents":[{"commit":"b276a9cdb52502a9abe1b0f3cf866d952f201b5d","subject":"Remove the Prometheus HAProxy exporter image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b276a9cdb52502a9abe1b0f3cf866d952f201b5d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 10:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 10:10:44.000000000","tz":60},"subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","message":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD\n\nChange-Id: I6b6ef656bc2009a2dc3cdfcb01956d50fba4b7b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5604f59207725a8529e0e1404c878a140bf594f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5604f59207725a8529e0e1404c878a140bf594f"}]},"branch":"refs/heads/master"},"97aaced02179be36dced01442377d97a66be5f7b":{"kind":"REWORK","_number":2,"created":"2024-02-21 10:38:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/909644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/909644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/909644/2"}}},"commit":{"parents":[{"commit":"b276a9cdb52502a9abe1b0f3cf866d952f201b5d","subject":"Remove the Prometheus HAProxy exporter image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b276a9cdb52502a9abe1b0f3cf866d952f201b5d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 10:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 10:38:35.000000000","tz":0},"subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","message":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD\n\nChange-Id: I6b6ef656bc2009a2dc3cdfcb01956d50fba4b7b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97aaced02179be36dced01442377d97a66be5f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97aaced02179be36dced01442377d97a66be5f7b"}]},"branch":"refs/heads/master"},"a05c19d7c0b26e8efb3dba6bd008783e3aaaf3c8":{"kind":"REWORK","_number":3,"created":"2024-02-21 14:02:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/909644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/909644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/909644/3"}}},"commit":{"parents":[{"commit":"b276a9cdb52502a9abe1b0f3cf866d952f201b5d","subject":"Remove the Prometheus HAProxy exporter image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b276a9cdb52502a9abe1b0f3cf866d952f201b5d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 10:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 14:02:00.000000000","tz":0},"subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","message":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD\n\nChange-Id: I6b6ef656bc2009a2dc3cdfcb01956d50fba4b7b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a05c19d7c0b26e8efb3dba6bd008783e3aaaf3c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a05c19d7c0b26e8efb3dba6bd008783e3aaaf3c8"}]},"branch":"refs/heads/master"},"ba43b2fbc8dbb53f87d6bec385bb530b80126359":{"kind":"REWORK","_number":4,"created":"2024-02-21 14:41:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/909644/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/909644/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/909644/4"}}},"commit":{"parents":[{"commit":"b276a9cdb52502a9abe1b0f3cf866d952f201b5d","subject":"Remove the Prometheus HAProxy exporter image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b276a9cdb52502a9abe1b0f3cf866d952f201b5d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 10:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 14:41:51.000000000","tz":0},"subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","message":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD\n\nSince 2.12 OpenSearch requires setting initial admin password,\nthat is not trivial/weak.\n\nChange-Id: I6b6ef656bc2009a2dc3cdfcb01956d50fba4b7b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba43b2fbc8dbb53f87d6bec385bb530b80126359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba43b2fbc8dbb53f87d6bec385bb530b80126359"}]},"branch":"refs/heads/master"},"756ee984c3c761514115c342e7cba9a3cb533d02":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-02-21 15:37:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/909644/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/909644/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/909644/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/909644/5"}}},"commit":{"parents":[{"commit":"b276a9cdb52502a9abe1b0f3cf866d952f201b5d","subject":"Remove the Prometheus HAProxy exporter image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b276a9cdb52502a9abe1b0f3cf866d952f201b5d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 10:10:30.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-21 15:37:07.000000000","tz":0},"subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","message":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD\n\nSince 2.12 OpenSearch requires setting initial admin password,\nthat is not trivial/weak.\n\nCloses-Bug: #2054586\n\nChange-Id: I6b6ef656bc2009a2dc3cdfcb01956d50fba4b7b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
