)]}'
{"id":"openstack%2Fkolla~909879","triplet_id":"openstack%2Fkolla~master~I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4","project":"openstack/kolla","branch":"master","topic":"kolla_rpm_repos_yaml","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-10-02 16:05:25.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-02 13:44:34.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-02 16:05:25.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-10-02 12:36:58.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4","subject":"Add support for rpm to repos.yaml","status":"MERGED","created":"2024-02-22 14:52:41.000000000","updated":"2025-10-02 16:14:55.000000000","submitted":"2025-10-02 16:05:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909879-kolla_rpm_repos_yaml","meta_rev_id":"28ac713652433f5d347c385b03c3e6cba86ffebd","_number":909879,"virtual_id_number":909879,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-02 16:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-02 12:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-02 12:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-10-02 13:44:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-10-02 12:25:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 15:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 15:45:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 12:29:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-25 18:29:48.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-07-18 14:28:36.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"REVIEWER"},{"updated":"2025-10-02 09:27:45.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"37484b41dcd6c06914319a7188c16a41ec27ce07","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-22 14:52:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dea041118730d9596996da96a87ff6f3da9d9df8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-22 15:25:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"830ce79a83892d492807ba8ed01a0ab56b6f6c9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:34:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d9f41dc076a42c28349291cab102313\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fd978d7cd3de43c8ad402c12fb79abe1 : FAILURE in 6m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/30be49167f2148c0b4f5d250dd641999 : FAILURE in 3m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ce5c0940dfcd4d72935cc1347f0eb2b1 : FAILURE in 7m 05s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a8fe9e2534a49eeb1da98c07fd18a52 : FAILURE in 2m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e8f09f56e6647b683cf5fe3be02c48d : FAILURE in 4m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99885ce2758146bdb86f42b81bbfb96a : FAILURE in 3m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9c6703c35ebb4d379635941024a38d36 : FAILURE in 5m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f2c6fccf57f94ece8ffd1e485a52d79d : FAILURE in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d88ba65fd5176aec5f8174ac131d17f32bd9282b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:45:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6059a951f2234b68a1556c3cf325ce72\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/888ac1cc340d463fb827850505ffbc10 : FAILURE in 3m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f06611d54cbb45e1b141174361b35908 : FAILURE in 7m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e7cc823edf0344868cab05484e7590dc : FAILURE in 3m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/16d4a61c7db84dd99df9593fbb9f1d0b : FAILURE in 2m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/086fb0f3baf14ec1ac099efcf9ee4895 : FAILURE in 8m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/30890a7cb3db4bf4bded3200912631c1 : FAILURE in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c9b78c965046118e18682b55b3709d : FAILURE in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c6516eb40e347f3b627f7ef876de2ec : FAILURE in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/192472ece2b149b28d9a4e50d63bdc48 : FAILURE in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23c7354e44354bb7a5d9146e34639132 : FAILURE in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5ff3a8abfa4c78b93faae38219c593 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/04d303f7d1f84460be2f5a24c65fff94 : FAILURE in 4m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2bdc1bd7d73746f1a91629d751848836 : FAILURE in 4m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cf5e838653b64b8a84fddeff424185eb : FAILURE in 7m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/16f64b3c6610405581a77dd4ca89d651 : FAILURE in 8m 38s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e6a30fb5b2e94be8a52db1f2b94cc7bc : FAILURE in 8m 45s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/edc599a079944c80a0fc14ca77110aef : FAILURE in 8m 34s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/2941ba549f1443149aab40edf7aa7a1d : FAILURE in 7m 39s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ae1364f91fd0431f925d719922cd07c1 : FAILURE in 10m 50s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/73ad07deffb244b2aa485be3c92b11ab : FAILURE in 10m 06s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/74ac2b573daa4aa4a62a5c3a72a4f00b : FAILURE in 9m 46s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4f59219ae5ba4008a9a2143107dd154b : FAILURE in 11m 49s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/23c838c097c84eed9da067bcf45e55c9 : FAILURE in 9m 26s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0452bf3587d24e7eb41a9f838832cb49 : FAILURE in 10m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8ea3633d04824d7d9d421a26b515de76 : FAILURE in 3m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/354e1b0834c642b1b5cc6911bec0fbd6 : FAILURE in 2m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/057b33bce0cc45979069a7c40ef421d1 : FAILURE in 5m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e98aaf1dadad476ab6f11bdae60a3bad : FAILURE in 6m 43s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f45cb4c6c95a444eb8c0880c5ccc370d : FAILURE in 8m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/67ad93f461a94c7facce05c3530decc5 : FAILURE in 7m 09s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c0c9f0b5c87a4474933ec17465820e71 : FAILURE in 6m 52s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/52cf3865febf4a03a6826f8233cf48cd : FAILURE in 5m 37s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8fc6d8e99ecb423ea0552f863e48d0d6 : FAILURE in 7m 36s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/820edc1471e1437fbb1b567aa2c48154 : FAILURE in 11m 13s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ebf96db0a17a4c6f956af140646bef65 : FAILURE in 11m 30s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d2d7376722f040468995c7dbfa1f921d : FAILURE in 8m 08s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2ff1e6812efc49ee82e5e514425f90ad : FAILURE in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"9ac448186607d2acdbc0eabdcd66278fe044a199","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-23 11:24:11.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"58392bbf01d3032d42e47e74773fbc8df798e3eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 11:48:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5744048633a405c9376444134020b69\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a35197ac901a462d8ff6ffd08cc5b0bc : FAILURE in 4m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1a62e2d1725d4f29ad66483152287e21 : FAILURE in 3m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5c4b746cb3e2417b8812c90722e2005d : FAILURE in 2m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e809664333de45408048badd54498337 : FAILURE in 2m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8543023c57ac48e683d6ca3b70ce53bf : FAILURE in 3m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/593b85c858bd4a04b91d5739c627bc30 : FAILURE in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1766768c15b54c5d99cf85a129eca9f6 : FAILURE in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e888c9df95b4747a0f6d9a335aca61a : FAILURE in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d1756ef936947ef8df8b83ecad92fa5 : FAILURE in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1426ffee4e214517a48be27da23c4242 : FAILURE in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39904eb8af8546299082e93359610db2 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/91f9138f87da441abaf8ca43fe53cb26 : FAILURE in 3m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e05230173ab74799a6735232fbfb19c8 : FAILURE in 3m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/723ecb86a5e7421abb72a0e4844d315d : FAILURE in 3m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ca8f7be8ac7142b39c4733445fe2dbfb : FAILURE in 5m 05s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ad36188872e54cff80f1c0dfcacd392e : FAILURE in 4m 15s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cfef9ac5b7ee427ca71a4fb7c55b3c81 : FAILURE in 4m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d85e1eb1adb94851ab48aa7072fc71fb : FAILURE in 4m 44s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7bd44639f4284712813a90dd778f9b2e : FAILURE in 4m 39s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/38b4d77a235d4dc69db4a8a9432e7f02 : FAILURE in 4m 42s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/20f4ca78f01f4213972b775960c8d0b8 : FAILURE in 6m 51s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/bbb25b0201b841f7b6c27d73ee407c5d : FAILURE in 3m 08s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/56e27602fc1a4a40a51106432ff9ede9 : FAILURE in 4m 55s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/44296356a75e42b4bef6972c10bd6b15 : FAILURE in 3m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/30d2c9b33a3a4c4682ca2ddd5f862d25 : FAILURE in 2m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9d4f29f4010b4a348c3ea310b957c259 : FAILURE in 2m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2cef85b871aa4775990e14e980c0c9f0 : FAILURE in 3m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b1b4229e013149b2bdb35d21399660de : FAILURE in 3m 31s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8dd2f25835a3401d87e789cea81bda42 : FAILURE in 4m 39s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/01fe17baa76a4879aba36556aad3dc60 : FAILURE in 4m 38s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8b2eab8051e146c68426b74ba31d3534 : FAILURE in 3m 40s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/001996c2a1224db8aa27885ae7557cb5 : FAILURE in 4m 53s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/1c386f1eee924c4db556ecdc922aa367 : FAILURE in 4m 12s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/193b6761e44b46f6913ad7f6f9867dbd : FAILURE in 4m 33s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/121ba9ca02624814aac0a852c3978fc5 : FAILURE in 3m 31s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e0c3283175d94995b0cfc60396e3504b : FAILURE in 4m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/eebff365a164499a865bc241ca4dd441 : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"0ff39a50a8ae1148e9a082aeac47bbeaf68cf030","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 11:56:39.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac74079296bf4ffeacd8547cab946d3e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/74eeebb530ab47b3939bc195cf8d12f9 : FAILURE in 7m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/17f16cc827e74f26a0978c5693a7ef66 : FAILURE in 3m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5e3147b1e5354e03bd99897680f67b80 : FAILURE in 5m 07s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87b1be413f894e8aaae7cff164d46821 : FAILURE in 5m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/63e4b20b8e004e03a93bbd271d9b9830 : FAILURE in 3m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f820c9cad8a4e6c904a2d0ce382a17b : FAILURE in 4m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fe4e7e786a5a481789f5e3f62535a803 : FAILURE in 3m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c14acc3f11294ae0b3a4d6d22a3f4c02 : FAILURE in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"487fae981cdfffa49a881fd37c709f8f6f4b69f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-23 12:02:42.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"facf88a417aedef1408255454ba194fb6bec0b2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 12:10:39.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c4f56718cbf4ce588e31c3e507087af\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/853cf92d947f4591ad1b4918df72b2cc : FAILURE in 4m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cc92c211ea5e42b1964e0bbe12fb18e7 : FAILURE in 5m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7a27f3f3e50845c09eba82c60e396a17 : FAILURE in 6m 36s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4326f251bfa5426ba987c23b57c22b92 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e941096b92b64af3ad43e8470657a541 : FAILURE in 2m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73e3daefcc414f79b103583880cfd85c : FAILURE in 5m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6632ab72fde04999910b680d7cb0f36c : FAILURE in 3m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c50a4f26ada741abbb5d6d0054fbca37 : FAILURE in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5063cb32fbd7fb82447338cb872132bd4c0751d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 12:23:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3952caa9e4fe4105bb1d5972fd20ad7f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0a93bb0fcfae46e489698d37ef718ee9 : FAILURE in 4m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/527b5408d58742dc82d65f32808e8fec : FAILURE in 7m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/40d30b731ded4856b5d75bed13852714 : FAILURE in 3m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cb4796a65f2349f7902083b244ac1e24 : FAILURE in 3m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f774544f7edc4322867f3d092dad6490 : FAILURE in 6m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8006b2fc7ae449a69e7661c0a4a61b00 : FAILURE in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a1399a28e149a38c716bedbc2ac5a0 : FAILURE in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5612e66a7c5747efa5ada90db082257c : FAILURE in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05e9821e5c73438ea81384a3fc076047 : FAILURE in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/906a844c2845417ca0143b26a0bf8470 : FAILURE in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7557116f41d4728925ee4b01aeb755e : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f8f5bb97b3884c2e9e757b7ff12b5ae9 : FAILURE in 3m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/81a632ffbe9f4179a74b64d28f34fb2c : FAILURE in 4m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7be4c3377b2048c3b9c274e7b6ba92fb : FAILURE in 7m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f066f5bee87949db974a27733c5ad0d8 : FAILURE in 8m 48s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/04cd14e6bfbe4494956705df149e4037 : FAILURE in 9m 11s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c6a04e205beb40faadc3e8420a3fa29c : FAILURE in 8m 26s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5f290b5a1bef4a109d1decf18de5c9af : FAILURE in 7m 58s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/910150cb94d84eedb5b222b798672c2c : FAILURE in 10m 14s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/a3b0f2bea9d8485483166516440f1751 : FAILURE in 9m 57s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9f1c160314754c5e8928b47775b3aa22 : FAILURE in 9m 10s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/975af19e52484c249f9d6341d3462cdd : FAILURE in 8m 44s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/02f8981153084a03888e6c5c0bf87e9b : FAILURE in 7m 45s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ba0cebc0e62343cd8d01ebc9a71b7a51 : FAILURE in 7m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a1825e1d2aca479daa36e6770b482135 : FAILURE in 3m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1435a1856838479c92ee65d8f7ab2af6 : FAILURE in 3m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/43f1bed341914987a045cd257b39a3c6 : FAILURE in 7m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a21b385fb9cd4b0cabcacc63f6d290f0 : FAILURE in 6m 26s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/a2889e3414c345c7ac1d35c26ba6a686 : FAILURE in 8m 32s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5c67e9654df148f3a0345a105de0f50b : FAILURE in 8m 07s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f14b209e37044f41b1fe84a7ab4bc3fb : FAILURE in 9m 29s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/3c5709ff1af643e0a070873e7aae727a : FAILURE in 4m 14s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2d331b51bf254f13adc700c5c9e6c71e : FAILURE in 12m 34s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/84d92ab9f6264e22b91349bf57d05ffe : FAILURE in 9m 09s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2197ca25457049b5a5b784c0270a4429 : FAILURE in 6m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/42d4c0d1bc94442bba9b6ea6c6a3450f : FAILURE in 8m 16s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e092fc0088ba45548f61cd59937a0e16 : FAILURE in 7m 24s","accounts_in_message":[],"_revision_number":4},{"id":"7ce3d52a9fac97c66c873411a3668a97e9e9ae54","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 08:42:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e2402a5982dee900946e8e276311d22adde5eeab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 09:16:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2a23345afe2ca749cce48059b4df7d6c3ec4dadb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 10:52:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"97acf99df9715e6b45e16787411dbb618b04e060","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 11:51:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"344649dd80446e53f652e0669d7d2a49ed8b0f86","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 12:17:44.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"83dfe1aeceb0a148d2ccf105015b72d42cd9de59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 13:42:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37d1d989ca2a4670be899b98b25c474e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2fe3bf6407a6466e8d9b7367faac6ae0 : FAILURE in 26m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/15cb7eab21044cd0860311646591db37 : FAILURE in 24m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cdf2e88153654429b9ae78160f891b4d : SUCCESS in 1h 23m 45s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1215529a410e4302a72c2b4f0664eeef : FAILURE in 3m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/016af579cf1447f29e58036467842a65 : FAILURE in 2m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e90d5175b8047c3b07fb158a55f5a37 : FAILURE in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2524eec3a2a3486bb26c0749f9728e95 : FAILURE in 3m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b2b90b06dced47e09243e9d784937507 : FAILURE in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"18b8d183ff86254907dcd117e503e9d13d2c7872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 14:22:43.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9efb48b0a6d423e9c4aae31001c07c6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e646416b49ca4cdba149df45f085a22a : SUCCESS in 37m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e11efc5a554e4f7bab5de940c6372299 : FAILURE in 44m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/76dfc1e345a547f7aa31a6bb67f64729 : FAILURE in 18m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/545fd61d827643f48797139bf578c83f : FAILURE in 24m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/de0a362c19394278a53c58a04c24ef2c : SUCCESS in 56m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c04d387b2c80451d9423c4d7fea57da8 : SUCCESS in 1h 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d56af1e12724da7afbee62ff1d5a2f4 : FAILURE in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f1374fdf6de448ca97d69028692a1ea : FAILURE in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d434f0333606411080ffde919d2a0bdf : FAILURE in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4142fa2c46554f4fa5051009752cfbae : FAILURE in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541bbeec90fd49b1a3e1a7c4618141a8 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0e3f6bfb9d50424bbcc8ed49099f508f : SUCCESS in 28m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d3b309ad82bb4f8ea41368b1edec54dd : SUCCESS in 30m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/342c299ae51e4e6a871b1e291ef200f1 : SUCCESS in 1h 19m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da3943eaac694ff4b04b8311677a1679 : SUCCESS in 1h 39m 36s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a7ed1eca28d14aca97cf4b8712be9cf4 : FAILURE in 26m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0d4b357b617148748e1e599fc290a703 : FAILURE in 1h 54m 17s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/621071ae5de34e0aa0d704a37a0556c0 : SUCCESS in 1h 23m 07s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/5dde27bfd15a47b0903be1afea119636 : SUCCESS in 1h 26m 44s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b02ecfd7302a4f8ea58054341c460288 : SUCCESS in 27m 43s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f143275814064ee391d02834b92bd98d : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ddc5812a0b064b2ea9dae823f4315988 : FAILURE in 58m 51s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/05297b77eefc4dc9b259b729fd2eae01 : SUCCESS in 1h 22m 26s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f2d55d189d8a47b7a75828438b6d6df5 : SUCCESS in 26m 10s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/c86640ed792f498fb8d18b73d198bdc7 : SUCCESS in 50m 59s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/83922f7dd41a4f0ebf093a2b077e1245 : FAILURE in 32m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7bf8fd42f8384beaa0b3dab0fc973a86 : FAILURE in 57m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/614894b1fd9e49b984be8b53f81c5f1c : SUCCESS in 1h 03m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3b9c2f3ccd6546a09d6c26c340a83776 : SUCCESS in 1h 22m 45s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/86d4ec058b5643b69b2fa5183eee2b7f : FAILURE in 36m 07s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/73cebb9238c949da87acf16b0bfdca09 : FAILURE in 1h 31m 01s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6d14f105cb834fdc9c90191260533dc7 : SUCCESS in 1h 12m 36s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9ba19d0d87604507a9efc97e749120fd : FAILURE in 4m 26s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/62068c6c602f4bf2a8507522a9442587 : SUCCESS in 28m 49s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/19c1900e4fb5490fa2af47e9ab76a456 : FAILURE in 17m 03s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b899c1d5b5974f648e50e005919861c2 : FAILURE in 53m 10s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/61e291a1a65748a0ba419e3109988ab3 : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a6c37040ee654e0ab9877cec6e06e5a8 : SUCCESS in 34m 41s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/ad09b3c1c4f0488a9e5afd888a1b2f40 : SUCCESS in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bd8c917b88a02adff086f28a76a55bd133509a58","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-28 12:47:14.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8cacb0f79f6e005eced6cdd41942bdcbcdd5799a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:12:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe0f6908ad2482f834fa25fc405086c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/97262d3a843541d6b974f5c68d52faa8 : FAILURE in 32m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7cdc099c952d40459eede1f06477f8ff : FAILURE in 25m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fbc325b55a9148dabc71ab16384b8c50 : SUCCESS in 1h 21m 05s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/280394ddd59046029b9d2076aee66f22 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6a577e5d7123404ba1096527465cbf5b : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/92d611655f564f12bd71dfba53622429 : SUCCESS in 4m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/491b529a15004747a2c8601e04ec4614 : FAILURE in 7m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9fec0e35a3b44a6b803a9d211d7659d1 : FAILURE in 1h 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a4a3edda952b339e6fdf679e108b2af19deafb93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 15:02:29.000000000","message":"Patch Set 10: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d11014a021c44b98fd0f73aa32488d6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/de18b06d96324ef584835c3d9fa9b275 : SUCCESS in 30m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/154236d8d69c4381b231c4a4a7673de9 : FAILURE in 41m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6da9eb64a42040c2b913e6a8e1d72a0c : FAILURE in 18m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fa9e2bb53f55412cb9bdb8c241ad6cbd : FAILURE in 27m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5d3e90ad06f04df484a3b881a07b437d : SUCCESS in 1h 05m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ced0b512c84d40d2ad0451231f3b2d74 : SUCCESS in 1h 39m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/829c61bd43b3495e8e15bb4f69695aa9 : FAILURE in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5bc9dc72f8f44808c9854c4389055e2 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3fa1b09583f4b57be50e0520ee1d9b0 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/560977ef1de54cc289e944819f8df57e : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/250d798cc4ed4774a1acc5299bd6f574 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b52beef57d37424786d04c97fffacdba : SUCCESS in 23m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2139b5dd62dc4b6bad63df4ba5216dd3 : SUCCESS in 31m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0cbbe5710732498e875388bf990cb8ad : SUCCESS in 1h 15m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3c1799f9d38446d6850e8ac204ec50b3 : SUCCESS in 1h 38m 28s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4f142c5ee1ac4392907cc022458c50ea : FAILURE in 30m 49s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/90268a879d524299ba1d7f16c1a93309 : FAILURE in 2h 00m 05s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5a18a4479b4240bf9f1bbb10c6cc7caf : SUCCESS in 1h 22m 43s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/f014a2f8a0794466829bab5156f59341 : SUCCESS in 1h 26m 40s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/41404007935f42a8bbfee776fc5beaf2 : SUCCESS in 30m 30s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/6f5cacde7e1946a4ad30031cfc37189d : SUCCESS in 1h 15m 13s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/903b3ddd45314c28980a1aa9dd820f87 : FAILURE in 1h 04m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/adfe6ebea37e427ea0e8e808426b8b37 : SUCCESS in 1h 31m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/114e2b221f0e43939600782c6fb48297 : SUCCESS in 29m 37s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/042c1c79f5d84194a9805cab9149ba34 : SUCCESS in 1h 01m 04s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/88f91df0384d41649cb57b4fc7b35c2c : FAILURE in 17m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e130f45fc4ca4cb1b99725528b6839ff : FAILURE in 1h 03m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5f21e46472b64149b39a9fd2a1d52e8a : SUCCESS in 55m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8d3c5673c9a945f0a234527566213b7d : SUCCESS in 1h 31m 48s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/81f1bf015ea544b489c3505627a4c662 : FAILURE in 28m 13s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2052a52568104ecc8189e4653b066009 : FAILURE in 1h 14m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fd89bc247299440c93a8f61df919503e : SUCCESS in 1h 10m 48s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6c51fbf2776e46da9a42eda4e7970e52 : FAILURE in 4m 55s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/249e14f5427444d49e2a6b44f445636f : SUCCESS in 24m 48s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/993149c2fdd14b0fa8c6c46ed24fd7f3 : FAILURE in 14m 59s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b6556493527c4efea1f8ed61f0f1e5a8 : FAILURE in 55m 03s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4005c51add524a919c2471582d44d4cb : SUCCESS in 1h 15m 08s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b3173f6e4b864d629992592fb9c1c1a9 : SUCCESS in 28m 47s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/084e9d1332794f0ea8c73b1a06025479 : SUCCESS in 39m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4783fbc6b7bec439b20aa64213bde4cb541a8a80","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-12 06:55:13.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nThe following files contain Git conflicts:\n* docker/masakari/masakari-monitors/Dockerfile.j2\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"bfb3c16012ce62a7c3fdfd98bbafccb8202b8cf9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-12 06:55:31.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"72ba5d6fe373acbbbeb4173c19752a5ec35aa120","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 08:17:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ebec3de95724353896a4860b2035d21\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/de0d273b18054fbf99d68d7941495204 : FAILURE in 54m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c7d77f0405743e681e2ab64f8db1bf8 : FAILURE in 20m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e357e347d53d47ce92e36febecc826c9 : SUCCESS in 1h 20m 27s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cd1a9ede28f4b628441cae82a0edd37 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42733220a002488a89e487a167637810 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9c52d93494cf44318ae3d89d6bf03c67 : SUCCESS in 3m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a1a6bdec8f3b46ec81d10a49cb905ab7 : FAILURE in 5m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c43c1ed08adb4885af8ded427726fe4a : FAILURE in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5fd7caf3214bf2d12e04876eb9cfc595451f3e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 09:52:09.000000000","message":"Patch Set 12: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f1a41f2e6a04aa6a364920a0508dd54\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f82999bbbfcb478890eccff2a3af8f81 : SUCCESS in 28m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/33eacfd7280941f7b92f07ccc03e3b39 : SUCCESS in 1h 11m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6c74fd1fb6d74a5d81619d53d3d12459 : FAILURE in 19m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3cb57232c5d640b1ba02eaa06d9ee431 : FAILURE in 23m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d7d83ee00810421999266c4703a135da : SUCCESS in 1h 00m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2c3d187c3535465c838e99ed1603acc5 : SUCCESS in 1h 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f5195ecce7483bb0dae1dd44db97ce : FAILURE in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e8927d5626f402faebc72926fd3301e : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/278aaa6013cc456c91e1fa560c5e4e31 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6455a09a99ef4edeb82cba57c8648f5b : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5cabb6101e49e1959f283a120cdf0c : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e1da2a5d374b47cd9ad5848895181314 : SUCCESS in 22m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6bef0be7460f41e0ade05f5ccdf82a73 : SUCCESS in 27m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f30d7ced26fa4c54a8a3173bb665b074 : SUCCESS in 1h 11m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cdce2cb84b9f46d588d7c7dc962ff395 : SUCCESS in 1h 35m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/60dd70cdf61d4024893695baecbab89b : SUCCESS in 1h 52m 20s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2cb0800ecedb484eae6f886cc2ff8ba4 : SUCCESS in 2h 39m 55s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/977f838505194b0691b40af22c74bc3f : SUCCESS in 1h 21m 50s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/fa3f7bb9cada426ab73ac4ef16cbb516 : SUCCESS in 1h 30m 40s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8a0c154f0f4c4b9b99d12f88ddf99f65 : SUCCESS in 27m 44s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e79f3a2196174f2db2dc2d240db96ee8 : SUCCESS in 1h 03m 48s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/946e3760ed8f418e88d14f12ae92689f : FAILURE in 1h 04m 28s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/bab75474a45a40ac910ab379905c8d9f : SUCCESS in 1h 26m 00s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b40c2d8504c74b80a8b5767f8ba9d2b8 : SUCCESS in 33m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/247b6551cf65438e90fceb336553e726 : FAILURE in 18m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7edd2188c71b4255b1d6ec865cbadbc8 : FAILURE in 1h 17m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/97a8e6b8779040d5a01f682fa9a52445 : SUCCESS in 55m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0477cfd87a3046bbb8ee607bd7117ee2 : SUCCESS in 1h 26m 23s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/efbe8b1466e94e72b28e3361488eb67a : SUCCESS in 1h 32m 39s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fbbb359672944743b80dc581ecc32608 : SUCCESS in 2h 26m 32s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/44ed32ba2d9f4841aa35d3d157871e0b : SUCCESS in 1h 06m 52s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/18453acba408457c8228e9c91d905520 : FAILURE in 4m 44s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7c7cb2dab8f243d58911427f11f68220 : SUCCESS in 22m 25s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/583158aa559545449e8f0198155b0abb : FAILURE in 16m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2bc2f2b54f1548a981c5274178e88a92 : FAILURE in 41m 30s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e11f129762c24ecd99ac3fea5517eabc : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5b4f6a95944f40fe9e9d6f8acbdf6eb2 : SUCCESS in 27m 12s","accounts_in_message":[],"_revision_number":12},{"id":"f3aa7b1017c97d1c09c32235600534372d3b8f87","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-12 10:30:39.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"f7c78bc8b14f725ac06c4551ba09a0c6dc0c41d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 12:16:22.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/381afe971136417b9b764da289773bdb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2e6e7e776eda4237a38c0217102ad507 : FAILURE in 35m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a19c16d21ce34faab885215e1a959e33 : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cbd39d821c814cd6804dc3a3e000a57a : SUCCESS in 1h 40m 12s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cff1fedf1f47402dabb3575315240690 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a37468fb4f4c48b58cc1b5a8b7f26567 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/41b94afda86d41f0b5954b5698d48b5e : SUCCESS in 3m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d153fbe236c54677883c8533310a1d9c : FAILURE in 4m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/616c810985294e05af36e2b809c84a2c : FAILURE in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"47cdca3ff6533f617e416f59063f7ce94985e132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 13:40:42.000000000","message":"Patch Set 13: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71f77b8acc0e4b0bafe3a9f96e563e21\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/67ae2eeeb7bc4a21b97b838510925c26 : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/13cb0ef9467049f5933c582cc238f49c : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d7627dfbc9024fcc8b3eee645d317ff9 : SUCCESS in 20m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/260bbaf0075f4ce98931e2f1ac35ad9f : SUCCESS in 31m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9f018ea10bdc4d6cba7f70ab9cedbe1c : SUCCESS in 1h 05m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2d7de268a9144f62a61481365a3c7a57 : FAILURE in 1h 30m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95869369f16f4c819f3bb7cc40a4bb7f : FAILURE in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4aaf8151a624833a1b64fc2953d4926 : SUCCESS in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4875a5ddd1ad4d3f8c286bc26566d256 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e25cc252158a4dd9aa86719938197813 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1396403136a34eef9133bfaf047adda6 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f6f36dba69c543e0b7906002a2df3a6a : SUCCESS in 32m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c4e394c95e0040c18ae554b40a3e0799 : SUCCESS in 32m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a5da4aca0ee7435a92ea0d018f4b8b1b : SUCCESS in 1h 17m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e50451227e6f4cfaacdecb58128ed842 : SUCCESS in 2h 09m 08s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ed2f06ff9dfb4380be8ad54b737422c7 : SUCCESS in 1h 49m 40s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/39b1ba962af84f1d90ae544461a5f821 : SUCCESS in 2h 54m 43s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/7d7c206eb0fa4b96a5d1a61dba6f3491 : SUCCESS in 1h 23m 17s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/954144c9818c4ee18c5a20bee7fdefee : SUCCESS in 1h 36m 52s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/1d881438e16343a5b006bcddef65fc32 : SUCCESS in 27m 30s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f04b90d340ef4f72ace929260b538912 : SUCCESS in 1h 06m 32s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5baa4045e1c2434eb1db73331635f720 : FAILURE in 1h 04m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/90ae036afa974f748d71e3ec3ff30b50 : SUCCESS in 1h 53m 09s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/922188db5d924054860008b0fe877ef3 : SUCCESS in 33m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fc20836deb3e432a912fdbbad60f26ee : SUCCESS in 26m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e5f59a97f38b44358d18b199918a2b40 : SUCCESS in 1h 20m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9b230944730a4b4daf8f209d1acfc8c9 : SUCCESS in 56m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0326bd828fe94698855ddb0af9811d9c : SUCCESS in 1h 28m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/058ef8100ea44e9cbfc033e9c83f6718 : SUCCESS in 1h 30m 51s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2f75aee1fe164eafb00b6199c25d674d : SUCCESS in 2h 16m 38s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d336fb2bbc42468bbaa7cfe4889d6793 : SUCCESS in 58m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0bce72bb2ed94b73aed93381eea5c008 : FAILURE in 4m 25s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e24b400accc84dd595668d99531347a3 : SUCCESS in 26m 21s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/78535a79a7b24b9dbb58e2b0ed1fbbf4 : SUCCESS in 57m 58s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/dd6f3ecfa2264b42809c1549452f9bd0 : FAILURE in 1h 02m 54s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e4a1c5e62199453995a4f47966be0a86 : TIMED_OUT in 2h 02m 57s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/58174f51a96b49e4a97fb3d3fa5f5905 : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":13},{"id":"f886aa5257f3790cd5d91438a98063d8b7fd5a52","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-12 13:45:41.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"bc1a68105f694c5b0544d3181d3da8d7cd0616d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 15:09:03.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b70ec8e917844f4b5aca1e587a18589\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9758c79326a849f5ae9de499a33d5da6 : FAILURE in 28m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/158f8a31816b41608450696c148f190b : SUCCESS in 21m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1f4e8d234783430a99cb9857dd9a632a : FAILURE in 59m 32s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e15716b083554a32a1df1709193680f6 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/82a2d04831294433948bf56aa3f6b1f1 : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f927c7ea5f3424795d914a4542e9b51 : SUCCESS in 5m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4d4491b302d44db383b6945ca5aa77ab : FAILURE in 4m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/840b973955064210875ff5d5c9cc1bba : SUCCESS in 1h 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3cd2475ea73f751143809541ca65bb2638d654a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 16:29:59.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c91c558d162c472dad4975aa65629e27\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/382d82971fc942f393b74556a2674ddd : SUCCESS in 37m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/db427c5ba7134c8bb8278fdb11eae03a : SUCCESS in 1h 04m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/84b6782e5cab44619fb5c3d477fa3fdb : SUCCESS in 18m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d401d1f7bf004043b1adb2283f7354dd : SUCCESS in 30m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/519b0e5f0d814c64937775ddb73d4940 : SUCCESS in 1h 05m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cb0c3ac4d98f477c8497a6573990dd52 : SUCCESS in 1h 45m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da4495571faf465eabf5eb502b2d4340 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42f777a0a18c445f85d7ce6f314b628f : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c60ee71b6fae418f80da4929a3ae274a : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4068e0574a574555ac73de4e22a8ebd1 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1824f2aadf42b5bdb245801140808a : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7412be641b9e483780d12c45e8274e83 : SUCCESS in 24m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ce1a97f6c14942c990c804fd2c71b573 : SUCCESS in 33m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/059f335bf3bd4755a0eb524c7c9a684f : SUCCESS in 1h 07m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7c80bcb74e594f73a712960ea19a09d4 : SUCCESS in 1h 40m 21s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7f35d428ae0542c09b83bbc3190d50ee : SUCCESS in 1h 52m 44s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a4d2e6498f68412a976b25e6a9577a60 : SUCCESS in 2h 38m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/08b9a5a1f10149559aa60bbed220cd79 : SUCCESS in 1h 19m 13s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/b8e8b94817af4ce0ab4301e207a7b174 : SUCCESS in 1h 33m 00s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9890c77a2c5c4300b07734ed1950909d : SUCCESS in 28m 08s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8f7d62b3ebd247fdb9b3673ea8ad38fa : SUCCESS in 1h 10m 00s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/38b6dc38753b46368ab0866f397453b4 : FAILURE in 1h 01m 15s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/365404c74c6243a7bad8c53ba976040c : SUCCESS in 1h 27m 12s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/db74468d005349c9bb573fce377f879a : SUCCESS in 33m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/25377437a1e643e18b73dc8f94a23010 : SUCCESS in 16m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/49f94c9722a84f0aab0776cb0b755851 : SUCCESS in 1h 00m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/50d603463b4f4bd5816a173bcd222161 : SUCCESS in 1h 02m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/097698d5e88c420ebc5902725e07106a : SUCCESS in 1h 21m 07s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/46a48f9e8a4541f9afaddcedb690f03b : SUCCESS in 1h 41m 32s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d192498df626456c87e50834a7192025 : SUCCESS in 2h 30m 03s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/90ff9a863f5848c6a5eba414e1243813 : SUCCESS in 1h 13m 32s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/c1c4a68df0fc4cd39654b8a787cef284 : FAILURE in 4m 19s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6a885bdd42684795805456fa0d93f93b : SUCCESS in 26m 28s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/19f5d0359f1341c6806838703d4ec68f : SUCCESS in 51m 32s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d347146ce19340fb99a7fbb25adff757 : FAILURE in 56m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8370d8db8f8d468b8f0f38b88f1109f8 : SUCCESS in 1h 18m 19s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/889b0112bddc4163b521f3180848c106 : SUCCESS in 30m 26s","accounts_in_message":[],"_revision_number":14},{"id":"08d90ea9d14337d1f99cb01fd0ff4901a1433f56","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-29 12:06:18.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"a4dc8a3d5db3c5241ae90f608a34fb9bdb2f7ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 14:44:29.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a37146ddc8f49359d0b094aff962ae9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a54ad059781847498db36e4b2ac25957 : SUCCESS in 26m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/27c7d321a2a341d4804123320236d7c5 : SUCCESS in 1h 05m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e28646a2e2614a659f4e186fe15acd45 : SUCCESS in 20m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/063219a826cc41b3b7b1b2012ab3b222 : SUCCESS in 33m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1aceac146ba7495991d49924524b12e8 : SUCCESS in 1h 07m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e9275b1261e945dd8bd36d9802493696 : SUCCESS in 1h 39m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece077c6f5d54ef9a886b1f1c80c6529 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c80b49e2aea1456ea641621c43611443 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fb66171aa63400db25141aeaa400350 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce47fc01f9774803a2b6e9a073978f9b : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec3d6002a36f4b22832d49634940474d : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef63d71552ab4d62a8706948ded6ab99 : SUCCESS in 5m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e9c7f4257f5b4611b8cf90d2d4cd7010 : SUCCESS in 28m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2ea9934e01eb4565bf2fb62649f14fda : SUCCESS in 30m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/55056b8a92324569bcff4834350e2012 : SUCCESS in 1h 07m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d0c3ced96b8043c4a1bfbf9362f6e617 : SUCCESS in 1h 24m 20s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/38f643ef9b524b99b8d8e006f97babbe : SUCCESS in 1h 52m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/72e8fe6374384091b6086bd9ba13249e : SUCCESS in 2h 33m 01s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ee20be0d2df345ebb7c06464c049da41 : SUCCESS in 1h 21m 27s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/a936c8a3ca9143399cd1233d32b46469 : SUCCESS in 1h 18m 14s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b0d89b4f455a40619913e9ee3672c04c : SUCCESS in 27m 17s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1089563f1edd41cfae8321472a9ae8e2 : SUCCESS in 56m 26s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/338e4642ebbe4f3297b132ba8d14757a : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ede19289a58b4032ad0f324ccd9339fd : SUCCESS in 1h 26m 50s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cab42b05e74448c88528f22922bd7445 : SUCCESS in 40m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40384459db764e8180acd22649f8ed5c : SUCCESS in 28m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3f9a428ea8904da8b15c5412af196dde : SUCCESS in 45m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2a54f527976c4947b35dc441d574e6bf : SUCCESS in 46m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7849b88e13f94899a459b680b050d056 : SUCCESS in 1h 07m 53s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c9bb3861fc6943e0a8fba00350cd5904 : SUCCESS in 1h 19m 26s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/839281c1961b404a95495a79f9d87cab : SUCCESS in 2h 03m 02s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/72dc720c017548faba9c4a04a17ec1f2 : TIMED_OUT in 2h 01m 18s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b347398add6e4a6c858d8517a3238d78 : SUCCESS in 1h 03m 43s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/1e09a2f094b84f0abf56fe5facc6cec4 : SUCCESS in 25m 24s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/8f1cb5b1b12945fab64b133339bc60ec : SUCCESS in 46m 02s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e74522da6f064b2890a7e709dfcec3de : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/20631275bc30451993443ce8df60a747 : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a3fae7fb47294aa8a0f2756915709c08 : SUCCESS in 26m 32s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a5ab1758eff4117deac33496aab5d6082f50a6ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 14:53:46.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db081db9207410bbe14580e54c60e9f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a4ef47b90f154183b707f7bf74d0a8c4 : FAILURE in 38m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2b14090403546e0a3befb62cd885a04 : SUCCESS in 36m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2d0ec1f30e140629c83b9658e96ab73 : SUCCESS in 1h 15m 29s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d7cd4c2562948a9bb31429abfd9f426 : RETRY_LIMIT in 1m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c42aad7640e2485daae9b76d109e0fe0 : SUCCESS in 2m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2305a60754944f42895400d92b0a2def : FAILURE in 6m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f327e47fb73946b8bc0ac73831785bf1 : SUCCESS in 2h 47m 05s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c16b7b9c4065aef71b3c22e60d50842d2e5e6b11","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 11:52:29.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"7782da164327de7f631351e3f104a457cd49242c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 11:52:38.000000000","message":"Patch Set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"1544050199e239cb88594cfd41d366d09085661b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:38:49.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbdab0ec12564be294d63a0137366cc6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b2c01cf3c0af4ff5b03f8e7bfa69767f : SUCCESS in 28m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/889f53eaa4644eb7a134333050d39773 : SUCCESS in 56m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3a5bf5f9538442b59783d284c7d1dc8a : SUCCESS in 16m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/71561e1a218d4ab990ed7f276e5a9288 : SUCCESS in 29m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2710271ef8344bc29f3f596c00cea543 : FAILURE in 7m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3e0cd88f5198470aaf7cd778e7e61620 : SUCCESS in 1h 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7496971263a4b9abb1112fa0ca9ac67 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/190c3cc53be445418bd98bbe80db59a8 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc15bbe8c99247e8acae9008526a52aa : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7eabca2bde3142f394cc0bac1d29b112 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2c969e3f3e1483b9ff581b09c56dab5 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6869364d37e74598953eba5ae1b786b9 : SUCCESS in 6m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/253523d1c13c4160b36ec8fba7553197 : SUCCESS in 26m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ab8acd91050844e9bbaf5b1c7b2a9fa8 : SUCCESS in 44m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/eea587821e63476e86bb018e9648a4f1 : SUCCESS in 1h 07m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f8996b1547ff4ca3a5e7731605757e66 : SUCCESS in 1h 25m 02s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f2efa3f64521482b92b6c8a0e895fa5e : SUCCESS in 1h 41m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c307dc288c5e42a3903ec40c00819ba8 : SUCCESS in 2h 33m 08s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3cd0d08b048b41408ae07fc1aa60ade8 : SUCCESS in 1h 07m 55s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/2523994592974b619b985910577ae3d0 : SUCCESS in 1h 32m 18s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2204a72074494302bc51b11a094b62b1 : SUCCESS in 26m 21s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d87b1e4bdc634af4a5789d042ff8d955 : SUCCESS in 1h 00m 17s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/89280d9be65441dc87710301c0e2be1b : SUCCESS in 1h 03m 07s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/53a6231cecee409ea5af736c531499b0 : SUCCESS in 1h 29m 25s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bf7fda79ddff41e3a5447cad02117815 : SUCCESS in 36m 37s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e28ed44a475747cda4c821100b242456 : SUCCESS in 18m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6ae0f589048340f6863ea19f631cb1db : SUCCESS in 1h 00m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ee6d47e52915473b962e11bb51a6aff2 : FAILURE in 35m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3c1284fb6a00403baac38ce26f109853 : SUCCESS in 1h 19m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/52fbba7eb96349618740f4bcc740acfd : SUCCESS in 1h 27m 01s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0d992682074c45a08b335ec12ee7eb29 : SUCCESS in 2h 13m 14s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/47d25de5d80d4d21be41f8e40af0d3de : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/28b6d870e5584f8ab92aeb099caa9e06 : SUCCESS in 1h 07m 49s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4bbcb53b26694e6182ffbe18605608e2 : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e2ba6c2e55de49f6853206e3fdc01807 : SUCCESS in 48m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/743afd4f30444cf5959cbca953f0023b : SUCCESS in 56m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fbe90aa982d1447a8fa345750cc52c9a : SUCCESS in 1h 04m 27s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/98826f265d86450fb7ec808ba9e2f682 : SUCCESS in 26m 47s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3c2aed88c1b2ac9bc8fc8313d617810592e3c950","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:16:12.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e087a484dbd547878514af167a920261\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/140f55245f004b4bbf07131d47e86b27 : SUCCESS in 2h 08m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d4b2ab6bdbd34982b151cf7c0e19c2b9 : SUCCESS in 1h 49m 07s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/675c32ee66204bdda998bb5913811304 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e4e12babab94942a5ca26eca4dd350a : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e750043bd4b3a8733fae607300ac09b86ab9f09e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-24 10:30:07.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"e3bcf5aff5080e600586681c5409c679101613eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-24 10:32:34.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"87bec84f0454a012f70e664f6e6c8b63e9a30479","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-24 10:32:59.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"c1f525eb472acfa9a5e8d41de88b41857aa5a201","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-24 10:33:15.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"fe6fd9d7f06f71f9c83ef7b67a753af360b777d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 12:08:13.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77867f799b1c4998a7097087fa07ff2e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a1945a00768f4c8abe2ba6684518eef8 : FAILURE in 13m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8de96902af7e4d3f85064e11479a3e45 : FAILURE in 13m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/33d63c1195f2488cb2a65d5f0bdb4b01 : SUCCESS in 17m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6832c235c40749cf90229a873541772b : SUCCESS in 25m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bc9644bb185b4f94af3acbf9a92012ab : SUCCESS in 54m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/18eaed9796e84e29b044acaa7a8bf78f : SUCCESS in 1h 27m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af78fe13f9b34843afe7f7c512ceac96 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e94becb1a54c4a9e83e50dba48a68c1f : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8735b81b3c2a4ef685327b207dd599bb : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4f2efdceb7e4d9681a24413e4aa3e92 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4438f41063f94e8aa93fb2ee017c120b : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a5ceb85c1b414ab183750dba54f553 : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/169a6ea3263546568cf859be2a142eda : FAILURE in 14m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c0238c889e8d4c96ac4c36a9be89c671 : FAILURE in 16m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e77f7cd570214d93925dbb7fefdcfe0e : FAILURE in 14m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/87d0204839e343eb834a05b96f71fb10 : FAILURE in 12m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/68b5e7558c9048acbb3729be53f85734 : SUCCESS in 18m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c7b0a9c70c6b4e579fcdf55d0486defa : SUCCESS in 1h 08m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fc076eb559fb41349fee618bb473212d : SUCCESS in 43m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/36b0e57f16fa44e4a89dfba7c64d9183 : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":21},{"id":"df3a4bdcbef6aaeaaaa42f5e2eff46146653f9b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 13:48:54.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea348892f234e888e05f42d0385597a\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d1f806a20e924b2fb99a679686455288 : TIMED_OUT in 3h 03m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cff499544eea47c2940173df80de6625 : SUCCESS in 2h 19m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e8b36cdaafe49f4bf769807ce205a28 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/72f05a7b770242abaaffda4755f7c8d7 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"b17f4e9a656f9be8419aeedd99c1e809080561b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-20 18:09:16.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"8cf985e9a16fcda26a15ce16dcfe52c33503b617","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-20 18:10:19.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"5fc1f5dbcb8a81ac74d86f55af24cc289b22ce16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 19:59:24.000000000","message":"Patch Set 23: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fed596d8f4d5461083bfb583917be202\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a563937fdb8a4db087e5f15c67391e78 : FAILURE in 11m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b179d4ace5b548fcbff27c7cfaad8f3e : FAILURE in 16m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ad3eded08c974c979939edd2eb9b0281 : SUCCESS in 20m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/47ae8cc36d7f4ca9bc36b779c3cd2ad5 : SUCCESS in 16m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/af27836d8e03443ca8d6990afaba727d : SUCCESS in 1h 00m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4102e9967c4c41ee811537caeeec95f3 : SUCCESS in 1h 42m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3465594ce504f98b13831f496339c6d : FAILURE in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45876908161b4021afb9e087d6534974 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/312d98a3f88c4d1d8f1f16d4a086b0c6 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b583fd11f8d040a98ecf17e64d5445c5 : FAILURE in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79742a21c4684c10ae0d7a94aefef51c : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d43f94d2974cb7a35cb8117630b7ec : SUCCESS in 6m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4e54dd3b41534f7e852fbafeffa32719 : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fed07962002344eb933e198f76183d2f : FAILURE in 11m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a627e5c4b3524a7f94f8862acfb6e8ec : FAILURE in 16m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/33945e5d62c54b409bccc658959c33b9 : FAILURE in 28m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8b764b9a93dc43a1b460cd308b4a896f : FAILURE in 12m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c370ac482a3c4585b79de37f8a905549 : SUCCESS in 21m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7a07c0db6f3f4df8957fe645fd55004c : SUCCESS in 25m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4200eb8127ed4f81948764fb72326055 : SUCCESS in 1h 07m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2b3f6d8343d042cd89db406da8cd68ea : SUCCESS in 1h 44m 28s","accounts_in_message":[],"_revision_number":23},{"id":"46b3a9316b2ce1f9ec72a16ede5296ef5208bd7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 20:25:29.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741d5c82388f45d79ad1859aa1df9bd3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9100ec6ab33d44a59c977804e1a00194 : FAILURE in 33m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/683b04d0ee13415db5366c53e88b6986 : SUCCESS in 34m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/46a54e7f7a674ef3a3aed8de310577dc : SUCCESS in 1h 37m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47306359ca6a4fe0b1443dd4362649be : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0634eda73d764430ab9126024c16a8b4 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d930986adf944a81b5bac68ce547ae9e : FAILURE in 9m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9a398d4615ca4497a9118daf26ed73d2 : FAILURE in 6m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1ed23caca9ad4eb6961f2eec96ec5338 : SUCCESS in 45m 55s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"01c81f9942d4a192d6460dd2b536b4a4afb32f22","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-22 08:57:03.000000000","message":"Patch Set 24: Published edit on patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"ae6d044c0bdef00ccfa4c748155efd73c2b8f769","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-22 10:07:17.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"72bdaaa2eea8e6fff9f7c88a62c568fcdf41c4cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 10:44:55.000000000","message":"Patch Set 25: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77674fd12b8640e5a6484e4ccdebae7d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2c0736a4a28548b18f419112f0de0464 : FAILURE in 20m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/19dc5876e43741adbe731ef610c4ff28 : FAILURE in 9m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8fdae663106d4aae9e3a1b108f38efd6 : FAILURE in 17m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a1169ecbbd314a25a6ef8ce3aea8e558 : FAILURE in 14m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d15f3e0d01bb4c64b2cd003340da9899 : FAILURE in 16m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f41da31811414190a7eab717b85b4474 : FAILURE in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c133fb7f36342229abd35ecbe41d06b : FAILURE in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e33bff0a150a4bcab7ac254474dba46c : FAILURE in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84368876226c4a9787dd492049bbd9b3 : FAILURE in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfd2e8f0c437483197ce2eac2602469c : FAILURE in 6m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22b37d803de24d448bd30d289ce98bbc : FAILURE in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/767442f39c804c70b0047219913af7b0 : SUCCESS in 6m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b4f26ef69d5e48bf9a0cfe3a221e1b1a : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/693aedc451dd479e81f5fa0aff4d5446 : FAILURE in 17m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9476caa9fe9440fcbdfd0fa91ff97a7b : FAILURE in 37m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9c9f4e9dea4b426abcce7d223c880ccd : FAILURE in 19m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2cf7ef23979549759da1e8e921d62687 : FAILURE in 17m 34s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/edd66d696b4d402c81c7201998949594 : FAILURE in 18m 08s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/18cec8b76b0f4d17bbf9f9beb701f1df : FAILURE in 20m 13s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/164408f0d143413196afe749dde30e52 : FAILURE in 23m 09s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/65138befd1144511840cd9a0fde7c2d4 : FAILURE in 10m 51s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7c857b53a2c54375a11b8856a1badf77 : FAILURE in 23m 22s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/388bc76a54c04b458197d0b1cc1eb4f1 : FAILURE in 23m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2d1f4de7170042c4b79f423a6026561e : FAILURE in 23m 12s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2595d0d5c55c4337aff86d69242d07f7 : FAILURE in 13m 02s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bc9d883f621f47298fe80e9bd8859153 : FAILURE in 24m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/933e7d535f794e4cbb1ae88cd16c5b7f : FAILURE in 24m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4be6ac931b1542c79f0994cef45efb32 : FAILURE in 18m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7064e4c7edd0480d97d649f97fae5677 : FAILURE in 14m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/65c19766a24048308470a660d23b9d11 : FAILURE in 24m 02s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5b67aa4590d14779bf2d4b7d3a6b202f : FAILURE in 21m 18s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/3d2a7b16e7f241269e4bd39d20570e83 : FAILURE in 18m 48s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f3a5889dad8e4c21b8f84166e11ec3ac : FAILURE in 9m 38s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2807a2188dc24ffe88e866ea5c550c7d : FAILURE in 19m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7f1e68274db04d0aa029413a40eed98b : FAILURE in 21m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3d4e2b622e914ce692db267fb2d335f9 : FAILURE in 17m 12s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bf8615ce390f4be1aaf177c9b1499e3a : FAILURE in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"a94dd68bacfce8e97a4c1311a4225c95686a6542","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:53:40.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2740420af3c5428bb3bc5495c6cb66b8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f865931c096548b5af973ead1e3c3ad9 : FAILURE in 32m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/459392e3ba9641b59c945448c20e437d : FAILURE in 42m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6eb36486fd2e4a8ab35192e5c24d6333 : FAILURE in 28m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e6f7c9a81d744f9fb7a3f273905655ad : FAILURE in 5m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/78edb13f09264b44be59a3fc6d2c4efd : FAILURE in 3m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4e0d28be765146d2b51e5003977ed4bd : FAILURE in 6m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/afbc272dc3c349ca955e2402a01ddbc3 : FAILURE in 6m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ef63b36bc7ae49e6acb8353be03fbd42 : FAILURE in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"63f0dedb3771831308c3ac1a6c64c25e10e11057","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-07-18 14:28:36.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"88742c813b2f1c8e215fc31e960449ecba9dc7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 15:03:33.000000000","message":"Patch Set 26: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c9c0abce7ce411ab66f465e8acc25e9\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/13677813852a467bbcc70310508bac57 : FAILURE in 12m 21s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/bc5a9da3494f490d8c0c9fef21b4f6d1 : FAILURE in 14m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9cc66dd587c9441f9d181473d809bd19 : FAILURE in 23m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ecd40d8320fc41298291f874487ed60a : FAILURE in 24m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3a8d16eef9e9407eb30575e1db70f41a : FAILURE in 16m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/72babd3680744b29b210be1fd6ebb3f7 : FAILURE in 14m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e401bf64654e609ad94e60b51f2a06 : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9af68414cce2461a803e2fba94a8ca23 : FAILURE in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45b83cb6a62c4d659366df0ea161cbbb : FAILURE in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2745f05e40264af5ad8c26cccc56b180 : FAILURE in 4m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5236c269b444bbba4177b72feeeb51c : FAILURE in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7507ab3f48d44ce382fa986909b940bf : SUCCESS in 4m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5f953953ba084bf2985b47f5a3062e07 : SUCCESS in 1m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5ef65a6134a7462cb3c3d4f88cea9616 : FAILURE in 18m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f77f88e06bf84c19a75479e09c0c4b08 : FAILURE in 27m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b2d58c731fd145c180b905f17808624b : FAILURE in 17m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c5101f0ea07444e29b9f023793fcd508 : FAILURE in 19m 58s","accounts_in_message":[],"_revision_number":26},{"id":"c07210d7e8166592f94a8b4aa61a94a27fe0e608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 18:19:31.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b8d833aef246fdbb01ddb6ca0a09c4\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/de5611aed5634db39f786e16537ec129 : FAILURE in 31m 08s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e6871e89ce4643fdbadc36f93e61aeda : FAILURE in 40m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7778be841abf4b66b848268b33bc7eb7 : FAILURE in 1h 00m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ffdf6eba1f484d90aa451ba0324a88fc : FAILURE in 44m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ab93e969b8d4ecc8f4cd3e55e32db30 : FAILURE in 3m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0ea03efcc89744719513bbaff22fd934 : FAILURE in 5m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/387162271bb641198bfc537f5ca9c9ae : FAILURE in 8m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/57fc3cf3916c42b094e8045538c6bc18 : FAILURE in 1h 24m 29s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"72289f7a26fe1daaf8abe9f693ebc79d2838db96","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-26 15:56:33.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"4e27c4d1aa474376591e9f9812029d1f7abfe874","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-26 15:56:38.000000000","message":"Patch Set 28: Patch Set 27 was rebased\n\nThe following files contain Git conflicts:\n* tests/templates/template_overrides.j2\n","accounts_in_message":[],"_revision_number":28},{"id":"7a2122c4fabd0275c9ee56041c47dd20cd7e30d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-26 15:57:12.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"912b1090156e43ace05db3eb2c1227dfb1ae15b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 16:08:15.000000000","message":"Patch Set 29: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57abb322d44545d49cff8cf948be7571\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/93b37db5162f4dfe81c16eca6d290d91 : FAILURE in 2m 06s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/a0050cf01f60416287fa3ee84a507159 : FAILURE in 2m 50s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/21f0a009150143b784dd17eb91ba5f48 : FAILURE in 2m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7e4963408444f25af5a5a13d0cefd93 : FAILURE in 1m 51s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7eaa7aa51d4e463195bd20a4ff5d39b8 : FAILURE in 1m 44s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/23e3988d580d4b6286921c6e4b7933d8 : FAILURE in 5m 09s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/c2fb0cb6d8094ed6adb6d0f7c19e61a5 : FAILURE in 6m 48s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/0fd28e5e195e4a6dab96ff5ca728b629 : FAILURE in 6m 51s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9c56a51a96d46ba91eaa73a60009e0d : FAILURE in 7m 07s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/399ad41b9eed4f48883d06084aa74db4 : FAILURE in 7m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2c73d0e513cd482abc69530662b3adc2 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2132554295749f6957016a15afe4e6b : FAILURE in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/309b5bc409bd4a64b49b6a2dd71c9576 : FAILURE in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/215ef8877bc3432688b110dbe4efe536 : FAILURE in 1m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29a740bda7354628961a137260bacc80 : FAILURE in 7m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fef7c2ea4bf4782a580fd3912ce03f3 : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f6fad428464e38b05c01dbbae03a62 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":29},{"id":"b9257f193450ca947d1902d3a6b5776f40392965","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-26 16:19:06.000000000","message":"Patch Set 30: Published edit on patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"2f8e7d2735e71dc4fe2d08c9131699d8b40372e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-26 16:43:08.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"6688dad77c0e38e6ff49553a33a23dcfb35cd806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 17:18:22.000000000","message":"Patch Set 31: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff630eb1948e4202b796aee67dee04e0\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/82238ba0a3f6489da7402dc92b78b2c4 : FAILURE in 6m 28s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/8a0f4e10119a4f6d823b7fceb0cc8307 : FAILURE in 17m 04s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/82d6695ec0e14514b2e3036ffd2acc52 : FAILURE in 30m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa051c762f254bf18c5f5afabf13672b : FAILURE in 24m 52s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4eea7a40ae4a4f739d7eb7e8351aa176 : FAILURE in 29m 49s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/638cf2192d7140b3a1d8cab456f2b19b : FAILURE in 8m 49s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/7ba87155caa14c3594ef9692e043d1ea : FAILURE in 14m 18s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/a6885523dfdc48eeaee61ed3950cbfcc : FAILURE in 13m 48s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2568bf9052494adfbd5b62353852c8e2 : FAILURE in 18m 56s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/9599154d761649cba1a9d6019f9c73e5 : FAILURE in 19m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fc9cf779d71f47f19ba37d3400cb824d : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ca74fe426f1497cb039d17fd58d5e6c : FAILURE in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48142e9ca2f74dbb8bf14273dcbe467b : FAILURE in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ba24e92c7d74dabb42845dcd9e9b3d6 : FAILURE in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0ae103cb5d940a8af5223f5b64eaaa5 : FAILURE in 3m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b60e11c641c430790a5e3677ccc4b6d : FAILURE in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f81297468ed49cbb718635cffcf7553 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":31},{"id":"465154f83c651457edefbb3f7e8b720450198bbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 19:53:03.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c3a19351064b0d94749b2c5e87129e\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/07e21362e52d454da8889b9bc458b34d : FAILURE in 9m 40s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/1887a1f496de48e49143abf31275b9c3 : FAILURE in 39m 25s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/8d7c3a486d504bee8fb623d3562c9e74 : FAILURE in 30m 03s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/d9d9d094704241368d5e73ca0938e457 : FAILURE in 17m 18s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/d38d969ab37b48d6925f9ca4175c6c2e : FAILURE in 35m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e0b432272004ed4b10142bb84e56f6b : FAILURE in 6m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d840b3e25f24556b39177b837b1c9d1 : FAILURE in 4m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e6313dcca7e44a199c5bed3cadbdc3b : FAILURE in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"307ef3480a9d62f706c00bc8e28a48c73af883ce","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 07:51:51.000000000","message":"Patch Set 32: Published edit on patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"c7188169a941358b1619800b0f327c8c3b2e9f85","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 08:01:32.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"599c1af95909754cb16af83bc9169ca7fa397c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 08:12:18.000000000","message":"Patch Set 33: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a16da05ba584323aebdc6af514a232d\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3c66f72538af452b8ad08dc052ba3198 : FAILURE in 5m 45s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0986362f26ae4401bd45b4089c0bf09e : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0935d850b8a047dfa09dab3f0d639efe : FAILURE in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71c818e4647940048353f3ce4df27675 : FAILURE in 2m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c32494952dcf42e4b87818898a9c1546 : FAILURE in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9bdfb4e73794753b1afcda0666c2b98 : FAILURE in 3m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0223a1980484d278b16fc30491b17f0 : FAILURE in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c486078b8e440d977dde8cc17751d1 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":33},{"id":"df38a156640d1ec951aa1eace2f276108ebb7d98","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 08:15:36.000000000","message":"Patch Set 34: Published edit on patch set 33.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"4d6a37a5f64e94bc540769be3d6818405cb79549","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 08:17:38.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"15fbb48f7e08bca77ebcfef627308cb2d1a1b801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 08:27:23.000000000","message":"Patch Set 35: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f47cdc18346b445690659c29b33b6e72\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/5698b293bfd548cd9d14465740823f33 : FAILURE in 3m 58s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2c54954d0f2548dc8de6812e81f50ec1 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77549aaa74bb43ae9ca20e79966fb14c : FAILURE in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eca9f726ff33480cabda2a0c06b861ee : FAILURE in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b59946e5dbf64865aafac838b1fcba0f : FAILURE in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f4360b3f77646cb8f1134b40da05508 : FAILURE in 4m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22352ff62cb54de38adcabf32a35f112 : FAILURE in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806d29d0e3834547a18334d58ea290b4 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":35},{"id":"ea5669995811d2b5d3e687ac2f8d4b2370e70a2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 08:28:52.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/275e4e6b212b4dbfbca32d0005b41fa7\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/95a5542d9d844b31ae7f8ca216c4d208 : FAILURE in 8m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd03a634a98b4264b55dd6a9aa1a387e : FAILURE in 3m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de4626f8527b47c69624e3b880a22d9b : FAILURE in 4m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/98a560b1392142cda1d10ccfefd60dbe : FAILURE in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"92255094cf09ef785339bd303fc61822da1d5ce7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 08:34:17.000000000","message":"Patch Set 36: Published edit on patch set 35.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"242cd28e222a441e279e48479f224afc687b1f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 08:43:14.000000000","message":"Patch Set 36: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8faa39a5ca54c44b0e0ed834e2a86d2\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3f29467392dc4d448c9055cf95d93085 : FAILURE in 5m 12s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7dd508df2fd74c7a996b91d2948879b5 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a008f89456d147b3b1d23a590208cc91 : FAILURE in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecf01e9e7d524694ae12021dd2a09e6f : FAILURE in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bdc14adfe654ce8b2ebc0edb81d81d6 : FAILURE in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5d2cc49d717442aaa5de243269f8904 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19eee233a02b4cbe9857835e2e6cf78f : FAILURE in 1m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddca51229e01457da4b6dc89a07585ae : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":36},{"id":"c3fa8f1720ab66119a1260d2fd04211d15ad3152","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 08:44:32.000000000","message":"Patch Set 37: Published edit on patch set 36.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"e6335d78fc7828f4acb42850873971d6a2bc5a33","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 08:56:24.000000000","message":"Patch Set 38: Published edit on patch set 37.","accounts_in_message":[],"_revision_number":38},{"id":"143a5479fb6bd72475f9a67ea0e56698e6789c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 09:09:14.000000000","message":"Patch Set 38: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38ea1ace18df4bcf8a1e641b22c88b50\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1d90a1351eb44d8d89da87cb090ab0e9 : FAILURE in 7m 55s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d74ff38a47dc44818f9aebb9091bafa3 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60290e135d894f9eb2184a32aa788bee : FAILURE in 1m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8154b4b4ff65455b9de58ffc993b90b2 : FAILURE in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/716ef62f03f24c89a3b57aa0b89e76b1 : FAILURE in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0ad18b70a694d4499dd20619f79cc45 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1138450fcf69429db821a2db4a15c395 : FAILURE in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024311ceecbd4445b6b6224b50c14af1 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":38},{"id":"c3c05c007a514cd6950fdd750e27c9ac550fb4a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 09:24:22.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb27f908b8242f6a99e5ddd8debb5db\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/10de72b84cbd403cab00fb26ff0e3216 : FAILURE in 21m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89d63a362fa046f78b16b836fc8901c1 : FAILURE in 4m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95753d40fc1b41ce8e1e490e1004939f : FAILURE in 6m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b6939494da34ccfa572df24b6284896 : FAILURE in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"699e57b356eb739b862e0ca11988685e33438d2d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 09:39:13.000000000","message":"Patch Set 39: Published edit on patch set 38.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"39c6ef513e0217e5c755b7fd84651471232e3554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 09:49:33.000000000","message":"Patch Set 39: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5f629f8ce1941a38032c1267ec7856a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/15a4b6eedce24844a2da0b6b255db887 : FAILURE in 5m 01s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2a35c4c67972484f9c370c2105c27e8a : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb0631ca2856418881343e91e7b75237 : FAILURE in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2726012fe7a4fb08db542199a10fc14 : FAILURE in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f9ada8115bc4c449e42a7e9f0c69e21 : FAILURE in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0258dc00cbd144f5b1270b66d5c11bea : FAILURE in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba981ab9323e44d3b66ed25265028118 : FAILURE in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb2473ccb5b46cb9950874b4e5f9dba : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":39},{"id":"ba273cc284ea1f6b75f88c5600ec7af388f61e8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 09:51:29.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e954c265b28f4575be235b7ada055779\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b44e3402ecb140df9aa873abe03ac3ea : FAILURE in 9m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ccffc2cb10964ee79b90d3170e201d66 : FAILURE in 3m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7bf1854ccac7408495f9562ef2d15357 : FAILURE in 3m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a2d79e13b312487fb633b1324c7f4bdd : FAILURE in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"b8e40bef849bdf8e2b1e4094cb707bc1ff617dea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 09:55:33.000000000","message":"Patch Set 40: Published edit on patch set 39.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"67c8a0821d33945270bdba2efd3a3226d9ea24dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 10:02:06.000000000","message":"Patch Set 41: Published edit on patch set 40.","accounts_in_message":[],"_revision_number":41},{"id":"da87836387b7ed01e9e548ff30cc919da966c3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:12:48.000000000","message":"Patch Set 41: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbb25ce04dfd49d1bccbed21a07a71aa\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/edd5c73a645b4b418eb68f6db19d07e2 : FAILURE in 2m 29s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9bad6b588a5f4c9792775da44a236ff6 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123e3a607f4d4254a96e4a9e486c0ab3 : FAILURE in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9442354bfba34071b9c02b7b3d044058 : FAILURE in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78970ec47c1745b3a55b100626e28efd : FAILURE in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63a2632342ca4a41a53d6eb28152c6ee : FAILURE in 4m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4efb1a1207344cc39ab5e164cf69f38b : FAILURE in 1m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa55ed53b43b425f8784bce7cab020b9 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":41},{"id":"8db085c63ff98410a96c446de909007e128989e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:13:22.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ef21e533664f64811212ebcde608c0\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/77deac00e0b54e71b5388890ca1b2390 : FAILURE in 4m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4b973c069af42898c45c9d323bcb369 : FAILURE in 3m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7fe9f70ad3ec48b99f64bc9d1ae33f0e : FAILURE in 5m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/41e19dc51cab49b3b747af6cbf618665 : FAILURE in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"c241a18542d878719fff190d1f32ff9645ddf620","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 10:16:20.000000000","message":"Patch Set 42: Published edit on patch set 41.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"8b794779f4d1902eb70ca7f20c252223e7ee1c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:28:24.000000000","message":"Patch Set 42: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bea8675310849758a8b70ca49dbfeab\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/9e3f5c0c8d444abf92cc0e6ce28c93f2 : FAILURE in 5m 08s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f601c4235ca14f0ea47e16a705fefc31 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451c7f2bfb2f4bdcbcf3dfa5e88e1fa5 : FAILURE in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/185be50b247f4855a54da1ed54f49378 : FAILURE in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d30cefddfc094ab5a976c962158b3134 : FAILURE in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a46c02e4996460e81e30baee2c2fa9b : FAILURE in 5m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90dcc43b06244d44a0f6d89c31374d7d : FAILURE in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c84aa27095449ebbe6b4f17ba3a18cf : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":42},{"id":"1cafd63c758329d118ffdf999a3fbad43bf80c62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:29:32.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7a315854af4cf0a1a960738f0ed231\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/2d31aa7e29e14700840f5eb63791230e : FAILURE in 8m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/48c66e02729740b58c4920eafb7870af : FAILURE in 4m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56c3c6a56fed496fac8f594ec579e5b7 : FAILURE in 4m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/11bdb2f3fd084cc39f3e2063d24de611 : FAILURE in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"af7fd216c83819e3ca9d09f5b7d39c99c1725d65","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 10:30:05.000000000","message":"Patch Set 43: Published edit on patch set 42.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":43},{"id":"1251dc25af7e15b0706b3a2806db9fe34a2cd17b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 10:39:25.000000000","message":"Patch Set 44: Published edit on patch set 43.","accounts_in_message":[],"_revision_number":44},{"id":"cb82b21e5cba693e859c88bf92bde689996a5700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 10:50:50.000000000","message":"Patch Set 44: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33845ab217194272b8800ace3544f18a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3c3a21a407594efb99f9b06120d74261 : FAILURE in 6m 36s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e882c63b96e64f3f9719b9c3a8097fad : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08bcdc0c02e4c1bb13856803e6d6d32 : FAILURE in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57627489c37f43c5a878039c61156d79 : FAILURE in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11b2cd2580104937848e8280a04eb684 : FAILURE in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7f01abf3f4e485b8604f4a718da2e32 : FAILURE in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/663e338db70f4a21a85900237764b5a3 : FAILURE in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d04f7d755349bfbff24165082897c4 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":44},{"id":"0eecf716e767b3ac8bde5565115682848cd11be8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 11:04:41.000000000","message":"Patch Set 45: Published edit on patch set 44.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"36e4be4e1182af501db9d86b69a143012048e918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 11:17:13.000000000","message":"Patch Set 45: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44f0d1baf75f4837b9fd77f6cdc29d82\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c263c08e1d7a4acf8219ab14289d7f5c : FAILURE in 7m 31s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e8c5522a7f044dceb0749786ee2585c5 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6340275c7740158a86b6eea5969dff : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c00c4a590e14804a73353391fa095ec : FAILURE in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55a257a556814f6fa28ffcee7ca60aaf : FAILURE in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b5aada6cef04be2a8658d14e689c232 : FAILURE in 4m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/021641e9c0614bb4bf7e65bd03f2343d : FAILURE in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640a381545d7467daa9c01e8683b397e : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":45},{"id":"589fb6c7182f458cb0df48b3407ee49ebfffca13","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 11:20:24.000000000","message":"Patch Set 46: Published edit on patch set 45.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":46},{"id":"4b82cb4475fff7df67d8a544e5b01332ede4ef68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 11:20:58.000000000","message":"Patch Set 47: Published edit on patch set 46.","accounts_in_message":[],"_revision_number":47},{"id":"c979850691b35775e58e2c08920ee0ef79b489ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 11:23:13.000000000","message":"Patch Set 48: Published edit on patch set 47.","accounts_in_message":[],"_revision_number":48},{"id":"8724b1b2a0124e4bc52d290d5b7e9e84bd4b3189","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 11:25:27.000000000","message":"Patch Set 49: Published edit on patch set 48.","accounts_in_message":[],"_revision_number":49},{"id":"1aaff40fa8039556e362bbb7053c38e5c96cbefb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 11:26:18.000000000","message":"Patch Set 50: Published edit on patch set 49.","accounts_in_message":[],"_revision_number":50},{"id":"6e25db7887a1d5a71667c0ac94ba1fd908382ae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 11:35:48.000000000","message":"Patch Set 50: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/876718d8907b47d89311943a6e8e1116\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/4aaea3a75d824371b6d5689e2a218a1c : FAILURE in 3m 55s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6a49e95bb8b14c5d991a9e3cc45738a9 : SUCCESS in 1m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc56273afc7748908c27d13b25b35c7f : FAILURE in 1m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dff3f1e66acc4d1c8895111357c43e1a : FAILURE in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e69bb79ba34c4eee94b27893b395d6ae : FAILURE in 1m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/561e02c79f884b00be2b64bb350ef223 : FAILURE in 3m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49066988c7d7469ebb9953cf5cedc589 : FAILURE in 1m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a146746e8c485c9baff6cb3b3f5351 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":50},{"id":"15463942bfe32e0421bee14be30dd6c249e96d1e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 11:54:17.000000000","message":"Patch Set 51: Published edit on patch set 50.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":51},{"id":"942d58bc8e1d0f30cd3f9defce53b9a3799e4bec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 12:03:56.000000000","message":"Patch Set 52: Published edit on patch set 51.","accounts_in_message":[],"_revision_number":52},{"id":"d9a7c06c02e0d85c8a54fd20a71ff95bd5b7e984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 12:15:15.000000000","message":"Patch Set 52: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/328d3d4845624441a8e4e660120f0062\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c4be1e72907f44339f9602490bb97c88 : FAILURE in 5m 20s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/faa6bf09b14c46ccb268cd58c277c730 : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8dbc62f743f45eab94074bf1337fff4 : FAILURE in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78e586574e6d4653aad7ded63cf0ff57 : FAILURE in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31254527eced47f3bb6a73385b56a472 : FAILURE in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb37c8a7291b4fb5936f16ae6f995816 : FAILURE in 3m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cf2ac2dcf6942c0ad4d43156f224571 : FAILURE in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087ac67c8165409eb62ebc5d7d8d41fc : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":52},{"id":"aefa7a9317e4d03a2a9503bf6c0dbd05a08eb1b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 12:24:29.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b119d30efdd4dd19b1394a11ab7adc5\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/f6bda7b2359b4285bbf9523c2f1a9efd : FAILURE in 6m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d82974bf2a343578a7ba696a3a45324 : FAILURE in 4m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ca631bf43c8d468fae52e35c35e985e4 : FAILURE in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c8ae6f3944844ff4975a74e07bcc5126 : FAILURE in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"c4625beff390cfd76a4b740eea6f7561860b3781","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 15:22:17.000000000","message":"Patch Set 53: Published edit on patch set 52.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":53},{"id":"554597d86598908be892ee32e5fe459c30990def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:44:33.000000000","message":"Patch Set 53: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c172d44d8fd451da6db0cf7ae1bed31\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c01f7bc41e744726bb0e8a18f67ec942 : FAILURE in 11m 19s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b2c46f82beb401c9aa8bd16137448ae : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbcb7c506490403ba3304ef937bd8705 : FAILURE in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b43411e86309404885ca1579d32cfbfd : FAILURE in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2df58bf79a084529bf09fe8d1d725ee3 : FAILURE in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c7757f9ae264d5191e2778d899dfdaa : FAILURE in 5m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9748b0e8f0334d2db5d13568ab661077 : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d209516727054ac19b8fcef5e6e61e90 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":53},{"id":"f0a62324589b780a2ac75f82e2535cca6fd7be01","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 15:47:02.000000000","message":"Patch Set 54: Published edit on patch set 53.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":54},{"id":"f68381b74a870cddeed63c6227cfb3693f799bca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 15:51:30.000000000","message":"Patch Set 55: Published edit on patch set 54.","accounts_in_message":[],"_revision_number":55},{"id":"fced0b6bd253f693101a81725f71d6564913e259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 16:05:27.000000000","message":"Patch Set 55: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f8774d3cbcd4f12a82578124beac745\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/8058e296e5254fa0aff8511e032fc230 : FAILURE in 3m 20s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f6fc8227ebad4f0ba545922d1dc77ed4 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d020f0f4f8441e6b64fa49cf8d1237c : FAILURE in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33153a4e794b4b5e87e9b3b7ac3a137d : FAILURE in 2m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78ce5379fb6740b08875c506b8e4f98c : FAILURE in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9470468285514b0b8cc799caf3ea3d66 : FAILURE in 4m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d639634fdb7a4e1d90e438018a211811 : FAILURE in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3008bbd0d8584f009465d9f8108b0a96 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":55},{"id":"ae7b53b66666fbd2d16a2bdd8219210163d50f88","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-29 16:06:50.000000000","message":"Patch Set 56: Published edit on patch set 55.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":56},{"id":"f404ef314577e27ba40795a6fc811c414ef363a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 16:23:29.000000000","message":"Patch Set 56: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b1b251153ec496686cebf9fe8c55e24\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/55be7263e42b4fc3ab0e01c6fa74c0f0 : FAILURE in 10m 10s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1c9aae2b206e400d903dd9171d93ee09 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c55fe6756f47cf956160cc5aee8bef : FAILURE in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4c93ead1357451f8d4ae136f2b6cb17 : FAILURE in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d1c4fd4ea924fdf92d06f4077722791 : FAILURE in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c1fe0276bf24d8da291b076e5c3970c : FAILURE in 4m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f42d79160e7540ccb61b4052cdebad25 : FAILURE in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d9ed54c5f3e4eb98bd6e7d15d699f43 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":56},{"id":"843938abbcfe1484e8ee3f52ac44d18356c38d25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 17:06:34.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f85f043e04984ff0b3991ba8d463da8b\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/7c7ee104d92f4a5787590f458cf9530d : FAILURE in 17m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea365aca36754459a4aacf0aac8ebed4 : FAILURE in 4m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e181c1272888448d854c2e80da786bdd : FAILURE in 4m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b0663a214b17468c8d47ed0dd9c97efd : FAILURE in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"2c41ed8ce2f48c265ef9fa4266c75af796a95541","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 05:50:42.000000000","message":"Patch Set 57: Published edit on patch set 56.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":57},{"id":"f97d91d8d809c2182aea45ea4d03ef8e7d9b4310","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 05:56:54.000000000","message":"Patch Set 58: Published edit on patch set 57.","accounts_in_message":[],"_revision_number":58},{"id":"b203aa7d2ff919f851ac69516faa7a92bf5c644c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 06:07:12.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b491e5e5fbda49c78bc53e6c7f39fac9\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/4f362425dc1f4178adf99db98dbd242d : FAILURE in 7m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56526a660d3a4514b5a6b2ba0779c889 : FAILURE in 4m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e414c11e92fd4263be3d2306472be99f : FAILURE in 4m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5788d09fa0824458acf2cdb69d784f9d : FAILURE in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"0d99960f7b7c03ab5de8ef73ed7159aed387f9eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 06:08:35.000000000","message":"Patch Set 58: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a6f456fdaff47c8afdbe170092d4465\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/7c90747269b5463593b2a10d0b985530 : FAILURE in 3m 45s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c570071fb43249c794da2037f0162915 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecae8fa953d74cd7bf9e64eb6ba33711 : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79be056e34f64346be810936de385c13 : FAILURE in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e96316d2af1b401f80abdf9d7394ad23 : FAILURE in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8760b436caa54a3b87b619c2669cf9ed : FAILURE in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22ba385d1fb54387aa1bc8420b213bf7 : FAILURE in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcdf6d83ba98470f8d8050118b0ab3a8 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":58},{"id":"d1c6f9dd149780cabcc78547c7f368f57f6cc24e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 09:03:55.000000000","message":"Patch Set 59: Published edit on patch set 58.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":59},{"id":"fa1ac67e7897e8d5446093a9f79b4f2250b00891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 09:15:30.000000000","message":"Patch Set 59: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33187631d8384fe0ba5183961042199b\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/c4d85b2993e545a6a756963809b80897 : FAILURE in 4m 49s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dd926c20869c46e8970fb2a4c8222688 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ed718e6924437b9a835a69713ac062 : FAILURE in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98b0a9f4dd8c4bae97a7190f2337134b : FAILURE in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed01511155f54877978e21bfc0e28ce7 : FAILURE in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93149746c5184e5199bcc44e6ccdc9e3 : FAILURE in 5m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93948879cdb3404ca3ac8b78571685a1 : FAILURE in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ee913e8c4a430da8c551eba09c50b0 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":59},{"id":"bb43b9e6264e7b0034aedea73bc23eb71b81d4d1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 09:22:19.000000000","message":"Patch Set 60: Published edit on patch set 59.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":60},{"id":"253b2ef47a4aecf215c8c08f62ac0a3f8392d848","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 09:28:35.000000000","message":"Patch Set 61: Published edit on patch set 60.","accounts_in_message":[],"_revision_number":61},{"id":"d573d90027559aa843dd32002369c2abd9e90187","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 09:36:20.000000000","message":"Patch Set 62: Published edit on patch set 61.","accounts_in_message":[],"_revision_number":62},{"id":"79501d3ee329a523d8b9971c3db6464ee043778f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 09:37:11.000000000","message":"Patch Set 63: Published edit on patch set 62.","accounts_in_message":[],"_revision_number":63},{"id":"70a06906890008f63f1bad445249171f61558fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 09:46:29.000000000","message":"Patch Set 63: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e61f833b6e44d128021f75ac9fad7c5\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ed3b0b6e17554b9196a435d1579e183c : FAILURE in 7m 45s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/57a5f82d278b4e1c98c277241519fc85 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbc503053b324a238a703406d73ad7ed : FAILURE in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a404559975c4e3b920ac181e25a815e : FAILURE in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a552ceb394e24dcea0b44fdbc6401e99 : FAILURE in 1m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e871fb164e9d4dfc82036283daab2ce9 : FAILURE in 5m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6894bdc3c824e77b3b4f7e11fd92267 : FAILURE in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7a419a3b2774ebdbcf1a7913d0165cf : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":63},{"id":"021e94b922d05b73154226896ff4201451cf014c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 10:12:17.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebc619b541c1476893f5c254850fecf9\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/ff4efb3440714370acd4f677c0d26b60 : FAILURE in 16m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42abeb8e3d664a4789df6d13e6c0f82f : FAILURE in 4m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f7b09d28960457cbcfb33dc4abd1eb6 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/109ebe9779cc43cabe28044fd04caeb6 : FAILURE in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"174354b90ea9cc821067f0820c5ac491e5c30803","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 10:23:54.000000000","message":"Patch Set 64: Published edit on patch set 63.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":64},{"id":"d2efda590146291c660779a00a546cd97db19269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 10:36:44.000000000","message":"Patch Set 64: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dd44ffd681648c2b2a01f4c93e8ba12\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1e91b67863224958a60744a047fc62a9 : FAILURE in 6m 38s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/aae77aef84ca44109b6e77e41962b636 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1918baf37c94187b7d116dbef7534f0 : FAILURE in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfce8367d1974ab2b9aad6895474d75e : FAILURE in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed27d5ad3db14fbe846ed8a95a77f697 : FAILURE in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce72935a1393422d880b3c407655d75a : FAILURE in 5m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5928b53cec1f4dd8aba29d4b73050143 : FAILURE in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5c7a73490a404c955f78cd491fe1d2 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":64},{"id":"058068c863ef3d365ed1c63f2716feba37c29e23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 10:42:41.000000000","message":"Patch Set 64:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8b03d03e286469dbb1a5144fb2ca95b\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/3f9d495a2a8f4c64811af2e07bee8f86 : FAILURE in 16m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3dcf3e2b9e3247aca3fc377d7a8d0fd6 : FAILURE in 3m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3c0a19172bc4159aa627caec8d1dfa9 : FAILURE in 4m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/983649ae6b3d442f894bc87cb26d3546 : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"ef9e3eae635dc323b7dd5d6d118b548efa7369dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 13:12:11.000000000","message":"Uploaded patch set 65.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":65},{"id":"e969a30adf52fe2ad7490acb34959fd61699588a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 13:17:39.000000000","message":"Uploaded patch set 66.","accounts_in_message":[],"_revision_number":66},{"id":"fdc4dc0dfdbea85f14f6e37f29ee36dfa7ab3d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 13:30:38.000000000","message":"Patch Set 66: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df42aba9680f44498a8e46a75e9ed008\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/31f8106ba18c45a98ab6f5494aa6699b : FAILURE in 7m 34s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3c6364f9829c476a830193d128d9d9c4 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a801fafd3a43c0928b3a6bb5b25e64 : FAILURE in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e14faa3a895a426eb559fc8d9b6799cf : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5326468372684b8d8ace23c917ad5440 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1533e52f50a47f09afe68d7c0aa682f : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e18ce6defe540bbab2c23a605021b7e : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0220ab85774ed4a260f8f8332656fe : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":66},{"id":"b31979c361ef1f173ccdcb2cd8275197f19ec13f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 13:32:08.000000000","message":"Patch Set 66:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f1ec427955401b85af88b4e9c93d98\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/4802276fa26c4960841cb61cc6e6db49 : FAILURE in 12m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7234694d61f94014a4a2cdc3747bde21 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6853c245a484474ea07acde544babbd5 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c9e005972e54823a76a260b1b6a027b : SUCCESS in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"f8ae543b869b23b8cab1a2174bfdb06530da23f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 13:33:47.000000000","message":"Patch Set 67: Published edit on patch set 66.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":67},{"id":"bdc0cffabef3c5ab8dc6ef62ac5c03694bf317cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 13:46:10.000000000","message":"Patch Set 67: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a091f323c494d20b0bade3358ffcb40\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/12fbe2a1418a413c93a51506ddc201b8 : FAILURE in 7m 35s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/79b3300017b349f590c799234e76c3be : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20584e8697aa45ea8fd1448acf6611ea : FAILURE in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76464f8a6f4b48f3818e17524130c004 : FAILURE in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b225caffa754b1bbec243404c3c0ebe : FAILURE in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02ef0733930c4836a733d79e944d23d5 : FAILURE in 4m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb17b40475b14cfdb3c85c10a3bb4fff : FAILURE in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f54e5bbc7eb14fdbbd395ab146e0e55a : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":67},{"id":"dcde124851c40754f7ec757b7ce1b43ac97da301","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 13:57:07.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51901004560842c29b61b6efbb8e55db\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/2af6a2df4dc24d0cb4abfcbbecfc0e09 : FAILURE in 20m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/66cf6038dd0e407b85e0c2322a6a020e : FAILURE in 4m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7c865a5fcca4294a52c8eca7046175e : FAILURE in 5m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/412cad9f7f9e4499917e4dce502eb167 : FAILURE in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"ef822abfbd80fd1416021668b36d388eeaef3db8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 16:56:04.000000000","message":"Uploaded patch set 68.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":68},{"id":"aabc5588805169d9b19cea8dec0b55f216e1f22c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 17:09:54.000000000","message":"Patch Set 68: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48bdf6f29cb345108014291ac8aa1c35\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0194a6c3a63a476290bd84a16d343de7 : FAILURE in 8m 35s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6bf00d5c3df34ca9b7721487bbd49bf1 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e97367cb8934630bddff9e8c102a40e : FAILURE in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e508d302c128423592d9cfaf7ba22750 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46d76e99d08e45559a6c5c95f2c022a3 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/921cba77c9204403be0278693618adc4 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8823990915a434aa710ef44d44b272d : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37b587a4ab7474a85e64d886e48d8cc : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":68},{"id":"822d1b405b6b8c4ceaa266964ecf87b181d203ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 17:18:15.000000000","message":"Patch Set 68:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a9ae372e560494889f534d0b68e8336\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/366798a61efa429b85ad31af6664b8fa : FAILURE in 19m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/209b2ad93cf34cb2866b5d720122f037 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/91f19085d57f4d45b6674d207bef880f : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2529e08eaa0948e1a7b2026fb3dbfa18 : SUCCESS in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"758edefbaf26ad939ad63ba9dc696d5c71d9da2b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 17:19:24.000000000","message":"Uploaded patch set 69.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":69},{"id":"711bdf3627519b715745852d6f17ba08ec88af07","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 17:25:33.000000000","message":"Uploaded patch set 70.","accounts_in_message":[],"_revision_number":70},{"id":"87f019d3e6e7789ddc218590bd556f7f7086795f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 17:27:57.000000000","message":"Uploaded patch set 71.","accounts_in_message":[],"_revision_number":71},{"id":"24797763cd8a77e44e65d61f9d81eccff210bba6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 17:36:19.000000000","message":"Uploaded patch set 72.","accounts_in_message":[],"_revision_number":72},{"id":"a8863d2293391efacccd783f584e5843f904ad42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 17:46:58.000000000","message":"Patch Set 72: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/826537b03b0242e981ef86e9277a9034\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/e7b7b378c82a4c56b60dbe6d6c48599c : FAILURE in 5m 13s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/58478786f9da4289a1ce8eded7e8f856 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38bf63e10df490ca226e41df2681534 : FAILURE in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3113630d43b944f7a3a90390318d65fb : FAILURE in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e3498cb5fce4b20a554044d8a743284 : FAILURE in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/309e5f12276543cb8d6f1f763fe1e071 : FAILURE in 6m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31a0846ef718441aaad7744054399373 : FAILURE in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26195ce1a9ea44daa4d98286c309884b : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":72},{"id":"c009da47719ef22780dddb5a0b566fe3edcdcb09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 17:53:58.000000000","message":"Patch Set 72:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a8612b53c94444b659933053e65fed\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/aa6aa34984e34d488d76c006f9427cb1 : FAILURE in 6m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89034c3fa8584d93ae2a6b325daa55f3 : FAILURE in 3m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb2fb581953e484999b75552be3c77ae : FAILURE in 3m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a747ee775ab54e5b98462d12f90579d3 : FAILURE in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"d36023d1c38bc673fcb3d51fec89f866769b1f53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 17:58:57.000000000","message":"Uploaded patch set 73.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":73},{"id":"825b28215d7ede5fafd854b3d91d31b012428159","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 18:02:54.000000000","message":"Uploaded patch set 74.","accounts_in_message":[],"_revision_number":74},{"id":"c87cb563046351c0736b0a5dcdbf820f8ed911cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 18:09:16.000000000","message":"Uploaded patch set 75.","accounts_in_message":[],"_revision_number":75},{"id":"a264e513092668868ad6f350c3fa8ea3dc441234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 18:20:07.000000000","message":"Patch Set 75: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff41d7419dd546818881ee4a8bde55da\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/101b04acfc9741d5b1180fd68dc2320a : FAILURE in 5m 14s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0e5595d008914eaebb7ee242ce96a0c4 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33b72a4bda24102bf363ca15c802a22 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71dc277ca2a14451964080481bbc7666 : FAILURE in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee776d8def3f430b988406680d326353 : FAILURE in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18a00cffe7e14099a2ea3e1454ce14c1 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0383690af3554ae2b2d9647718c89d39 : FAILURE in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3950d8a5eb354fea8b720a836747cea6 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":75},{"id":"d3faa610c4fb70fc5134e84986e0f0aeb31da653","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 18:28:06.000000000","message":"Patch Set 75:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87213b0400a14213ac3ee548809b3a90\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/18dbcd1301af4e2eb236e7444d5ca2b2 : FAILURE in 5m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb5b8f33568f4923ae7c92ee9095ae14 : FAILURE in 4m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc97273bbbe746758c602b9b8a7095ec : FAILURE in 4m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9151bf71a9b40d090d6e606b4e33179 : FAILURE in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"167fca5168828c322d7b9699b1c068e096fe1bcb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 06:25:32.000000000","message":"Uploaded patch set 76.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":76},{"id":"d2c4e0eadfa5f3ac8af428b990f6dd72ddf729b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 06:39:23.000000000","message":"Patch Set 76: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11af8a9da78c4cfd89ff6222c53d8733\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b2af3052b93c427d8db2fda21a611477 : FAILURE in 11m 14s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a1e4144dfcbf45078ba50114e29e74c5 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d46aa1bf0b6486ab26653d24179acb9 : FAILURE in 1m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51bbe9b88e6448548ef5bf17174c3d1c : SUCCESS in 2m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30fb80835cbb4a91a14275560f789264 : SUCCESS in 1m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/49c78a3257444e36a4c326c39e9aab37 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bc3b0f724c140a3890d79aa4dfdc2de : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e11d162b3149ecb444bddd1b785fa0 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":76},{"id":"d4efe280c5db8c460493223f27c33ae35d4e985d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 06:42:19.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ef7d37f9ff7457593fd4420007f3a9e\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/0929071c752c4370b3e5409912686731 : FAILURE in 13m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/493afb9eea2f4e31860f0cef539b7cd4 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b3c7ddc895948dd99925143f4b8862f : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/45e01337baf443ba8ada50de5963f11e : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"9b0e67b0644b320deb55ca71612e2e4da929d649","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 06:53:48.000000000","message":"Uploaded patch set 77.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":77},{"id":"15a16a8cf5a0c099e170d369758ee2597b720ad5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 07:01:08.000000000","message":"Uploaded patch set 78.","accounts_in_message":[],"_revision_number":78},{"id":"a3b6feb76e9a877604bceb4608f3fa2576ee58f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 07:28:45.000000000","message":"Patch Set 78: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c748710980fd45d5a1eb6172548f6dab\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/599b68be6245497eac53fd77e7f1025c : SUCCESS in 16m 33s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b8d46c1632e844a4a698b2a1db62bb38 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fce29721b164ecf8131c0c650193bc0 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bb50991fb8a45fda6aa9faf1868badd : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79a1f51509d54c668b52c1f2a630369e : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5352d89ed3604a2db48c0218b74783cf : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dd6939f02d34439a7b890470d388564 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40c92eb612b4fc0b699800aae6dc50c : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":78},{"id":"b207f192ddaf4cc18c8706c4805092b64285ab05","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 07:33:45.000000000","message":"Patch Set 79: Published edit on patch set 78.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":79},{"id":"bcae6f0264c8cba43a72115bb463d15f1f96eaa9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 08:01:24.000000000","message":"Patch Set 80: Published edit on patch set 79.","accounts_in_message":[],"_revision_number":80},{"id":"817b969b4a51456ae9547d1ae66ed63652975165","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 08:27:16.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":80},{"id":"718b2e09d013551b749447fd29c167156d3a71ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 09:05:57.000000000","message":"Patch Set 81: Patch Set 80 was rebased","accounts_in_message":[],"_revision_number":81},{"id":"e13ac17f8374951c2a1a6a6f66e37c4e0c35e102","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-01 09:06:32.000000000","message":"Patch Set 80:\n\n(1 comment)","accounts_in_message":[],"_revision_number":80},{"id":"79712bbce4a60259d27e8162b5f3d49dcdf28b51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 09:13:32.000000000","message":"Patch Set 82: Patch Set 81 was rebased","accounts_in_message":[],"_revision_number":82},{"id":"766e9646e6f49b73c06f78f86500eb1ca464bb7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 10:04:22.000000000","message":"Patch Set 83: Patch Set 82 was rebased","accounts_in_message":[],"_revision_number":83},{"id":"5214c7c13a5611102651fbb84e8c7a22a3f2aa67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 11:40:47.000000000","message":"Patch Set 84: Published edit on patch set 83.","accounts_in_message":[],"_revision_number":84},{"id":"ff10bd7b5c872b768c1d56c9adfb41fad6fbbc03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 12:45:40.000000000","message":"Patch Set 85: Published edit on patch set 84.","accounts_in_message":[],"_revision_number":85},{"id":"d86135d94491798b5bdd05cbb8db05460c10e676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 12:46:02.000000000","message":"Patch Set 86: Published edit on patch set 85.","accounts_in_message":[],"_revision_number":86},{"id":"2003ef1cecc5d47e878508f69fc6720f409c2277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 13:02:30.000000000","message":"Patch Set 87: Published edit on patch set 86.","accounts_in_message":[],"_revision_number":87},{"id":"23a7b249733d9f578e5c5da5f8c4b53d528df200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 13:30:30.000000000","message":"Patch Set 88: Published edit on patch set 87.","accounts_in_message":[],"_revision_number":88},{"id":"2677de51a23f04f635bd9db57c2ef82f2c28a50f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 13:37:39.000000000","message":"Patch Set 89: Patch Set 88 was rebased","accounts_in_message":[],"_revision_number":89},{"id":"eccae42ae38e16e441e653700d3033f4b5203138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 13:44:49.000000000","message":"Patch Set 90: Published edit on patch set 89.","accounts_in_message":[],"_revision_number":90},{"id":"c75c16ee526849cfc7dafe5a59a9f85b82603aea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 13:54:57.000000000","message":"Patch Set 91: Published edit on patch set 90.","accounts_in_message":[],"_revision_number":91},{"id":"67097ea3ff4341cb4001c5eb5e3717736e983fd8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 13:55:15.000000000","message":"Patch Set 91:\n\n(1 comment)","accounts_in_message":[],"_revision_number":91},{"id":"e2f65b2e7a9fc932a235bc97cb2b316c535aede4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 15:27:46.000000000","message":"Patch Set 92: Patch Set 91 was rebased","accounts_in_message":[],"_revision_number":92},{"id":"3be43ef35537a87f669edab190cccb0fa960cffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 17:51:46.000000000","message":"Patch Set 92: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e51b569ce884855a871502e43804da3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/44bf7f9192854708a79c8c3babfc3991 : SUCCESS in 16m 17s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/660b7b8170f8494097dc2b2d717a2c4f : SUCCESS in 9m 53s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/16520876dd184ac9bb20c6616e30349c : SUCCESS in 15m 06s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed6489e52caf49a389302a5e4bc9ae7b : SUCCESS in 17m 28s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f9eff952f83f4cdaa25ea1cb64a98277 : SUCCESS in 14m 47s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/79c5b477c1ce4c5db15516a268f87975 : SUCCESS in 34m 00s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/92359867634949b88d3d47365ebd6472 : SUCCESS in 42m 54s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/46a4af99442e4f108eff99b1d315923f : SUCCESS in 1h 14m 13s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5525990a95ed468c9cf1a10fd79475e1 : SUCCESS in 53m 00s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/83590989e7864551a681227688c15734 : SUCCESS in 1h 17m 12s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/b54af223f4c24124bd35b36fe813bf68 : SUCCESS in 1h 44m 49s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/8d2cdb65914b47b3b8e51f21116973f7 : SUCCESS in 2h 12m 48s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/1cf3bea1ff924e999bc3cc9b7a559f0d : SUCCESS in 2h 04m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/fa00e311dc4c4ae49d31ee3ffee25789 : SUCCESS in 2h 22m 11s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/d0240bbedd3c4741a1a75a4b4a29a89a : SUCCESS in 55m 50s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/2ff9ddfba53d4df998b4a22413d7cb5c : SUCCESS in 1h 22m 04s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/464b3f8f6b464a2faad3bb0c4f88529a : SUCCESS in 24m 40s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/ee82cdad1fa947caa7e8e8efaf57d5e3 : SUCCESS in 15m 45s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/9990061303f24f22a96bcaf008f7e69e : SUCCESS in 29m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/1cff47f413544b5eb0931d8bbf164d70 : SUCCESS in 47m 18s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/d5b598fe2ec14b139b3e5feb1a4604e9 : SUCCESS in 47m 18s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e081c888880e456396c5ae169d442b0b : SUCCESS in 51m 48s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/2994ae23342f4dbbb1ab582e42f2c5ad : SUCCESS in 1h 21m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/07172553c7934a24b8d3862533df3ae1 : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d5e208d71e09413eb4bb42894f48e97a : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588c8c57a2c64b3cabb0c133be622769 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78e0d6d6b0b64511981bd021640fba7d : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f0e446184164bfc8e0098834ed2e46d : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea8b179c315d4cbd81cc9ffa8b614e49 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9b9384ffa1f47a480e1db765f474f95 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34a662f95b2478a982736a5e954934f : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa560ba125424c6eb7a7eb53c279a773 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":92},{"id":"a0bee268da52848582510b23b67207c15cb5b39f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 01:01:12.000000000","message":"Patch Set 92:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a99e39c5d34c43c98ed637b849430cb3\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/137ed9c8e4b0497890004a6356efc270 : SUCCESS in 32m 45s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/78d6f5a73df5414db2c3d2e4f5f87f0a : SUCCESS in 49m 30s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/def2d45c143841ca97deaebd938f2ab4 : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/16a61f13541240e5b3ce8efa2781cdfd : SUCCESS in 1h 09m 14s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/605a59e6c45545128ba8ca6cfbef4b66 : SUCCESS in 59m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0204c237db3a49e488c500eea261f674 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9f69a6ab6194135bcac18dc72e7ba93 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/555f346bd7844daabdea11187e9f5100 : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"8fdb8f69e6a1b63db912393e59393005c41ab47e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 08:37:12.000000000","message":"Patch Set 93: Patch Set 92 was rebased\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml\n* tests/templates/template_overrides.j2\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":93},{"id":"72a58e5d7223f59ff8c90b89789850f53acc9612","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 08:38:30.000000000","message":"Patch Set 94: Published edit on patch set 93.","accounts_in_message":[],"_revision_number":94},{"id":"c78726c6531a5509c321fff573503546b35b814b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 08:41:32.000000000","message":"Patch Set 95: Published edit on patch set 94.","accounts_in_message":[],"_revision_number":95},{"id":"1fc9a02e0541b2b01f488a534541ff34c0710fa5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 08:42:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":95},{"id":"a764977ba5b4c92546ab8ed66d5c02a0cba3dc33","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 09:20:47.000000000","message":"Topic set to kolla_rpm_repos_yaml","accounts_in_message":[],"_revision_number":95},{"id":"321cb9247e3de26084eb4fb7fa17d68af4f8c48e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-02 09:27:45.000000000","message":"Patch Set 95: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":95},{"id":"95a7d40253a1db2fc15cd45135535b94c5fc5807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 09:29:45.000000000","message":"Patch Set 96: Published edit on patch set 95.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":96},{"id":"4f5201659ce250623470e20290e2870dc370a7d5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 09:29:57.000000000","message":"Patch Set 96:\n\n(2 comments)","accounts_in_message":[],"_revision_number":96},{"id":"21cdde4aeac352446d75814c524d95879001b252","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-02 09:46:47.000000000","message":"Patch Set 96: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":96},{"id":"38fa2d17fe28d36a5913cc9a9d56f1ed4281e07b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-02 09:51:50.000000000","message":"Patch Set 96:\n\n(1 comment)","accounts_in_message":[],"_revision_number":96},{"id":"b567a29242238bf80a3b121e659582ce691ec17a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 11:15:31.000000000","message":"Patch Set 97: Published edit on patch set 96.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":97},{"id":"8e094d072e28ab083ed0b00a7391517a922b30f8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 11:16:39.000000000","message":"Patch Set 97:\n\n(2 comments)","accounts_in_message":[],"_revision_number":97},{"id":"bf9d6a1f3c9dd397d4095a20e0661ff90b7f69a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 11:39:07.000000000","message":"Uploaded patch set 98.","accounts_in_message":[],"_revision_number":98},{"id":"80f8307efc78a0b0dd053190dfb905805b5fc02f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 11:42:11.000000000","message":"Patch Set 98:\n\n(1 comment)","accounts_in_message":[],"_revision_number":98},{"id":"0482ce666e85300cd37172461e61fefe758e8584","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-02 11:54:42.000000000","message":"Patch Set 98: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":98},{"id":"c672f61a6c177cf57b3a8f90751665a0d7c77bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 11:56:27.000000000","message":"Patch Set 99: Published edit on patch set 98.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":99},{"id":"61118bd73edb719d17186ccb799e7fa49f4ba9ad","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 11:58:13.000000000","message":"Patch Set 99:\n\n(1 comment)","accounts_in_message":[],"_revision_number":99},{"id":"5604b4f793470c8e3f9a3cb6af7b75711c672891","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-02 12:09:04.000000000","message":"Patch Set 99: Code-Review+2","accounts_in_message":[],"_revision_number":99},{"id":"926ab989bcfd53d4cbf95be7f80b967751ed71ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 12:24:40.000000000","message":"Uploaded patch set 100.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":100},{"id":"fae52cca6930207c73f3146c3cd9235d1f162702","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 12:25:15.000000000","message":"Patch Set 100: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":100},{"id":"e7532380356457eeb4ce69af6c5f5f09267e8e82","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-02 12:36:58.000000000","message":"Patch Set 100: Code-Review+2","accounts_in_message":[],"_revision_number":100},{"id":"42d09e9bdd607ca2cae56236da50176c207fcbf4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-02 13:44:34.000000000","message":"Patch Set 100: Workflow+1","accounts_in_message":[],"_revision_number":100},{"id":"b10e143227f3e4d9e90ca323c5749ab36f659cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 14:58:10.000000000","message":"Patch Set 100: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8afedc9c378d4ac2936b313c1e0dd5af\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ef1cf5bf1bf248e7a9c229ac8da2560b : SUCCESS in 19m 36s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/0c1d459bc3e74fc699497471e14c216b : SUCCESS in 20m 08s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/65fa67b66a094bd0af3a3533cda3667c : SUCCESS in 28m 52s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/004468e69c3b4ff9b8971486627f72c5 : SUCCESS in 21m 18s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/43d682d37f794a63b4b4ac69a776cdc3 : SUCCESS in 34m 34s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f2279163ce44136ae5d7e6f20b79715 : SUCCESS in 19m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c927f5340124a2393e0475c0ad8a8c0 : SUCCESS in 21m 37s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/ddb35bd7661c43358ad35263b37f5ff9 : FAILURE in 1h 23m 16s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/98329b761c9c41018576fc5bbe9e571d : SUCCESS in 2h 26m 36s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/0d72c6364bb44c0ab8641b791900691d : FAILURE in 1h 21m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/ff0cb8f18af94f1cb99bdc4d39eae83d : SUCCESS in 2h 32m 01s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/be6677aca10a4c6eb3912bddfabccf9f : SUCCESS in 59m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/d7ca0dc1f39647a2a6ff7f945bdda2bd : SUCCESS in 1h 20m 36s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/10183b9c01eb4589b004e32edc6e5e5b : SUCCESS in 27m 20s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/6e2a703407cf4006828291b99431a6d2 : SUCCESS in 25m 03s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/9f5fe394b8ed4b7c8cf59a7d18a1e78f : SUCCESS in 25m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/0ad3bf913e8f4f17a7a6388db97a3e96 : SUCCESS in 1h 03m 06s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/4662d884b18d4b18bb6c48297388bd81 : SUCCESS in 57m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/92e7e04197bf466dba02e5119343a7c4 : SUCCESS in 54m 19s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/d2258cd197a9433bab0dff4ab5d317be : FAILURE in 1h 26m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d3676791bb0642869130589ad2e65332 : FAILURE in 1h 24m 03s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4ac8279484e04fdcaf8e3ba074bd65d3 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ff529e70ee46629470568a4ce1492c : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/162142309b964ebcb1c4e3a92a4b5f56 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/432036c2bed8445c888375fb6405fcb0 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd92b91688484ee6a305e35d2188f7d8 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d23807c6661842a7baee8aa36d40d3dc : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fa540a07a2494891aebf921174d7c3 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17dcc76967e64de89de95c79dea895ec : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":100},{"id":"ba2349085dcdacdfaa967c35dba8b7955d4336a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 14:58:37.000000000","message":"Patch Set 100: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":100},{"id":"667bb76a971e4cd34842b5bc09d686364a147152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 16:05:25.000000000","message":"Patch Set 100: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd205a9e46d14e9387dd945edd31fd8d\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/42b1ba28fec24d458534c87474fa0e07 : SUCCESS in 19m 53s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/cf8dfb00e7684c0ba1070091a4ce3dd2 : SUCCESS in 26m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e1a615c46b4649adbf273c0ec3dce216 : SUCCESS in 23m 03s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ebf8c042eab04d9fa8961d51679e9fcc : SUCCESS in 1h 03m 54s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9928d4f438714befb65c7ceb46829130 : SUCCESS in 15m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0077dbe0d9a641f5a310e0f38284f2b2 : SUCCESS in 24m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d7670f4da3a44d42b1aa513833e60530 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd31ac54e41f45fb94b32fe422572f5c : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f05837976a214701a32bb374b733cd12 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/247112e3bd604f528725ae11e8e48ab0 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e93629ac10141bb866a787db634f8c2 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d417250ef2c432da9eecb4548e69bfc : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":100},{"id":"286cd9c7ea385f161e3ab9573412aa0e31f859c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 16:05:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":100},{"id":"784de3067751aab686dab2a0bd119018ebfbe11b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 16:07:47.000000000","message":"Patch Set 100:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/585f1dd033d4448aa495c5a3a9a8f8da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc016759c10a4b1d81d056b34cedb847 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23bdb1aef83846a3a3fbc061f600504a : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":100},{"id":"28ac713652433f5d347c385b03c3e6cba86ffebd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 16:14:55.000000000","message":"Patch Set 100:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3edf1c324d6a47d78d03a330670d1092\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/17cc56eb01e04c30bed940894fe64225 : SUCCESS in 42m 52s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/5621e33097fa431c913f64dd4a20928e : SUCCESS in 58m 25s (non-voting)\n- kolla-build-rocky-10-aarch64 https://zuul.opendev.org/t/openstack/build/4db162c3c0aa432fb74f50bdeaa9bf00 : SUCCESS in 45m 15s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/2ae052c773e3446db28194c0eb6a6ac7 : SUCCESS in 45m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e24bc8b3f9e14777856c6524a6e44514 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a0f74ae0ddfc4b1badc760975ead4b5b : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b1044dd28aee41a799161972505cd17e : SUCCESS in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":100}],"current_revision_number":100,"current_revision":"806602e488240deb87bdf1c8ce00973ca8ec914d","revisions":{"1c559bf1a9e00334ce959601c350e156bb3220ca":{"kind":"REWORK","_number":1,"created":"2024-02-22 14:52:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/1"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:39.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c559bf1a9e00334ce959601c350e156bb3220ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c559bf1a9e00334ce959601c350e156bb3220ca"}]},"branch":"refs/heads/master"},"e6ec6d0bb54bf81ba351f0733fc594332dbee622":{"kind":"REWORK","_number":2,"created":"2024-02-22 15:25:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/2"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 15:25:01.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6ec6d0bb54bf81ba351f0733fc594332dbee622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6ec6d0bb54bf81ba351f0733fc594332dbee622"}]},"branch":"refs/heads/master"},"fdf4eb070848e0891c7e1ee610a4778f6df3ccba":{"kind":"REWORK","_number":3,"created":"2024-02-23 11:24:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/3"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-23 11:24:11.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdf4eb070848e0891c7e1ee610a4778f6df3ccba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fdf4eb070848e0891c7e1ee610a4778f6df3ccba"}]},"branch":"refs/heads/master"},"ad538464717188b0fde658bf5de2a073e53710d9":{"kind":"REWORK","_number":4,"created":"2024-02-23 12:02:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/4"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-23 12:02:42.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad538464717188b0fde658bf5de2a073e53710d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad538464717188b0fde658bf5de2a073e53710d9"}]},"branch":"refs/heads/master"},"50428fe3efed8b9a33c03e22061f07cd81eefc2e":{"kind":"REWORK","_number":5,"created":"2024-02-26 08:42:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/5"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 08:42:20.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50428fe3efed8b9a33c03e22061f07cd81eefc2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50428fe3efed8b9a33c03e22061f07cd81eefc2e"}]},"branch":"refs/heads/master"},"6c6393be46deea90135044f43f6c734bcc4250ba":{"kind":"REWORK","_number":6,"created":"2024-02-26 09:16:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/6"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 09:16:10.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c6393be46deea90135044f43f6c734bcc4250ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c6393be46deea90135044f43f6c734bcc4250ba"}]},"branch":"refs/heads/master"},"999a1e5b11cdf0af8e82416b093fda460678b240":{"kind":"REWORK","_number":7,"created":"2024-02-26 10:52:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/7"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 10:52:31.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/999a1e5b11cdf0af8e82416b093fda460678b240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/999a1e5b11cdf0af8e82416b093fda460678b240"}]},"branch":"refs/heads/master"},"107aec794e322280b53038d0a3d442ee33ad4765":{"kind":"REWORK","_number":8,"created":"2024-02-26 11:51:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/8"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 11:51:41.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/107aec794e322280b53038d0a3d442ee33ad4765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/107aec794e322280b53038d0a3d442ee33ad4765"}]},"branch":"refs/heads/master"},"461b1bdec82ff7e14076a5d3a263bda9bd134f47":{"kind":"REWORK","_number":9,"created":"2024-02-26 12:17:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/9"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 12:17:38.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/461b1bdec82ff7e14076a5d3a263bda9bd134f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/461b1bdec82ff7e14076a5d3a263bda9bd134f47"}]},"branch":"refs/heads/master"},"7980e14a2bea54f8e75c8ff839a5fbf6fa6370a1":{"kind":"REWORK","_number":10,"created":"2024-02-28 12:47:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/10"}}},"commit":{"parents":[{"commit":"756ee984c3c761514115c342e7cba9a3cb533d02","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/756ee984c3c761514115c342e7cba9a3cb533d02"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-28 12:47:08.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7980e14a2bea54f8e75c8ff839a5fbf6fa6370a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7980e14a2bea54f8e75c8ff839a5fbf6fa6370a1"}]},"branch":"refs/heads/master"},"570bd662655c501565c626d32208267a2d9993a2":{"kind":"REWORK","_number":11,"created":"2024-03-12 06:55:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/11"}}},"commit":{"parents":[{"commit":"f18b9142ade246624467ca71310d28ce5743bd10","subject":"Merge \"openstack-base: Pin tooz to 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f18b9142ade246624467ca71310d28ce5743bd10"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-12 06:55:13.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/570bd662655c501565c626d32208267a2d9993a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/570bd662655c501565c626d32208267a2d9993a2"}]},"branch":"refs/heads/master","description":"Rebase"},"057325d3092f94d72adda4368a439fb872d32b2e":{"kind":"REWORK","_number":12,"created":"2024-03-12 06:55:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/12"}}},"commit":{"parents":[{"commit":"f18b9142ade246624467ca71310d28ce5743bd10","subject":"Merge \"openstack-base: Pin tooz to 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f18b9142ade246624467ca71310d28ce5743bd10"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-12 06:55:31.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/057325d3092f94d72adda4368a439fb872d32b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/057325d3092f94d72adda4368a439fb872d32b2e"}]},"branch":"refs/heads/master"},"6be8655c4dd2397311bedeadf6f04f7c57b6831a":{"kind":"REWORK","_number":13,"created":"2024-03-12 10:30:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/13"}}},"commit":{"parents":[{"commit":"f18b9142ade246624467ca71310d28ce5743bd10","subject":"Merge \"openstack-base: Pin tooz to 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f18b9142ade246624467ca71310d28ce5743bd10"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-12 10:30:39.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6be8655c4dd2397311bedeadf6f04f7c57b6831a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6be8655c4dd2397311bedeadf6f04f7c57b6831a"}]},"branch":"refs/heads/master"},"cde5c148cc742e7999470d692ee6c97cabfaf134":{"kind":"REWORK","_number":14,"created":"2024-03-12 13:45:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/14"}}},"commit":{"parents":[{"commit":"f18b9142ade246624467ca71310d28ce5743bd10","subject":"Merge \"openstack-base: Pin tooz to 5.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f18b9142ade246624467ca71310d28ce5743bd10"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-12 13:45:41.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cde5c148cc742e7999470d692ee6c97cabfaf134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cde5c148cc742e7999470d692ee6c97cabfaf134"}]},"branch":"refs/heads/master"},"cbb7f57fa75f63e5b43106c4b2b76f89d5f9e5b0":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-05-29 12:06:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/15"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-29 12:06:18.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbb7f57fa75f63e5b43106c4b2b76f89d5f9e5b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbb7f57fa75f63e5b43106c4b2b76f89d5f9e5b0"}]},"branch":"refs/heads/master","description":"Rebase"},"d6e6654d7b38b77cdeda9c35b6eb240785121628":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-08-28 11:52:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/16"}}},"commit":{"parents":[{"commit":"ece67cd2aec82d661ba355ce6bc6e41bb88001b6","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ece67cd2aec82d661ba355ce6bc6e41bb88001b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 11:52:29.000000000","tz":0},"subject":"WIP: Add support for rpm to repos.yaml","message":"WIP: Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6e6654d7b38b77cdeda9c35b6eb240785121628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6e6654d7b38b77cdeda9c35b6eb240785121628"}]},"branch":"refs/heads/master","description":"Rebase"},"72f5c235e7faf18befcea8b0f069603eb3af3416":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2024-08-28 11:52:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/17"}}},"commit":{"parents":[{"commit":"ece67cd2aec82d661ba355ce6bc6e41bb88001b6","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ece67cd2aec82d661ba355ce6bc6e41bb88001b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 11:52:37.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72f5c235e7faf18befcea8b0f069603eb3af3416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72f5c235e7faf18befcea8b0f069603eb3af3416"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2971b943cd72f57120a0d67f7abe8373f5d303dd":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-09-24 10:30:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/18"}}},"commit":{"parents":[{"commit":"8f0a3b5871d20d31390aabdb5110a89451140720","subject":"Add missing iso packages to ironic-conductor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f0a3b5871d20d31390aabdb5110a89451140720"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-24 10:30:07.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2971b943cd72f57120a0d67f7abe8373f5d303dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2971b943cd72f57120a0d67f7abe8373f5d303dd"}]},"branch":"refs/heads/master","description":"Rebase"},"a89933d7d3e0df17c07feb69368cdae5d817938f":{"kind":"REWORK","_number":19,"created":"2024-09-24 10:32:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/19"}}},"commit":{"parents":[{"commit":"8f0a3b5871d20d31390aabdb5110a89451140720","subject":"Add missing iso packages to ironic-conductor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f0a3b5871d20d31390aabdb5110a89451140720"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-24 10:32:34.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a89933d7d3e0df17c07feb69368cdae5d817938f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a89933d7d3e0df17c07feb69368cdae5d817938f"}]},"branch":"refs/heads/master"},"1ca5e3902142f741d0801a1695fe4fc2ef7305a9":{"kind":"REWORK","_number":20,"created":"2024-09-24 10:32:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/20"}}},"commit":{"parents":[{"commit":"8f0a3b5871d20d31390aabdb5110a89451140720","subject":"Add missing iso packages to ironic-conductor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f0a3b5871d20d31390aabdb5110a89451140720"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-24 10:32:59.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ca5e3902142f741d0801a1695fe4fc2ef7305a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ca5e3902142f741d0801a1695fe4fc2ef7305a9"}]},"branch":"refs/heads/master"},"580550835634dc9cd6add8381dd442732f2cb494":{"kind":"REWORK","_number":21,"created":"2024-09-24 10:33:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/21"}}},"commit":{"parents":[{"commit":"8f0a3b5871d20d31390aabdb5110a89451140720","subject":"Add missing iso packages to ironic-conductor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f0a3b5871d20d31390aabdb5110a89451140720"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-24 10:33:14.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/580550835634dc9cd6add8381dd442732f2cb494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/580550835634dc9cd6add8381dd442732f2cb494"}]},"branch":"refs/heads/master"},"fbac582225ed9e8599140b37a3f19823656b175b":{"kind":"REWORK","_number":22,"created":"2025-05-20 18:09:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/22"}}},"commit":{"parents":[{"commit":"8af831c75afd2d6fa19123748d4237c56cc07b33","subject":"Merge \"Revert \"mariadb: pin to 10.11.11\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8af831c75afd2d6fa19123748d4237c56cc07b33"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-20 18:09:16.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbac582225ed9e8599140b37a3f19823656b175b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbac582225ed9e8599140b37a3f19823656b175b"}]},"branch":"refs/heads/master","description":"Rebase"},"dbad00b8f3aa4dbd99e4f3b43b01dc138d395f63":{"kind":"REWORK","_number":23,"created":"2025-05-20 18:10:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/23"}}},"commit":{"parents":[{"commit":"8af831c75afd2d6fa19123748d4237c56cc07b33","subject":"Merge \"Revert \"mariadb: pin to 10.11.11\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8af831c75afd2d6fa19123748d4237c56cc07b33"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-20 18:10:18.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbad00b8f3aa4dbd99e4f3b43b01dc138d395f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbad00b8f3aa4dbd99e4f3b43b01dc138d395f63"}]},"branch":"refs/heads/master"},"b8232bc75381e1c8dc629f0557a35d259919d186":{"kind":"REWORK","_number":24,"created":"2025-05-22 08:57:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/24"}}},"commit":{"parents":[{"commit":"8af831c75afd2d6fa19123748d4237c56cc07b33","subject":"Merge \"Revert \"mariadb: pin to 10.11.11\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8af831c75afd2d6fa19123748d4237c56cc07b33"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-22 08:57:03.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8232bc75381e1c8dc629f0557a35d259919d186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8232bc75381e1c8dc629f0557a35d259919d186"}]},"branch":"refs/heads/master"},"d1f16a2a23c8ef6f6d8a44cc3529682b9bb3b103":{"kind":"REWORK","_number":25,"created":"2025-05-22 10:07:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/25"}}},"commit":{"parents":[{"commit":"8af831c75afd2d6fa19123748d4237c56cc07b33","subject":"Merge \"Revert \"mariadb: pin to 10.11.11\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8af831c75afd2d6fa19123748d4237c56cc07b33"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-22 10:07:15.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1f16a2a23c8ef6f6d8a44cc3529682b9bb3b103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1f16a2a23c8ef6f6d8a44cc3529682b9bb3b103"}]},"branch":"refs/heads/master"},"f749a4c6efce0bb65655dd78ce4dff81280e0e5e":{"kind":"REWORK","_number":26,"created":"2025-07-18 14:28:36.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/79/909879/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/26"}}},"commit":{"parents":[{"commit":"5d2dfc0eb9fac7d22a04665173fe650b61985d19","subject":"Merge \"Move to CentOS Stream 10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d2dfc0eb9fac7d22a04665173fe650b61985d19"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-07-18 14:28:20.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f749a4c6efce0bb65655dd78ce4dff81280e0e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f749a4c6efce0bb65655dd78ce4dff81280e0e5e"}]},"branch":"refs/heads/master"},"93e0c433e4f2b49da43c6adc206e4e64d0930067":{"kind":"REWORK","_number":27,"created":"2025-09-26 15:56:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/27"}}},"commit":{"parents":[{"commit":"5d2dfc0eb9fac7d22a04665173fe650b61985d19","subject":"Merge \"Move to CentOS Stream 10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d2dfc0eb9fac7d22a04665173fe650b61985d19"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-26 15:56:33.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93e0c433e4f2b49da43c6adc206e4e64d0930067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/93e0c433e4f2b49da43c6adc206e4e64d0930067"}]},"branch":"refs/heads/master"},"90fe66e14ba78186a763514e1c5944f1ff6b8482":{"kind":"REWORK","_number":28,"created":"2025-09-26 15:56:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/28"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-26 15:56:38.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90fe66e14ba78186a763514e1c5944f1ff6b8482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90fe66e14ba78186a763514e1c5944f1ff6b8482"}]},"branch":"refs/heads/master","description":"Rebase"},"7aa9e2c1d9683dc89e71f3249751a0c22a595890":{"kind":"REWORK","_number":29,"created":"2025-09-26 15:57:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/29"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-26 15:57:12.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aa9e2c1d9683dc89e71f3249751a0c22a595890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aa9e2c1d9683dc89e71f3249751a0c22a595890"}]},"branch":"refs/heads/master"},"d5c7362ccf65a251d2be76e90e021782ba155b86":{"kind":"REWORK","_number":30,"created":"2025-09-26 16:19:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/30"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-26 16:19:06.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5c7362ccf65a251d2be76e90e021782ba155b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5c7362ccf65a251d2be76e90e021782ba155b86"}]},"branch":"refs/heads/master"},"98a500147504ef351ab3355fe55ca3e27a9f0adb":{"kind":"REWORK","_number":31,"created":"2025-09-26 16:43:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/31"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-26 16:43:08.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98a500147504ef351ab3355fe55ca3e27a9f0adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98a500147504ef351ab3355fe55ca3e27a9f0adb"}]},"branch":"refs/heads/master"},"117020623b53a6e3f3267a3a5654558fe507b87b":{"kind":"REWORK","_number":32,"created":"2025-09-29 07:51:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/32"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 07:51:50.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/117020623b53a6e3f3267a3a5654558fe507b87b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/117020623b53a6e3f3267a3a5654558fe507b87b"}]},"branch":"refs/heads/master"},"1bd6f2d64879a884c8e44e6da191774299908d82":{"kind":"REWORK","_number":33,"created":"2025-09-29 08:01:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/33"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 08:01:32.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bd6f2d64879a884c8e44e6da191774299908d82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bd6f2d64879a884c8e44e6da191774299908d82"}]},"branch":"refs/heads/master"},"5ebf9a53b747d759a1811ade1eeef6c540548899":{"kind":"REWORK","_number":34,"created":"2025-09-29 08:15:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/34"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 08:15:36.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ebf9a53b747d759a1811ade1eeef6c540548899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ebf9a53b747d759a1811ade1eeef6c540548899"}]},"branch":"refs/heads/master"},"78f9742a101ffe67342859c77d9994b367b96edd":{"kind":"REWORK","_number":35,"created":"2025-09-29 08:17:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/35"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 08:17:38.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f9742a101ffe67342859c77d9994b367b96edd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f9742a101ffe67342859c77d9994b367b96edd"}]},"branch":"refs/heads/master"},"9f6b9937e46d2dc2a88c21e6fe1bc6b3c2d9e645":{"kind":"REWORK","_number":36,"created":"2025-09-29 08:34:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/36"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 08:34:17.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f6b9937e46d2dc2a88c21e6fe1bc6b3c2d9e645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f6b9937e46d2dc2a88c21e6fe1bc6b3c2d9e645"}]},"branch":"refs/heads/master"},"e36302ba5a15951c80756de038045954a1bbc1e4":{"kind":"REWORK","_number":37,"created":"2025-09-29 08:44:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/37"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 08:44:32.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e36302ba5a15951c80756de038045954a1bbc1e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e36302ba5a15951c80756de038045954a1bbc1e4"}]},"branch":"refs/heads/master"},"39db1139129f0cded36c4d3f6253af4d07ca436b":{"kind":"REWORK","_number":38,"created":"2025-09-29 08:56:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/38"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 08:56:24.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39db1139129f0cded36c4d3f6253af4d07ca436b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39db1139129f0cded36c4d3f6253af4d07ca436b"}]},"branch":"refs/heads/master"},"35edf46b37a718139a247c30f8960d2b1e3981d8":{"kind":"REWORK","_number":39,"created":"2025-09-29 09:39:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/39"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 09:39:12.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35edf46b37a718139a247c30f8960d2b1e3981d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35edf46b37a718139a247c30f8960d2b1e3981d8"}]},"branch":"refs/heads/master"},"ab95fb1a34292c45ab1b480c2c0d4fc836f562ad":{"kind":"REWORK","_number":40,"created":"2025-09-29 09:55:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/40"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 09:55:33.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab95fb1a34292c45ab1b480c2c0d4fc836f562ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab95fb1a34292c45ab1b480c2c0d4fc836f562ad"}]},"branch":"refs/heads/master"},"a02b484bc9b20cd0f9e604d9368e3cfec3323d63":{"kind":"REWORK","_number":41,"created":"2025-09-29 10:02:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/41"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 10:02:06.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a02b484bc9b20cd0f9e604d9368e3cfec3323d63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a02b484bc9b20cd0f9e604d9368e3cfec3323d63"}]},"branch":"refs/heads/master"},"e6928728fc591e23828fba2c50418426aa86147c":{"kind":"REWORK","_number":42,"created":"2025-09-29 10:16:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/42"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 10:16:20.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6928728fc591e23828fba2c50418426aa86147c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6928728fc591e23828fba2c50418426aa86147c"}]},"branch":"refs/heads/master"},"7075ea28364ed4db5c4237a762592bd5341a0454":{"kind":"REWORK","_number":43,"created":"2025-09-29 10:30:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/43"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 10:30:05.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7075ea28364ed4db5c4237a762592bd5341a0454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7075ea28364ed4db5c4237a762592bd5341a0454"}]},"branch":"refs/heads/master"},"3df2bd890589a11c13f38fdabe60a7aca3aae692":{"kind":"REWORK","_number":44,"created":"2025-09-29 10:39:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/44"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 10:39:25.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3df2bd890589a11c13f38fdabe60a7aca3aae692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3df2bd890589a11c13f38fdabe60a7aca3aae692"}]},"branch":"refs/heads/master"},"e9e9377ff52ab9fd832adaf3de76228f5a87eeff":{"kind":"REWORK","_number":45,"created":"2025-09-29 11:04:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/45"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 11:04:41.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9e9377ff52ab9fd832adaf3de76228f5a87eeff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9e9377ff52ab9fd832adaf3de76228f5a87eeff"}]},"branch":"refs/heads/master"},"b69a852e39e93aedf6378044471568a3939c2bd1":{"kind":"REWORK","_number":46,"created":"2025-09-29 11:20:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/46"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 11:20:16.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b69a852e39e93aedf6378044471568a3939c2bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b69a852e39e93aedf6378044471568a3939c2bd1"}]},"branch":"refs/heads/master"},"e576715eae1f1a1504bc49ac6072d520e2226954":{"kind":"REWORK","_number":47,"created":"2025-09-29 11:20:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/47"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 11:20:58.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e576715eae1f1a1504bc49ac6072d520e2226954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e576715eae1f1a1504bc49ac6072d520e2226954"}]},"branch":"refs/heads/master"},"6ffccb608cc51af342f77470bad523609bec68de":{"kind":"REWORK","_number":48,"created":"2025-09-29 11:23:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/48"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 11:23:05.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ffccb608cc51af342f77470bad523609bec68de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ffccb608cc51af342f77470bad523609bec68de"}]},"branch":"refs/heads/master"},"c8dc2ecc0804fe0bc317ba289c7b076356d9296e":{"kind":"REWORK","_number":49,"created":"2025-09-29 11:25:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/49"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 11:25:26.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8dc2ecc0804fe0bc317ba289c7b076356d9296e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8dc2ecc0804fe0bc317ba289c7b076356d9296e"}]},"branch":"refs/heads/master"},"ad3917bc64ac2880e97702f99eb874025821f7c8":{"kind":"REWORK","_number":50,"created":"2025-09-29 11:26:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/50"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 11:26:18.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad3917bc64ac2880e97702f99eb874025821f7c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad3917bc64ac2880e97702f99eb874025821f7c8"}]},"branch":"refs/heads/master"},"ee1fd524a71a9931628b2ce296b75981871e880f":{"kind":"REWORK","_number":51,"created":"2025-09-29 11:54:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/51"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 11:54:16.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee1fd524a71a9931628b2ce296b75981871e880f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee1fd524a71a9931628b2ce296b75981871e880f"}]},"branch":"refs/heads/master"},"f664f2b4d416bd4105ddefd5935b1f04b1adb497":{"kind":"REWORK","_number":52,"created":"2025-09-29 12:03:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/52"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 12:03:55.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f664f2b4d416bd4105ddefd5935b1f04b1adb497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f664f2b4d416bd4105ddefd5935b1f04b1adb497"}]},"branch":"refs/heads/master"},"62a381d8123753743523bee9dd5ac3b2dc02043a":{"kind":"REWORK","_number":53,"created":"2025-09-29 15:22:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/53"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 15:22:17.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62a381d8123753743523bee9dd5ac3b2dc02043a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62a381d8123753743523bee9dd5ac3b2dc02043a"}]},"branch":"refs/heads/master"},"a33b9c1fa83997b2753808bd8920919f02ed623d":{"kind":"REWORK","_number":54,"created":"2025-09-29 15:47:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/54"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 15:47:02.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a33b9c1fa83997b2753808bd8920919f02ed623d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a33b9c1fa83997b2753808bd8920919f02ed623d"}]},"branch":"refs/heads/master"},"a96d23e85dc99b7a64d1d6ca3ef4edd39154eb1f":{"kind":"REWORK","_number":55,"created":"2025-09-29 15:51:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/55"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 15:51:29.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a96d23e85dc99b7a64d1d6ca3ef4edd39154eb1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a96d23e85dc99b7a64d1d6ca3ef4edd39154eb1f"}]},"branch":"refs/heads/master"},"1e63a804659ef241ff4b5e755565856790ec7712":{"kind":"REWORK","_number":56,"created":"2025-09-29 16:06:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/56"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-29 16:06:50.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e63a804659ef241ff4b5e755565856790ec7712"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e63a804659ef241ff4b5e755565856790ec7712"}]},"branch":"refs/heads/master"},"72864ad06e70e87f19b069c8b5a82f12e7cba6cf":{"kind":"REWORK","_number":57,"created":"2025-09-30 05:50:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/57"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 05:50:42.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72864ad06e70e87f19b069c8b5a82f12e7cba6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72864ad06e70e87f19b069c8b5a82f12e7cba6cf"}]},"branch":"refs/heads/master"},"a1ec28dc27d864efc3039be7b2587b0f0b0aff37":{"kind":"REWORK","_number":58,"created":"2025-09-30 05:56:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/58"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 05:56:53.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ec28dc27d864efc3039be7b2587b0f0b0aff37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ec28dc27d864efc3039be7b2587b0f0b0aff37"}]},"branch":"refs/heads/master"},"43cd47721b69fbdd67495b33f3faee883fe05d81":{"kind":"REWORK","_number":59,"created":"2025-09-30 09:03:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/59"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 09:03:55.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43cd47721b69fbdd67495b33f3faee883fe05d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43cd47721b69fbdd67495b33f3faee883fe05d81"}]},"branch":"refs/heads/master"},"6d017b2e02d159a26f9b23a17b4442185f39ef24":{"kind":"REWORK","_number":60,"created":"2025-09-30 09:22:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/60"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 09:22:19.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d017b2e02d159a26f9b23a17b4442185f39ef24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d017b2e02d159a26f9b23a17b4442185f39ef24"}]},"branch":"refs/heads/master"},"30a9432c7da69daea02e7f320af5b9f2a847dbe8":{"kind":"REWORK","_number":61,"created":"2025-09-30 09:28:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/61"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 09:28:35.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30a9432c7da69daea02e7f320af5b9f2a847dbe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30a9432c7da69daea02e7f320af5b9f2a847dbe8"}]},"branch":"refs/heads/master"},"be2e236a3b3070264163c6d21053ba013103a963":{"kind":"REWORK","_number":62,"created":"2025-09-30 09:36:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/62"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 09:36:19.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be2e236a3b3070264163c6d21053ba013103a963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be2e236a3b3070264163c6d21053ba013103a963"}]},"branch":"refs/heads/master"},"4796ecd4d460fc13bd68e841740722377b2e3190":{"kind":"REWORK","_number":63,"created":"2025-09-30 09:37:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/63"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 09:37:11.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4796ecd4d460fc13bd68e841740722377b2e3190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4796ecd4d460fc13bd68e841740722377b2e3190"}]},"branch":"refs/heads/master"},"9eef1952f5c3efb9493c9f577a3544f17087a73d":{"kind":"REWORK","_number":64,"created":"2025-09-30 10:23:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/64"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 10:23:54.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eef1952f5c3efb9493c9f577a3544f17087a73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eef1952f5c3efb9493c9f577a3544f17087a73d"}]},"branch":"refs/heads/master"},"e8162e947f8ab489aacba115eddc6a1cff7a3f19":{"kind":"REWORK","_number":65,"created":"2025-09-30 13:12:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/65"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 13:12:08.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8162e947f8ab489aacba115eddc6a1cff7a3f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8162e947f8ab489aacba115eddc6a1cff7a3f19"}]},"branch":"refs/heads/master"},"66b5926f89afd8e03e99bad6eb771f85d67ef4f3":{"kind":"REWORK","_number":66,"created":"2025-09-30 13:17:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/66"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 13:17:33.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66b5926f89afd8e03e99bad6eb771f85d67ef4f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66b5926f89afd8e03e99bad6eb771f85d67ef4f3"}]},"branch":"refs/heads/master"},"124822f13d62ef8f811418be02cd3b1986d55ac5":{"kind":"REWORK","_number":67,"created":"2025-09-30 13:33:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/67"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 13:33:46.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/124822f13d62ef8f811418be02cd3b1986d55ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/124822f13d62ef8f811418be02cd3b1986d55ac5"}]},"branch":"refs/heads/master"},"ed94e8244428f96309ab94bd400baed1b4704133":{"kind":"REWORK","_number":68,"created":"2025-09-30 16:56:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/68"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 16:55:58.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed94e8244428f96309ab94bd400baed1b4704133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed94e8244428f96309ab94bd400baed1b4704133"}]},"branch":"refs/heads/master"},"52748a1e973955a4de041fa0548db9787ed9f6d3":{"kind":"REWORK","_number":69,"created":"2025-09-30 17:19:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/69"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 17:19:18.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52748a1e973955a4de041fa0548db9787ed9f6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52748a1e973955a4de041fa0548db9787ed9f6d3"}]},"branch":"refs/heads/master"},"67b1c76360a89217348d5ae2b2b8a6051db664b4":{"kind":"REWORK","_number":70,"created":"2025-09-30 17:25:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/70"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 17:25:27.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67b1c76360a89217348d5ae2b2b8a6051db664b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67b1c76360a89217348d5ae2b2b8a6051db664b4"}]},"branch":"refs/heads/master"},"9997e5d51e4b2da59b29387ad4fd63382ee6d4e0":{"kind":"REWORK","_number":71,"created":"2025-09-30 17:27:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/71"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 17:27:50.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9997e5d51e4b2da59b29387ad4fd63382ee6d4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9997e5d51e4b2da59b29387ad4fd63382ee6d4e0"}]},"branch":"refs/heads/master"},"6575a424c97e053a35a9013d35eb0ca916420429":{"kind":"REWORK","_number":72,"created":"2025-09-30 17:36:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/72"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 17:35:26.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6575a424c97e053a35a9013d35eb0ca916420429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6575a424c97e053a35a9013d35eb0ca916420429"}]},"branch":"refs/heads/master"},"8fb63de6363d4d855fe51337dc403bc78a91d5f5":{"kind":"REWORK","_number":73,"created":"2025-09-30 17:58:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/73"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 17:58:49.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fb63de6363d4d855fe51337dc403bc78a91d5f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fb63de6363d4d855fe51337dc403bc78a91d5f5"}]},"branch":"refs/heads/master"},"d57c6dd42507f0bbfd3769fc9a69f94218515720":{"kind":"REWORK","_number":74,"created":"2025-09-30 18:02:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/74"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 18:02:48.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d57c6dd42507f0bbfd3769fc9a69f94218515720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d57c6dd42507f0bbfd3769fc9a69f94218515720"}]},"branch":"refs/heads/master"},"37b12bd381f457355ad658151f1b772e46907dd9":{"kind":"REWORK","_number":75,"created":"2025-09-30 18:09:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/75"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 14:52:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-30 18:09:09.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37b12bd381f457355ad658151f1b772e46907dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37b12bd381f457355ad658151f1b772e46907dd9"}]},"branch":"refs/heads/master"},"f06867561309a73b647477819b87e9eb8405eda4":{"kind":"REWORK","_number":76,"created":"2025-10-01 06:25:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/76"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:30.000000000","tz":-240},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f06867561309a73b647477819b87e9eb8405eda4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f06867561309a73b647477819b87e9eb8405eda4"}]},"branch":"refs/heads/master"},"37b0690ccb7b4d52a3901c04cc4dd880acbd1059":{"kind":"REWORK","_number":77,"created":"2025-10-01 06:53:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/77"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:53:42.000000000","tz":-240},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37b0690ccb7b4d52a3901c04cc4dd880acbd1059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37b0690ccb7b4d52a3901c04cc4dd880acbd1059"}]},"branch":"refs/heads/master"},"a83b87a3cb28389d032e9ccaf85d0e7e3b0ea5ea":{"kind":"REWORK","_number":78,"created":"2025-10-01 07:01:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/78"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 07:01:01.000000000","tz":-240},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83b87a3cb28389d032e9ccaf85d0e7e3b0ea5ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83b87a3cb28389d032e9ccaf85d0e7e3b0ea5ea"}]},"branch":"refs/heads/master"},"790a929413b9d9af3a8d3aa73eb0d4efe00708c3":{"kind":"REWORK","_number":79,"created":"2025-10-01 07:33:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/79"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 07:32:30.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/790a929413b9d9af3a8d3aa73eb0d4efe00708c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/790a929413b9d9af3a8d3aa73eb0d4efe00708c3"}]},"branch":"refs/heads/master"},"691e7c3a023dd71181be2cc3e1b437ca22e71e65":{"kind":"REWORK","_number":80,"created":"2025-10-01 08:01:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/80"}}},"commit":{"parents":[{"commit":"8f462c7f0549f47354213e7d32568b879ca872ee","subject":"Merge \"CI: Rename jobs according to changes in K-A\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f462c7f0549f47354213e7d32568b879ca872ee"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 08:01:24.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/691e7c3a023dd71181be2cc3e1b437ca22e71e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/691e7c3a023dd71181be2cc3e1b437ca22e71e65"}]},"branch":"refs/heads/master"},"717dccb20ea34aad412dc324f573f187b441b81e":{"kind":"TRIVIAL_REBASE","_number":81,"created":"2025-10-01 09:05:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/81"}}},"commit":{"parents":[{"commit":"78b10910efe1bd8dc6915e71e5e1b54ae8ae00e5","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78b10910efe1bd8dc6915e71e5e1b54ae8ae00e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 09:05:57.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/717dccb20ea34aad412dc324f573f187b441b81e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/717dccb20ea34aad412dc324f573f187b441b81e"}]},"branch":"refs/heads/master","description":"Rebase"},"17a15f5ccd93c81b652244836523ad4f140075cb":{"kind":"TRIVIAL_REBASE","_number":82,"created":"2025-10-01 09:13:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/82"}}},"commit":{"parents":[{"commit":"32fd0a3ff6e624560018c55b76a439ebccb6d400","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32fd0a3ff6e624560018c55b76a439ebccb6d400"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 09:13:32.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17a15f5ccd93c81b652244836523ad4f140075cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17a15f5ccd93c81b652244836523ad4f140075cb"}]},"branch":"refs/heads/master","description":"Rebase"},"c1ddd0cba857b14dcfac49cfe28506f6a5d7beed":{"kind":"TRIVIAL_REBASE","_number":83,"created":"2025-10-01 10:04:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/83"}}},"commit":{"parents":[{"commit":"9c528faa23581e62c85a293bf900c723f43b954c","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c528faa23581e62c85a293bf900c723f43b954c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 10:04:22.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1ddd0cba857b14dcfac49cfe28506f6a5d7beed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1ddd0cba857b14dcfac49cfe28506f6a5d7beed"}]},"branch":"refs/heads/master","description":"Rebase"},"ce2d2fa74fabc99cecd66741696853e1d58520ad":{"kind":"REWORK","_number":84,"created":"2025-10-01 11:40:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/84"}}},"commit":{"parents":[{"commit":"9c528faa23581e62c85a293bf900c723f43b954c","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c528faa23581e62c85a293bf900c723f43b954c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 11:40:47.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce2d2fa74fabc99cecd66741696853e1d58520ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce2d2fa74fabc99cecd66741696853e1d58520ad"}]},"branch":"refs/heads/master"},"0f61f1b9b6033f5cc88bae39afbcfb3f5b25b5df":{"kind":"REWORK","_number":85,"created":"2025-10-01 12:45:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/85"}}},"commit":{"parents":[{"commit":"9c528faa23581e62c85a293bf900c723f43b954c","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c528faa23581e62c85a293bf900c723f43b954c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 12:45:40.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f61f1b9b6033f5cc88bae39afbcfb3f5b25b5df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f61f1b9b6033f5cc88bae39afbcfb3f5b25b5df"}]},"branch":"refs/heads/master"},"32f8af9dd938135caee60716472169f98fd6f5cb":{"kind":"REWORK","_number":86,"created":"2025-10-01 12:46:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/86"}}},"commit":{"parents":[{"commit":"9c528faa23581e62c85a293bf900c723f43b954c","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c528faa23581e62c85a293bf900c723f43b954c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 12:46:02.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32f8af9dd938135caee60716472169f98fd6f5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32f8af9dd938135caee60716472169f98fd6f5cb"}]},"branch":"refs/heads/master"},"7c81ed3540b7212fa3240feb4ae01736bfebfb98":{"kind":"REWORK","_number":87,"created":"2025-10-01 13:02:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/87"}}},"commit":{"parents":[{"commit":"9c528faa23581e62c85a293bf900c723f43b954c","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c528faa23581e62c85a293bf900c723f43b954c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 13:02:30.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c81ed3540b7212fa3240feb4ae01736bfebfb98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c81ed3540b7212fa3240feb4ae01736bfebfb98"}]},"branch":"refs/heads/master"},"35239311ac6c1d12b2a26321dec3df8b0dd97a08":{"kind":"REWORK","_number":88,"created":"2025-10-01 13:30:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/88"}}},"commit":{"parents":[{"commit":"9c528faa23581e62c85a293bf900c723f43b954c","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c528faa23581e62c85a293bf900c723f43b954c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 13:30:29.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35239311ac6c1d12b2a26321dec3df8b0dd97a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35239311ac6c1d12b2a26321dec3df8b0dd97a08"}]},"branch":"refs/heads/master"},"735537d868d3cbd5fca44c3d6240645497be17d4":{"kind":"TRIVIAL_REBASE","_number":89,"created":"2025-10-01 13:37:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/89"}}},"commit":{"parents":[{"commit":"28aff00693c8f3f9007ddeeecda44c9367f5fd66","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28aff00693c8f3f9007ddeeecda44c9367f5fd66"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 13:37:39.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/735537d868d3cbd5fca44c3d6240645497be17d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/735537d868d3cbd5fca44c3d6240645497be17d4"}]},"branch":"refs/heads/master","description":"Rebase"},"67b5ce5ad4445869d07240883aa9febd152ece5c":{"kind":"REWORK","_number":90,"created":"2025-10-01 13:44:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/90"}}},"commit":{"parents":[{"commit":"28aff00693c8f3f9007ddeeecda44c9367f5fd66","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28aff00693c8f3f9007ddeeecda44c9367f5fd66"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 13:44:49.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67b5ce5ad4445869d07240883aa9febd152ece5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67b5ce5ad4445869d07240883aa9febd152ece5c"}]},"branch":"refs/heads/master"},"5e2d42511ca382faddff7ce88325af90c3f5dfc3":{"kind":"REWORK","_number":91,"created":"2025-10-01 13:54:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/91"}}},"commit":{"parents":[{"commit":"28aff00693c8f3f9007ddeeecda44c9367f5fd66","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28aff00693c8f3f9007ddeeecda44c9367f5fd66"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 13:54:57.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e2d42511ca382faddff7ce88325af90c3f5dfc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e2d42511ca382faddff7ce88325af90c3f5dfc3"}]},"branch":"refs/heads/master"},"abd2379c373396968b883f775372a3648fe3806c":{"kind":"NO_CHANGE","_number":92,"created":"2025-10-01 15:27:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/92"}}},"commit":{"parents":[{"commit":"eb22697f164b20a49ff1d66f279c61f6318121ba","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb22697f164b20a49ff1d66f279c61f6318121ba"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 15:27:46.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abd2379c373396968b883f775372a3648fe3806c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abd2379c373396968b883f775372a3648fe3806c"}]},"branch":"refs/heads/master","description":"Rebase"},"31efa4a417714337c23361292b3b8b4dabaa2280":{"kind":"REWORK","_number":93,"created":"2025-10-02 08:37:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/93"}}},"commit":{"parents":[{"commit":"17727e148120c43a5c7056b33e4277592376ccca","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17727e148120c43a5c7056b33e4277592376ccca"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 08:37:12.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31efa4a417714337c23361292b3b8b4dabaa2280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31efa4a417714337c23361292b3b8b4dabaa2280"}]},"branch":"refs/heads/master","description":"Rebase"},"a6fea594ff9601eaec926588c7c8402bf4463964":{"kind":"REWORK","_number":94,"created":"2025-10-02 08:38:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/94"}}},"commit":{"parents":[{"commit":"17727e148120c43a5c7056b33e4277592376ccca","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17727e148120c43a5c7056b33e4277592376ccca"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 08:38:30.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6fea594ff9601eaec926588c7c8402bf4463964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6fea594ff9601eaec926588c7c8402bf4463964"}]},"branch":"refs/heads/master"},"829ce5ebab8efeb0eb745d9a9c52a79145ac33d3":{"kind":"REWORK","_number":95,"created":"2025-10-02 08:41:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/95"}}},"commit":{"parents":[{"commit":"17727e148120c43a5c7056b33e4277592376ccca","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17727e148120c43a5c7056b33e4277592376ccca"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 08:41:31.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/829ce5ebab8efeb0eb745d9a9c52a79145ac33d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/829ce5ebab8efeb0eb745d9a9c52a79145ac33d3"}]},"branch":"refs/heads/master"},"707c725141e64f2e2477a022946fabc9c299bc26":{"kind":"REWORK","_number":96,"created":"2025-10-02 09:29:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/96"}}},"commit":{"parents":[{"commit":"17727e148120c43a5c7056b33e4277592376ccca","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17727e148120c43a5c7056b33e4277592376ccca"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 09:29:45.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/707c725141e64f2e2477a022946fabc9c299bc26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/707c725141e64f2e2477a022946fabc9c299bc26"}]},"branch":"refs/heads/master"},"d9c7fa1cd9cb858c5e4ca2701c6c38e1b82840d9":{"kind":"REWORK","_number":97,"created":"2025-10-02 11:15:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/97"}}},"commit":{"parents":[{"commit":"17727e148120c43a5c7056b33e4277592376ccca","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17727e148120c43a5c7056b33e4277592376ccca"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 11:15:20.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9c7fa1cd9cb858c5e4ca2701c6c38e1b82840d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9c7fa1cd9cb858c5e4ca2701c6c38e1b82840d9"}]},"branch":"refs/heads/master"},"be5e11dae55e7539fb84f13f9e514eaaf930cf3f":{"kind":"REWORK","_number":98,"created":"2025-10-02 11:39:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/98"}}},"commit":{"parents":[{"commit":"17727e148120c43a5c7056b33e4277592376ccca","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17727e148120c43a5c7056b33e4277592376ccca"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 11:38:42.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\ngpgcheck \u003d 1 is default in docker/base/dnf.conf\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be5e11dae55e7539fb84f13f9e514eaaf930cf3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be5e11dae55e7539fb84f13f9e514eaaf930cf3f"}]},"branch":"refs/heads/master"},"cb14c8af2cbfc418d97440ecd9ab8372d89a6fac":{"kind":"REWORK","_number":99,"created":"2025-10-02 11:56:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/99"}}},"commit":{"parents":[{"commit":"17727e148120c43a5c7056b33e4277592376ccca","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17727e148120c43a5c7056b33e4277592376ccca"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 11:56:27.000000000","tz":0},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\ngpgcheck \u003d 1 is default in docker/base/dnf.conf\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb14c8af2cbfc418d97440ecd9ab8372d89a6fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb14c8af2cbfc418d97440ecd9ab8372d89a6fac"}]},"branch":"refs/heads/master"},"806602e488240deb87bdf1c8ce00973ca8ec914d":{"kind":"REWORK","_number":100,"created":"2025-10-02 12:24:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/909879/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/909879/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/909879/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/909879/100"}}},"commit":{"parents":[{"commit":"17727e148120c43a5c7056b33e4277592376ccca","subject":"Move repo enablement to respective rpm blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17727e148120c43a5c7056b33e4277592376ccca"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-01 06:25:23.000000000","tz":-240},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-02 12:24:31.000000000","tz":120},"subject":"Add support for rpm to repos.yaml","message":"Add support for rpm to repos.yaml\n\nAdd the ability to configure repositories in the same\nway as we do Debuntu ones - via repos.yaml.\n\nRemoves .repo files from base - since these are not\nneeded there anymore.\n\ngpgcheck \u003d 1 is default in docker/base/dnf.conf\n\nChange-Id: I7df7d452c45c57c434d5b4ad15a87963e4d6d0d4\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/806602e488240deb87bdf1c8ce00973ca8ec914d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/806602e488240deb87bdf1c8ce00973ca8ec914d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
