)]}'
{"id":"openstack%2Fkolla~909915","triplet_id":"openstack%2Fkolla~master~I372fe0bce4fc502eacce214d24c5bddd198bc33f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-03-11 13:22:17.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-03-11 13:22:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I372fe0bce4fc502eacce214d24c5bddd198bc33f","subject":"Fix aarch64 builds","status":"MERGED","created":"2024-02-22 17:30:18.000000000","updated":"2024-03-11 13:24:38.000000000","submitted":"2024-03-11 13:22:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909915","meta_rev_id":"c85d4328f21ddb801cb00e19ca1af6ed01dcc029","_number":909915,"virtual_id_number":909915,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-11 13:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":2,"date":"2024-03-11 11:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-11 11:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":1,"date":"2024-03-11 11:42:47.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 17:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 14:38:45.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-03-11 11:34:10.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-03-11 11:42:47.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"8d1f95f21fd4b978963f9ee4f4f21354ff59df8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-22 17:30:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69abcd78aecce0e3bb36108e18820c6cc6787809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:52:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6ee67495e964d3eb348d21edc3ae99d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b7d89efefcb4fd59bdb1ed916de3bb2 : FAILURE in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66ca561c2018486ba614aefef24fec0b : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee965e79bbf749d28d3805d7e5c2da0b : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c26a174be47c44afb2a0b968bf1e7311 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93838c3e15c469eba35bf9d44d09625 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0da4e50468252b00b178aea1b69ab39e619a7e0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:25:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46d0df7596d34ca9bcc4f4397bdb9c83\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0f14d0915b33402893549b9db5e52343 : FAILURE in 1h 30m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c1c3b8708f1f4ebab58458b0a8dc22b3 : FAILURE in 1h 28m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6bbfc5052ba74fce8cdcd5e8fb03f3ba : SUCCESS in 1h 53m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70a9d06b5b804caeae7ce8dfa640d773 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5a153ac3ccb43978fe93c75577a5e80 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/53622296c5b047f2a1ce4df6d1b830cd : SUCCESS in 4m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0cb09d5404a84555b0015c331dd1e694 : FAILURE in 39m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7b7c249569cc4f32a4178ac8bf7b9ba5 : FAILURE in 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c41b54235e214e2fce141c40b5501d6585d3debe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-04 08:49:42.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8073ea5524446957bc5b67e2d86fbfb0f81e6751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 08:56:28.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1bd51c5bb3c4d3d8f3d1f7b245777d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10671d4410b0498297bfea57c54aa828 : FAILURE in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67b8ab1019934331bba4e1b093c34f75 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8aa44338bcd34e8480bbf08c31cbc1a6 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bcf0ec165f64616bef1a8801542edb2 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/829b290767c546f89a7ce260c13e8b10 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"14388c54bfd3da631c7e44a10be39ea59c8ee1f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-04 09:44:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3f28f506ccbd81b612981e50129f5b8444cdbae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 09:55:00.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcbc3ebedb1e4a13864ef113eb2890c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64d2f3cb29ab46f2b0f32f3d1b43d9bd : FAILURE in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78a8bdb9936d4914843e472ac0817bbe : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0da4784de4ca44b8b18cc9b96912f34c : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5704beeecfbf491c96a4f369d383ad26 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e79a93de81407092bd7f8710e71fe1 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3c14d5e72969de032f4758ac7341dd641dfb8368","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 12:11:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1981265d9b434e5c9a6548a45ce78d4d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cc01e09a1b424198a09545efe02e7673 : FAILURE in 18m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8bd253806be64b878503f0f5b098afb2 : FAILURE in 2h 24m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5c5ecbd2091149a38ac39e9a1f9f8a44 : SUCCESS in 1h 11m 13s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0445d2a49b1746559cae36ac62e63ec3 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45fe3595a7814e279abdf95c5d43c616 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/80684735807d4ea4b944fa34e752449e : SUCCESS in 4m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3c97dd7531384667aecde4f2f293685f : FAILURE in 20m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ada790478d648af88b471750f4d30f6 : FAILURE in 2h 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8e540311630faff0d945f39af5a134eba6f67e52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-04 12:42:27.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9d33014c0d7a16d5839bddfdabc12676ecec36f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 12:53:27.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acc0d3e18c904ffeae6c98d38f171536\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5aaa51199d14b98832fd6cc5b807790 : FAILURE in 2m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab24619305814d198d66f827b72429ce : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5899d5e49f2448f821ff2abb9afd0ee : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfc0ab0904d14bb2b8195ae8cc1faccc : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fda9833942464fa7f57c6381cd205a : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":4},{"id":"76fac9efe669858ce43be5abc2e3566730912387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-04 12:57:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c5092216aeaf6923ac25c483ba2cb8eb440890da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 13:08:49.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d180b71118ad45bdb2e5e10526ce7936\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f22afcfcc3b441fbdcba1d97d085bff : FAILURE in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdac32e823264560ad486c252baabde5 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/630890cf387b4775b75c5eff3e3208c1 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24f179465b67440ba6634bc540be79a3 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c664f1bece04757976199e594d9be7b : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":5},{"id":"e515fffb7d5824f26591372b908b1b8905165b3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 14:13:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaef2f2ef4ed45b1984dcb907d70e580\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b76f559762634d32a24892183ffb52cb : FAILURE in 32m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c0b8aa734946499bae8efc722944f9fb : FAILURE in 1h 11m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6199149d66c64799bbb6f3f48a12901e : SUCCESS in 1h 14m 14s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37fdd0a2ca2a46da86befb3ded776bdf : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a6ffcbdb46a4960a26ac6fe15f7e11a : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a872996b68045efa19e2ff1d3a4c097 : SUCCESS in 4m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/45eb5a20fa7545b5a01cebe02d963bb1 : FAILURE in 23m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5a664c872c2b42bd847008b086c33aaf : FAILURE in 1h 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"42667cf64bf46f2ad82b0e61750d47b82d532a56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-07 12:38:49.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0cc82c6a04686cb98645d8761a31364140109f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 12:49:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39c16bd09b084bb0bc06cb8e1bd06d69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e3c98353b314305b3cc268a124adf21 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e8ff330fdea4c70ace95f23a99147b3 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f861a49ed0f54a55b1bcf9ff69702152 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4641889c4e04192bf49428d97689f86 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbed57ff3d1f424a96ddbe180b7a709a : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":6},{"id":"c016f696d819403e222da6805ca1f4a484bac7bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:33:05.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a0a1bc3cf7c479dbb0b594e652f32ac\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6635114da5f74d2b86bc9acba2955bdb : SUCCESS in 1h 51m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4d8f02b85b4f46a58c8b4a24e44141a6 : SUCCESS in 23m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/907186a43e5d4bcba1687073c3a62e99 : FAILURE in 1h 51m 07s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2129856afc2437f80b4dd48c8c9f420 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b5d9568aa4ac445a96b27bad93d4f687 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/295fd7cbdfa248c09143001c866e08ba : SUCCESS in 3m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/00ab5f700db345799f59c3eb3432bd92 : SUCCESS in 1h 35m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e8cfc2b3003e48fc98092b7ca3c7ee8b : FAILURE in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"640dc738a1e2cc207a97ae553d116c07a0b0cd82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-07 14:37:23.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"0e80e6d4ef9a308a5769e66b76f376c265984043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-07 14:41:33.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"c43d49ec38f93b0e5536c40ac599aba6c919fccc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-07 14:46:56.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":8},{"id":"f1a62100ee5dc55321dd5df8734d43a21e1bbbbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:51:41.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb9fdbdf0a474e0ab68204a52ba0041a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/faab4ff1659046e19f69be848b27b9b3 : FAILURE in 4m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/40da5dcc5b3b426186fb8281321d7774 : FAILURE in 4m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/15406c2fe8a341c0b5d22d16d5813a9d : FAILURE in 6m 38s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb877d0b66bc4efeb1bcd67fe45dcebc : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d39ab1df5bbb45dd86d2ab51048eb3e7 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d6f305dbcc4e427cad98770d1dbaa342 : SUCCESS in 6m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/61046bc1899c4797ad2c915e1336cb2e : FAILURE in 4m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/97143983873f44738ae0a30b2e1ce42a : FAILURE in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e03570f1c656d01fe550a8573cbf3c991bb1412f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 15:15:55.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cafe6a9310e64f3db0b3412571d79dc3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5a3af8da496f4c28b98355f733d51574 : FAILURE in 5m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/001c385f29624ae4913cad8735861029 : FAILURE in 7m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a7c5fc8466894e6abae575491c86c364 : FAILURE in 6m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7a84e56f5a2f4dfc9bc721451b4452c6 : FAILURE in 3m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/41f73885ba434ed7b5b535780d3d69ed : FAILURE in 8m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/036eef3d42e6483bbd30ec0b49580727 : FAILURE in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadaa28ad84c4ed9ba0456057ebd684b : FAILURE in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f87051ee3a401bbfa0ca898b054616 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77940a0beb934adca8c05c050fbe3be7 : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66229f33a9bd419589e4b84187a1cd68 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcb985aeeb944648721d6da3d777acc : SUCCESS in 6m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ac39433dc1b14a80a952fb0521299d75 : FAILURE in 4m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6031b23bd27c44bb8c1695e5f0adb695 : FAILURE in 5m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9a41c9f394cf4f6c939d27f2142a787f : FAILURE in 4m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/71d857059e554db188174995d6205cd1 : FAILURE in 8m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0c5af63bf9b14db89d071c0382916ed2 : FAILURE in 3m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5fff61bcd84a44efb711957e94b3d46d : FAILURE in 3m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cb3db2634d6c43a19d6c614782e54730 : FAILURE in 6m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5aff7a849c6f40a29a21830e06a070ee : FAILURE in 8m 08s","accounts_in_message":[],"_revision_number":8},{"id":"8960b5911b2383cbb29d286f6deb6f23f4c41501","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-07 17:00:26.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"a6da8259e9787c472151149f1db2116a9c538e17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-07 17:21:19.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"0aa17a10543f4e693cc65286803dbd005bde0e2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 18:09:11.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4dfe7831aa44469a01f95175779975c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8049bc3903384768bc80b5bc3c580de5 : SUCCESS in 21m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dee3d51c14594d5683761c6f03db14d3 : SUCCESS in 22m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/97f8e45f57d241889c622723c75e3e0c : FAILURE in 46m 03s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbd506f47aed47718472a2896005fa9a : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95d8ee27fd1d48d88ea7bb5a39df5209 : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ebfec04fdc1d43f7b3d4f89040ffb6cd : SUCCESS in 2m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/415f1d08e88c4649b1bb8b0c5c66dbb4 : SUCCESS in 22m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a31a841aeebb48adb02bcb638a0433fa : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"70cd2dcff9e9d7f475dddd31441e4f631136db55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 20:24:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/095e531e9e534ea487a374bb0e9a205d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/63d1050504e740dfba712ec1214011da : SUCCESS in 24m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6a65b5a2f6ac4fcf95ddf2347ab3403f : FAILURE in 54m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7e64d53b391748949fe363c50e732b56 : SUCCESS in 16m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c54888fb9a604d81a07d03718101ea1a : SUCCESS in 29m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9c96334d7d5f44e28078c5fa80eb2cc5 : FAILURE in 43m 31s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/779a81e52a9146cfa31112563188a536 : FAILURE in 1h 39m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa286967c9a042f08910aa0f38eecb64 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/213400b712964ae0a06931b334b34473 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62c553734096481e81aa7d9a73057e8e : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8ac0b14caa046008000dd4c10444ea7 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bd57bf4c0c427381adca03f4f10401 : SUCCESS in 4m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc7622a5e93c4b84b6f007fe86b49884 : SUCCESS in 23m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fbd3f20e6091487f86789f4899929e49 : SUCCESS in 33m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c0c31f3d1d9e440a877d1ed9c99d0c17 : FAILURE in 48m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5eb224ebb0914230a77ed730778b5929 : FAILURE in 1h 19m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e67d4c12eca44090b6d315dc46e365fc : TIMED_OUT in 3h 01m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/53a9c41de35c43dda9d470239591dadf : SUCCESS in 51m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/eaf5e95eea844c5b9e4dcd009021dc75 : FAILURE in 30m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ed733ca76473424bbd1e3a83285d06c4 : FAILURE in 1h 21m 19s","accounts_in_message":[],"_revision_number":10},{"id":"dcf52e9cbe478e8332064edb4517c30504c76ea0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-08 06:00:30.000000000","message":"Patch Set 11: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"8da775855b49c5de537287ff266877b15e9f4701","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-08 06:00:33.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"00ec1f1ecdfb938fdaae9cbf8188f6d6a855f975","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 07:07:47.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/912030 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"18dfb4ed3cb372e5d0ff112e899e52a6218d6abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 07:07:48.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/912030 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"7490739399d2771f429cc0e31deae7bf9d589b3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-11 09:33:44.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"cc58c3f3d2afed89e81e4bb6501939e800bb16fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-11 09:34:14.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"19ab36be92febd3604f987c0f842617e1785f3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 11:20:55.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88419812f63940c8ade7de4c5b968ac2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4e1ba70d11884c02994c36dcdae8724d : SUCCESS in 22m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3b1fd894375442dda1f2be03a9dd4948 : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f7759479819842c5bc167b3ab4f6fc0d : SUCCESS in 22m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/494e6b03d30a42b1832446f46266392d : SUCCESS in 28m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2668c5a4b0f94b01a626a7a6037bacb9 : SUCCESS in 1h 06m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e5aef1c61ac84b62aa0ea6f8709ef3f3 : SUCCESS in 1h 26m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82058e96eac4a5a8c028bd34c83a283 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/042839f9ab594803a12f94ca24b44e63 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16d2607a1bc34908933bf100c94b5128 : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/342785558fe246d688e847701049a22e : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d120158115d94a3995ed89007ee16686 : SUCCESS in 4m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/89fd6acda6e947169af1058bcf3e9880 : SUCCESS in 24m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4f76c032b2914338b74b9cb0ee772188 : SUCCESS in 29m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3f0c6102267c4b07be8a29d066737e18 : SUCCESS in 1h 06m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/28cc4baecdca42ea95c272c3d9cc5b18 : SUCCESS in 1h 29m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a88b6c47f6e54400958cf63283918f28 : SUCCESS in 17m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4cefe5bd5b684c269896532249ced5c2 : SUCCESS in 59m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/65b7c77d27e940ad8cacb18e0a5ee132 : SUCCESS in 54m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4c2b605ab51840668f76dd53fe61df7e : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":13},{"id":"a117a0b8d79a5a32a2c2e986e22a54e9492365cb","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-11 11:34:10.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"7a47758f817cf5a198d6aedc682cc521df59097e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-11 11:42:47.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"7969b307e06e11ec46234f1fd918657a133d0cf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 11:43:08.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"d2b190bcc2ed951b7468252ea7fedd240c9f0ae0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 12:15:27.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae8478b622964d10b2607997c06150c7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/212d726cdaac463786f23afbfeeb9bd5 : SUCCESS in 2h 15m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/20a909f6005240df8d9f7b27d8fdb257 : SUCCESS in 1h 55m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d9df39615d60435a8fb98d903aa12646 : SUCCESS in 2h 34m 23s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c8e18b0065e4815bc3797b0ddf3be8d : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03ceb5404ec94e0a9eab34593cf9e509 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/92415b3730244d7a9652709b5b97a99e : SUCCESS in 4m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5ffdabc56e634bfa8c00218fbee892c4 : SUCCESS in 36m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/45fe2aa4a6a54247b5247b16ffa66071 : FAILURE in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3c535013444b6084e76b14ea1dc22153f92894d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 13:22:16.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d0862a94844f87b9fe00ffb5a78245\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/01784121d2a646efa1c43c846b02f5e0 : SUCCESS in 18m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a919404347b04efb83b60f11ac96e371 : SUCCESS in 27m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/540d1af0af1e48ea81c4280e617eb95f : SUCCESS in 1h 03m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5e0724af63b14578a01c6db01ec8774c : SUCCESS in 1h 21m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ac51c5e86f3450bb169f6250544ac74 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc3a8bc3d853486d8654759e4f8c7601 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cebe4c5cfcbf464ea2e290f55edba6f9 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3952b3411fc84838bdacea8d28003bba : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7ce676b361a04a1693f2b053d8eb4933 : SUCCESS in 23m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ac9bdc796fb74b3ba5aee1310a8b88fc : SUCCESS in 30m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/865c3213f5954f218e624b87ea898363 : SUCCESS in 1h 06m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6bf37125d73a469fbbc4315b088ddf5b : SUCCESS in 1h 36m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9f685daf89194d3483073447ecebf78b : SUCCESS in 18m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7d44e03fe290466b9151be02d1beeb90 : SUCCESS in 1h 00m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/80051a60292d48a3819ee8f968a40b2e : SUCCESS in 48m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7f0205f45bb5443abc8868b58cb2b1bd : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":13},{"id":"36dd941306277c0aab27a18265c1022280278d04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 13:22:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"c85d4328f21ddb801cb00e19ca1af6ed01dcc029","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 13:24:38.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/691432d99cd74886927e3dd712df6764\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/823651c6b0694c25a9b64471538ccaac : SUCCESS in 43s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"3795887e5f3dc1a7d3532dca274b8db64cf47d2d","revisions":{"95f1e5cf8623744ed5a8aac68e03d5d5a53587eb":{"kind":"REWORK","_number":1,"created":"2024-02-22 17:30:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/1"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:16.000000000","tz":60},"subject":"WIP: Fix arm builds","message":"WIP: Fix arm builds\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95f1e5cf8623744ed5a8aac68e03d5d5a53587eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95f1e5cf8623744ed5a8aac68e03d5d5a53587eb"}]},"branch":"refs/heads/master"},"b279888728f0a7e5b9f7e157da883f7d1780409c":{"kind":"REWORK","_number":2,"created":"2024-03-04 08:49:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/2"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-04 08:49:42.000000000","tz":0},"subject":"WIP: Fix arm builds","message":"WIP: Fix arm builds\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b279888728f0a7e5b9f7e157da883f7d1780409c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b279888728f0a7e5b9f7e157da883f7d1780409c"}]},"branch":"refs/heads/master"},"0b316e31d6b3187dce2e4dea02ce77ef3b1b0f9f":{"kind":"REWORK","_number":3,"created":"2024-03-04 09:44:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/3"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-04 09:44:33.000000000","tz":0},"subject":"WIP: Fix arm builds","message":"WIP: Fix arm builds\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b316e31d6b3187dce2e4dea02ce77ef3b1b0f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b316e31d6b3187dce2e4dea02ce77ef3b1b0f9f"}]},"branch":"refs/heads/master"},"eb759ce7700a8dd56eeca0039cf04b7cddaa87cc":{"kind":"REWORK","_number":4,"created":"2024-03-04 12:42:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/4"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-04 12:42:27.000000000","tz":0},"subject":"WIP: Fix arm builds","message":"WIP: Fix arm builds\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb759ce7700a8dd56eeca0039cf04b7cddaa87cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb759ce7700a8dd56eeca0039cf04b7cddaa87cc"}]},"branch":"refs/heads/master"},"2116b676c615e3bcb5cf6d220c18f5cd9aeae42f":{"kind":"REWORK","_number":5,"created":"2024-03-04 12:57:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/5"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-04 12:57:25.000000000","tz":0},"subject":"WIP: Fix arm builds","message":"WIP: Fix arm builds\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2116b676c615e3bcb5cf6d220c18f5cd9aeae42f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2116b676c615e3bcb5cf6d220c18f5cd9aeae42f"}]},"branch":"refs/heads/master"},"f830fcea5732e3f76e77d57634336fa3e59955d2":{"kind":"REWORK","_number":6,"created":"2024-03-07 12:38:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/6"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-07 12:38:49.000000000","tz":0},"subject":"WIP: Fix arm builds","message":"WIP: Fix arm builds\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f830fcea5732e3f76e77d57634336fa3e59955d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f830fcea5732e3f76e77d57634336fa3e59955d2"}]},"branch":"refs/heads/master"},"c960dc481e8ef2ab05fe79c34bcc6cd3b53ddaae":{"kind":"REWORK","_number":7,"created":"2024-03-07 14:37:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/7"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-07 14:37:23.000000000","tz":0},"subject":"Fix aarch64 builds","message":"Fix aarch64 builds\n\nThose are failing now due to fixed setting of debian_arch variable\nthat was omitted in the general patch.\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c960dc481e8ef2ab05fe79c34bcc6cd3b53ddaae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c960dc481e8ef2ab05fe79c34bcc6cd3b53ddaae"}]},"branch":"refs/heads/master"},"522daeaa776e5897cc467765c399d1e43f5b7295":{"kind":"REWORK","_number":8,"created":"2024-03-07 14:41:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/8"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-07 14:41:33.000000000","tz":0},"subject":"Fix aarch64 builds","message":"Fix aarch64 builds\n\nThose are failing now due to fixed setting of debian_arch variable\nthat was omitted in the general patch.\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/522daeaa776e5897cc467765c399d1e43f5b7295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/522daeaa776e5897cc467765c399d1e43f5b7295"}]},"branch":"refs/heads/master"},"cba9f9ab309780af29c3d7a3c40f348edcf6b4fa":{"kind":"REWORK","_number":9,"created":"2024-03-07 17:00:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/9"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-07 17:00:26.000000000","tz":0},"subject":"Fix aarch64 builds","message":"Fix aarch64 builds\n\nThose are failing now due to fixed setting of debian_arch variable\nthat was omitted in the general patch.\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cba9f9ab309780af29c3d7a3c40f348edcf6b4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cba9f9ab309780af29c3d7a3c40f348edcf6b4fa"}]},"branch":"refs/heads/master"},"19d965b8d8eca4d674e847b293fea73657399aa6":{"kind":"REWORK","_number":10,"created":"2024-03-07 17:21:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/10"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-07 17:21:19.000000000","tz":0},"subject":"Fix aarch64 builds","message":"Fix aarch64 builds\n\nThose are failing now due to fixed setting of debian_arch variable\nthat was omitted in the general patch.\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19d965b8d8eca4d674e847b293fea73657399aa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19d965b8d8eca4d674e847b293fea73657399aa6"}]},"branch":"refs/heads/master"},"78dfb65d6338c513a3e2ece13e06e70f64d0bbb9":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2024-03-08 06:00:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/11"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-08 06:00:30.000000000","tz":0},"subject":"Fix aarch64 builds","message":"Fix aarch64 builds\n\nThose are failing now due to fixed setting of debian_arch variable\nthat was omitted in the general patch.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/912030\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78dfb65d6338c513a3e2ece13e06e70f64d0bbb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78dfb65d6338c513a3e2ece13e06e70f64d0bbb9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a7fd5972652a3b83599d213787eb3ba84bd9f599":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2024-03-11 09:33:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/12"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-11 09:33:44.000000000","tz":0},"subject":"Fix aarch64 builds","message":"Fix aarch64 builds\n\nThose are failing now due to fixed setting of debian_arch variable\nthat was omitted in the general patch.\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7fd5972652a3b83599d213787eb3ba84bd9f599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7fd5972652a3b83599d213787eb3ba84bd9f599"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3795887e5f3dc1a7d3532dca274b8db64cf47d2d":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2024-03-11 09:34:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/909915/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/909915/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/909915/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/909915/13"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-22 17:30:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-11 09:34:14.000000000","tz":0},"subject":"Fix aarch64 builds","message":"Fix aarch64 builds\n\nThose are failing now due to fixed setting of debian_arch variable\nthat was omitted in the general patch.\n\nThis patch also adds a validation on debian_arch cfg option\n\nChange-Id: I372fe0bce4fc502eacce214d24c5bddd198bc33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3795887e5f3dc1a7d3532dca274b8db64cf47d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3795887e5f3dc1a7d3532dca274b8db64cf47d2d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
