)]}'
{"id":"openstack%2Fkolla~910006","triplet_id":"openstack%2Fkolla~master~Id70ea504acc37b3c42e2e7b1400f566b0dda9421","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-02-24 06:31:39.000000000","reason":"Change was abandoned"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-02-24 06:31:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id70ea504acc37b3c42e2e7b1400f566b0dda9421","subject":"DNM: test rmq","status":"ABANDONED","created":"2024-02-23 16:23:06.000000000","updated":"2024-02-24 06:31:39.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"1896cca430727f56cc1455bfd2faffde49827338","_number":910006,"virtual_id_number":910006,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-23 20:55:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 17:05:26.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-23 17:52:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-23 18:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c3758f826aaf638b64b39ab7213a49e74b726a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-23 16:23:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42e4aca00219083a88e59978d7f7006666b6a2bd","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-23 17:05:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4dc82f245a053601af56f2d70352178c9670b5d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 17:52:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd77e50d06014ef787398ad61b199944\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/98a8b97682064d99955383b69fb82b96 : FAILURE in 25m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a79060aa34ac4912811027095bd623c2 : FAILURE in 23m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ce03f794ea724f74a68fee52fc2acdbf : FAILURE in 40m 40s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eecdf93ca7fe4548a8959def77dd11ae : SUCCESS in 2m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c3a79e2c01b4c33a8e9c51fe1d2798b : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6a77f72770742c892f66a2684122f2b : SUCCESS in 2m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/717ea379ef164c4789e16165f0c4fe72 : FAILURE in 1h 27m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7829c7cf673b4719b8d69326b937dde3 : FAILURE in 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d6b6900fed9d4ad8dc1222e0c1f829fa464563c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:11:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2f3ace5a4f24c5982f3d473bf8bfc59\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/352a5ec678db43d9a1211fe3970f5a1b : SUCCESS in 34m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/aed65eb7ec7b44c79d7b7d60fafdc4f0 : SUCCESS in 1h 12m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9705e8ca1491416da98e125a6a6643dc : FAILURE in 23m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/06619ec0cbfa41e3baeeacc13a7d25fe : FAILURE in 29m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8be8a3ea4c7e4b47b539f23199d8f18e : FAILURE in 22m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/87625baabaf84098bbeef0d80d109e21 : FAILURE in 23m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18b8e52d516b40cdba6a8beece3a2341 : FAILURE in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e50f39087584dedbf8241dc68dbc770 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98179dcd6bf940aab516dd11b3eaf011 : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52aeb0d8751d479a8e74dff27550ab76 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c403f790414f82b53f0ace6c59db26 : SUCCESS in 4m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/73240eeb39ce4199b82b5e3045ef7722 : SUCCESS in 23m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/66d159dd6d344d799065c91b7d939150 : SUCCESS in 33m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5c078696b95b415582b37c3047c823c3 : SUCCESS in 1h 15m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/941bc2e9bbd24c8faa5160804a171018 : SUCCESS in 1h 31m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7ef6c350a146447b92b161378785c4f6 : FAILURE in 20m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fb859e47832345bdb5b175e732a2cdc4 : FAILURE in 53m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f2b08741bad449abaca98c35f2cc47e5 : FAILURE in 15m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8530e3a2880042f6ad99efb093edd6ec : FAILURE in 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ca0129d9339a17367e12887591673f835f88444b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-23 18:19:53.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2eeaafe52104687bf4c4efd9655f59fad8f667f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-23 19:09:28.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0345f424175153c54af716d97a0eba29fc96914f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 20:55:49.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f5584e44bfe4617826ac1d272927677\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d001e34e5f5943f0a25cf2958811fbcf : SUCCESS in 28m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9a16ffd9ba194beaab68736f4203a028 : SUCCESS in 53m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e407845710c8400ca7bd2c4d6b397d1f : SUCCESS in 14m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/445c10ba9ae14ed28b0c02e6641432f8 : SUCCESS in 32m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cd43b9ac65714471a2c9f8fdaba82db7 : SUCCESS in 1h 05m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/456e675bd7f24e63b048f706e12f59c9 : SUCCESS in 1h 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a52003908af4973a47710b89e67acb5 : FAILURE in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57cf6c8c67c24edbade86e07033ef67c : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b52e9af8d33147b3b90401d8502b56eb : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5934b26c8b7c4d19ba2d3d3a47bb4c3f : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b1ceed090d34d6b8f39c805f41b16d5 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c5b7a42e5da349f79ea62be9db50ff37 : SUCCESS in 20m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5b15d729815740438812bea34e9c78fb : SUCCESS in 30m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/02c0eeeb6db343b4b52eb86140c212b0 : SUCCESS in 1h 09m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a2061cfd609d437890c1b07186953f58 : SUCCESS in 1h 36m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e976ecd32fa34aaba28760dd3873969c : SUCCESS in 20m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/db51cb85081142f483d0b7aaaa5c6fa9 : SUCCESS in 1h 04m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8a180807021d4406927e61e0ebef83bc : SUCCESS in 54m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f8a34d0bf84d4d5da46eb1e8636012d3 : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":2},{"id":"20705e6ae6e7ced27dec1f1dcf13d2702bf309a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 21:21:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/245e9340f14249149fe58d33e297b855\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/523df87bc9c44f0b9c56046e3cb3de91 : FAILURE in 1h 37m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2b302557294d47298b45f57e84b7b1f6 : FAILURE in 16m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c306f52615e847d4bf77367ea43760ab : SUCCESS in 2h 08m 32s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/339ea1e91eaa43d8a1a96f1a338c6449 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f16fa58dea7a422f986231f3ffdd51dc : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/88f4fea06a1e421d9df4f2443df6e92a : SUCCESS in 2m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/82663ab2e995476fb706497ae40dedb5 : FAILURE in 1h 25m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/00a1e87c54c243e38bcdf8dd00248a1d : FAILURE in 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1896cca430727f56cc1455bfd2faffde49827338","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-24 06:31:39.000000000","message":"Abandoned\n\nfixed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aab0ad152dd6dd81f104a62fdc96fddde0152f65","revisions":{"68205f584887792f8b8624e2c5896843d8c82b6e":{"kind":"REWORK","_number":1,"created":"2024-02-23 16:23:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/910006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/910006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/910006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/910006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/910006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/910006/1"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-23 16:22:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-23 16:23:04.000000000","tz":60},"subject":"rabbitmq: Mark unbuildable on Debian/Ubuntu","message":"rabbitmq: Mark unbuildable on Debian/Ubuntu\n\nThere is some issue with the ppa1/ppa2.novemberain mirrors [1],\ndisable building for now to unblock gate.\n\n[1]: https://github.com/rabbitmq/rabbitmq-server/discussions/10599\n\nChange-Id: Id70ea504acc37b3c42e2e7b1400f566b0dda9421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68205f584887792f8b8624e2c5896843d8c82b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68205f584887792f8b8624e2c5896843d8c82b6e"}]},"branch":"refs/heads/master"},"aab0ad152dd6dd81f104a62fdc96fddde0152f65":{"kind":"REWORK","_number":2,"created":"2024-02-23 19:09:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/910006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/910006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/910006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/910006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/910006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/910006/2"}}},"commit":{"parents":[{"commit":"0246b32d70da39f6e837974f339e845f0c8b9e8f","subject":"Merge \"Move curl sources to sources.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0246b32d70da39f6e837974f339e845f0c8b9e8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-23 16:22:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-23 19:09:27.000000000","tz":0},"subject":"DNM: test rmq","message":"DNM: test rmq\n\nChange-Id: Id70ea504acc37b3c42e2e7b1400f566b0dda9421\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aab0ad152dd6dd81f104a62fdc96fddde0152f65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aab0ad152dd6dd81f104a62fdc96fddde0152f65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
