)]}'
{"id":"openstack%2Fkolla~910148","triplet_id":"openstack%2Fkolla~stable%2F2023.2~If66f95e52b1897d5b2d319fb1a43bde5ed9db751","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-02-29 18:14:30.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-02-29 13:47:50.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-02-29 15:51:13.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"If66f95e52b1897d5b2d319fb1a43bde5ed9db751","subject":"toolbox: Bump ansible-core to 2.15","status":"MERGED","created":"2024-02-26 08:27:59.000000000","updated":"2024-02-29 18:15:31.000000000","submitted":"2024-02-29 18:14:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910148","meta_rev_id":"bc274040f4c82ca2128f980c2ff2cde7f77d10da","_number":910148,"virtual_id_number":910148,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-29 18:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-02-29 13:59:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-02-29 13:47:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-02-29 15:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2024-02-29 16:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2024-02-29 16:43:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-02-29 12:17:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-02-29 12:16:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 09:36:27.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-26 10:07:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 10:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 11:18:23.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-02-29 12:44:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-02-29 13:59:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-29 16:43:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f9511b0c888102a203966b16ba0474f99ff2a80a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 08:27:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bb5a4d2da15bb18aaf3cad31df28f5a377fd4d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 10:07:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43237389e58b40d5ad64f9522f6901b1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a28ba2180d3f40a984dc9d0789a83916 : SUCCESS in 27m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a0e585fe5ed847db9ad0d412695f7df8 : SUCCESS in 40m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4d0adb4321304478a93f7e5a88f42ee6 : SUCCESS in 1h 37m 58s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3fa885b1c2b4e62a9e287492bdc4a4a : SUCCESS in 2m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70d59ea6e6b243ef9fcc8eec5c2752b4 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a9792a8134f4dfd835d2f92d4d48788 : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a96926e0a2ab4ea1b758fcb121f2ce54 : SUCCESS in 5m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/82023f2850114a85b32a21c60a428cb7 : SUCCESS in 27m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/367773bc2f37412abac92cc4d2ecc66a : SUCCESS in 1h 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46146c668e2ad2e5c7a83844e82f63760269fff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 10:20:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4987a76da2a4e5a956079fdb1553911\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/37f6ef833a6c43cdbf258a47f7ab3c59 : SUCCESS in 22m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/015a4bf3318a49d48defd602f3f770b9 : SUCCESS in 1h 09m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/726862ef40344622a5bb8b8c621f8bc2 : SUCCESS in 21m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8c3a16026ba24c5081bf15f2dc974640 : SUCCESS in 25m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/671448247bfd45c09780bc347d18b63c : SUCCESS in 1h 05m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/70aef52ec0a54557a786fc5dc3ab0ff7 : SUCCESS in 1h 42m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251931290d254396b229ca7a5637d6bc : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e51d50854ea846ff8681cda5f0d234f9 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/676f5e9f94f24fa689d395f4de61ae9f : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93eedeeddc4042e7a3283fef93b84474 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4cfd7496d6b443092fc4b646d4dc49b : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdb3510c0ba3481f81382af7ffa3e80e : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e31cc63959d4f7da436a290e2c9101e : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e6dd2872af3b46d5ac75f8fa068eb08b : SUCCESS in 22m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/755f27633dd34b2994d35b8ca73b1c10 : SUCCESS in 30m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8675dcf91647460fba9435f552a52e59 : SUCCESS in 1h 06m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/118ebcfbce1446b5a85855024a66c6df : SUCCESS in 1h 41m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f435b98c8d1846b98acd9ebab0cd9ff3 : SUCCESS in 18m 01s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6f314a43df45499bac6d50c12f328667 : SUCCESS in 1h 05m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f4b67cc7bc2e45c19727913475d803ab : SUCCESS in 58m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/421e0645f8b0413caaa30689d3c67546 : FAILURE in 32m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6bc7b3bbc2824fbc1840154ec860332a3a518aef","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-02-26 11:18:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a80dc250fa010c6c9aeeb0a216bef6d8f2c6fc78","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 12:22:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69ef177963b4042b435977b8d2ced26c209df10b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 13:00:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeae26fa1ebf49aab2dad7ae4d7cda4a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/998908110f784f5e86f46ce95c995177 : SUCCESS in 1h 13m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8b7d74af566c445b8783a5e6f7329a88 : SUCCESS in 1h 05m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/005271a501954e30ac2934042edfffc4 : SUCCESS in 1h 40m 11s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74d26396ed934ba995478627b8601b85 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5105cae90b24e2595781c478c1b8454 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b261886089e24e7198232cc14b2faa22 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce944fcab4694f42b64740daa11a4997 : SUCCESS in 3m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b200a4dbbddb4f4c913ab49f3056b0ec : SUCCESS in 20m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8e74ba293ff14fe6a344e9720393e3ac : SUCCESS in 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"648272d5dfc09a9f8e27e01da10fe5335d26be35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 13:01:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85c7f2d7a2dd4557a829f0412bc9b738\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/316584b8d5984d038e063d092564cb3c : SUCCESS in 22m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4287d7f5ad474ce59ad97f2ffe51f659 : SUCCESS in 1h 05m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8aaf4e53c698424896e4b6b03d37740c : SUCCESS in 17m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/baadb7cfe7934fc58875b400620c2b6c : SUCCESS in 26m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bf8089e04bc543448c01484b9b0d14a9 : SUCCESS in 1h 02m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8f518debefc04a66be299f2b81233d02 : SUCCESS in 1h 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363535e189464526b0249a3c2e4bbe34 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a37e61939c804bd4b4787dfa84457944 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06039f03ccba4adc8576bf9d513de374 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a33159fe944448e4ae59384661b05c88 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7d542bbed4d4c019db515a8a574d432 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4eec7de170b646ffb023863f9b54f9b3 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d709db92824d6995a54f695e40ec66 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7fca4c74647d412493f212ccefeddd11 : SUCCESS in 19m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e56ca97c3210467db3dfcbe9d05424de : SUCCESS in 29m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/02210536767f4a3d94b9648d3fd17293 : SUCCESS in 1h 07m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1428038d1e1d454b96fbc74ddad1f5db : SUCCESS in 1h 37m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/75747b31ecb04ac2859ef21ecb856d05 : SUCCESS in 18m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fce60b6c4fc64a6b926acfe57dd56312 : SUCCESS in 54m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c2b7249334b84185aeb7864629943197 : SUCCESS in 56m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9a524c3b4efe4472a6bde49187a1dbf6 : FAILURE in 29m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8e36dd18e0a3619344ca5f040afa0c853255aabd","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-26 17:03:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"622601449809277e4e4767beb1c75a4585653b12","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-28 12:29:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a335124ae5e23fb00612b3ecdf5de6b0732d6a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:01:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29d79c75fd5468096fc97023f4c9cdb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0a540035ea384a85a3c2cadf2c05ebd7 : SUCCESS in 1h 26m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/311ca4f399c8425ead77506d45904f0b : SUCCESS in 1h 28m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/97d7168da54b4a7d9f882ae415f71804 : FAILURE in 1h 17m 32s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5994e75103444d4fb5f7b310bce782da : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6cc2d4f328b7457d890aaf01069d489c : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac0205a944ac4fe3aac7d3382252d63d : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e71747130a89428f9dbe3d5b8dbda4ad : SUCCESS in 5m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e5b3508b5c96473d839d4cb687cc5a91 : SUCCESS in 1h 19m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/907a81eb5c0e466b937df8c76db9c661 : SUCCESS in 1h 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a52e6b14be282372336f4866d7fb7a814d1e729d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:24:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e809b9a743d4bb8ba5a6c556458bbe2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/219df3365ee748a5860933320ef86515 : SUCCESS in 20m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/54986cfdd93c44598dc1cd3e1181fd0e : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cc2dfae7e9e1465f87c56f52e80eba8e : SUCCESS in 16m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/59dabffebf2c4854bcf5f8e3872bb2a8 : SUCCESS in 29m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/95eaa1e6253646e99873a22df05757fd : SUCCESS in 59m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1604d629a6fb4afe85feac069c86cd8c : SUCCESS in 1h 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7932613bf27e435ab83776be0a1ff7a9 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb97d5ed6351423e997729f9495ff05e : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0838881c2851499c97be205fdf4ec155 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b25a627e5484b3db10141a9716e64d1 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5e2e4f913ef4c52a62147e4b4390de0 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8374d97e20c5440abe7762fdf6f13d1a : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f63adb697a4464bb6b3b6058fa844c : SUCCESS in 5m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ca0aa40b43dc44d5bf1bc0735cac127a : SUCCESS in 20m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5af6bc6d460f4cabb44aec4f30182b91 : SUCCESS in 25m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7ff45703cbd445f7ababbb08c644cdd4 : SUCCESS in 1h 13m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f3a38ca27dd54b42a934a4808fc26b1b : SUCCESS in 1h 30m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cbc306c337c0425b993587a78408127d : SUCCESS in 18m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7203866dbb614688b5791344f243208d : SUCCESS in 1h 06m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dd2dec91f9ee4a4ba6552d35ce1fbed7 : SUCCESS in 51m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/06544cf730dc44679f0b43cf2e5d6180 : SUCCESS in 1h 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8fddf47e92f77a73f2f2f3d723c7511c497baa65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-29 12:15:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"996689252cff6c763d71f111f1f909d8f2a2b01d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-29 12:16:08.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e46d09f413948543a291d6c964a38c0cfac5021","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-29 12:17:10.000000000","message":"Patch Set 2: Backport-Candidate-2","accounts_in_message":[],"_revision_number":2},{"id":"edd4c446fdec43e11b427d67cc217fa2c219f352","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-29 13:47:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"39bda8c3bc0a651755ce5136cf1d0c0bc481ea2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:51:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a51cf7d0326452a92e4841c271289fa\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3b0aa4b89c2f4bcb89c638fb1418855e : SUCCESS in 19m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/32a65b4af227481997b82130cad8d9ab : SUCCESS in 16m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cae37f0279ed478e97eebc70fbac4b9f : SUCCESS in 1h 04m 56s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9858ac7985f54b11a59423216af0337e : SUCCESS in 2m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fa4d64386ff4b13a08ec61e80465c69 : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10b589e3ec94417689c48dd343288b27 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/58482ca859364043963a1ea6d9a1d332 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/15ea96da22304544ba713a49fa67085e : SUCCESS in 1h 28m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/281d719db24b4f0cadd3d23370b9bbf5 : SUCCESS in 1h 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"20261af716539c33af24e6db28f95971fcd1f2a8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-29 13:59:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"42fa051d7cf1f6055d7b3df2a5290abe14a97611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 14:02:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/469187af52e74b86b1bbf2c97cea73e7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6c662471b1054aa69471c54faf1c8580 : SUCCESS in 24m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/93afecf81b3a40d89808f8c657567bbf : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0f7897231b3c4d9a95b706de10dd4ed1 : SUCCESS in 15m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d2c483b408164f0588ea3cbae0864800 : SUCCESS in 29m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f82fe4ff9afd4483801f71c5028e9c4b : SUCCESS in 47m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b5e2bbbf655b4c4894ffa586b026771e : SUCCESS in 1h 24m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e10a75259044d02980cca960ec50ccf : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0edffeb6e6754d8eab1a44a28d03286f : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be57c896ac784a629abb5a310ab68e3b : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ea181f30639457e8823899d48c525dc : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29f23eb8811549cf98d0f4358080784f : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c78c75d1790e48378de51305c60b166e : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/710389f2537449b192ae2c115965f101 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58c15cb6b3e84761bd0f5c164df897fc : SUCCESS in 3m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a73b6ea2852e4078887e8cce28545021 : SUCCESS in 20m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/84fe719fbd8e4b80b75510e2c46f4d6f : SUCCESS in 25m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3566ddb6f190427390af54b47fc82cd5 : SUCCESS in 1h 04m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/07cacd94cdc44e5eb31b3a7df9447fa7 : SUCCESS in 1h 45m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/929d4ff0d30f42e685d7527c1e2ae3cb : SUCCESS in 18m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d832d14cfd164769ad8b5abbdc49ad02 : SUCCESS in 1h 04m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0ed0ea01db0542b7af01ec6757d77423 : SUCCESS in 52m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/65df9ab444e3437fb6b4a7f67e528907 : SUCCESS in 1h 33m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f4346db41212be40d67ddbb6bcfa84e822e6cbdb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-02-29 15:51:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81dfb942bf42481c0c0563561aa72c0026f5440d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-29 16:43:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"16f5315a4271b9bf4f710ed95c0f7c3649b3246e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 16:43:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"016408f0f7e0f59a70536134cf057cc8a028b0f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:14:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18ced6744f245558b552773c6f5b34a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/98fff8d6027b44768e6251a7c639fc8d : SUCCESS in 19m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e69368a8309b4abb986aaa1d619fbdd5 : SUCCESS in 33m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/111ce91bbc914feda1cafd273174b953 : SUCCESS in 57m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/062d5d6dd0bd417fa4f52778db3681df : SUCCESS in 1h 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f16b00003124db0a0ceef0a22747a86 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd07c8e9875b4b9ab37eb531ae61b62c : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1536512c1111480bb30ececeb58e32a2 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c858d8d8980f445f9061d70e952db803 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94ce8d08368942e5a5b131ec14d315ea : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3d14b0c45754e198ac0fefe9fefc547 : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/26483d96951d4e338f46ba022b467736 : SUCCESS in 21m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4ed78c7addc84044a0cef4e01110112a : SUCCESS in 27m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/03ee2edb34ef43e3a3d58ecfc3b3fbba : SUCCESS in 1h 05m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/faeee0dcc6604dc1b28c4b7161ca3a9a : SUCCESS in 1h 19m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6436c88aee184af9aed02e51cad99d8c : SUCCESS in 20m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/114b7d1727774642a9ded8ab6e39faca : SUCCESS in 1h 03m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/df9f00d049b84d6aa1204482926cb825 : SUCCESS in 1h 00m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5168492f70d74771a4eefb550078aeda : SUCCESS in 1h 23m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2d86672ea0fd818eeb2e5f5558b076d5d4101608","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:14:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bc274040f4c82ca2128f980c2ff2cde7f77d10da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 18:15:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dcb789a05994224b5ce8638fa8b3467\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e5642419b24137863fcd42403be956 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dd8c61b7dda4889ad59272c98b0437f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"954356392466da8d677b9a156017171c3d2ba411","revisions":{"6737f054bed98852e31bacc5c984d242420c852a":{"kind":"REWORK","_number":1,"created":"2024-02-26 08:27:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/910148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/910148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/910148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/910148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/910148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/910148/1"}}},"commit":{"parents":[{"commit":"3927238ade1ead641d387f9edfdfb3789a429142","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3927238ade1ead641d387f9edfdfb3789a429142"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 08:27:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 08:27:56.000000000","tz":60},"subject":"toolbox: Bump ansible-core to 2.15","message":"toolbox: Bump ansible-core to 2.15\n\nThis was forgotten in 2023.2 cycle.\n\nChange-Id: If66f95e52b1897d5b2d319fb1a43bde5ed9db751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6737f054bed98852e31bacc5c984d242420c852a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6737f054bed98852e31bacc5c984d242420c852a"}]},"branch":"refs/heads/stable/2023.2"},"954356392466da8d677b9a156017171c3d2ba411":{"kind":"REWORK","_number":2,"created":"2024-02-29 12:15:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/910148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/910148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/910148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/910148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/910148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/910148/2"}}},"commit":{"parents":[{"commit":"3927238ade1ead641d387f9edfdfb3789a429142","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3927238ade1ead641d387f9edfdfb3789a429142"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 08:27:27.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-29 12:15:44.000000000","tz":60},"subject":"toolbox: Bump ansible-core to 2.15","message":"toolbox: Bump ansible-core to 2.15\n\nThis was forgotten in 2023.2 cycle.\n\nChange-Id: If66f95e52b1897d5b2d319fb1a43bde5ed9db751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/954356392466da8d677b9a156017171c3d2ba411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/954356392466da8d677b9a156017171c3d2ba411"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
