)]}'
{"id":"openstack%2Fkolla~910181","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I2bdeb77e3133123053b05adcbde3a426c60f3304","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-02-27 10:19:07.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I2bdeb77e3133123053b05adcbde3a426c60f3304","subject":"toolbox: Improve retry loop for ansible-galaxy","status":"MERGED","created":"2024-02-26 09:45:26.000000000","updated":"2024-02-27 13:53:59.000000000","submitted":"2024-02-27 13:52:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910181","cherry_pick_of_change":907543,"cherry_pick_of_patch_set":4,"meta_rev_id":"ab1f6b30e655194c66cb73bdbce2a47893ac0752","_number":910181,"virtual_id_number":910181,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-27 13:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-26 09:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-27 01:18:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-02-26 09:52:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-02-26 10:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-02-26 10:18:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-26 09:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 09:49:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-02-26 09:52:15.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-26 10:18:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-02-26 11:34:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 12:06:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 01:18:55.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"}],"messages":[{"id":"6963f325968a95d97505d10c57110e47ecc3079e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 09:45:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8b95ba26cd3acad465258660fcf5b8f9d8f15d62","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-26 09:49:09.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"c04a8996c795452779d6de5f72eae7be6f813af9","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-26 09:52:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a71c7f2023e9ef37f5e8034e92545d1bb2b169a0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-02-26 10:18:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14d6c62dc8aa8a6d2a227976bfe1d7e22d609b82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 11:34:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8115975a0a142999ad8e3d31d56ce10\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4060409174ea47838c63b88ee99b32f2 : SUCCESS in 39m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4063cdf1eaf841259efac8ebcd9f91ad : SUCCESS in 39m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3716168a0384487db9e2cc95e58d3103 : SUCCESS in 1h 47m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d58e7a2122d49398d3e0764093b7562 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a04a3e0a6f544b69786f74e055da808 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99af8336016941afbf051887c3534ae1 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a276e10dded4221bf2dcf181020a348 : SUCCESS in 5m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/43232f5567bb459f815c587374d671e0 : SUCCESS in 1h 26m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a679178658d84bce84cd122592da0a79 : SUCCESS in 1h 29m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4aa42fdd5d30f99a5ca179bffd20a2021ddf5ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 12:06:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e683b1e0f91c4d8eaf65ccf783b46a83\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/43c38e804ee3462db5fa110e1e099f19 : SUCCESS in 20m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1bdd1115ad734e0dad8a49cbf36cef99 : SUCCESS in 1h 07m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9952d476957049d4be6f0165f374f90d : SUCCESS in 21m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dc2ec00786124a29b7c489e506eae014 : SUCCESS in 28m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d536fc7659f94c969ab26a2439fdf0a5 : SUCCESS in 53m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e7a952aa23734a6b9a5e57a1f80c64f8 : SUCCESS in 1h 47m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6c2de176d249fb9bf9cd4b251a3cbc : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c0c3fd9123e4fa78ba86588f14ebae5 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6b91f52d62b404d9d8c14bd260ec4ec : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a1e09081cf24176b20fb24c7d8369a2 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f4a431e56044edfa1a82e8e303f383c : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e56a2cb059946fd994b347bebb72fb3 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a31bad48e4943e58c9a6be08d7f7fd9 : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/266fcce0cc9e45379c370235393a0f62 : SUCCESS in 20m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d697ecedd39e4e99ab11e476c861e7d1 : SUCCESS in 19m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/aa2ff7ba36c64441a64d18b8e6568528 : SUCCESS in 1h 07m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/69755a076feb413cbef3c2826e55de66 : SUCCESS in 1h 23m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8613af26ba144f71a1c9cad6867f3e61 : SUCCESS in 22m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e7123ec4b58642c1a0e55b8bbb41b226 : SUCCESS in 50m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cbd72bb22f8e4251a3e6524b5be5b996 : SUCCESS in 44m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3da3e324456c4273bbdeb3ab110564fd : FAILURE in 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"709861bd6759c0ce97c3c18484c28e42235416ae","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-26 12:07:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f77af01eda6651f0049b8478e67f2bfa0b405cf7","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-02-27 01:18:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1b262a0877850fe92c5fde14b866f68d548fd2bc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-27 06:44:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6e7344de4a6e0cc3a87cb43ba9c3308576c4295","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 08:02:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fecc7f1bfe95478181d270a237866e9a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2cd2089a24a849f0928823b369921be0 : SUCCESS in 17m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/86a668ec44ed4189b48d558c0e8f2534 : SUCCESS in 1h 03m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e8eace15d1bd4c5a9260f8b2391ee89f : SUCCESS in 1h 16m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46d08f5f764347be9ff00a3aa28e35fc : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f89d621475242d08122566335611465 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58a300f0fd1d44ecab43dd0381abaa10 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8526d55b1f674a8fa8f0beb2fbfe3961 : SUCCESS in 2m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/35cca0fa4008467a8dfda67bc02b70af : SUCCESS in 59m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/708242abc6064468bca2ad0897cbf0dc : SUCCESS in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fe14a8b96b405b848c689884833f89b1b9bd4e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 09:48:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3ca0d0bf1104f5b8e589fe7e6900b4b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b7090cca2f544e1e84ebdfa5fff87a38 : SUCCESS in 22m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/815bf11cb4f648afb27b9a2d523fdd65 : SUCCESS in 1h 06m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fa45924f64b243068a566ef964ee71f5 : SUCCESS in 19m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a88e43a395fd4261b1f460c97ac333e3 : SUCCESS in 32m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/977495222fd749d697804fe9b4d63be4 : SUCCESS in 1h 06m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b3baa1d7ae384354b86f189375835ff5 : SUCCESS in 1h 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8330be4b21342ec88a5bc175f535261 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5f198d726224814b42214b8a1271d02 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a0b206f6f694ec1803aee23ae9b822c : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04d7764f91d1418d97f35224796415ad : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae389e06e47b434e867a2df5665b297f : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcc5e5180252457f91f41d0f5b4de5d1 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ccbe9e5f9d64a5d8766343ab6c11542 : SUCCESS in 5m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3fb017334b5d4262a8c2a6f8f6bd1a17 : SUCCESS in 26m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4d794f065ffa4636a5cf70dd6008c823 : SUCCESS in 25m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bdaea949feb54047ad02205b59bf7118 : SUCCESS in 1h 08m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1a5e0cdb30ff4b46a820a47547d523af : SUCCESS in 1h 21m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/af95f6cdb1a2480fb3f5692cbb8e9c71 : TIMED_OUT in 3h 02m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/424d485500d6411e9b766eeac5a23726 : SUCCESS in 56m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0930840ed9ba47b9b9852f2729020dc7 : SUCCESS in 57m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fc5d231db3ed4e798472c0c52e941244 : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f39b8203cad92b1944c77dad81ae00e061508d42","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-02-27 10:19:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc190ec5cf414b11eeede9afcd9f17082dedf611","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 11:29:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc72bb75d924edfa10f62e471d08a75\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a9f1fc4b0fab4770a57c8afdbedcba13 : SUCCESS in 20m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7882c35d4e0f4e21ab9efafbc0c9c852 : SUCCESS in 18m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bb963def9dd94648a8bb942a3f5cb5a1 : FAILURE in 1h 08m 49s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25ccb07f171e47e69d4975d53cd464e0 : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c383a04f76d24b7984feb6e518dfba99 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3eb8d7cd58f34443a77e0068ae8a91e3 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12df5204ebcf4d1aad62333ba523630a : SUCCESS in 4m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/da065d6249294c18abc8012ab4142617 : SUCCESS in 21m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c348f4930cef45eeb209d9a6c28a1c9f : SUCCESS in 55m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"274dcee66cb2b78c27caff74300d98e5feb8d2d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 12:09:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0420b7803a4c4289ee04228f797182\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4d140e6b20a6452ab3eb6ffd6cdb18d7 : SUCCESS in 24m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d74a21dcd5f04eaa9073a08108a2c3a5 : SUCCESS in 1h 08m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25f40774a76446e784210207de704835 : SUCCESS in 19m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bc0d0d50becc47bea5cec5bdf7091779 : SUCCESS in 28m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e0b520083ee04fa1b54593f605e57b56 : SUCCESS in 1h 09m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b94233d5589b49cdb4a6dd34cba64332 : SUCCESS in 1h 23m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4640528b51d74c4c983077e42677c3cb : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3fc4eb75d5c4c648e3dbefbf494a5ec : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f290cb500946405497af0c39688f9144 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01da0ea242e347e0a325de1dd5e78210 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ce53c7b30704baa9442db9a24e35205 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7914ec38a5104ba5bad9f4407cdb8d86 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8e542ebefe44e3bc89888b22b06c46 : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b08d609eb53459a8effac1ae86d511c : SUCCESS in 20m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/597c860f138a4c36b0c1137460895ea3 : SUCCESS in 32m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6986e056ae1f4564b1075be862f30c55 : SUCCESS in 55m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/41218ce16a144ab98fccdd4bfaad0df1 : SUCCESS in 1h 43m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/324a39189ebe4fa7a334f6deacf9670e : SUCCESS in 16m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a078e1f8cb004da68083476788fa1947 : SUCCESS in 58m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/674bb94892aa4953b09752f413397a4e : SUCCESS in 59m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5ca1b4088839491c877a1c44117ee840 : SUCCESS in 1h 24m 34s","accounts_in_message":[],"_revision_number":1},{"id":"22a6f90a13c951066bcbbb5f8a24c3b1ebe263f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 12:10:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9daea67e7d868dd4aaad5f7c5a351bbb0889de31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 13:52:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a59ed72b129454bbed415420496ce26\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/664fa225c9944c93b2b529e74ab05fca : SUCCESS in 15m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/92ff57bc43614995ad9d404565090bc2 : SUCCESS in 29m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bfbb3c1f85664bcf9238c2654bb79fbf : SUCCESS in 1h 01m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0fd79876427b4d8fb162c63772473dfb : SUCCESS in 1h 38m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1148a0d8c50647e1a1a5283b688204ed : SUCCESS in 2m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42e5efd2eeff4dda9834a04b7e63f59a : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16e538be915744a3bbdd23d667d4fcbd : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81f53bea3c484d7890609d62713d1f2c : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c6ab20fb62403185a739b82007f08c : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/83f613eb0f9d45fd87e530195246a0e5 : SUCCESS in 32m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/923bb85f7edd4251a4b0382e2b46edb5 : SUCCESS in 31m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c02900bf22724dd1a3fbfee1e1bbfd04 : SUCCESS in 1h 10m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9f6a9dc7a8f9450296214e1eba1907c1 : SUCCESS in 1h 30m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f649e00f59fe4ca3ba090991a687db56 : SUCCESS in 23m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c89e82495df440cb921af71bced3de01 : SUCCESS in 1h 03m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e98cc5dac6ac468288cd1a2da20a0338 : SUCCESS in 49m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1e81e07994b741ad89d25e24b94e65cc : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9c7d168c80744103da916dbfcd878931b3e3dcc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 13:52:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab1f6b30e655194c66cb73bdbce2a47893ac0752","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 13:53:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f9f49161b024b03adb8d3fd1e5f58c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f3773dadab459f895f79097aeb96ab : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0418b57672d7d62bba8cf62ca63980697950ca00","revisions":{"0418b57672d7d62bba8cf62ca63980697950ca00":{"kind":"REWORK","_number":1,"created":"2024-02-26 09:45:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/910181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/910181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/910181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/910181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/910181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/910181/1"}}},"commit":{"parents":[{"commit":"3927238ade1ead641d387f9edfdfb3789a429142","subject":"opensearch: set OPENSEARCH_INITIAL_ADMIN_PASSWORD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3927238ade1ead641d387f9edfdfb3789a429142"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-02 13:26:02.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-02-26 09:45:26.000000000","tz":0},"subject":"toolbox: Improve retry loop for ansible-galaxy","message":"toolbox: Improve retry loop for ansible-galaxy\n\nCurrently if it somehow fails for 5 times - we get exit code 0.\n\nCloses-Bug: #2052293\nChange-Id: I2bdeb77e3133123053b05adcbde3a426c60f3304\n(cherry picked from commit 94bd013ed55fd79bc4bfa33121c823927ec40b56)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0418b57672d7d62bba8cf62ca63980697950ca00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0418b57672d7d62bba8cf62ca63980697950ca00"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
