)]}'
{"id":"openstack%2Fkolla~910719","triplet_id":"openstack%2Fkolla~master~I4ceb1bfcde9f01cf83e48b96f3178696482c881e","project":"openstack/kolla","branch":"master","topic":"sources-updates","attention_set":{},"removed_from_attention_set":{"36839":{"account":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"last_update":"2024-03-06 11:50:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4ceb1bfcde9f01cf83e48b96f3178696482c881e","subject":"Changed the versions and checksums of some packages in the sources.py","status":"ABANDONED","created":"2024-03-01 10:37:11.000000000","updated":"2024-03-06 11:50:33.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"332be834c66faca4fd4c3e5eedfac885c4e4a5d1","_number":910719,"virtual_id_number":910719,"owner":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-01 12:08:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-03-04 08:41:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-01 11:57:58.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-03-01 12:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-04 08:41:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c91988f91f3159e8adb5d6d416c456866c5c7fde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-01 10:37:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a25bef03b449dc4b8397444b37b7220c7a925a13","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-01 11:57:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e21c5595d60194acceeb775dfb3e0525f798134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 12:08:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fabbcea3e80e405fa27ed6ca1d10626c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bc9bf22a84ec4cb5b014a17eedf8b4cc : SUCCESS in 36m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e455f2876cdd426dbfb629d7faf4a79f : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/94db0d46687a4dfca9faf85a3cabdbea : SUCCESS in 17m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ca6656f5e2e44b50a53e6c6ce9a6851b : SUCCESS in 27m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3fecb12f2959465ab00168bf4461b848 : SUCCESS in 1h 09m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e547c8c5ee8e4e6181930156afbefd30 : SUCCESS in 1h 30m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8041bdb032e1463994c388c4edc785dc : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/981e0c7efd1e418398fe99b9731a82e8 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/363b5be2f6b0448da20d83b471f2ca2c : SUCCESS in 4m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28268df0d1124535a8a96d59eba7e2d4 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f21bb0c2493468e977c7f69846dcd46 : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e8a9f717f27c4b91a1971ecfe6f1a410 : SUCCESS in 22m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/04bdfdee64f8483ab49b0a43b3274d4b : SUCCESS in 39m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/581c91759bf6439db8a52d092d1e1a8b : SUCCESS in 1h 04m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/15c53b93838e46009990274f09eb654c : SUCCESS in 1h 15m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7630c81cb29944618158f0ca6d0718f7 : SUCCESS in 32m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7ea2db25b5324f78ace22356632b3c5c : SUCCESS in 58m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ac91193498b54cbb8ac6356eb5e6a18a : SUCCESS in 1h 04m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/887f6545c7b34b359cd3706114bf0dd3 : SUCCESS in 1h 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c928e2b3c9b42243057b9e66009df2b7a9c16877","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 12:09:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e32744c12848af97bb2e8afc001a42\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/527f7f45520643408832063e7ab5fdc3 : FAILURE in 22m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/22fd0e47e12a44d7b392cc42c9430618 : FAILURE in 17m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a3eba43d688a40ef98fc9d4dc6713a3c : SUCCESS in 1h 30m 22s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2c475c8f83f4c1b9b6385a5046f5aca : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36aede72e20540fcb0e0884a739f6d45 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0534a57d18984343a6b5c1d032a0e0ef : SUCCESS in 3m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f3eb72faac8842b3a8d158f5692698f9 : FAILURE in 1h 04m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a06204b3ea9140b3a4d3bbdedff655f5 : FAILURE in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6faac07d5392ee748fa40bac9c6420926860dd3f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-04 08:41:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"332be834c66faca4fd4c3e5eedfac885c4e4a5d1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-06 11:50:33.000000000","message":"Abandoned\n\nWrong first commit made. I am therefore opening a new review, as I can only change this in a very complex way. I was not aware that a separate commit has to be made for each version.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe16e760f794c7fd06272a9bdf83e363d3a2f1b0","revisions":{"fe16e760f794c7fd06272a9bdf83e363d3a2f1b0":{"kind":"REWORK","_number":1,"created":"2024-03-01 10:37:11.000000000","uploader":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"ref":"refs/changes/19/910719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/910719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/910719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/910719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/910719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/910719/1"}}},"commit":{"parents":[{"commit":"a699013a6a366844ac04a1c2d0b1e23284cb23f8","subject":"Merge \"Use openstack-exporter 1.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a699013a6a366844ac04a1c2d0b1e23284cb23f8"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-01 10:36:22.000000000","tz":60},"committer":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-01 10:36:22.000000000","tz":60},"subject":"Changed the versions and checksums of some packages in the sources.py","message":"Changed the versions and checksums of some packages in the sources.py\n\nChange-Id: I4ceb1bfcde9f01cf83e48b96f3178696482c881e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe16e760f794c7fd06272a9bdf83e363d3a2f1b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe16e760f794c7fd06272a9bdf83e363d3a2f1b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
