)]}'
{"id":"openstack%2Fkolla~912340","triplet_id":"openstack%2Fkolla~master~Icadf6c140cf903b75b7e6b8e5d205ee64427102e","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-11 14:38:30.000000000","reason":"Change was abandoned"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-03-11 06:39:07.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"Icadf6c140cf903b75b7e6b8e5d205ee64427102e","subject":"Fix upper-constraints [tooz]","status":"ABANDONED","created":"2024-03-10 09:33:05.000000000","updated":"2024-03-11 14:38:30.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f1ac2b5b54be912dcfca53674f0b137940fd7abe","_number":912340,"virtual_id_number":912340,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-10 12:35:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-10 10:45:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-10 12:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-11 06:16:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"aa159deb194074b5a8291905c45a72bc3878e250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-10 09:33:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d486ecf042b9cdc71fb15fba55ac7e4f40ceb252","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-10 10:45:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704b6ba45d0f4d6f98ba814d31b06aee\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/31b4654227f0422680eca8891bbf7003 : FAILURE in 20m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/78f7bac03a8a416e90cd792bab9449db : FAILURE in 14m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/77b5c20e75984d5a941e5eec04a4a108 : SUCCESS in 1h 09m 31s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ea9cfe2472f4fa19f63730e42f1bc98 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f011e51102924bec991fe7649a320d6c : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/23fb39829adb4b06ab9e37bf34421449 : SUCCESS in 2m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cf77270453724916afcb0541000f64ef : FAILURE in 1h 03m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0749518bb2bb4574af3388df730c2b5b : FAILURE in 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cda099fbf2294dc439c8d9edeff58a1b9a91526d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-10 12:35:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77d47578696a40349eee5a5e9a1ffdc5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d9b96aa553b940388202a09dbaf2d9b6 : SUCCESS in 20m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8904a14e42b3487f8b3711b26c243d92 : SUCCESS in 53m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d9a9860e92ff4b3b93e688e6818f7cd9 : SUCCESS in 19m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/912815787bf64cbabc7dfa4786c8b81c : SUCCESS in 28m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c861b321daba41ccaba49a328f5593ac : SUCCESS in 45m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/28b3d5e5546b4d98af3dee47aa072082 : SUCCESS in 1h 38m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f1ea9e5877403a8c5837629ca6d8a1 : SUCCESS in 2m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e00d2df4d6c4c879720dada189d577e : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79f43c876198444594d2210a8080ea62 : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2ef327c44c344f2af18aeff2a641da0 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2d9deeb4c143f382871e1c012d80a5 : SUCCESS in 3m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1c09ceb5634141caad871c01a02c139c : SUCCESS in 22m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a642b62757346efaa8abf9ef92fcc34 : SUCCESS in 28m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/da8cda7559844660b146165715091972 : SUCCESS in 1h 13m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5824b6c602d84a67a8e8d464423b54d5 : SUCCESS in 1h 30m 19s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/170f84c617864391854b55a006f670ca : FAILURE in 49m 51s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/29c26e471e5f49e99e4dacc0138d1ec4 : FAILURE in 2h 27m 08s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/05029dba498e4d69845c8f251a138eda : SUCCESS in 1h 17m 41s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8196015784c94bfc9a1053a691e278e5 : SUCCESS in 1h 27m 59s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/fa45d617ab264f12a8ad129ad3e908ae : SUCCESS in 49m 31s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2c35205f291d4597b38fd625e3b34f6d : FAILURE in 47m 24s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/e0103c5425544dfc8369a24351044fbc : SUCCESS in 1h 35m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ce718d2f4cb340e88d9ef49000fdf91b : TIMED_OUT in 3h 01m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/633979403ffa498ea882a9bba8402f0d : SUCCESS in 56m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/df197ff66075486b907ce03961438071 : SUCCESS in 46m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3aba0ab899064094ab1df6c40530d5a4 : SUCCESS in 1h 03m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5d9f3041e23344b7a02e477565208188 : FAILURE in 43m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/64c93637f6634838b388e187431d4117 : FAILURE in 2h 15m 23s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4872319650794f718690662a372da4ea : SUCCESS in 59m 31s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9d097a6c97e34566a075ded9006d7c36 : FAILURE in 4m 03s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/eebe8398a54341fea42f3942a0f0f60f : SUCCESS in 41m 21s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e0bc53f2b3314ec6af234657930e86e5 : FAILURE in 41m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2009e601996b496c9816e8ef8df545f3 : SUCCESS in 55m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63365dbe79ed9b57dfa201b9b1d380e3955f9a3f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-10 23:00:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea46a7b41320f1404b280bc0925191b634490725","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 00:11:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/947cf0751fd04accaf084b6e31af108c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/34e92da046704e2cb3645d7b77ad2b57 : FAILURE in 23m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7f4d26d84a354c88901c24ac57b0c5b3 : FAILURE in 16m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/80fc6443f33b4ffd99f76adf5b0219d7 : SUCCESS in 1h 05m 31s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcf27efeee434678997a1e1f050aa08c : SUCCESS in 2m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5caaaade8b6421698c14c36eb469284 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ad2cb26c55f4f05b58e92abeb44b173 : SUCCESS in 2m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/487fa1068dee4dd5bfc3ff12e45df413 : FAILURE in 21m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/89837f204e6546b0b827137f5fccdfd7 : FAILURE in 10m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e10d06529df32b85c3dae1323d06258d7a567d39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 02:04:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a16db8839054d6ab84301d7ca4cd2ed\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fd6e894f6fcc4ac9bcee288672795c8d : SUCCESS in 30m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/09b8a575b2e54752a65b1a728f973449 : SUCCESS in 1h 09m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9cbf9192ceca4c07b9f68782c48841e8 : SUCCESS in 18m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/74fc15a0ccdd4d4283790082a577cb9b : SUCCESS in 31m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/74fc1ed99767467fa6dc95825d866346 : SUCCESS in 51m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/28a31f9fbcaf4de4980e4090ccbc8917 : SUCCESS in 1h 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154f88ebb2c54ed4b6f83c4bf681e2f0 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2aa56a9ddff34cd6977e314affac95cf : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2dee7e6610384965a07e004becd84f61 : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/492c905a92d5488bb72d063ff7ab08f8 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638cd8c245f64a9e95b04bf8d674e80a : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5a27babda07341068082b063c95b8e94 : SUCCESS in 26m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/96683d88cc8f4fd5bde0ced2d872d92f : SUCCESS in 31m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6cfbc15c373a4f8bbc7843de7ef71cf4 : SUCCESS in 1h 08m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2dd10c2e9e364ac5862fc47085d0478d : SUCCESS in 1h 31m 59s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/9ecee94e201d40ce82e9676a4de9c799 : FAILURE in 49m 16s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b0c24720589a4a498b2216a871638210 : FAILURE in 2h 29m 57s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ba8edfc006b54c5ca5f5c126dd9444fe : SUCCESS in 1h 19m 44s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/9e9ca55def2549c5b18f93346be944a0 : SUCCESS in 1h 25m 07s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/05bd1fde353e471fb5905fbe3046f909 : SUCCESS in 55m 16s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/afe5ae4aa8ce40fdb75ac77d10537504 : FAILURE in 1h 00m 03s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/6ab4dfd95b2c40229a0d44e37ae1239b : SUCCESS in 1h 20m 28s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/859d1b9a3efd4a548aaa30f20f53163c : TIMED_OUT in 3h 01m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d85115c64da84117ad2eff6b816ea5dc : SUCCESS in 45m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cb0225ac3d25410899855580fa2eebb6 : SUCCESS in 43m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6d2e8741041941288195f39e0848c510 : SUCCESS in 1h 28m 39s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ab5ebd205ad74b4c8d285a7bd5122d3d : FAILURE in 49m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/25ff8c7813f843f1a4b364440d5e70ec : FAILURE in 2h 07m 45s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/129b4d688fae4f6d94ee568e1b69aeae : SUCCESS in 51m 34s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/5cb6c967153149f1bf2a03725e3a5271 : FAILURE in 3m 55s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e297a3781a4841f0bccfb8a8e9650d35 : SUCCESS in 43m 45s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b4a6201620da46ecba5467571d8d38a5 : FAILURE in 41m 57s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/55833c3b98b94aaca220726b7b4edf59 : SUCCESS in 55m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"138e88fbc8db34e6a43bf20dd9c6b29b850b21b8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-11 06:16:07.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"295bd9d72ba6f49a1cc50a1fe237b405ef9eea1d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-11 06:39:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1ac2b5b54be912dcfca53674f0b137940fd7abe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-11 14:38:30.000000000","message":"Abandoned\n\nabandoned in favor https://review.opendev.org/c/openstack/kolla/+/911703","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3830c29e47f2621be5537aba39c7d7d47962fd44","revisions":{"3830c29e47f2621be5537aba39c7d7d47962fd44":{"kind":"REWORK","_number":1,"created":"2024-03-10 09:33:05.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/912340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/912340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/912340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/912340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/912340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/912340/1"}}},"commit":{"parents":[{"commit":"a699013a6a366844ac04a1c2d0b1e23284cb23f8","subject":"Merge \"Use openstack-exporter 1.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a699013a6a366844ac04a1c2d0b1e23284cb23f8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-10 09:31:30.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-10 09:32:51.000000000","tz":60},"subject":"Fix upper-constraints [tooz]","message":"Fix upper-constraints [tooz]\n\nTooz was released, but upper constraints still\nnot updated. This patch fix this.\n\nChange-Id: Icadf6c140cf903b75b7e6b8e5d205ee64427102e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3830c29e47f2621be5537aba39c7d7d47962fd44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3830c29e47f2621be5537aba39c7d7d47962fd44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
