)]}'
{"id":"openstack%2Fkolla~912682","triplet_id":"openstack%2Fkolla~master~Iccb3a2ade93279d5978a465cf6bf2db2326adc03","project":"openstack/kolla","branch":"master","topic":"cadf","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-06-20 08:27:05.000000000","reason":"Change was abandoned"},"36216":{"account":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"last_update":"2024-06-20 08:27:05.000000000","reason":"Change was abandoned"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-06-20 08:27:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iccb3a2ade93279d5978a465cf6bf2db2326adc03","subject":"Add CADF audit support","status":"ABANDONED","created":"2024-03-13 00:27:56.000000000","updated":"2024-06-20 08:27:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"01bf32dd201d754c40e66483ed6e2ad0126b10cb","_number":912682,"virtual_id_number":912682,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-13 22:02:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2024-04-28 14:36:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-04-29 01:52:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 03:39:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 04:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-13 05:44:12.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-03-26 12:28:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-28 14:36:26.000000000","updated_by":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"reviewer":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"state":"REVIEWER"},{"updated":"2024-04-29 01:52:26.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"3907691fd9bd868bb6bbd51f6f34eb0e4cf087d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 00:27:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"866650bdeef4ea5e4d4bc7adb83da8b8ca91553a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 00:30:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"11dc8a5cc01301871e1ce26b0c0a80bef95f1079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 01:34:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6e266ae7cf1703b02ed4f2f40c2d93d350edce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 01:47:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"10995a44642b1dae2e0fda0f8b951d01b4f4b714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 03:39:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92f97d7dbcab47519194b5183873c114\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b9bb527dabc240e5af85361be344d1b2 : SUCCESS in 20m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/797fa6f601c640889e7a21136ee1ed8d : SUCCESS in 16m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/57bd808d5d694ae392ad2e18c4c117f7 : SUCCESS in 1h 50m 19s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad905b85a23d4cd9af3cc85314b5c663 : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/197cc932426543858c294debb6477a21 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3688b9807f984d8e9466e1bf33bc810f : SUCCESS in 2m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/64c8328753ea4cdeba8d1e04a065e6bf : SUCCESS in 1h 31m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/22c42cfc37f445569616a47e6d00eeb9 : SUCCESS in 15m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"46408db688e50d20736dadf869b9492bd38429e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 04:58:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/930aa22e37164a77a557516aa7d8e1c6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a427742251de4ea5a68b2d9268467056 : SUCCESS in 25m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4259d299e4b14d888e33fddfdd1340ca : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f817d9d5ccaf4352acfe8f5a09c6e570 : SUCCESS in 15m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e35d5670f7ba49048f94cb7d3247de6e : SUCCESS in 25m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c3b3c65283054216bf5a3f40925c54a9 : SUCCESS in 1h 08m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/deefaa25d0864868a7d986fc18744fc0 : SUCCESS in 1h 31m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500f465dfb26410dafe2652bde0b0e31 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a22d04c4af145a3a3d718451979f9b6 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43ef45ccab8e4175a865d103dfffb522 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98077150a2a04b379cecff7123525d38 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d99c9cf9f54d62ac262a2a831210bd : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/78addf26a06f45d1a0e4145720f14c6c : SUCCESS in 26m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6bcd384a54ea44dd9a9a8af15e6f84bb : SUCCESS in 32m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/85fb75e0b0924432939a68ce3706d91f : SUCCESS in 1h 07m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ff731a76249648bf8e50a4908fc8971a : SUCCESS in 1h 36m 09s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b712a95a1fe9449ebb32581d5651fe15 : FAILURE in 1h 17m 04s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dd3639a45b7e4023bab79ccc93ba188e : SUCCESS in 2h 52m 28s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/877d4ced390740179858a30cd0486162 : SUCCESS in 1h 21m 10s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/4ede46ee6b324067a70baa260aca01d3 : SUCCESS in 1h 28m 49s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/fd24e80c5bb145e2b8a46ae48189f609 : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2546a024b5b44aa99cf72fc06d74c76a : FAILURE in 1h 04m 49s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/adac817ce10b41eaac72b81ff8210ced : SUCCESS in 1h 30m 13s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/22e48eecdac841f3a4688638e2aa7a34 : SUCCESS in 52m 23s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dc6ac0ceda0d427391be6437ae833cc6 : SUCCESS in 21m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2590768fa06c4166ad85a4147560012e : SUCCESS in 1h 05m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aa432f86cc984aa1a900b82a1ef63abd : SUCCESS in 1h 03m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/edddef1d0e6842d887ba4455dd4b7a62 : SUCCESS in 1h 19m 32s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/601908bb86664a6b8b80526df8fb3c2d : TIMED_OUT in 3h 03m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/607e917910e5460098e2c5c0807e7949 : SUCCESS in 2h 34m 00s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9ad0c8fe8fd64732ac1d3bbd7b19e274 : SUCCESS in 1h 00m 15s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/3fb3a86074744365b798cdf1a9188a1d : FAILURE in 6m 12s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/14bc68ad9b044cfd915ef6e691339e8e : SUCCESS in 47m 36s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ea6694db03d44dc0beae76d00a08a27d : FAILURE in 56m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9b47972e1c7d493db3e10226be3f4a47 : SUCCESS in 1h 15m 06s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/b2a7d1007ad24f22aa070300d266a85f : SUCCESS in 36m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cf0a4336939d399408a5c4af0a116ef2b8695f23","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-03-13 05:44:12.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c59f4f97879bd4224232b1840a132f986d187ac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 14:09:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9a3f668c685931c1386cacfdd2af45367964b510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 14:18:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"249bb32ef9e093d033e407647e2fba6d2705264f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 14:22:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d9e664083bcc590c3244ffe92fadd2e23f009c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 14:46:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e363fc37d66795961ad3704556edf434c93e17ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 16:29:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b064ad46f7364e14ae6ee6581ec1ba2a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/26c8c810912c4e9399875ef55725f14d : SUCCESS in 1h 13m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f40fbabb8fa04da3984dccd62667f903 : SUCCESS in 17m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9e8144505b794093ad4eeda8a61dcdc5 : SUCCESS in 1h 41m 07s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f013c4925a314941a1d230e71ab7cee1 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03231f4e02af443c91d319b6308c6b21 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ce18356ba754f76828d386f3ef7d511 : SUCCESS in 3m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fa8bd95e66854d2c9b2f028b716ba100 : SUCCESS in 1h 06m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e9dbe0d55e2b4292bc43831231dd2780 : SUCCESS in 14m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0dc84f8e4759ada621545bf33e2bdb43739e5b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 17:39:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/730ae2ee6d524eb8ba29fbd894fe77bb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d658312924164b9696f0a78c68c74173 : SUCCESS in 25m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0396e491036c473d9ef8ddeebb169724 : SUCCESS in 1h 12m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/99d3ec97a47c40fc8457bd6b6a505fc0 : SUCCESS in 15m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7ee322f826bb4b41af790a32e129b3db : SUCCESS in 25m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b004d8a8714f4c04a87fd101ac48a1ef : SUCCESS in 1h 03m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/818a148a3c7c48a2bf12c8f5818c5112 : SUCCESS in 1h 31m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745f8b42c0e44a0497ffdb79114e136f : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/191fc6278d994b6fbe4544f0ab435a95 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c799b6d91a942e18e406dc518ca3ea7 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/435fcab088b141cc8c82555960962e7d : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42602dba811e461c91e29da55e2c50ff : SUCCESS in 4m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d9430c54940545d9aba99281f1f25fff : SUCCESS in 24m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b0b4cfcc3f0249bb91dbeca1522235f5 : SUCCESS in 28m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/808478cf069a4017b0985d69b03e57b9 : SUCCESS in 58m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5432ff6e203f4ba998279be8ddec1a05 : SUCCESS in 1h 43m 04s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f6ac595bba8c42babd7107bcb7b1aa1a : SUCCESS in 1h 44m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0b276134fd694c2fafbed4b4ab118fdd : SUCCESS in 2h 17m 15s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/cf3c621542394c64994dc85a0b6702e7 : SUCCESS in 1h 24m 26s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/558bd128e5c8435ab8a764fda93f1853 : SUCCESS in 1h 19m 48s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a1aaaa6965174ceeb2aba4d6a46e4741 : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5a5578d1b81c4a2294568e494c79d6d8 : FAILURE in 54m 57s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/cbd5b68543b24f108bb9016698084202 : SUCCESS in 1h 32m 35s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/e5ea5fe3bac249c68bb8dc610fe610a8 : SUCCESS in 1h 02m 59s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/001578bbf3bf4bf98699c8e09bd908a0 : SUCCESS in 20m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a709a861a2dd48ca8c4428d935406056 : SUCCESS in 1h 00m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/767dbaad6dfe45fbbcee0021852a6f11 : SUCCESS in 49m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c5e69b67a09c492fb3d24fc746089e1a : SUCCESS in 1h 21m 02s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/54b5eca7539d468ebc40229405f48e76 : SUCCESS in 1h 28m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/aff46f842e784944af948ef451edf09a : SUCCESS in 2h 41m 37s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/b33a5845494345e096312654e8ede3c3 : SUCCESS in 1h 08m 19s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2611b1314de24f39b34ea40c15cf682d : FAILURE in 6m 17s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f9333b4c0c804af78e626950052b6d2b : SUCCESS in 53m 15s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ee2b9c81aa8f455db8aeeb52bf2cf439 : FAILURE in 47m 01s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/7c8a55571f024f048e7a1b22b9c6eb52 : FAILURE in 45m 33s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/28c9a3f08c484ae5af48b681752fa551 : SUCCESS in 39m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f5a8f33050ce86cc1a8ff090bd06700bf041cb7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 18:20:46.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"0a8cdfe1b540907b6136c2afb20e8aad70f4dc36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 19:14:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d3f1296df318ab619b2fced558334648a8733639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 19:15:40.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ec22a34b04414ca53c9aad48dbd17f06f278167e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 21:48:06.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/440c75009261407ebb140d4f78070fc0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d05ff2777ffc470c8d9dd5c3b3bc9211 : SUCCESS in 1h 50m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/43a8fc9ffb114df689f7b89b5e8c6408 : SUCCESS in 15m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d33f25dbda6a4ebc8a59dc6a8f922d8f : SUCCESS in 2h 30m 27s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b090e5d4dc104ab79560177eb1ef792e : SUCCESS in 2m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff0a8ded8660400fa2d7669263e2cba9 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e3796c653c184b2e892098bfa43cbc2c : SUCCESS in 2m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cca761312d124599a4e60bf8de69cff9 : SUCCESS in 1h 44m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0c3f6546a5544f07889bb674fecfcc22 : SUCCESS in 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"91c20e2d7db96d97e98e3721d7b733c71b45d4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 22:02:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df16505187ae4a8aabd5d06021693a8b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1f3414c6e92c47f18053d8a158fc6714 : SUCCESS in 30m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/15486423af914ef09b5c0b7e6d3ef335 : SUCCESS in 1h 11m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3a73704d07004651b56156b432383381 : SUCCESS in 19m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6053e28e47794f74ad69e80c78e101e6 : SUCCESS in 29m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fdb75d20cd9746908d6f4eb97f528392 : SUCCESS in 1h 08m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f4a4f0b2080a450c9e88d999b8411a9f : SUCCESS in 1h 34m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e31865c453547268bc0e8212c236302 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee6d90eeac074cb399c7cf49f0886e97 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6bb81fda7d34b12845d033e619df147 : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f13159710a048258cbe1ce823f89919 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2297ce46c947789755d42c7670f195 : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b6d2eb73c1a6406ba9d0173fe5e656e9 : SUCCESS in 28m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d23e13d22a3549fca2960805fff3ae03 : SUCCESS in 33m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1dbe379582ae49bfbe5099a6c237db42 : SUCCESS in 1h 07m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b9b7b9da77124e1da9c92bb2aafb10f1 : SUCCESS in 1h 43m 58s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b497801ea79b437f895d46b3fe809b9d : SUCCESS in 1h 49m 49s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/344a549af41b44bc80f6fa2f6bce4bab : SUCCESS in 2h 31m 05s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b67a192d506c4d378e3dd21ad1ce6c7e : SUCCESS in 1h 30m 10s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/bfb4f6508b5e45a2a2d6b215a353e82e : SUCCESS in 1h 35m 59s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/72a884e275db40419a921591283b5663 : SUCCESS in 1h 07m 55s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b773fc706a0844bbaa9c0f3c78bb647d : FAILURE in 1h 01m 37s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4403e7d966d74204af92bd987a89e4de : SUCCESS in 1h 33m 55s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/be58f56c336641abbd3c110769a4c54f : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0fad83c22114a048fcfe8eb98809527 : SUCCESS in 23m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e7d03c1e5fa1452aaee7288f1144511f : SUCCESS in 56m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0b34109c88dd4896af7c4e46fa585129 : SUCCESS in 57m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dc5fecea1bf64e15ab74874a7847414d : SUCCESS in 1h 11m 45s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c5f7c16ae45d49699ea644bad00d6de2 : SUCCESS in 1h 20m 53s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dd7d1fc76d8d4e6093a026014dd3fa34 : SUCCESS in 2h 30m 20s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/91f2fc6c5f0f46e0a231f31831d5eb58 : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/d181cff76497455d9b21e77f4bce8386 : FAILURE in 4m 10s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ad1d4504c97b43b1bfcc61fdfbc0f760 : SUCCESS in 57m 16s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/23d498d5114140da8b5074dde2f7d408 : FAILURE in 44m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5a68ff870b184c6d887dfd51c03329a1 : SUCCESS in 1h 04m 34s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/d6dab44596a5440eba609c83d90e2a38 : SUCCESS in 38m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"440b650fa55fbf866ebc9e27e5c4af9551cf3371","author":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"date":"2024-04-28 14:36:26.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"cdd37a9cd6caf8cbe784a4d97a82ca34b49ceadc","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-04-29 01:52:26.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"01bf32dd201d754c40e66483ed6e2ad0126b10cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-06-20 08:27:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"9774456619be6fb05812294db24e312e5b108294","revisions":{"e011895a607dc391d7a00d26e6a37dd7d3bf4d41":{"kind":"REWORK","_number":1,"created":"2024-03-13 00:27:56.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/1"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-12 23:54:55.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e011895a607dc391d7a00d26e6a37dd7d3bf4d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e011895a607dc391d7a00d26e6a37dd7d3bf4d41"}]},"branch":"refs/heads/master"},"c3f553e26643c1dac89898dd390d89d28deb08df":{"kind":"REWORK","_number":2,"created":"2024-03-13 00:30:27.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/2"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 00:30:05.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3f553e26643c1dac89898dd390d89d28deb08df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3f553e26643c1dac89898dd390d89d28deb08df"}]},"branch":"refs/heads/master"},"2e4057d4a2cdb0646c6dd9c26e879b2b5b626374":{"kind":"REWORK","_number":3,"created":"2024-03-13 01:34:43.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/3"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 01:34:33.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e4057d4a2cdb0646c6dd9c26e879b2b5b626374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e4057d4a2cdb0646c6dd9c26e879b2b5b626374"}]},"branch":"refs/heads/master"},"f9d4a4257631f8e08405ed2bd8de614e94c06040":{"kind":"REWORK","_number":4,"created":"2024-03-13 01:47:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/4"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 01:47:25.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9d4a4257631f8e08405ed2bd8de614e94c06040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9d4a4257631f8e08405ed2bd8de614e94c06040"}]},"branch":"refs/heads/master"},"0a632ddaa6ef74c7b8c4ba655524f77c1aac474d":{"kind":"REWORK","_number":5,"created":"2024-03-13 14:09:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/5"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 14:08:28.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a632ddaa6ef74c7b8c4ba655524f77c1aac474d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a632ddaa6ef74c7b8c4ba655524f77c1aac474d"}]},"branch":"refs/heads/master"},"3da0d036ca33a89a5672c1a0c2ce43cbd0994b1a":{"kind":"REWORK","_number":6,"created":"2024-03-13 14:18:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/6"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 14:16:37.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3da0d036ca33a89a5672c1a0c2ce43cbd0994b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3da0d036ca33a89a5672c1a0c2ce43cbd0994b1a"}]},"branch":"refs/heads/master"},"6524f1825d23727a7b6e6cfddf0ea2b9b67532ed":{"kind":"REWORK","_number":7,"created":"2024-03-13 14:22:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/7"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 14:22:01.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6524f1825d23727a7b6e6cfddf0ea2b9b67532ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6524f1825d23727a7b6e6cfddf0ea2b9b67532ed"}]},"branch":"refs/heads/master"},"6ee6c6cb50772ccc70b81a92446b403711348574":{"kind":"REWORK","_number":8,"created":"2024-03-13 14:46:21.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/8"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 14:46:11.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ee6c6cb50772ccc70b81a92446b403711348574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ee6c6cb50772ccc70b81a92446b403711348574"}]},"branch":"refs/heads/master"},"84640161a634f958f18a943a2c97dbab17a0acd8":{"kind":"REWORK","_number":9,"created":"2024-03-13 18:20:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/9"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 18:20:33.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84640161a634f958f18a943a2c97dbab17a0acd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84640161a634f958f18a943a2c97dbab17a0acd8"}]},"branch":"refs/heads/master"},"cd46a07e0de012a64786c977489be6caf6665c63":{"kind":"REWORK","_number":10,"created":"2024-03-13 19:14:14.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/10"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 19:14:05.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd46a07e0de012a64786c977489be6caf6665c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd46a07e0de012a64786c977489be6caf6665c63"}]},"branch":"refs/heads/master"},"9774456619be6fb05812294db24e312e5b108294":{"kind":"REWORK","_number":11,"created":"2024-03-13 19:15:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/912682/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/912682/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/912682/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/912682/11"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-05 08:52:35.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 19:15:31.000000000","tz":60},"subject":"Add CADF audit support","message":"Add CADF audit support\n\nThe Identity service utilizes the PyCADF [1] library\nto generate CADF (Cloud Auditing Data Federation) notifications.\nThese events conform to the DMTF (Distributed Management Task Force)\nCADF [2] specification. The DMTF standard offers auditing capabilities\nfor compliance with security, operational, and business processes\nand facilitates normalized and categorized event data for\nfederation and aggregation. CADF notifications include additional\ncontextual data regarding the resource, the action, and the initiator.\n\nThe Keystone middleware library provides an optional WSGI middleware\nfilter that enables auditing of API requests for each component of\nOpenStack. The audit middleware filter utilizes environmental\nvariables to construct the CADF event.\n\nThe problem arises when activating API auditing, as it requires\nmodifications to the respective api-paste.ini file for the given\nservice.\n\nThis patch introduces a script capable of modifications, thus\nenabling/disabling CADF auditing and configuring ignored requests\n- all based on environmental variables.\n\nOne might wonder why api-paste isn\u0027t included in kolla-ansible.\nThe reason is that for many years, nobody requested it in\nkolla-ansible, and it doesn\u0027t make sense because it rarely changes\nand is not really user configuration in its nature.\nTherefore, it is logical to provide them in the kolla project and\nonly change them if necessary during container start.\n\nThe only api-paste files included in kolla-ansible are masakari\nand cyborg, where it is clearly evident that nobody maintains them\nbecause they are outdated and I am removing them in [3].\n\nThe script is simple, straightforward, and single-purpose,\nconstructed according to the documentation on configuring the\npaste deployment pipeline [4].\n\n[1] https://docs.openstack.org/pycadf/latest\n[2] https://www.dmtf.org/standards/cadf\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/911087\n[4] https://pastedeploy.readthedocs.io/en/latest/\n\nChange-Id: Iccb3a2ade93279d5978a465cf6bf2db2326adc03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9774456619be6fb05812294db24e312e5b108294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9774456619be6fb05812294db24e312e5b108294"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
