)]}'
{"id":"openstack%2Fkolla~912731","triplet_id":"openstack%2Fkolla~master~I26e6c04ad0ba15ad4aa90f95259dc41cad439e73","project":"openstack/kolla","branch":"master","topic":"proxysql-bump","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-26 13:11:26.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-03-20 14:11:33.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-03-19 09:48:42.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-03-26 13:11:26.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-03-26 11:24:40.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I26e6c04ad0ba15ad4aa90f95259dc41cad439e73","subject":"Bump proxysql to 2.6.x","status":"MERGED","created":"2024-03-13 10:35:04.000000000","updated":"2024-03-26 13:12:33.000000000","submitted":"2024-03-26 13:11:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912731-proxysql-bump","meta_rev_id":"f0cc5aac5a8af67c3f8f038dba5b707944867917","_number":912731,"virtual_id_number":912731,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 13:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-03-20 14:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-26 08:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-19 09:48:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-26 11:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-26 11:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-03-20 14:11:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-13 12:06:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 13:19:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-18 11:53:42.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-03-18 11:53:42.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-19 09:12:53.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-03-25 12:02:02.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-03-26 08:56:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"3eb3f13e29cd41252a0a429bbe97154ca7b669db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-13 10:35:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9a482d64843a44cb2a9bb23285c14cb32ece050","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 12:06:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e354727923a4e5694420dd3d3d49864\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bf71af3c9282429ab57b1d4599908b3e : SUCCESS in 27m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1523d88265254b4c97881ed06cdc2894 : SUCCESS in 16m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1f7f3cee46a942f29e397e2717cdae1b : SUCCESS in 1h 29m 28s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7f96910ecee4476b0110e77ad92c35f : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2dc7699a9b184f6dbbf2eeb9858763b8 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0869361c4abd43ed99f130ff69ec4d54 : SUCCESS in 2m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3cb21e128958459a832a5d2f0f5954a8 : SUCCESS in 59m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5d090693834a4680aa051b4b3d177518 : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"459d99225ff12466c5105542b325934d8dfed859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 13:19:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f581886a7bd04ac1b1d94a90744f58ca\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/23131f095ca54de49224414fa1f777e8 : SUCCESS in 25m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e639916572424c93b0c0391e67843041 : SUCCESS in 1h 07m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b2f00de760804789a440c3481ed5bc4b : SUCCESS in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/69573db046724337b7159837224069ac : SUCCESS in 32m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b47318000cfd4dc1b37b67d5bc2ff3a8 : SUCCESS in 1h 01m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/62cde0503edb489690d15763567cd5e1 : SUCCESS in 1h 35m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6eff0508f641f7bb691a87b4e6229d : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af8c13f00dbe4c5aa1b3b37f2b3f4ac2 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d4b30ee9d8149108419d507defd893e : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70ac4c647f0e4c0bbeeecf643b9e88e7 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892e1e21778d42ed9b02450a70785e31 : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9e2f9510b5524a08adf7d22f206edfd6 : SUCCESS in 20m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b48327451dc7464abbb192c3985213ec : SUCCESS in 33m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/448cdbf09ec645fd849c72d900819a8b : SUCCESS in 1h 08m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0251a56aeb504d62904aeb18cde97d4d : SUCCESS in 1h 29m 53s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1a95838cdb38435b8207c6c54b10e5e9 : SUCCESS in 1h 47m 28s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ea87f292a38847ed83eff99022b60d5f : SUCCESS in 2h 38m 07s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/eee7efa233644d659d5dd0b8025eccf4 : SUCCESS in 1h 12m 36s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ae6d5bddc8d245aba084236bc4f48c77 : SUCCESS in 1h 26m 34s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c58e07503b824e00af4db7aa27150881 : SUCCESS in 25m 19s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/055b7b267edb4db9b65aec3fee0d0972 : SUCCESS in 1h 04m 35s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ff80396625c84fe3b547300bcc170bfd : FAILURE in 58m 45s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d2ac82a508a848e6a4ddfc0fc97c92a9 : SUCCESS in 1h 22m 15s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c3c6930b45954c7ca8b56b1f52066433 : SUCCESS in 30m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/05722c22f98449d7849616994c28d9cd : SUCCESS in 25m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a10aa8ebf35b44d88928b6a3e915c77c : SUCCESS in 1h 04m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/43713d14eced41c79db7eb509a15082f : SUCCESS in 51m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/729fe241e4314f34a035ef937be9b0db : SUCCESS in 1h 12m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6c5a35e2657c44d883253d546c51ead5 : SUCCESS in 1h 44m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6fda95ab28aa496285dd522cb3e3f24b : SUCCESS in 2h 16m 17s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8a5c64f30e87467a932722c8d55ceb7b : SUCCESS in 1h 04m 40s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a1ff903b65744dfe9bc1b3930eaf309f : FAILURE in 4m 27s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/0b3e30dc673d41b4b21867e9bae300de : SUCCESS in 24m 08s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b558e8b8846a47899c30389ab0b6c30a : SUCCESS in 1h 03m 43s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/950e7115f0e24f2da44a24ac1d2ba21d : FAILURE in 51m 11s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/bc83450802a146e5805cc2c6ee7c8add : SUCCESS in 1h 07m 41s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/83aa2f8e9d2945b888de980f6b3fdc40 : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4678ba80be69e0f3e26a9aab45543c27adfb6976","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-18 11:53:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"98cda0c7ffd3e0d1a4bce2f5c2833b0815cab072","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 14:09:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6433200faa5746f68222d7104f2a260a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6b4cb1b15d4c40c8b8400ac327141cb4 : SUCCESS in 48m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/55fdc89807b24f01b54ba902d29f89d5 : SUCCESS in 2h 14m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9511e9ab103e4661a7b5d072fe7ad36f : FAILURE in 1h 24m 02s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/288b512b55dd4cc2bbdc97d8873c41a9 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2c8c6647cb54440bc71439fd5a6911b : SUCCESS in 2m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1399c627911f45519158b611ca4f50cd : SUCCESS in 4m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f484bc5b9e734a328885a431cd41e5d9 : SUCCESS in 44m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7594ba658e164be39f9381022e8f3556 : SUCCESS in 1h 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3794cbfbd2022964a32c5fe01e90b7f892a86606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 14:40:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d1ce75933024d74a704e078a288e2b0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/604f9c5c61b7480ea991ff401f47107a : SUCCESS in 23m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/69b0643b719c448caa36815cbb798029 : SUCCESS in 1h 04m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/146c33cc0db242dcb8a7aa8d5f3bc68c : SUCCESS in 19m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cc601bd46a994be698da3a7ad17d914c : SUCCESS in 29m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4da8e1ce8d734ae1b4c271bf32c45c1c : SUCCESS in 1h 00m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7cce3d848593471898bc9e6054cd3186 : SUCCESS in 1h 39m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868930c4022f43ffad3a4827f7b6ffe0 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a40023be8184fbdabeb0b86e5d7fe81 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b29bd1a0ba44682ae779c328b9085c5 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6347f869384340cd963debceeeecb820 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc71e8b676e4fb1a34821ad1f9d14bc : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/831c1c4287ff4136a6cf96af6cca988d : SUCCESS in 2m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/30c44cfff7224402925c44e739c4d5d9 : SUCCESS in 24m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/31b7146df2a446959a61fcfb16e51953 : SUCCESS in 31m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e38226600cfc4b35bb14228f6a1c6e34 : SUCCESS in 1h 09m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e89713b399949f1a58279ffaea9f662 : SUCCESS in 1h 45m 53s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1877b687b30c402cb09496c8c4a62b69 : SUCCESS in 1h 54m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/90a17b4d78624de592b8d7e8941ea0fa : SUCCESS in 2h 38m 28s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/01c76aa18700430ca74f23b0df109862 : SUCCESS in 1h 16m 12s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c1da53cd28984384be4cabbcab447f97 : SUCCESS in 1h 31m 32s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/208b415d8ce44a8d8c57331f91ae912a : SUCCESS in 29m 49s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/7f31448e397646c68047cbb1f2f5bc68 : SUCCESS in 1h 07m 14s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a0f4dee8cc0844b4b29a1e658711be19 : FAILURE in 1h 04m 22s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c1947fa984274dc3812465223cffc9c6 : FAILURE in 1h 07m 03s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/edbf912f230840dba2c9b62174870ad9 : SUCCESS in 32m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1434d09b7ed646ff86e33c85ea89f06b : SUCCESS in 15m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a7ac9e78f10a43a38c4d298a87a622ce : SUCCESS in 49m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2a9da0258d694a2788d6290b7f5eb9d1 : SUCCESS in 52m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/820af3d97385455bb2eebc2cf3fa4176 : SUCCESS in 1h 21m 21s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4e27a8ec12c343c6ab1a660a2e8f07cd : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/da9e8b1aa1f541239d72679560f1eb2a : SUCCESS in 2h 21m 12s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/be35df60de91465fac378044f01a9ee0 : SUCCESS in 58m 36s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8958deb3387141c78c3ee42396db25de : FAILURE in 4m 49s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/09ee5bd3ace34b3a9f26f3c3b42c54bd : SUCCESS in 19m 54s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ad5754b1284646119fda6d1ed8ad722e : SUCCESS in 47m 17s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c827f75d2df643f8b01fc35930681830 : FAILURE in 51m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/316411dfaf2d493bb55d18eac33baf75 : SUCCESS in 1h 08m 15s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f2ee767207a547e5b0774872eeec3935 : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6a557cc0dc50ec810021588574397d082f257389","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-03-18 19:32:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9d9cdfec501f177a303a74a83024c32a47ebe036","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 20:48:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbfaa37d332b499cb0d4512ae80b1a1f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f85c9b354d0f4ce193898721f1e43dc7 : SUCCESS in 16m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79ff753e99574234bad171bfeb1f4d70 : SUCCESS in 39m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f8dadbfaca1d470c92870e4928af0e48 : SUCCESS in 1h 15m 05s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b85cfe60e53a493e8155b13221576f89 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04b7f2e1f62b473cba02c68ab2897c11 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a4d1a2cf51348ad959ee5539e55a3af : SUCCESS in 4m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f02847d9c134422d8a0d172222fafaeb : SUCCESS in 20m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/03f6717a15404c9b8e298ee7d8c147dc : SUCCESS in 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cad6ef31e9172ce06b195a0405f9bfe995d0d66a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 22:22:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/387cc31e1a7040b5ad90d0ee93a9dcec\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/218a175c0b004deaa91640c464d74871 : SUCCESS in 26m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f958da3afef04a9abbb7cb2248636500 : SUCCESS in 53m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1d665f880c6f42968ed25c1ca37b22c0 : SUCCESS in 18m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b31c9fb4dc864908bdc3d4c6db7586e0 : SUCCESS in 30m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ae54866c1bc54dd1ba6a8d8e344dd977 : SUCCESS in 49m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/42e333d3d5674fcf84026c3c3ee1c5f6 : SUCCESS in 1h 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b7b263c52164a6b9a9ac6b1b9d437ac : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1d47f4b508349dca2ab4279583e2ffe : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70de34219e98429eb8d47741b80c1d69 : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c1c4d16f7bb4b619f58c5fbc74b7b81 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080dfbfce4904ac9972dd822fb876b46 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/452bbe9599f74c95bbfc7e4d432b9e24 : SUCCESS in 3m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b9ea8d2688d4c9cb44c03c9011c641e : SUCCESS in 23m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9304b7926d0a444fb443c2d31e4459ab : SUCCESS in 30m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2351a0403afc4ffd9b848dd9fa3ace2a : SUCCESS in 1h 10m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c401453ef3e44795a7ca463c33c9dbd9 : SUCCESS in 1h 40m 57s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ce1b32b3abcf4b4aa554cb4e7efb336d : SUCCESS in 1h 52m 21s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/626cf07155674834b8788386df939d76 : SUCCESS in 2h 39m 26s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ad60de68f5ee445d9544ed47be42c423 : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/458bb83ebee64550aa6a774754313d6b : SUCCESS in 1h 35m 44s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e3053f0477a94881b6695449efb4b05d : SUCCESS in 23m 28s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9f73c60ca3bb4719bc65c73d6830fee7 : SUCCESS in 1h 04m 39s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6bd5f68e6aa342d3a8ca23e650501cbf : FAILURE in 1h 02m 28s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/259e0227652d458ea1d2bdebea55ff63 : SUCCESS in 1h 25m 19s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/14b33e5d310f437eb7eda27df2cb2abb : SUCCESS in 32m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a2604a5573764993847b6285cc4db030 : SUCCESS in 18m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7abda4141cb444fa864395352814e61a : SUCCESS in 54m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9aa660eb76384b58b29d91029405a523 : SUCCESS in 52m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a4c199994f2a4ae191f06b281239763e : SUCCESS in 1h 11m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/cfbc8539bf954816a5eaa2809555ff26 : SUCCESS in 1h 14m 16s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8b19cce4466340fe9c114ce78b5355e0 : SUCCESS in 1h 58m 48s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/27a0f4d35c3e45c29fd2d870ffde0cbd : SUCCESS in 57m 28s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/fbd46a36ee49494d80c44a2c2ae16289 : FAILURE in 4m 24s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2b0aaecf044d4d128980aed5ce71e757 : SUCCESS in 18m 46s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/54912bb2d48141d690f14af9d666a054 : SUCCESS in 41m 49s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/aac6cecf713e404db3578b92e44763f8 : FAILURE in 38m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f3f5eee84dcc476ba9e4af011bbcdbb6 : SUCCESS in 53m 21s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f23934fd86fd41789a56ef2df3bef3d0 : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":3},{"id":"3c9e0209599fa4c8ef4bca2b00918a21a603f2d8","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-19 09:48:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"08d2c713483355c270cf9bfcf3408b9e8c49c481","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-20 14:11:33.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"cbc5a462956a05f216bd7c8f0715034f03bda02b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-20 15:57:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"582f659d1f481e2d4f46ca9773186e24cf4a7754","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-26 08:56:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"95a5600496b713bfc1f2b1080fa7b6b4e2bc1628","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-26 11:24:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"53efd09fe7ab669789e26a96caf3552f58a57440","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 11:25:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bd2529f400e9cd10e39004e38dd82ace5a762875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 13:11:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a5bd27224a846f6b0a8998df82fb195\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/160fbbf841f04082928ded8ca04f0565 : SUCCESS in 15m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/38a57c4970614d12ab344e1a63fe2708 : SUCCESS in 23m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4d98ba2c8ce144dea62b98239c49ed10 : SUCCESS in 58m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6d04284166f8453fa60269be66b6750a : SUCCESS in 1h 28m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad5520d9bc540ff99dde87fe4c0ce54 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea36e5ee39c34ecf98cc43296716df27 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f5bb882915440a2872aa766629b74fa : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ea30e35f7348a19055166a92c7ed7a : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a03fb7432a7a473f8012fbb10a5ad5b8 : SUCCESS in 3m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a5b20fdb127484282786830035dd4a6 : SUCCESS in 23m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ac4b6c59daa24901b0dbf71ac4d85008 : SUCCESS in 37m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4151a3081b5a4ac6898e3af252fe7d5c : SUCCESS in 1h 12m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/386a6bc8b18b4de99a8408620c026b16 : SUCCESS in 1h 40m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/99ea6ba94a3e4858aca463b14425cb36 : SUCCESS in 17m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f1f3b4fb86ce4c0ca2289859cb9ddeb6 : SUCCESS in 1h 06m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4f0f22dfd3e5464d95b5e3ac3d3c9800 : SUCCESS in 54m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b1c1f828c69f40c1985d601885699acc : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ec1d0c4bf5c9f0ddf37340d53c4b7e6201c40c60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 13:11:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f0cc5aac5a8af67c3f8f038dba5b707944867917","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 13:12:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4e759bc47648adbd78a976c0009cb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e28e61e20a4e058a0f5e590a067447 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85a7d88277244eda94fc28b2b3431f7a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2bfdf4e7c9d620a4d1397ae5acbfaaff9ec95beb","revisions":{"4e184db14a3709319904a9fcd0b4ece09494ddad":{"kind":"REWORK","_number":1,"created":"2024-03-13 10:35:04.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/31/912731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/912731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/912731/1"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 10:34:19.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 10:34:19.000000000","tz":60},"subject":"Bump proxysql to 2.6.x","message":"Bump proxysql to 2.6.x\n\nThis patch bump proxysql 2.5.x to 2.6.x\n\nChange-Id: I26e6c04ad0ba15ad4aa90f95259dc41cad439e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e184db14a3709319904a9fcd0b4ece09494ddad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e184db14a3709319904a9fcd0b4ece09494ddad"}]},"branch":"refs/heads/master"},"31b727d6e832ce1f1062cc5d70c9e01f69b367c3":{"kind":"REWORK","_number":2,"created":"2024-03-18 11:53:19.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/31/912731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/912731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/912731/2"}}},"commit":{"parents":[{"commit":"ff798d757bf1fbc684348f32bbf6ebde13ffc3bb","subject":"Merge \"mariadb-clustercheck: Add a script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff798d757bf1fbc684348f32bbf6ebde13ffc3bb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 10:34:19.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-18 11:53:10.000000000","tz":60},"subject":"Bump proxysql to 2.6.x","message":"Bump proxysql to 2.6.x\n\nThis patch bump proxysql 2.5.x to 2.6.x\n\nChange-Id: I26e6c04ad0ba15ad4aa90f95259dc41cad439e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31b727d6e832ce1f1062cc5d70c9e01f69b367c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31b727d6e832ce1f1062cc5d70c9e01f69b367c3"}]},"branch":"refs/heads/master"},"2bfdf4e7c9d620a4d1397ae5acbfaaff9ec95beb":{"kind":"REWORK","_number":3,"created":"2024-03-18 19:32:29.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/31/912731/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/912731/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/912731/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/912731/3"}}},"commit":{"parents":[{"commit":"ff798d757bf1fbc684348f32bbf6ebde13ffc3bb","subject":"Merge \"mariadb-clustercheck: Add a script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff798d757bf1fbc684348f32bbf6ebde13ffc3bb"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-13 10:34:19.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-03-18 19:32:20.000000000","tz":60},"subject":"Bump proxysql to 2.6.x","message":"Bump proxysql to 2.6.x\n\nThis patch bump proxysql 2.5.x to 2.6.x\n\nChange-Id: I26e6c04ad0ba15ad4aa90f95259dc41cad439e73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bfdf4e7c9d620a4d1397ae5acbfaaff9ec95beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bfdf4e7c9d620a4d1397ae5acbfaaff9ec95beb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
