)]}'
{"id":"openstack%2Fkolla~913316","triplet_id":"openstack%2Fkolla~master~I30ee6ed32c536d0e3e90d0933d6049b316ad1263","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"36839":{"account":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"last_update":"2024-03-28 19:55:27.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-28 15:14:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-03-28 19:55:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I30ee6ed32c536d0e3e90d0933d6049b316ad1263","subject":"Updated letsencrypt-lego to version v4.15.0.","status":"MERGED","created":"2024-03-15 10:41:12.000000000","updated":"2024-03-28 19:56:26.000000000","submitted":"2024-03-28 19:55:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913316","meta_rev_id":"a2723a140a5273bffa3fd3c44b7f8921775433b3","_number":913316,"virtual_id_number":913316,"owner":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 19:55:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-03-28 15:10:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-28 15:14:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-03-28 15:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-03-28 15:10:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 12:48:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-18 09:06:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-03-26 12:27:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-28 15:08:32.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-03-28 15:10:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e5957fbcdfb06c7be2e6a9175daef4fda9608989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-15 10:41:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddbc858a795a035ea8c9a0c3b7f4b5c63fa794ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-15 11:02:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cd6e4fda8d9883a17d3eb9166a13abab3279c651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 12:48:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f23533062a34161b6e7b82b604b289d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3a1a2a2caea94bf2ad4b2d20987d2260 : SUCCESS in 25m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ebe56b46ab0a4149a1f6cf5f30f68581 : SUCCESS in 1h 10m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4160c80b0c164d82b67a20dc62db631b : SUCCESS in 19m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/80b4a9880aa64a169c726cef7c728d5d : SUCCESS in 31m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/209bd8d08f3d4ef48bd0d318f80338eb : SUCCESS in 1h 03m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/be9a9a9d8d1048368f2f585e18247da2 : SUCCESS in 1h 36m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a492ed376c464670bb7efe8c845681f2 : FAILURE in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3637d663ab6c4731a3ed5ba81a540e67 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f2a09d7e8124b66b0fe7e937f4084a7 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d23da1b07684e8497ae909d979c76b2 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf833c999bd4530a8e1a091a02f9eee : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90762e4009574639920506029704f1dc : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/294d708059d4499f96daa9bf55e7d2b9 : FAILURE in 41m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e06989f766bb43d69a782ff39f5c4dcb : FAILURE in 44m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1834ee9ea60b415e9b8e6da1e319259f : SUCCESS in 1h 08m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/881106add21b4abfbcda40c8974108c6 : SUCCESS in 1h 44m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bc0819d1fb2a499e9390adb069151b4a : SUCCESS in 18m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e6f5867adad04c1e8e7a9c9961e59936 : SUCCESS in 1h 02m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c9019850999643bd8c9b54365905ffb4 : SUCCESS in 50m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8c449204e1d842cbbe0d1f0941b6917a : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":2},{"id":"09e3b70b547fb2f895c62a76098bba6e6262abe2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 13:21:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be9836a14cb241b6b6b9d2e65979a1fc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b63ed070547e40de9e741b5f48fa6e4e : SUCCESS in 2h 03m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc6ea2e75f3d4d769467b870140fba21 : SUCCESS in 21m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/74b330bd589f4b58b8dbca874ee76d38 : SUCCESS in 2h 17m 11s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e98a4177b384bfd98ae1e8749d07232 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9852353016647d9943a7804205f60bf : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/58ed4e287a714968925e6c7419afbb9e : SUCCESS in 3m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ea8a577e9bf944efab8bad4258e0e8cc : FAILURE in 1h 45m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6f14a19282f942fc892e2cae0cd15630 : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"449e319eaa803fb931bff587be654e862b367903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-18 08:12:25.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a46f3566ba55ac26ba550d175922025bdc5ab12a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-18 09:06:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a4080cff063a7d01b9a20d5334f9d98d68e37419","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 09:46:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe31d11b4c04cec97676c597914012c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/004caaaee13641dbb5ef98592ca9a773 : SUCCESS in 35m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/89b175792925420f9d368937c33c6edc : SUCCESS in 1h 17m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/19fd3ad08e114a43a1fc258fa97e1d24 : SUCCESS in 1h 31m 00s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daaed30e14ad437b970a4e89c47e7e99 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/215d06a424474e00bfb5dc96bffeef3d : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c2e473d088b4c74b42c5a00bf63e3e9 : SUCCESS in 3m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/86d789f541af4f52a3764f1e4899eca8 : SUCCESS in 35m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/76eb626ec1f04d43829118fcbfaa06aa : SUCCESS in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0b84c9a34dbda6d293de37f5b8f7b7dc7b2ceabd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 10:31:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ff62a0a971b4a57a10dc6c942441c16\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/57e3a24bb378407ba0d315187d5d1d32 : SUCCESS in 22m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4fbbb4368f7c4830b225ad2e6f2bddf0 : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6328dae4e99a4cb29bdfaa9173e439c9 : SUCCESS in 17m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a31ddf029d03490184139d24d903d0e6 : SUCCESS in 26m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d4ed8c24235140f0a754c02581c736b0 : SUCCESS in 1h 05m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a371b8d740374d4a94c7862f86e7d1b3 : SUCCESS in 1h 43m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080a79a374544824a8cebdca47c5bcd6 : FAILURE in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c281c42ac07249a58093f92e27b7ef10 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1ed584faf1849c9981f196cd93d04e1 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df6ad6349bd74f85af92ae7a8b0311fd : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce24989f96ed438696747095e7ff07e2 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aebc301651eb492e8e3c0759bec2e45b : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/05a74374e0934966aea91f569034b96c : SUCCESS in 20m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/26fcb1a0f9384504aeadbcf78c201b51 : SUCCESS in 33m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9c89a3b6f78448b9d78717416bff26d : SUCCESS in 1h 11m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0d34c66ef97d4ef288c3afeb413912cf : SUCCESS in 1h 33m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/582e11ebf2a74b3eb2649804b3801816 : SUCCESS in 17m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/176f78ed3b1646f08bc38da6a94b45b2 : SUCCESS in 57m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/15317f8dada54e81a51c4edb3a07fa77 : SUCCESS in 49m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0cfedeededce4e8db85055fd0bf35eea : SUCCESS in 1h 31m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a935bd8864349275698a73b7bc3ce79615f99ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-18 11:50:01.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"430f5d022c8d08e3d7d74a85124630fb25ead636","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-18 12:06:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9498b610b65454a2492d819106a62592c034927c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 13:29:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c03fa353ce84955b7f2d1b3ba4e2252\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5dd1f067713d40c58b75a9cb8f1f7a4a : SUCCESS in 24m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7100adc2c95042f2a17fe2b72fbbc17b : SUCCESS in 1h 04m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/98ef3b34554c4ddfb638e5277dee707f : SUCCESS in 18m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cca2a1f2de854c1aaefe141656b82483 : SUCCESS in 36m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/16d6e9e90bcd4162919d7d072b3b20d8 : SUCCESS in 59m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d5e3e276f3f04f298979925d69034da6 : SUCCESS in 1h 30m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b174c5b00134ad8b76e6d93ffd9d576 : FAILURE in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc529e7bed894fa69a859885971b0094 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb3f295139ee494290491c946478adca : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/324a09f48c004accba991f799c41cf26 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f941067e57d342ca90a876cc7814c33f : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3af27ff6bb80487b8be6917acbc7e5ec : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b521985aa6cf46a4812cecf9a223a3ce : SUCCESS in 24m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7329a80c19d44156bbecb0c6b72498e9 : SUCCESS in 38m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c5e12d588f97437bb3ae2b581817c012 : SUCCESS in 1h 12m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4daaa7a8165b472291cd29f6560e484f : SUCCESS in 1h 37m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d08f898c77b0469d8ca88974c3445c37 : SUCCESS in 20m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fbd617c9482145c188ed561a9320ddcc : SUCCESS in 1h 02m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f980fd0510a146ee87c84895b83cab91 : SUCCESS in 54m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b27be342fd7b445eb671db4b05e91fba : SUCCESS in 1h 26m 02s","accounts_in_message":[],"_revision_number":4},{"id":"908c85b533c9a87ace7713c6504a3d8102f8028e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 13:54:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f42a989f58634871a2c051e075e64a95\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/87e71c6d801e4c50ae5d0895516ea5b4 : SUCCESS in 39m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b76bfad17a864cc98a5bcf022f1e43a0 : SUCCESS in 2h 02m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/db93ec34c723483396bb92f56da2cb13 : SUCCESS in 2h 03m 28s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3710d90f6f784567badbc2ea10566b45 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0cceb39270343a19d5b3b2182cc76be : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/df54bfc9e1a94f46b17ae701f405e2a7 : SUCCESS in 4m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d4fe005a831d4a77aff669a23c7ac7ea : SUCCESS in 43m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9922e1631c8d492f902ab6dbc5831f9f : SUCCESS in 1h 58m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"711e6de4e22052b3d1bd0394cfef6bc3df82e233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-28 14:55:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d80200b23987db8ff93a72db05c9e24eb64db3a3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-28 15:08:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4f627668c24445449d4874fb9c1a06658866a65f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 15:10:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"dff981ab3f536070a99e4249178bda83e277e345","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-28 15:14:25.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a9c6387e1be0d7258898a924f50804ae9244669a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:41:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb64fde0f10464fa672412a6f263d57\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d43bd97adf8b4d4aaad64e4eaf48868c : SUCCESS in 23m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f31d992f04d047f6a889a562015f3509 : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/20b6b2488aac4fe6b779736851024b13 : SUCCESS in 17m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/26829214b6624f71a054ae7e36fb5915 : SUCCESS in 28m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a0ce8a37147d43cf80df170bc2846b87 : SUCCESS in 57m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/81cac411654d491db9990a803786e0d3 : SUCCESS in 1h 32m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3bf7ef61b94a8fa2d78ae71ed3f453 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e52db3637c4008b1ebd93cf029ff4c : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04f7a96fb9af4044a3c07709ce89d646 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4bf349d63f44e4fb65e0193da8b55c4 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae5817458efe4f6dbf023be5fafadc05 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffba364b3a814c93ac88d93699843b16 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2732780855704cab8188782e6d9fc186 : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/13968803fd6d4c88bce0c13b9f3e5e2c : SUCCESS in 20m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/43093f5af8bc4ca08917a709a8a7dacf : SUCCESS in 34m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1b20cede23ce4dceb8e3cabad007a68b : SUCCESS in 52m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1d079ad0d0784f009b03c1c2e66cd475 : SUCCESS in 1h 39m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/af9c0f0246b04d738f6783e720f009ff : SUCCESS in 23m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ab57d9e6879d4ae4ae35d9967190186f : SUCCESS in 1h 14m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4f53d6868ee8439b92d14c55cc9b0f17 : SUCCESS in 49m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e0781fe2884142a994a5e92c005b1396 : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":5},{"id":"06ea006e7ff018f82a8ba1ff0d0b0e0d2b7c50c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:42:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ae76211026fd0c46f0fa914f022125b2f3927290","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:51:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9fc19166d224daea01faff03bbd8341\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3a8b01bbfcf04978a16eb2ba54e36ba9 : SUCCESS in 43m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/852fece560cb4bb6abce17b04e1510da : SUCCESS in 41m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5e8cf655d90c41588f2f130f9bfb5b19 : SUCCESS in 1h 30m 19s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4582ddb9268b40d5a1e285bd318415cb : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa03cd186f30436bbdf8da72c5a55bba : SUCCESS in 4m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9cf07b17633641c8a5e126976947d13c : SUCCESS in 36m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d9c35d1eead54470b815a6df35476cef : SUCCESS in 1h 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d2af69f0af18b7f55694aab2469850f77e9df63e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:55:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5140bc5e9ff341ed8359457b1569ba9b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/362c928ee5be4739bf88ed29366ff6f7 : SUCCESS in 15m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ead752a7e4cf489d94980a24b915d065 : SUCCESS in 31m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/100e670cc2c9480bb5a115cecf0bf225 : SUCCESS in 1h 03m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/64ca8681435b41f0ba3c03895b33a844 : SUCCESS in 1h 22m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239e91b7dc1042238644106cd9620ff8 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c1282b1e88d49feb7cc99679efc5a13 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9661588c36734e32894a6d80a4d3ce7f : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad713c9eff34605b127facee5ffe372 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ab9b133c6e14b23ba1fbbb9513c2d46 : SUCCESS in 3m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/959cce0bf54d499885d54889b711e415 : SUCCESS in 24m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ad8c9a7a71b340699ee1d1ac7ec43c12 : SUCCESS in 30m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/125b513cb6784fca8aed861e6866c529 : SUCCESS in 1h 07m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0ae804129b4b49f7b85861e9df56b4a9 : SUCCESS in 1h 32m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ff4e80dc582149af99bb433740db27a0 : SUCCESS in 19m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/157569d0c8ec4b7294f9a95ae72e47db : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e56c0b6c168d42788667cb19f2afd7ac : SUCCESS in 52m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7ce29091b86a41f785c141273f01b652 : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":5},{"id":"1df38598b928f98224b32f34301c94061f304413","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:55:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a2723a140a5273bffa3fd3c44b7f8921775433b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:56:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ca62774171641f3ac22fb80dbd982b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b4a8ab1c994d44905b28cb1e5b49d4 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f34041526724b7e87e648ced0b9a747 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4e0a68665b90d6a34de2c1754699427cebabd061","revisions":{"c16893f26e875a16c6b8f4f3cd20964f1da9836b":{"kind":"REWORK","_number":1,"created":"2024-03-15 10:41:12.000000000","uploader":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"ref":"refs/changes/16/913316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/913316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/913316/1"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 10:40:08.000000000","tz":60},"committer":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 10:40:08.000000000","tz":60},"subject":"Updated letsencrypt-lego to version v4.15.0.","message":"Updated letsencrypt-lego to version v4.15.0.\n\nAlso added a releasenote.\n\nChange-Id: I30ee6ed32c536d0e3e90d0933d6049b316ad1263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c16893f26e875a16c6b8f4f3cd20964f1da9836b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c16893f26e875a16c6b8f4f3cd20964f1da9836b"}]},"branch":"refs/heads/master"},"c57e0f054f12f4d993a47ae48c7f1cfd4babc453":{"kind":"REWORK","_number":2,"created":"2024-03-15 11:02:05.000000000","uploader":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"ref":"refs/changes/16/913316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/913316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/913316/2"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 10:40:08.000000000","tz":60},"committer":{"name":"Marius","email":"wernicke@23technologies.cloud","date":"2024-03-15 11:02:05.000000000","tz":0},"subject":"Updated letsencrypt-lego to version v4.15.0.","message":"Updated letsencrypt-lego to version v4.15.0.\n\nAlso added a releasenote.\n\nChange-Id: I30ee6ed32c536d0e3e90d0933d6049b316ad1263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c57e0f054f12f4d993a47ae48c7f1cfd4babc453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c57e0f054f12f4d993a47ae48c7f1cfd4babc453"}]},"branch":"refs/heads/master"},"41107a8050a394a75775687f4b99b35c2a4f0328":{"kind":"REWORK","_number":3,"created":"2024-03-18 08:12:25.000000000","uploader":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"ref":"refs/changes/16/913316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/913316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/913316/3"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 10:40:08.000000000","tz":60},"committer":{"name":"Marius","email":"wernicke@23technologies.cloud","date":"2024-03-18 08:12:24.000000000","tz":0},"subject":"Updated letsencrypt-lego to version v4.15.0.","message":"Updated letsencrypt-lego to version v4.15.0.\n\nAlso added a releasenote.\n\nChange-Id: I30ee6ed32c536d0e3e90d0933d6049b316ad1263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41107a8050a394a75775687f4b99b35c2a4f0328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41107a8050a394a75775687f4b99b35c2a4f0328"}]},"branch":"refs/heads/master"},"4a42e7ac2ac642fd2002373b0b6aaee3e7414382":{"kind":"REWORK","_number":4,"created":"2024-03-18 11:50:01.000000000","uploader":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"ref":"refs/changes/16/913316/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/913316/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/913316/4"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 10:40:08.000000000","tz":60},"committer":{"name":"Marius","email":"wernicke@23technologies.cloud","date":"2024-03-18 11:50:00.000000000","tz":0},"subject":"Updated letsencrypt-lego to version v4.15.0.","message":"Updated letsencrypt-lego to version v4.15.0.\n\nAlso added a releasenote.\n\nChange-Id: I30ee6ed32c536d0e3e90d0933d6049b316ad1263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a42e7ac2ac642fd2002373b0b6aaee3e7414382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a42e7ac2ac642fd2002373b0b6aaee3e7414382"}]},"branch":"refs/heads/master"},"4e0a68665b90d6a34de2c1754699427cebabd061":{"kind":"REWORK","_number":5,"created":"2024-03-28 14:55:39.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/16/913316/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/913316/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/913316/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/913316/5"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 10:40:08.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-03-28 14:55:31.000000000","tz":60},"subject":"Updated letsencrypt-lego to version v4.15.0.","message":"Updated letsencrypt-lego to version v4.15.0.\n\nAlso added a releasenote.\n\nChange-Id: I30ee6ed32c536d0e3e90d0933d6049b316ad1263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e0a68665b90d6a34de2c1754699427cebabd061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e0a68665b90d6a34de2c1754699427cebabd061"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
