)]}'
{"id":"openstack%2Fkolla~913325","triplet_id":"openstack%2Fkolla~master~I01170e91b867e80fdb49eba021b36dd0ed86522a","project":"openstack/kolla","branch":"master","topic":"prometheus-memcached-exporter-update","attention_set":{},"removed_from_attention_set":{"36839":{"account":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"last_update":"2024-03-28 16:35:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I01170e91b867e80fdb49eba021b36dd0ed86522a","subject":"Update prometheus-memcached-exporter to version 0.14.2.","status":"MERGED","created":"2024-03-15 12:04:12.000000000","updated":"2024-03-28 16:37:24.000000000","submitted":"2024-03-28 16:35:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913325-prometheus-memcached-exporter-update","meta_rev_id":"a546ed7cd15dd0066205bd4358ecbdea4df3f4ff","_number":913325,"virtual_id_number":913325,"owner":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 16:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-03-28 14:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-26 12:25:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-03-28 14:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-28 14:47:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 13:55:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 15:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 12:25:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-28 14:41:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-28 14:47:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"07dc047e6acbe76b7ebc83743acde54e68e4dd0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-15 12:04:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"822f7adf0e6d9c45e4b6d24d1dcaa95efc466dc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 13:55:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b69f10248864e6d80da75b0d8b91665\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b0a39af74c884f4b810e7cb44f8efa5a : SUCCESS in 36m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/59385efe076a4f8698f4698b8be791c0 : SUCCESS in 1h 49m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6994a8af7aaa4d829bddd3444d4d2598 : FAILURE in 48m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b050f294b4d147438f8948ca2d902fac : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5bf8e152e6a483ea49f372d7ce75a2e : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ebb4153114f041fe8c8047373b120563 : SUCCESS in 6m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/901cbdd6688d44948e6b976140b70dc6 : FAILURE in 41m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7127d79213e84d1faa962bf9da3291ae : SUCCESS in 1h 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50fb6129285d5dcb8dca873d73aa99be51b11c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 15:11:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f5545028e9445c99f9ff26672e485f9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/22fb2ff8cf4a4568b334c946af285edc : SUCCESS in 26m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1f03b82d12304c27b7aef53552d694fd : SUCCESS in 1h 05m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/82a879553a6e4476bdcbb7e175c137f5 : SUCCESS in 18m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/204ef901d7824d40b10c6d97dec62416 : SUCCESS in 28m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c35b6836f57c409f8a17831dfc9e1c19 : SUCCESS in 55m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6d02f4b80a7c456f89697b34dec019f2 : SUCCESS in 1h 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939982aa9ae144b38a83a2a76ad488e7 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f987a98ea6b74870898114d5e057a325 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/665acc119f304458af21132b2344d61a : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7e35784c1a14662bf5ebeff841b3cef : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2132f2287546b1a493fb6bb95877a7 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aad2f60e503d497b96cf7b5c1ca9dfce : SUCCESS in 2m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4954576624e044d3b0742aa286f09922 : FAILURE in 42m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9fde19a14a0f43af8796d31577ba7ecd : FAILURE in 49m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a6b11065630f48e5a8ac03fffd6d5de0 : SUCCESS in 1h 04m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/979858098d0e4e5196e610fd009f90e6 : SUCCESS in 1h 39m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/86d60b16ad7047b19b084e1510db421b : SUCCESS in 21m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/daaef050317d4ac4b199f01da31edefc : TIMED_OUT in 3h 02m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cfa56ec8a49d47f8af2b6db32bf114ce : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8f63527e0d684e20b91ef5d417869324 : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3f6dfa8cd956e6da9ecac8d76772967402a4fa14","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-18 11:59:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44301ae7bd48c3fd7d276a35d56533b94fb17356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 13:45:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3e52002410648db8e62ec2e8fd555aa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2b360501478248109e60a4ed0aaec62c : SUCCESS in 23m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/707aeaf886984e548f8a2b2061dddf39 : SUCCESS in 1h 07m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e73e39daf4bb4e8289597b389c39a720 : SUCCESS in 20m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f2e9d7a53b334d7a8bab71ff0224ed22 : SUCCESS in 30m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cf504f1f423b4d1e85a7cde5eff53807 : SUCCESS in 1h 04m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7bc2d6079214431a934cef13e7fd0094 : SUCCESS in 1h 40m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e58e66ebcfbd4d91aeb25ee047c7b245 : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6fa8628c5d34c3dbdc8e6a682a7129d : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/042ee289c0674084981c7ff14385c16f : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c317e50f4db40ec8629723542890edc : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd814be53c24eac96cefeadff1b8ee9 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/513a1dabde5c4570939533d7700b1733 : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b676fd11ee6b45b8b6a36cd5ec042768 : SUCCESS in 27m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6f6897561aa2456495185c41e4f24dc6 : SUCCESS in 29m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3cd9eb533b1c47368d6336dd3940d5c7 : SUCCESS in 1h 08m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/982533dcdb114ab5900a629f2f78ab95 : SUCCESS in 1h 37m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/65f9c984b67847e4a71e1b7b57faa36e : SUCCESS in 20m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f69245cea68e4e6e8d1074e483162d4f : SUCCESS in 55m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f5424d7b05894ff3a81b81dfe24976af : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1f7b3e9906784b36a3effeb15adb4e22 : SUCCESS in 1h 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c19686c500966219403e3842d2bea3c244d696e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 15:13:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f4f27397b694b14b7df280ec8cce57a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2ab018755ebc47249ca03773c53f83a1 : SUCCESS in 39m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42bc67bcca124a69827c1417aab06a28 : SUCCESS in 42m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1bfeeb9d4064d75bf3279405f018aba : SUCCESS in 1h 21m 18s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/051f3bf322504974ba05df0cbb061043 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/202e0e04aff44a60a74552899c3e38f7 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fdf6803487d845b9be69523f7bb6e422 : SUCCESS in 3m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e4a7de09f4f740e3a5cbc96626c6b297 : SUCCESS in 45m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5b1fec849a474ea6ad840f22827cd599 : SUCCESS in 32m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"738d4c08c2982273f6cf6d74edc8731d16e8dd77","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-26 12:25:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9edde87dc80174a7c0a24e428333f4f17b6bdb22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 14:41:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9bbf1ace07311850cbe0767acb1124f6f4a6a953","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-28 14:47:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43c7191b8786c61b9df21da578ec1d229bda462f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:48:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8f7009f5ef6a0eb65b4cd49762b2c4e471ac5e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:35:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b4c57bd3bf9455abc520d68f29c372a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fb006c3f75fd44d0aff93e04c85550bd : SUCCESS in 20m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f47c5b1b65854c27b150eea7f8783f2f : SUCCESS in 28m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6ae560d03e534d11bec125366d326249 : SUCCESS in 1h 08m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5a524764736c438eaa2a64a2e9c2cfc4 : SUCCESS in 1h 34m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bc496ce7e847999e19b1bb5479b0c5 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f16c9c230ca4c74adabe733f498fc30 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f283750e5db4f7eb06966f22a11e5e3 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ab7f43e1724603accdb23c72eee17d : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5af7f5fc40124d76ad8650edeb48e376 : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/58c0a4805496420ca4ebfb6d0fdd18a1 : SUCCESS in 25m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d373d9b498db4b90be1e921a64d76f6c : SUCCESS in 34m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8b2076a2643c465bb7a96549f6e153d0 : SUCCESS in 1h 17m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/213ae5508ef4406e82d10cf1575c7c01 : SUCCESS in 1h 31m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8baa540e6e5743c0b15ef01f8414652f : SUCCESS in 25m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e27dd2faa212407a828dbcde4188e55c : SUCCESS in 1h 25m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4d060d69ea9d47ac96ecc952ea158cb8 : SUCCESS in 50m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ee51165ab7ce47d388e220e030e0f73c : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"89d0523a348419e8d95f771ab4cdef00aa172cc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:35:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a546ed7cd15dd0066205bd4358ecbdea4df3f4ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:37:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4771838dda6f4e378e2a598f8da07f8e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1327775a3f4028b24ecb586a619b71 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8a028eb1cfe44a2917bade8365da55a : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec8b04b454c665e5ac5ad1ce817d0a10b8380606","revisions":{"ec8b04b454c665e5ac5ad1ce817d0a10b8380606":{"kind":"REWORK","_number":1,"created":"2024-03-15 12:04:12.000000000","uploader":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"ref":"refs/changes/25/913325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/913325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/913325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/913325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/913325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/913325/1"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 12:03:11.000000000","tz":60},"committer":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 12:03:11.000000000","tz":60},"subject":"Update prometheus-memcached-exporter to version 0.14.2.","message":"Update prometheus-memcached-exporter to version 0.14.2.\n\nAlso add reno.\n\nChange-Id: I01170e91b867e80fdb49eba021b36dd0ed86522a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec8b04b454c665e5ac5ad1ce817d0a10b8380606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec8b04b454c665e5ac5ad1ce817d0a10b8380606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
