)]}'
{"id":"openstack%2Fkolla~913326","triplet_id":"openstack%2Fkolla~master~I2b45abc65d4e4d65a9f07a16a64d4e7cf57b54b5","project":"openstack/kolla","branch":"master","topic":"prometheus-mtail-update","attention_set":{},"removed_from_attention_set":{"36839":{"account":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"last_update":"2024-03-28 19:50:01.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-28 15:13:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I2b45abc65d4e4d65a9f07a16a64d4e7cf57b54b5","subject":"Update prometheus-mtail to version 3.0.0-rc54.","status":"MERGED","created":"2024-03-15 12:13:44.000000000","updated":"2024-03-28 19:51:02.000000000","submitted":"2024-03-28 19:50:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913326-prometheus-mtail-update","meta_rev_id":"de441bde68efd55693717429e989cc4800998734","_number":913326,"virtual_id_number":913326,"owner":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 19:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-03-28 14:42:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-28 15:13:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-03-28 14:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-28 15:18:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 13:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 12:25:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-28 14:42:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-03-28 14:42:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-28 14:47:31.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f53f4db1e60ac0585e6cfedd0c16b9f475d0dddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-15 12:13:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"108382dc94bded8e44a67a4ca37ce581be259dd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 13:55:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99350c910b854633a42779d45be39363\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/836a6d38b9f4442b84d5e5239763b4d4 : SUCCESS in 26m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ab39ced007e74db09badc4392f547da0 : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aa25aca098714427bb2f0257118082a4 : FAILURE in 15m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/12139b4a0fce4b96969ae54ea1847f17 : SUCCESS in 29m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ad1ee23ab57449ba9cee258994a7b888 : SUCCESS in 1h 10m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cb5b8131aca1472c8240e417f09e7cc3 : SUCCESS in 1h 36m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5b3d6c6d8e4068b3bcafff7a4aa2c7 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f91203a2a2b452ca864b8854957fe29 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9702f0f8a3d4cd9ac5efb564ba8234d : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbc9b4c75312459ba12671f1983cb577 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fefba2a4b6f463eb577e480b3fc7fc1 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/397dfdd989044500b72384c0c58875a9 : SUCCESS in 3m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/420ba4e9df76464d86e9e286732f1ada : FAILURE in 35m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/35acf71f9dd7428eb061b0b2d668377c : FAILURE in 47m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/124c110f05094b1d8d3b69f0be8d0b0c : SUCCESS in 1h 10m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ffb2114a1c9d439491c1d35ed5c1da14 : SUCCESS in 1h 29m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ec172e074753417dbe8f84f1478dd7db : SUCCESS in 20m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e3ecd18517a14f4b94b46dfa767e1c2d : SUCCESS in 1h 05m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8c06e9a021e943e8a7142d5247d8bf64 : SUCCESS in 47m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d5e7b49fb434451792640054ab62e5ce : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2fb3142e4c3d6e808a03162a437c7886d868feea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 14:27:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d96c53f58f7463cb4985625c4b458e5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/10a33484e7f14f40bc95c8cd1edb1287 : SUCCESS in 32m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5fdaf6a64e42454b938f9e9df66b8dfc : SUCCESS in 30m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fe7d0f82053a4299ae26dddf400640e1 : SUCCESS in 1h 27m 07s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1bbc3587d014e7895fe92cce919e1dd : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2d34f76f1ff48cc9baeef6cfd34e76c : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4316ab6ab677476a9db1780739faaacc : SUCCESS in 8m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9096a7765fab46c5b69baf5f6ced67b2 : FAILURE in 1h 23m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f5b3771ca63a47c7911be4be8848dd6b : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fcf8fdb75542e47373a8b69b96c24327fedf45d7","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-18 11:59:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccca43b941f747b1d95eb8c33bbc7db376cf73be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 14:18:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a48521777a8409eb96aaf32297745ec\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/439cf75db7ed4b0881af7fde9135924f : SUCCESS in 1h 47m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6555a3b79ed742d69232a52e3d5f1e91 : SUCCESS in 1h 56m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/172b146db51d4164a538bd7f6614a78e : FAILURE in 2h 14m 59s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4db4d85cbc1f44a48a6714cc6c217354 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb1588cc942244648ec8ea148c0cd427 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8efd578a2b594330a8ab6019cbf07088 : SUCCESS in 3m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/08561ab48e6f4337b11a5fbdc51bc646 : SUCCESS in 43m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/803e879a681e4417a0005ac059e3159d : SUCCESS in 1h 54m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74a8d8115ec8a1d180996c8688d249b371f371ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 15:06:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cff64bd69e748b2bbf72c7237015cba\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/34c2de663c504f3d94039b1164814c24 : SUCCESS in 26m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/512a4bad62634f4fa412595365f9df40 : SUCCESS in 1h 05m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2f6b61f06a484f059b6bec4ee07eedba : SUCCESS in 18m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7121dc9bd37e4c578ca702334ce202f8 : SUCCESS in 31m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/430164fd296044b89373cb78107fb4e2 : SUCCESS in 1h 10m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/422560924f9840799c13701672a24eac : SUCCESS in 1h 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a464b52118cb470cab8daf4af471f038 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cb32a5e9f3b4ab79d7a2a935e0ce578 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4cadb72abab45b0a5a4d41fbb82ad0f : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e752f47d3c214e98b24acd6405e30052 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d61eba9182459c8862822d46891e8e : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec9dc82493241db9605e3fd50b54c03 : SUCCESS in 3m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e429d61739ca4cba84a2c5c8d555f544 : SUCCESS in 24m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/82a05c45e4484b43bd57805cb50eb8d4 : SUCCESS in 27m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/caf64bd538aa4bd283978a91d878a2ad : SUCCESS in 1h 27m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ea9281de00d24c499d5498da86260e23 : SUCCESS in 1h 35m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/03170bdafc884f95b639fce7cd4526e3 : TIMED_OUT in 3h 01m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fbe33c338901438daf430b64255320d9 : SUCCESS in 52m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b66d4899e6ad44eab7df2cdbf7526b4f : SUCCESS in 39m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d6f9349ab862461ea5039ce3e6c774a8 : SUCCESS in 1h 26m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bede6a54f2d2a10ae99b89714a7d67707451e317","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 14:42:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9c1d8081ae3730a63f870f0d36cf3b18dd92a25","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 14:42:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de00ff18601f2ef1fc85be548afbc75bcd2943c9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-28 14:47:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5ec3375bf2008c3f09e8563768ef5a24889e404b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-28 15:13:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e97c537a02b70d2093db59b86bf889da0a8f0b3b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-28 15:18:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"050b0a953ba25f8e435b10ae47578a07e4a462d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:31:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51247e560dc649c685d73a470fe20e15\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3b47546b69f44044a5f4d3b1c32f1952 : SUCCESS in 20m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ca34811b5ba5431d95c9beb4c1351444 : SUCCESS in 1h 01m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/64f2c9d5353b4270a6d0f334117db319 : SUCCESS in 20m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5257115ffb614eb6be5246c835162f48 : SUCCESS in 27m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2adfbc555efd4c02aa8dc76cc0e7a2ac : SUCCESS in 56m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6ab69f899d2045c19edb3e76557f28b8 : SUCCESS in 1h 21m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4ffc6277624712a15a8c0ccbf07c3a : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3559cd3a228b453c9bc90786981b2ce7 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb042f415f8b4ee991f2ef264ae87b0e : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9095a2636b5a4753ab2cd3cb17ed9303 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b00819ef46514ecdb99cb1d0cea54695 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ab7487059d4488a9e16beb03da7367d : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2de22d9032ec46fa9752e107341f6aea : SUCCESS in 3m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4645205110704c9589b8af6d675d26a8 : SUCCESS in 21m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d0d90726fa4444dcbb81f9f5a7f519e9 : SUCCESS in 28m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/eda494ae1dd143d78f733f9f0d2aa06f : SUCCESS in 1h 05m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/42a29649414a4aab992d455f967429a8 : SUCCESS in 1h 43m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3622a683be1c4aec95774651722f40db : SUCCESS in 23m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/400ab70fd833487bbfefbc631f1f1bc0 : SUCCESS in 1h 02m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/da94fe51d16640cb9b513d54f9692658 : SUCCESS in 56m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e3721dedcc9b4b39937a9678cd17d860 : SUCCESS in 1h 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f89003b2027d8b6fa779cf125c1bf289208d6aa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:31:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc85749c8758e473226d95a92166eaec820e20fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 16:44:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf21b6144624732a14b3fa7ddf12e7d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f47f62a8b05648d7bc802ed3c14f9c08 : SUCCESS in 1h 47m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3d1b4c76f7364458ad7b86c6c88e96dc : SUCCESS in 1h 59m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0ce8042be6124f35910e1cf41a5d92e3 : SUCCESS in 1h 35m 25s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c637459f6b4482f9446211a4c10b974 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb48613944864698b0911c9555630462 : SUCCESS in 3m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/53cbab51a3dd4aaba44338636494aed5 : SUCCESS in 30m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9de307be8c7e4017a5b9ab166fcbb8ce : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"032f22c9eb82ea7aa498ae6e3e9c913e80b1d150","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:50:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93796db40ba94a92a019c514a0b4e11c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8f31a5e091ca4ea980d7e8a85261c91c : SUCCESS in 14m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4b9a8ef29b144482b7deeaa74a2794c5 : SUCCESS in 33m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/600a18d5fd154b9aaa94ae2762e49d87 : SUCCESS in 1h 13m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fcc0e7f816144a2faceca7de63721779 : SUCCESS in 1h 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7621f87942914fceac67bdf29388c56a : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df02c782a54147d28489cd3a21975ac0 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1fe280a474f4732960a2bc65244b94f : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be410d0f5f664cac979670ba53b1e0b4 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd6ad02ff749498e8bcd7c14b9c53ff5 : SUCCESS in 3m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6980fec23d3d46c0b2857a7452e5d17c : SUCCESS in 21m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6b4c233e8f924c1b9dcc86c7a314fe1d : SUCCESS in 29m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2ed1cb2dd42c4a958bdf8f12c1e8c737 : SUCCESS in 1h 20m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/34f9baf33a8a4f6ea5c3e89453eb22b9 : SUCCESS in 1h 26m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d63170073ee442b2b2670027c490c0bf : SUCCESS in 18m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/def2e6fe396744818d59a0daab66db2d : SUCCESS in 1h 05m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f8a8d09fde0642f1a8898e459902a7d8 : SUCCESS in 54m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5b623c924216496f9f13957b49b9434b : SUCCESS in 1h 21m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e0ce97829137bfa1da12acf900906bbfaa5dd132","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:50:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"de441bde68efd55693717429e989cc4800998734","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:51:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a48e2e32a6f7446997f8a72ceef20e68\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf9404304294cd3974dc6a905ccf8ea : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e0c4cd1d5934d03a97d6c9ad0f6f3c9 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2cf1ef5976c90b23a5c91127e5603a74f925886","revisions":{"f2cf1ef5976c90b23a5c91127e5603a74f925886":{"kind":"REWORK","_number":1,"created":"2024-03-15 12:13:44.000000000","uploader":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"ref":"refs/changes/26/913326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/913326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/913326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/913326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/913326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/913326/1"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 12:12:33.000000000","tz":60},"committer":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 12:12:33.000000000","tz":60},"subject":"Update prometheus-mtail to version 3.0.0-rc54.","message":"Update prometheus-mtail to version 3.0.0-rc54.\n\nAlso add reno.\n\nChange-Id: I2b45abc65d4e4d65a9f07a16a64d4e7cf57b54b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2cf1ef5976c90b23a5c91127e5603a74f925886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2cf1ef5976c90b23a5c91127e5603a74f925886"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
