)]}'
{"id":"openstack%2Fkolla~913327","triplet_id":"openstack%2Fkolla~master~Iec2cc1421e9e145714f45599712ef062d715477b","project":"openstack/kolla","branch":"master","topic":"prometheus-mysqld-exporter-update","attention_set":{},"removed_from_attention_set":{"36839":{"account":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"last_update":"2024-03-28 21:58:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iec2cc1421e9e145714f45599712ef062d715477b","subject":"Update prometheus-mysqld-exporter to version 0.15.1.","status":"MERGED","created":"2024-03-15 12:16:37.000000000","updated":"2024-03-28 21:59:13.000000000","submitted":"2024-03-28 21:58:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913327-prometheus-mysqld-exporter-update","meta_rev_id":"16b3420287f6ca6f48e56df8bf0981d8f343f982","_number":913327,"virtual_id_number":913327,"owner":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 21:58:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-03-15 12:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-26 12:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-03-28 14:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-03-28 14:48:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 12:24:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-15 14:23:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 14:24:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 12:26:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-28 14:48:17.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9b75281ea7d27caa3ae2655be9358c8f41c1c1cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-15 12:16:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28853853ea3077e70dd7890d2677fc7fd271093b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-15 12:24:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6bcc03c12007896bea801752ae630c3502390ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 14:23:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d08b0c267db64c8db96fd2fcf06006bf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/046ff66978f0494dbcf236aa44111b74 : SUCCESS in 42m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c69e06887a8d45638cacf82ce996eb48 : SUCCESS in 39m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ebd429a6b8a4f3faabd7d35f4857818 : SUCCESS in 1h 27m 43s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b72b9c78cb34cd89f1f967c8a5fdae8 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c403964169b43cf81bbc081cfae8aa1 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e940d86cad64884a9b40cd35b305d9c : SUCCESS in 3m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5e7d5a80978a401b85916c968e4ba95d : FAILURE in 40m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e0a687bc1b5545b4a45905e6a63e046e : SUCCESS in 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a38eedfb7b31156dcfdeb983126142c02a921671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 14:24:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9371e52950ae4c51be7e0d2bd8f2851e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/13e31e12cbe84ba494645aac3df8111c : SUCCESS in 25m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8ba9b1546f674fd9afd5ca949ebed1f7 : SUCCESS in 1h 05m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/46881393dece48b0a9433570de30b10b : SUCCESS in 17m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1abfa5e8c08247da8eb164cadd878ac2 : SUCCESS in 30m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fe8157b6fb2e48998f7c7865664c1c76 : SUCCESS in 58m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/038dd630a38245a4a6dcd39f5b90e571 : SUCCESS in 1h 31m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1365c2a5b2af47a6ad46e9235bb6e8be : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62b1da29bdd244c1a3a0a0fd92af925e : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5315ce70dc245e6834109b3fc4e13d1 : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bdeb6774e254fe482f5eb44924abbb3 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed56db865b4e496c8529add04dae371f : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f124ea91a8b644bbac67c418f3109a78 : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3dc122ab977f492c8a17661aaf54afcd : FAILURE in 36m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/817366f934084e989bbfb4c201efe3e2 : FAILURE in 42m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bf24e71d83344758a1e78bea9645b62e : SUCCESS in 1h 10m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ddcada4e396640bfacdf5c98c3f04844 : SUCCESS in 2h 00m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f29953c42f7447af94f4cb959d83028d : SUCCESS in 18m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f32bc147d8074d8d85af6b72be137291 : SUCCESS in 46m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fa1c3e63a42d4890815ba3424e2b1f85 : SUCCESS in 55m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/39fb9aa563e840c397672f981daf5115 : SUCCESS in 1h 44m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4920cdec17b97b8ffb16a746a454cb9aa30c1d25","author":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"date":"2024-03-18 11:54:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd719ebce28f2359c6ded38683ab837d77e51922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 13:31:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f73ca5607b540b692e4b40b46063e41\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9136da1d2e524308b41772584500fb77 : SUCCESS in 24m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9c3023ce678f4a57ac9224b0a17f9284 : SUCCESS in 1h 05m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6461fd24daec47aa91fc3f46c32e2d27 : SUCCESS in 18m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f563b5e7dc53469192bb27ac97288185 : SUCCESS in 30m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c5d22ee6ad184346a4f3386f2dbaeab4 : SUCCESS in 1h 02m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/90438c1dd73c4decb694c5dee141736a : SUCCESS in 1h 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e176830d0443b09c41089f16dbd057 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c1c4dab12ec4fb8813f1dbf58fb6238 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e039d9afdf44e9bb0ba8e6a63f1e266 : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7d80daa595e4efe88bc443a128b175b : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e90ede68dc44fa4906df48e13f2b13b : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5caa1208407a4059b8ac7c036e68ecfb : SUCCESS in 2m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b119c42c5f5347ffba3f00de7c4cd5b7 : SUCCESS in 22m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0f3dc55a27074ae881b3a645fe236edf : SUCCESS in 30m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e3283a3a45044a3bbec3ac4486ef5dbf : SUCCESS in 1h 02m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f233ad51cc1f401da75a7cf20f21d46e : SUCCESS in 1h 27m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/77608f6b10fb43cb849e356e39158c30 : SUCCESS in 23m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/17a6f0dd958a4898893acd06c1463a60 : SUCCESS in 1h 00m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/af057a4f927d4e03a44f2d53ea1d9445 : SUCCESS in 48m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/87175176fdff4df8aa063232a3092810 : SUCCESS in 1h 20m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3a714151a684848bd7654b0763737f5bb2f99f25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 14:05:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da80bbe83fa24d5d8ae0b6a803d6ecf0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ab237e340e774191b62779e28c64e122 : SUCCESS in 2h 01m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/92db83a39fe74082ab11b21a7f292983 : SUCCESS in 2h 01m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/144209f7228e4b388f1ae28fc44a88ed : SUCCESS in 2h 09m 13s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/796fe37d35a84ef3bf836bfb28d15059 : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/392aa542b5e34c52951fff7b057984ec : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f4fa6bab55d848719314a7ebf503f126 : SUCCESS in 4m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/431c4f3c98214c4c9adc016c344e2385 : SUCCESS in 2h 00m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6f4b4f591a00415695ee1e75f327427e : SUCCESS in 37m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"909721d6663313d74180fa00d47276592c371d2c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-26 12:26:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"15979c4b9fbd0179791ed69a548b2470e0a084a7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-28 14:48:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54882c142f5efa952ae5fa5ec420f8abda640145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:48:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcae97fdefc3209ceb22359557a3ef838f30f398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 18:02:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea6d5292689c4e86ae068469143a67e3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/685ddacc0f324ded84618b20807693f8 : SUCCESS in 18m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/588c9aad097849f09f1f55cb5474dbe6 : SUCCESS in 27m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0f2cdaaa54b14b6093f6bd9833b5cc35 : SUCCESS in 59m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/72dd547edd224613b20d8a0f691fef4a : SUCCESS in 1h 33m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc7b5f3355247b3ace71591656fab22 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f25f315080c4ad68837da6563d9aa30 : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27278479f3a64641bdae72521cecc949 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3615db53577f42869f0cbd07c0873f91 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/268c1974722f4265a36ab175848a76d4 : SUCCESS in 3m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/92f43c4ee4f34f0ca7c612b5cef36419 : SUCCESS in 25m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3a4d168f6960433787700844774bc84f : SUCCESS in 29m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ad6d6004a16045568991bf432eee1be1 : SUCCESS in 1h 09m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/77b321da945043a1904115a77b825c78 : SUCCESS in 1h 39m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/92e6f29423d745ceb6e9aa445c0409b4 : TIMED_OUT in 3h 01m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/824109d95fe840518f87d5d3269eb420 : SUCCESS in 1h 03m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2dbc4474d86b4d08bfd52b9b11c43b37 : SUCCESS in 48m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b9ed9d3e2add4bccab7c604fa924ca6f : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ea6544e74a518555573f2789a58f998bffcc8e5e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 18:23:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48358208b78b96c9add658430eb9f48fd54b8dab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:33:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3afd1eef3893406eba1295ea7e820c82\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0afbfcf0ed5f4f2e8d51f140a15f9106 : SUCCESS in 26m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a186f46abf344ac59cc9a5325ec46aaf : SUCCESS in 18m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e492cc2b87c409c8b87a2bce13ab330 : SUCCESS in 1h 09m 20s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82659625bf064a8295d7c3ae35c7c450 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/54b298e0c1284295a05b16c8e6571fd3 : SUCCESS in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e950e2d86c224029afa1b86e3a52b460 : SUCCESS in 22m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5210dfb491ff471295292cdcd9984d8e : SUCCESS in 18m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3191551c9089e75cff9b9c8aeaea053b8e18c1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:59:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da2264ee273b427fa8a1ef39221f6803\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e1a380472a4142b0b13a38179cb79acf : SUCCESS in 24m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b5bb9119f1e44885bfbe8aec46a815e4 : SUCCESS in 57m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/71bf0c6b104d4034892e8629e39800cf : SUCCESS in 14m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/df65dd53e5ee49fa911a21a0dbb7ab6a : SUCCESS in 26m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/05694b6ed2bf4545a61a8c277d5d85da : SUCCESS in 1h 24m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0223836f99d148798b601b7ec39d8238 : SUCCESS in 1h 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3d70c1457144dca4b09201e866a90a : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bee8d7856fbf411f9a4d448429f19de3 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04e7216410904c4f871578f7ad090224 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f9157d3b9a243fdb91abfe45dc444dc : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4faa20a00c0643919cf47fe2b6fc8a38 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c6b67e608c404daeac3d4bf0892b69 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d83ba97a49942888d559b79baa74aee : SUCCESS in 3m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d46eb7019637499fbad450b6738ea6ad : SUCCESS in 21m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e7d7ccba5e684ff8bce9f85a3f30e46b : SUCCESS in 29m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/71096211a8fd4205b23a28cd82e1de02 : SUCCESS in 1h 03m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b351b6215b46445e976cc732d9d4dca1 : SUCCESS in 1h 31m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a7be5ba71f6a4ae2878cebea4292e3ea : SUCCESS in 17m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/42e8aba5bda64bfb98426335e7facdd2 : SUCCESS in 1h 13m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/86cb573931994860b8fabcc0ca66c31e : SUCCESS in 49m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e0e0f82166994ceca2a390708bcd664b : SUCCESS in 1h 16m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4f216e3f0505909aaafb0d3b4a1a8a90d8f6e8c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 19:59:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc77e451eba7d07806b0cf9739cf8770fca739e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 21:58:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/325515f4af334248ae4263ec9efc90af\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/785575120ef14c56989fc42f5ee7731c : SUCCESS in 17m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/10bd65ba880e4462bca7cff1b153ed76 : SUCCESS in 30m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f902422f22364e69afb2e2aefab22a25 : SUCCESS in 1h 05m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a712b62f0c0d48cb9d82f16cc556104e : SUCCESS in 1h 37m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c35d41474a4b07982673c7d3e8f84e : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf23a082dbc149fa99a594e866fea5e9 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc1e168b870f4c228ceb97d95b1dbf6c : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d713f85b7b4648378189a24f9a98ceda : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18de44d2071c4140ac9ef7cbb02da2d8 : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0521f55a042b43fb97c96b75f84d5661 : SUCCESS in 21m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/98bc00c8a85b4d7981e7dcdaf164902e : SUCCESS in 33m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/63ce9e1841cd40efb624d5a2b7a66217 : SUCCESS in 1h 11m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b16b7135fd564ab3b4d595dda7aa7ddd : SUCCESS in 1h 52m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fdb37969267544b6bd9fb3682d3ae0a9 : SUCCESS in 19m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/57f8a2e007444048b4b9fef20bcb77a6 : SUCCESS in 1h 04m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/03969d341ac7483ea941846c73dea2f4 : SUCCESS in 53m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c18df1272f994e37b22942da5b83a69f : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9f0095848e8ac30c748b6a94827aeb7ba4cb88f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 21:58:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16b3420287f6ca6f48e56df8bf0981d8f343f982","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 21:59:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650acec811604225a94cbf1a3c8f5353\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc863618a9a4ea7a4eda73d05ef8530 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb1520f34f5441b7b66774bd929203f7 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"737624c7801e20b402d9c94fdea4a136036276ff","revisions":{"737624c7801e20b402d9c94fdea4a136036276ff":{"kind":"REWORK","_number":1,"created":"2024-03-15 12:16:37.000000000","uploader":{"_account_id":36839,"name":"Marius","display_name":"marius","email":"wernicke@23technologies.cloud","username":"rhizoet"},"ref":"refs/changes/27/913327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/913327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/913327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/913327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/913327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/913327/1"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 12:16:04.000000000","tz":60},"committer":{"name":"Marius Wernicke","email":"wernicke@23technologies.cloud","date":"2024-03-15 12:16:04.000000000","tz":60},"subject":"Update prometheus-mysqld-exporter to version 0.15.1.","message":"Update prometheus-mysqld-exporter to version 0.15.1.\n\nAlso add reno.\n\nChange-Id: Iec2cc1421e9e145714f45599712ef062d715477b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/737624c7801e20b402d9c94fdea4a136036276ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/737624c7801e20b402d9c94fdea4a136036276ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
