)]}'
{"id":"openstack%2Fkolla~913786","triplet_id":"openstack%2Fkolla~stable%2Fzed~I7f1ee493cf07909b5ed8fab89133c0eabcef2793","project":"openstack/kolla","branch":"stable/zed","topic":"bug/2058484","attention_set":{},"removed_from_attention_set":{"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-03-20 15:20:08.000000000","reason":"Change was abandoned"},"32398":{"account":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"last_update":"2024-03-20 15:20:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7f1ee493cf07909b5ed8fab89133c0eabcef2793","subject":"Fix bug prohibiting images build:  * Plugins archive can\u0027t be build correctly, missing key.","status":"ABANDONED","created":"2024-03-20 13:47:40.000000000","updated":"2024-03-20 15:20:08.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d9024094086f54e2332a2bfe7ea3665acc608ce1","_number":913786,"virtual_id_number":913786,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-20 15:05:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":-1,"date":"2024-03-20 14:08:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 13:55:23.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-03-20 14:08:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-20 15:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59842bce2a6d4f8d3dd5364889e5cb28a83f1d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-03-20 13:47:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18834e37357b61d63daccaaab7d5836c39dc65eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-03-20 13:51:27.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"25c97a749140bbfc3eb4ccaefbb940b5ef3d0c4e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-20 14:08:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5f5a7c1c4cc480a05dbc0ca2376bddd904b3467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 15:05:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1f9fefb5f8c479690bc5a6293615be7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/debedc50ec094683aee667e401c82f63 : SUCCESS in 21m 37s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/52933f5e927246dca80c9178a5a74624 : SUCCESS in 51m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2646f1942563464b991a3e0df0b02a8e : SUCCESS in 24m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/530def8cd7254131b53f7ced37c4bb66 : SUCCESS in 1h 08m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9567d05212634298b37fb9bced3bbe8f : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e370d3fe9c9341ae91dca7712525a289 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/976d0d118ffa413f8f70a58e7b7a0290 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/249f5ed6451a4ee9ab5c340318ccd6d0 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb086d4450524cf1b297d12e30717a08 : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/526873dc85e645dda0b7cca3edad1348 : SUCCESS in 24m 14s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1e0d6f73c2d747acb94908c25ea411dc : SUCCESS in 1h 09m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e088eb03e7c94039b02c6834b884d91e : SUCCESS in 21m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/46332785edf440ad8b990c4cca348a24 : SUCCESS in 47m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6b8da9571dc67d4f411cfc4569a2aea036bc1bf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 15:15:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a690623c39d4ecc9569eaf4c37bd13e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1a7ca62193f6404aaefbf13ee165e05f : SUCCESS in 1h 17m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/93542e9bfdf74077967d2c40420ead38 : SUCCESS in 13m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/01da3834933e47ccba7cef3b0b35d652 : SUCCESS in 1h 22m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd014057d5754fee8c54afb97b905fbd : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5c4ff83ee494f60a4b09a0ec56e6111 : SUCCESS in 2m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/515f6cc596ba46dfa8db37f897d49ca9 : SUCCESS in 1h 02m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/00e7485fb2c044a5b2ca180965464a14 : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9024094086f54e2332a2bfe7ea3665acc608ce1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-03-20 15:20:08.000000000","message":"Abandoned\n\nCreate patch from master/MR on master.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32cc62a7f30b95512e47d531bb56fdb85a52eb6c","revisions":{"9f4e97f5aff760641bc9d117d81760b337634edb":{"kind":"REWORK","_number":1,"created":"2024-03-20 13:47:40.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/86/913786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/913786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/913786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/913786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/913786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/913786/1"}}},"commit":{"parents":[{"commit":"9eb088b63020ba22b1950eb6b248150721098b9c","subject":"toolbox: Improve retry loop for ansible-galaxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eb088b63020ba22b1950eb6b248150721098b9c"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2024-03-20 13:31:53.000000000","tz":60},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2024-03-20 13:47:35.000000000","tz":60},"subject":"Fix bug prohibiting images plugins archive creation.","message":"Fix bug prohibiting images plugins archive creation.\n\nCloses-Bug: #2058484\nChange-Id: I7f1ee493cf07909b5ed8fab89133c0eabcef2793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f4e97f5aff760641bc9d117d81760b337634edb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f4e97f5aff760641bc9d117d81760b337634edb"}]},"branch":"refs/heads/stable/zed"},"32cc62a7f30b95512e47d531bb56fdb85a52eb6c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-20 13:51:27.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/86/913786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/913786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/913786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/913786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/913786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/913786/2"}}},"commit":{"parents":[{"commit":"9eb088b63020ba22b1950eb6b248150721098b9c","subject":"toolbox: Improve retry loop for ansible-galaxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eb088b63020ba22b1950eb6b248150721098b9c"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2024-03-20 13:31:53.000000000","tz":60},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-03-20 13:51:27.000000000","tz":0},"subject":"Fix bug prohibiting images build:  * Plugins archive can\u0027t be build correctly, missing key.","message":"Fix bug prohibiting images build:\n * Plugins archive can\u0027t be build correctly, missing key.\n\nCloses-Bug: #2058484\nChange-Id: I7f1ee493cf07909b5ed8fab89133c0eabcef2793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32cc62a7f30b95512e47d531bb56fdb85a52eb6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32cc62a7f30b95512e47d531bb56fdb85a52eb6c"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
