)]}'
{"id":"openstack%2Fkolla~913788","triplet_id":"openstack%2Fkolla~master~Ib0e78581e6c34be650e59d8a044fe21b06ab45a9","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-03-20 14:31:29.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-25 12:28:25.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-03-20 15:37:59.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-03-25 10:40:31.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ib0e78581e6c34be650e59d8a044fe21b06ab45a9","subject":"[release] Use UCA Caracal","status":"MERGED","created":"2024-03-20 14:12:12.000000000","updated":"2024-03-25 12:29:26.000000000","submitted":"2024-03-25 12:28:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913788","meta_rev_id":"e6a258c1382ffe977b6fa3fb6f93bc808285ebf2","_number":913788,"virtual_id_number":913788,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-25 12:28:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-03-20 20:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-03-21 02:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-03-25 10:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-25 10:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-25 10:40:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-03-20 20:37:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 14:12:29.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-03-20 14:15:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-20 15:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 16:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-21 02:46:30.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-03-24 20:48:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"fb075c95b042abeef5fcd971187a7d82e9a344b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-20 14:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48f81814956d1ea841b09c59aac04744b655e2f7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-20 14:15:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"481186c81abe6d7ddf6f502ba88ef5c846c3dbc2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-20 14:21:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92b1a159692202209f6b60516f97da212a7cc167","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-20 14:31:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"776a162d12faa10d51bcdcd1bd5c872276087485","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-20 14:31:35.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ae312fc13b6215c1a9c6430f609c09c73762ca27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 15:33:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e54fe9043eb4432197ebd1c11f1955ca\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c76c5c5041ef496f8fc692fbd4b5a131 : SUCCESS in 22m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/975eeb743da54a0f97e49635b69ec404 : SUCCESS in 1h 19m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8f14c32931004e4ea41fd6745cf18657 : SUCCESS in 1h 01m 25s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f455eaa5cbe445aaa47ba0fbbc8f559 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d95cc805c5ca4c278ee473c8327372f9 : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c7733365832b47bfb8454b8d79df6c99 : SUCCESS in 4m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/842f560e2bd44a6bb361c192f6583496 : SUCCESS in 22m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/76dbf24000fc4fc79b2a0a7173f62079 : SUCCESS in 1h 09m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3db2126db6df11c3e0502dd15a35cb52198d1927","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-20 15:37:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1aa797ed926c29f3fce5e375e1f9463a42ccf22c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 16:15:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ac7ac95d8d14adf9ae56787cefc2de3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1de8f0e93af349d1ab5a1fc708bec87c : SUCCESS in 24m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b79b833d1cdf4eea958ffde0389b2518 : SUCCESS in 1h 03m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/885e67f1ebbd495ebe31ecb6e48c25a5 : SUCCESS in 20m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6ac808ad90b147a08027534c1b18a842 : SUCCESS in 22m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dd0129dc4b164844b0123d1376da7751 : SUCCESS in 1h 10m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8c435e7a1ed947bcb5314f72943415b6 : FAILURE in 55m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821e5e20ce4445089eaa161f5de7e571 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa8a510c0b754c6eb4093d9ac87571ef : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9aaefa11a56a4366a1486bc1b8ab6c06 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80ec4043af6d477280f0a778340b702a : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9fd345be6b94b9aa293f875e53c8279 : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3af7e279793b4da897d70d4bf715ace6 : SUCCESS in 23m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/efdc709f35614157a41bf2999e4bceff : SUCCESS in 31m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/583ad19160404f9484cf6e2c5c30bea9 : SUCCESS in 1h 05m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2788a53f75ca4264aa1655cf65cf0955 : SUCCESS in 1h 36m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5f9456002648499a8f63a09b692eac20 : SUCCESS in 1h 51m 01s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fe98c5c2fbb14deeaa57503afdf26cef : FAILURE in 44m 40s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c69eb859886b460f8d41668025ea6a52 : SUCCESS in 1h 25m 58s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/aa5d5269a56846b7a887e3bab14d07bd : SUCCESS in 1h 33m 44s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b3be243353be42a8a8b84dc97913e7c2 : SUCCESS in 25m 31s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/47304bb243a54531af38f17eeb2c521e : SUCCESS in 1h 08m 59s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e5131812899f4e4b8451ed3d8915a845 : FAILURE in 1h 06m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1c5808d67ccb4237babe8d8befa79739 : SUCCESS in 1h 33m 47s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/515bd4981bb64521ab9c71976dc06416 : SUCCESS in 24m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e713cef453584b54a3784752c21767d7 : SUCCESS in 21m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6f11cf69c2324473b759eaeead31b1bf : SUCCESS in 55m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7a62b6150c6849c3a1f63eb333ff4021 : SUCCESS in 50m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1bb124cae1594570830866e9cb818920 : SUCCESS in 1h 13m 18s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/869b1e9d6b0a45129517111c64e3bb0c : SUCCESS in 1h 31m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/46fb403844e745eda6803f2db91e951f : SUCCESS in 2h 02m 31s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/81d3938f989d4fcca31e727960c91850 : SUCCESS in 1h 06m 17s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0487877e94dc42259b091ad3dd0f2ee2 : FAILURE in 4m 04s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b0d402f74f7a4c2291cb963e5e17a18c : SUCCESS in 22m 16s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c033b988f51f46a296a61beec776497b : SUCCESS in 52m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/75923abe037c4a269f69e2f23f1a0371 : FAILURE in 49m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/10d6f39980a847d1add0719ab52ec803 : SUCCESS in 1h 06m 18s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2e7dc4b5a5a94d18ab8f3ae37198b606 : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8b6e80d87b04471eca64fb2f4f5bd351191776bc","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-20 16:56:46.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d1adf3f50acf198637ea7a9ca6ee9a19afec6df","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-20 16:59:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66c7b7e6d319d70bf3a3b9b2e1693b291a6c01ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-20 20:37:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1ab07ad65b195a761f6a3041640cf7ea53154f83","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-20 20:38:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f08b4ea63751fa5434f1ea3e09e73cd7c69f844","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 21:44:08.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85ef530b3024893822d34ea8110eaa5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/be451549b5d9431cb4a0d1004d3ffce7 : SUCCESS in 26m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/88011dc9982949e7bc528ef4b247ee01 : SUCCESS in 20m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5d182d391b1541f090d0fb87af3bc668 : FAILURE in 30m 11s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5e888c698f8456a8eaf68f69b0de2b4 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3dfdd0be789f402f9a1032fc38af4a72 : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/31f07cb9444a4c759e1cc20360efe78c : SUCCESS in 5m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b3e280da322447148a0ce467257fefdf : SUCCESS in 27m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c9d02a101b424c6eab28b42d21e64895 : SUCCESS in 1h 05m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"721aaa70864f7c674509b8b636d9cbdd1c579211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 23:49:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8fdc48d735f450093877aa08a3bcad5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2690733ce9bf41fdbf77ef84c3aeec7d : SUCCESS in 27m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7b13d05183dd4ed69b391f73e659c225 : FAILURE in 24m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3eb09b7424a848f0bbcfe596b38a78ac : SUCCESS in 17m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8912eaa0035d45fd8fbb381870d324ba : SUCCESS in 24m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/85fa176880b94eba9543b4920e9f40c9 : FAILURE in 42m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f5d9a008a3d34752b14501481bc878d3 : FAILURE in 50m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ee24bc797f4d069c352f541140f91b : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81edd59d1ad64685bf250be2573cf8dd : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/faf49ef253b642299aa650ee12b31168 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fac2eb608ea4e52a9cf5af95de721be : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed51ac4011eb476d87a565ce5fed22ee : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d9cdcd46edd94dd2bbdc99c899162815 : SUCCESS in 22m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2e1d51cc51e24cb380ddc3dfc7b6c153 : SUCCESS in 26m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4d98f5f2598745ec9d4c7cdb82d5d492 : FAILURE in 25m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0e6dad9fdb0649d6a7ecdf61a7604d64 : FAILURE in 29m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d00f7b8eddc84254a1ec1cb75eeed28c : FAILURE in 52m 40s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3625aab33625442992318cca7c6d457f : FAILURE in 40m 53s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a0beffaae62c47c0af069e603a2ad649 : FAILURE in 1h 06m 27s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/a6ef49f928214ca5b874d2c31ac8d5a3 : FAILURE in 36m 39s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9d2012098ce3455eba5302a6287c7f43 : FAILURE in 19m 04s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/23416d5c0d34444288ec3983103979c0 : FAILURE in 45m 38s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/362319595b8047a589480198bee11f79 : FAILURE in 24m 23s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/bb74cb287c304681a663ff886b0b03d6 : FAILURE in 33m 42s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4ca9f1b718034ea3b27b891b3800767b : FAILURE in 22m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/726f25048971412891ecd619c573eedc : TIMED_OUT in 3h 01m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/430658e0aa944f6fbe702e630c78fe27 : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f46a104c9dec4b4e88282dae0d107d0d : FAILURE in 34m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/47d17db249ce49faa00592b3971ca652 : FAILURE in 43m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/dae5b82ff1564c119c2334820bced0c4 : FAILURE in 47m 28s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/416e6963395b4bc488d4ced1d8bfb839 : FAILURE in 1h 03m 32s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/89beeb5f3a2045a3bf2b24846b427cb5 : FAILURE in 1h 00m 05s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/7b12b1eb893b49a381a4aa5bd9e88473 : FAILURE in 4m 22s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/845158b5d6344a7cb99801c107ce3148 : FAILURE in 20m 50s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/72b96bd4ef4d4865949811dc3dc570fa : FAILURE in 35m 12s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/355938e3ca984dc5a0a58cd0319df59a : FAILURE in 44m 23s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5535cff8795940c5b02dec7f9a9931b7 : FAILURE in 54m 17s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/19a11ef93c464ea2bf691c4702e12c1b : FAILURE in 20m 37s","accounts_in_message":[],"_revision_number":2},{"id":"de63b1e0e4562eaac3ffd548576d6cec0f416ceb","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-03-21 02:46:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"82bad114a021c7ba2febaee79d60123651de6435","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-24 16:06:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2da8617ea2b60a5a959b94933769ea1432b47da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-24 17:30:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49bbe932d65643f3ba9d47c476ef4f8b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f40c9f4e27464387b63e0b535d6d1bc2 : SUCCESS in 26m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9c8137ecbcda4d9fb44f64888dd7a9bf : SUCCESS in 15m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6e130fbebc274146a3d7f6f5d1b15c46 : SUCCESS in 1h 20m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4f2e2c7238f4fb1aca743db8bd09979 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b94d5f10a17e43a0bc865e4fe678f8eb : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec17b09c0cc94c8b9402eb5a2837775f : SUCCESS in 2m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a6389e9026fd444db2c5d76aaa194e0e : SUCCESS in 20m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6f506a986ea945fc9755c1e7609903fb : SUCCESS in 1h 04m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41baa9d73b5961afe18591b4d25b8a6f054fd062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-24 18:35:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29a09835e870403b877cd707a71dd970\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c9d8fae9b9444d48a9a2250b5d55c8cb : SUCCESS in 25m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/254e73f3bcc24629935ed36e58c8a37e : SUCCESS in 1h 09m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6d32a355865d4f4f811d0ad4f7417a8a : SUCCESS in 18m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/496cdcc34e7f45d4b11d45a68e93d5dc : SUCCESS in 27m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a7b99bb115684bc1bf2736f82964e73e : SUCCESS in 1h 08m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e28421698ef240bebedfe9a8a98bc0cd : SUCCESS in 1h 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287eaac51c354232b83b5c0773ce6e4f : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/564f333fdc7147cfb42c2a39aa03fd0e : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe0ad84d39ca40aea96332395034a6d1 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac883efbd71d4bc0be850b75f5b23c14 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85e9ac3783524e6881f013b2e2741f2a : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8b43e473d14e0091bf0de9ad2fad00 : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/baa744ab5d0840d9b2c8d79b71bd19df : SUCCESS in 24m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/49be31c135bf44d49ff61b88f3529bec : SUCCESS in 35m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9ce9e05232245af8aff4f7fe237a90f : SUCCESS in 1h 05m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4bd7e422118547f99c665668813ff484 : SUCCESS in 1h 44m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/727ca314d7144eef9dd1a727a4232ae4 : SUCCESS in 1h 28m 34s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/47d3a71157f9494a96f267ca3a2e2675 : SUCCESS in 2h 08m 39s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1ba5ee7a46944653ab74de1d8bd4a7d5 : SUCCESS in 1h 08m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/a491143593ea49d19b0d206bbe4a1b54 : SUCCESS in 1h 14m 44s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/87f655eaf010489e9682c59e619a30a9 : SUCCESS in 22m 21s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/20e509f44f9a4488904106dc86749a9a : SUCCESS in 1h 07m 23s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2b24f16351844205b23c99b98377ee73 : FAILURE in 52m 01s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/578065075f57468da6c847a20e7cd587 : SUCCESS in 1h 39m 24s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/106eb803eb3546beb447199f33626162 : SUCCESS in 33m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82049be1d8af4d66a4d6b4cf2f56b003 : SUCCESS in 15m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2ad2346aec46423fa78e100d412ee905 : SUCCESS in 43m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b80c73b33c394967b9484cfe512dc8a8 : SUCCESS in 41m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/08adc39aeaf147e4a6b802f5aaffec5d : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/985acbf487b8469fb4f490daa80d5344 : SUCCESS in 1h 29m 25s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/436bc9a2b17f4fa68f4b93f8ec1b406c : SUCCESS in 2h 23m 31s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/02845f732e6c414087a7b37df0d4c2fe : SUCCESS in 48m 02s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/72539e4ca1e84e9dba84520a80758cac : FAILURE in 4m 08s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/dea9da5908f14f3ea9e137d6e54e01d3 : SUCCESS in 17m 12s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7474391081414c8b8b5c17a9ae0ad8e7 : SUCCESS in 50m 31s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/cd27c52ae8e4431aae016775d19a033c : FAILURE in 52m 43s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c29d42f43cf442778a886d30af458c36 : SUCCESS in 48m 22s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0ede4cb553834fc2b10de66cd6bc3e94 : SUCCESS in 27m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c7778feae28949d404d0d7db04475f0be1cf5702","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-25 10:38:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37b1f5edaec9094688c9155c0cb793ad56ce3dd6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-25 10:40:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"08862d3894fdfde5003981cc3955520209c77514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 10:40:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b87b8d7e927e4cc4d9189f247f2bef16302e8a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 12:28:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8174bdee6de4b98b2bcdba0369b0547\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e3a47d36eb86406287f5d289fa04bdcd : SUCCESS in 19m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d2837a5f17f44d7db1dbbd8d95b38439 : SUCCESS in 26m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d1649b4902c541878c9b04c82da56520 : SUCCESS in 1h 10m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1f83dd493b97493a9abf1d3cbcac0d96 : SUCCESS in 1h 22m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e49fc39ce4494df9a7e50031fa4ca4e6 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/486b9c6fc2fb48cda6ca858b31dfaf94 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b54df0ddfe64ad1b3b4807da778039e : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c36a0796e264a6c89fef67683cbf1c0 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/12effe2680d34f0b9c1a4c40517f05ef : SUCCESS in 23m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2a322fd1e2a64c5c835b3df2ca82b824 : SUCCESS in 33m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/aa2c0b30aa424bffa6131775d174f16a : SUCCESS in 1h 11m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f68fbe37efb94c9781acb915496c7953 : SUCCESS in 1h 42m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ded4bc32a6a447c0b52a5843d6a527c6 : SUCCESS in 25m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5a76109c893b4b8592dd3365d314374d : SUCCESS in 1h 11m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fbec734a560f42ccaa9fd8ab50eb20ec : SUCCESS in 55m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/690c040fc1f44e80843240e7a7af9fbe : SUCCESS in 1h 27m 00s","accounts_in_message":[],"_revision_number":2},{"id":"422438b68338b8779e9150c474b7b40f34b01159","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 12:28:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e6a258c1382ffe977b6fa3fb6f93bc808285ebf2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 12:29:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34ab1f2d1f5f45a1b5f5f56d1cf8d5d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eabcb79c12c422d83e47bfc3735ec9a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ccccec9b9e0fb8b39f2c70d52e897a56063dfdf","revisions":{"a0972455ad9df0f99a06b5d410772129282add35":{"kind":"REWORK","_number":1,"created":"2024-03-20 14:12:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/88/913788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/913788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/913788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/913788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/913788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/913788/1"}}},"commit":{"parents":[{"commit":"a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547","subject":"Bump rabbitmq to 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0e22caa8a131486de6cbcf0a4d9b13c6e7c0547"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 14:11:23.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 14:11:39.000000000","tz":180},"subject":"[release] Use UCA Caracal","message":"[release] Use UCA Caracal\n\nChange-Id: Ib0e78581e6c34be650e59d8a044fe21b06ab45a9\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0972455ad9df0f99a06b5d410772129282add35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0972455ad9df0f99a06b5d410772129282add35"}]},"branch":"refs/heads/master"},"1ccccec9b9e0fb8b39f2c70d52e897a56063dfdf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-20 20:37:10.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/88/913788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/913788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/913788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/913788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/913788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/913788/2"}}},"commit":{"parents":[{"commit":"ff798d757bf1fbc684348f32bbf6ebde13ffc3bb","subject":"Merge \"mariadb-clustercheck: Add a script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff798d757bf1fbc684348f32bbf6ebde13ffc3bb"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 14:11:23.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 20:37:10.000000000","tz":0},"subject":"[release] Use UCA Caracal","message":"[release] Use UCA Caracal\n\nChange-Id: Ib0e78581e6c34be650e59d8a044fe21b06ab45a9\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ccccec9b9e0fb8b39f2c70d52e897a56063dfdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ccccec9b9e0fb8b39f2c70d52e897a56063dfdf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
