)]}'
{"id":"openstack%2Fkolla~913790","triplet_id":"openstack%2Fkolla~master~Ia7bb63dd799d13d8ca53d1a81db7ccda057b1559","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-26 07:51:13.000000000","reason":"Change was submitted"},"36216":{"account":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"last_update":"2024-03-26 07:51:13.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-03-25 12:48:41.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-03-25 13:26:15.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ia7bb63dd799d13d8ca53d1a81db7ccda057b1559","subject":"[release] Change RDO\u0027s DELOREAN_DEPS repo to Caracal","status":"MERGED","created":"2024-03-20 14:18:49.000000000","updated":"2024-08-23 07:00:35.000000000","submitted":"2024-03-26 07:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913790","meta_rev_id":"049e7bf8080afa66280fe939be2d86f093d6ad72","_number":913790,"virtual_id_number":913790,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 07:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-03-26 05:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-03-25 12:48:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-25 13:26:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-03-26 05:59:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-03-26 05:59:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 14:31:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-20 15:38:42.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-03-20 15:50:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 17:31:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-21 02:47:13.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-03-24 20:48:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"13aeb30df76a8a2e5839a6299839e872a2c97095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-20 14:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"669c808f3058f6e907f99bbc8b03d82341aee65b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-20 14:31:50.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"92cfe88b6d7f53c2f8b02baaf1b57b3de64944d9","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-20 15:38:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a46c44628ee5b7cbc997c779b4e80ae101de1482","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 15:50:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba7430fc728e438a8215b589119fb5ba\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ebab86b131b045cfa1e7e347da8c29f6 : SUCCESS in 23m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9b02add1469646849a83c897cfedb88f : SUCCESS in 20m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/247f2c7b2a7b4ce1a89e17774607ff9d : SUCCESS in 1h 29m 21s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5eed90758fe54ca99b20831b06e362a6 : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3f011800cca4f188909ba9190357dab : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa3c9746a59144a6b2041375f0088b65 : SUCCESS in 3m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1a504fbd2f60421fb01b085c2fac1de6 : SUCCESS in 1h 05m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7b9e7949aba145f9917d0a5b692b1a42 : SUCCESS in 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eacde77a4d6f7d26e5d64914c7a0fdfdc34617b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 17:31:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17d2aff96af1468bbb8f323e20ca0a70\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8a466bdfd24247768c0f3018658eca10 : SUCCESS in 23m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5aa7142947b24cd880165eb76118d116 : SUCCESS in 1h 07m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/46dbbcbd31e7432daa5c1a3849138b03 : SUCCESS in 15m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1f937704c9f145ac88d20cd294464f5a : SUCCESS in 29m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9dc9539dd7b241b397c50e620c941bf5 : SUCCESS in 1h 09m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2a51c896d1b34411bd5174487e8d67fb : SUCCESS in 1h 45m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6421b43934ed499cafe8652972049bd8 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d27f221f344465f8694d83291f8153e : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/464a07dc031b43d6b8d0121741ebbd9e : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4e95c4fc2e443e89f4063d66ed74697 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e87148a9ea42c198218c0221e3400b : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3abb8dd969f54080878a894e93300f1b : SUCCESS in 22m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4dfc12d24f264eedbd485b3369c118fe : SUCCESS in 23m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4bd9aa7639da490a92e5943751ff3973 : SUCCESS in 1h 09m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3da76d3db1b041f88faff53d082f7045 : SUCCESS in 1h 40m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/74b9fb0fb5124ea9994f166bcf5327bc : TIMED_OUT in 3h 01m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4c4e65073fe2453c9dcadfbc375c9348 : SUCCESS in 51m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/04a87232259b42b09f162fed20e725d2 : SUCCESS in 1h 00m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/837e9cf1a7214d0cabffc35490874754 : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c9c2aa72c9ef59847cf988898e8b400f08547c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-20 20:37:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"45d721fb2b8961ed686cbbb04cfefe3c30445857","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-20 20:37:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b6d0391b667adcc1d10e9540aa7b936087e014c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 21:53:44.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27367aaf3ff6436399f288ce260a0c7d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/70084720712747479aeeec93983acb84 : SUCCESS in 28m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/232fe0b22ab9416aa83d443891d3c15f : SUCCESS in 1h 05m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0e74ee8ecec44a349b002110fca75f73 : FAILURE in 32m 31s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c26ca18c21b842c39c216e078e30eea9 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e82fe91ec1404d92835ab5c6a22855d3 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/00715123da6845e49f3773701aa12e8f : SUCCESS in 5m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/050f739ee34c46f89e6934cc65e1bbf4 : SUCCESS in 26m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6af1f2db29f440dfb2bee84e22f40138 : SUCCESS in 1h 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1addb915172832597f2b7de757352ebd5e66f79a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 22:11:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61d677beb1434ac492e5bfd151b51bc1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/891bcfa3450643f38457ca27bd7260cf : SUCCESS in 24m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3b3bf2bc76b148b5854503e83c781847 : FAILURE in 41m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7ec287174ed14466992ec5c759086df2 : SUCCESS in 18m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e488e7d33561473d805f473bcf2eff74 : SUCCESS in 37m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b0f320122cdb4bfebe6f2d391c11cb48 : FAILURE in 26m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ae8db8f761d44abeaf740735dbec9799 : FAILURE in 56m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f052d39ab8864cc7954afaa32f5182ac : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d756282cf444c85aec8d844b19f71bb : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc0969abffde4a16bce8a3223ce38f53 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a2ef2da48f14898a3758b75964e64aa : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba917e051ff04233a9e0c2854c2a45cc : SUCCESS in 5m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b35ff6719078437583e5ee812aec3aa5 : SUCCESS in 23m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4ea47206f5504285ad86f6c9344defeb : SUCCESS in 29m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/04000a531f4e470380a6e709a6d7710d : FAILURE in 23m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6baaa01693124625bf968e00504dfae4 : FAILURE in 28m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3f9b9abe670e4cfaa7107bd1a217296a : SUCCESS in 21m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7ae9464db3f04c3a83bf2810f4e50c8f : SUCCESS in 1h 08m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/23f9fe2a3b834db98d53d1e37476243a : FAILURE in 45m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fc55d10ccced451abc48625ffdf1316b : FAILURE in 46m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d8a79ae80df8f77067e7cef3591716a5271ea490","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-03-21 02:47:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e9c1d93aa8b43c67cc1c073241f79bcb8c045a51","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-24 13:34:53.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1025838681601488158f68954492d61a9eeb580","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-24 14:59:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c1a73233bb472b954c1fd3d8527b0a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a3c27cc8d7c34f6a9f35068ea44eeaaf : SUCCESS in 20m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8448b33577441bbbf95ea508e4c8598 : SUCCESS in 17m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/042e5887c834488981595358d4506af0 : SUCCESS in 1h 22m 17s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e17ae2d5815147c6a0e08988a40874f5 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/999758d524cb47fe9d1b815791758779 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6b68115393504cdb81f9a728d658b934 : SUCCESS in 2m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/671a32ea3fc74b82be0b8871707b23c8 : SUCCESS in 1h 08m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ef32cd7a99d8437e97e7a6478af21a07 : SUCCESS in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44a6d5c26bb240ed487baca062e13a695218243a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-24 15:17:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b08dabee6b4aba9a853e3dd792d150\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d89ae7138598478bbf272ecaaed69319 : SUCCESS in 23m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/46aa6b3894344eedb558eac134608294 : SUCCESS in 55m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d5ca29a070284e70bb37532142f6f11e : SUCCESS in 14m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fbd058f9eea94cfeaea0fcd08842c74c : SUCCESS in 21m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3019260d85cc4a2188549356e6456e3e : SUCCESS in 46m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8dd9fef205f04c5a8e9d449c9ec18e8a : SUCCESS in 1h 30m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2992663bb3a745d58294e0bb410a69a2 : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed67516264be43f7990299bea9fcadfc : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f8226684a8145468448688493af546a : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/994d367fe09c4f76aa5ddc647285869e : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f2a3d3c72be4bc6b271b0ebd25ec144 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06894d0738284f0dafe330d7e408d7b0 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d500a415e8ac44c4ba87c2394510f7c0 : SUCCESS in 24m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/20d96e01eb6d491c9c1400a0c1dcb4b1 : SUCCESS in 33m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1677aba4c9ac46fcb8b7cb367f08f924 : SUCCESS in 1h 11m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/81c51d5d387e43809f4dfc933a92df3b : SUCCESS in 1h 36m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/557bffea5c634eca9dd5bba6a33c89a9 : SUCCESS in 21m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/74dae13b874847a29222c6fd76032996 : SUCCESS in 49m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ca899d62073b46e0b9d335ba3c29e737 : SUCCESS in 42m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8df4e202fab641ddb5ebba32a06c92b0 : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6ad9002e1c6e279d04af62fc308aaab834d62919","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-25 10:41:49.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"3e4495d41a796153bc158f22126ac5bb841d2ff6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-25 10:42:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87e8442e5e86c18fb694318f10d63aba1d679a4f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-25 11:01:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ec1ad88a761260acc356fc8ac45cbd5a1d14c30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-25 11:43:48.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"281b907e4b6b23d6a6921883df0995160430f9ca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-25 11:44:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dd2601bda0bec23330d95b4a57ecd796bd574491","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-25 12:48:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6f5b454988f5b54d832c599c3b5c2d8784607221","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 13:08:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4666ed491b54ee1ab528eca47ad212d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/94fbb74f639d47c18d1e8365730ba5a9 : SUCCESS in 30m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/232f4933fb2b4754948159e3a5cddb03 : SUCCESS in 1h 22m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2ee6277871e41f9b56566ad1d04fe6b : SUCCESS in 1h 12m 23s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c94a81b055f14ee5ae966e4f4b16553a : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/694f711e9ccf436db39ead65fb9281f2 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5ff9e8346e0f4c788b437db3ee8ef073 : SUCCESS in 5m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0834df987e1747569cd8e068d9341ee3 : SUCCESS in 27m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/250fabc303614275b73ca303c25d8309 : SUCCESS in 1h 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7263fd893a6507a030bc6c6f112064b41d366974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 13:25:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/723165180ef74d25bd0351f8f21c5b46\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4b157d01f6014221af9f63b1d4668037 : SUCCESS in 24m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ddadfb580dcc48cab9080b558c2ef14b : SUCCESS in 1h 00m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6726f71b29fe4af69bfa8a72a9bb4040 : SUCCESS in 21m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4293cd6d1a55406ba8739a87f0b710d8 : SUCCESS in 27m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/545ba53975ed46089035206327a4a3ce : SUCCESS in 53m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fb3d27445e2543f58979068755d14db7 : SUCCESS in 1h 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33343e515244437dacfc7d0713b13cb2 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81f4d92e76664541bd20fd8eb636e79a : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5567adce1a174c8cb2c28704502dece0 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8701af96ae4486e87d78b86dfb22104 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2bbb77f4ccb46be928c0d1dd4cb4118 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9fa0a107caa41cc86c2203645d89199 : SUCCESS in 4m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/69c5b192ccb5425e92fdfe2eecf01adc : SUCCESS in 24m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/040a747e94a2437e8f2bdee96a7ce20c : SUCCESS in 32m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ccdde01690a64ff2b0852e8e9c05b4a9 : SUCCESS in 1h 14m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ff840a5170ac401f95d2b8eb14b9cccf : SUCCESS in 1h 39m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/03ad85bd977d424cb5620f6ce907ef56 : SUCCESS in 19m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/944cc445ddc04d39a8ae92c515890bfc : SUCCESS in 1h 13m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9eca0028638b43d39cd06ca5ec0b2e7e : SUCCESS in 55m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/29bb038c4e6d4ffeb1766fab9e69176c : SUCCESS in 1h 28m 12s","accounts_in_message":[],"_revision_number":3},{"id":"7df821c5bda4a325e88127264f04eaa37a5f3606","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-03-25 13:26:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6dce702f774920bd4bc3f9368fcaf978c74e50c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-26 05:59:36.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f282ce359543d1f77b12f0d3e435cfd7c6407960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 05:59:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"56b5e3192af8ea0fa4521ba352a6ac141b6db41f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 07:51:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88072305af88429aadc25b3f896b1705\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/86ee32f3e4ee4a3787217d958df4e972 : SUCCESS in 15m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/de85a9b9236f4235a100176ffc3366b7 : SUCCESS in 22m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c78ab4624f504ab3b3b20549914d3b33 : SUCCESS in 49m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/712e3a0efad24c1b8e3f03f980843606 : SUCCESS in 1h 08m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50febe7ff44744759f4b02b576cd145c : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3cc5d17c3914011a3c7b376ca641156 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0453b71d2114aef9df06eebca18115b : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/351691879a9e4098bf12adb1f5904da2 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0788a1ef12b34015b03c437aa228b250 : SUCCESS in 18m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d913b1918ac94aeeac4f3ca57698107a : SUCCESS in 33m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5e491b90d9b349fd9d9157b1734cf094 : SUCCESS in 1h 05m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2855250ac80a4856bbc08a2c70023a91 : SUCCESS in 1h 46m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bd7d3cc33b4d47759928c44ecb6dd32e : SUCCESS in 16m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f773def36e07437aa5172b42a7296743 : SUCCESS in 1h 15m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e85fc841f6fa4e95849d675e19767baa : SUCCESS in 49m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f9b6496960e3435fb85be924e0e6ed36 : SUCCESS in 1h 29m 35s","accounts_in_message":[],"_revision_number":3},{"id":"03069619bea1a26367317788abefde4b035df9c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 07:51:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f7ced56a0ee6e851c1fd05431b1f0eb1cc2e5e8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 07:52:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10bbad9051b4eb4a0289a553a04823b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dcb6a101f0e4fa69c8da2bfc8bc1f7e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3},{"id":"049e7bf8080afa66280fe939be2d86f093d6ad72","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 07:00:35.000000000","message":"Created a revert of this change as Ia1142a5d67009c28ab820652785deb45263a548b","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8102d0278255bcc42129d1f51bc78284e110f178","revisions":{"17fefae8702750f818362f7c9879bc04509acb2c":{"kind":"REWORK","_number":1,"created":"2024-03-20 14:18:49.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/90/913790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/913790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/913790/1"}}},"commit":{"parents":[{"commit":"a0972455ad9df0f99a06b5d410772129282add35","subject":"[release] Use UCA Caracal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0972455ad9df0f99a06b5d410772129282add35"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 14:18:19.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 14:18:32.000000000","tz":180},"subject":"[release] Change RDO\u0027s DELOREAN_DEPS repo to Antelope","message":"[release] Change RDO\u0027s DELOREAN_DEPS repo to Antelope\n\nChange-Id: Ia7bb63dd799d13d8ca53d1a81db7ccda057b1559\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17fefae8702750f818362f7c9879bc04509acb2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17fefae8702750f818362f7c9879bc04509acb2c"}]},"branch":"refs/heads/master"},"53f0925e1e4acb2394ee5cb4742168f6b2a81ca7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-20 20:37:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/90/913790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/913790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/913790/2"}}},"commit":{"parents":[{"commit":"ff798d757bf1fbc684348f32bbf6ebde13ffc3bb","subject":"Merge \"mariadb-clustercheck: Add a script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff798d757bf1fbc684348f32bbf6ebde13ffc3bb"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 14:18:19.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 20:37:35.000000000","tz":0},"subject":"[release] Change RDO\u0027s DELOREAN_DEPS repo to Antelope","message":"[release] Change RDO\u0027s DELOREAN_DEPS repo to Antelope\n\nChange-Id: Ia7bb63dd799d13d8ca53d1a81db7ccda057b1559\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53f0925e1e4acb2394ee5cb4742168f6b2a81ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53f0925e1e4acb2394ee5cb4742168f6b2a81ca7"}]},"branch":"refs/heads/master","description":"Rebase"},"8102d0278255bcc42129d1f51bc78284e110f178":{"kind":"REWORK","_number":3,"created":"2024-03-25 11:43:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/90/913790/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/913790/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/913790/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/913790/3"}}},"commit":{"parents":[{"commit":"ff798d757bf1fbc684348f32bbf6ebde13ffc3bb","subject":"Merge \"mariadb-clustercheck: Add a script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff798d757bf1fbc684348f32bbf6ebde13ffc3bb"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-20 14:18:19.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-03-25 11:43:36.000000000","tz":0},"subject":"[release] Change RDO\u0027s DELOREAN_DEPS repo to Caracal","message":"[release] Change RDO\u0027s DELOREAN_DEPS repo to Caracal\n\nChange-Id: Ia7bb63dd799d13d8ca53d1a81db7ccda057b1559\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8102d0278255bcc42129d1f51bc78284e110f178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8102d0278255bcc42129d1f51bc78284e110f178"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
