)]}'
{"id":"openstack%2Fkolla~913795","triplet_id":"openstack%2Fkolla~master~I7f1ee493cf07909b5ed8fab89133c0eabcef2793","project":"openstack/kolla","branch":"master","topic":"bug/2058484","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-07-23 07:40:42.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2025-04-28 07:59:38.000000000","reason":"Change was abandoned"},"32398":{"account":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"last_update":"2025-04-28 07:59:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7f1ee493cf07909b5ed8fab89133c0eabcef2793","subject":"Fix bug prohibiting images build.","status":"ABANDONED","created":"2024-03-20 15:19:04.000000000","updated":"2025-04-28 07:59:38.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f9b8e1b04f5fa37f3a16c7296aee833a348fadd","_number":913795,"virtual_id_number":913795,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-22 12:05:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 15:19:34.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-03-20 16:37:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 17:11:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 12:15:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"68c95c292fef3bacd7ef0305a9ddf67420a94a4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-03-20 15:19:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3ace069cc710e4d6fe1008e4c96aa86f9de9272","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 16:37:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efef963ea8914704bc1804b68f01753b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c36f2a1aef30458ebf49655e130433b1 : SUCCESS in 1h 10m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ff7513c701c24a8a98855ac11a250e91 : FAILURE in 18m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a7fa8c792d514b33ba061d06b38fbb0a : FAILURE in 38m 11s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/497fa82fe2094d57b4e1a1efd9c98d1c : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a67f8c41aec48e49da6e29a42b07f35 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/54b75c5dae3c4c879edfe51c58410171 : SUCCESS in 2m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/61bf9f6d482143208cb7d7231ccae2ab : SUCCESS in 1h 16m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dae93c2d06394cf2a1f05d454a33ebd1 : FAILURE in 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9336c99109c7d83a1854fe6deed225dfd1514e1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 17:11:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c74bca26a6b745b6af4398fad23190c0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6be43be2bd214762ab5cdd0505388820 : SUCCESS in 27m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c311232c537d4d079a05c1c7f2a712d8 : SUCCESS in 1h 11m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3083761a5d5844618394498094ac901f : FAILURE in 18m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/740edc4b696c413c946627acb0c852d9 : FAILURE in 26m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/213ffcbb209d4112ac79356593439a1e : FAILURE in 12m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e1c15cb2675c4ac290d36ea240363145 : FAILURE in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90f5ba2ee88d46f981e52c42fc32d285 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b291d2d526fe434d9cb6e6edbce48a5a : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b50e011b1d0407493dda4bafb4f32a2 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dacf896027e4ae588f658f84b1074ac : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2958e9d7aa084e0ba45d39d69d7da76f : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1eb5d3adca384c8391f71232155039ac : SUCCESS in 25m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e2bd6c9ba4c9439aaf5c644daf58d230 : SUCCESS in 28m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ff44858eab474865afabb73c1f15c533 : SUCCESS in 1h 08m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/35c1669731fa434489b1cf6764ab4b3c : SUCCESS in 1h 47m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3bc8b948d72042e3aa9f53a302b5aaf7 : FAILURE in 15m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8f6b47929a2c4e52b888722c76320692 : FAILURE in 1h 03m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/162d3ae751034ed6aa80683a2a278bbb : FAILURE in 13m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/139a4ce79e9648fc9916fa5f45a26004 : FAILURE in 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"040ead31060f223779ba0a25028596367192ffaf","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-03-21 08:09:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91ca4db6e6611bd8e7a81055401d5dbba268d75f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 09:07:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74dde55eb48f4a55bb53c73bdf507285\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/94ac7dcb42344855a85d7909fc91c807 : SUCCESS in 23m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4848c9b0c7b0407c94641e7848278415 : FAILURE in 32m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/77a433632cd1467689fd595369b77cc4 : SUCCESS in 19m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bdac3efff80345b59d604d42516b8b96 : SUCCESS in 31m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e02b0d6d24b84ba6a95dcdba154470ca : FAILURE in 29m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5c4d44dd20024e56adf25c56b78b8bf9 : FAILURE in 57m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89643f88a32491a88dfca1c59d50b7a : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6025d9cd997e4510b2b3ea45540dfaec : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f955dc11bfc4926b69b7808deed97d7 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bc0633689134f38800c06635a6f9dba : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aef2e6e8e0842aebc4f30ce8b579a95 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/417a4ad7db3b44dd88440d1e66b310c5 : SUCCESS in 23m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ecdcb903f8024684b8e5e7b4d69383f0 : SUCCESS in 31m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/43e3bb22cd1240b5a9c75caf2227b2b3 : FAILURE in 28m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/413f72f14b1942d0881f44d2795e9228 : FAILURE in 46m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/72f0806db0434406b8c7251739c56cf7 : SUCCESS in 21m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5fe0efbfe5104fe7be45a09848bb75ef : SUCCESS in 50m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1b0df049ec6f4b52848f251ab736be6f : FAILURE in 35m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/681ce5473cb145308ccb0abec95d3459 : FAILURE in 44m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7d1685c9ef2f72b15d5a5ccf4ebd89ea6371351e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 09:34:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8e0fcefd6024aedb163974265cc24be\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/26a1ba1a84d547c9a720b77cfd5678ec : SUCCESS in 26m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/720acfb008a547e096a9362186d92f5b : SUCCESS in 19m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4e0a09fbe7cd4bb28bae70a3acfefecf : FAILURE in 27m 28s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/891c18a797d34870aef041432db6cee8 : SUCCESS in 2m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c7272b2e0f04b1abe44120b44daea74 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2d368228d27e44dba9046e78d260733e : SUCCESS in 3m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f3f2809bf0484adfb7fc451fb2e1b3f2 : SUCCESS in 1h 23m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9602da1df4fc47c4bec201d1d093e841 : SUCCESS in 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4aa8e6239c900cb183513ee87ee029ac65a04a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-22 10:19:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e86be6644226cd65e0941e1be200634bd1e8a6c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 12:05:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/849b6f06d38c4f918b1a135332a7c263\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9ca3ef3f33164fada35747f9a0049b03 : SUCCESS in 26m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/34551b8764814032847465f2b6f89374 : SUCCESS in 1h 10m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1ae74fe68ba74a7c8303a551e71900a3 : SUCCESS in 19m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5b12fff39a714117b5435d0c53aa6e3c : SUCCESS in 29m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8c2122b27a4b4b9ba268e40c331d737f : SUCCESS in 56m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a0b581720d5549e1a46213ae8a080dc2 : SUCCESS in 1h 34m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911aae91325e4a258a72af688779524b : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d14bbba616c45a8ae8cc7997d295388 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/709a497d4fdc4eeabfae5c4ec14cf9c5 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a6e33131b5d4cf6b89ada3a17ec0bec : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c8a1f8927d4437fb7694580646fd9d3 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0450cdf6753c4bd29b726d22d1481302 : SUCCESS in 4m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a76a58fc4e464c149620f7cad261e7d1 : SUCCESS in 28m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/30eb51c4389446d1b62be239aa6aa62e : SUCCESS in 38m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d47dd89fdf37430698abf65f3f9775ad : SUCCESS in 1h 07m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f32d87c673d8407e8bfff0b51b194c53 : SUCCESS in 1h 40m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b0420021fd5b4476bea90878fe17a4f5 : SUCCESS in 21m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f6d6f1379447458b9630418251f837f7 : SUCCESS in 57m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7428d4a84f994063a10516ebb87d1329 : SUCCESS in 52m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6e697511b9424c08bda3962983247b54 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":2},{"id":"462b183bedc60ee8b3749ac074ec95ebf709a2ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 12:55:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03d4c86c446b4217a276b5cfb3f6c510\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/454ff8a3cafa4670992465973580a7b4 : SUCCESS in 33m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3620fe752add478cbaee80bab71ea645 : SUCCESS in 26m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5ef9af30f411432e8ca76743ebc9c3cd : SUCCESS in 1h 36m 51s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98b250ab8bdf49feafc73793e709db94 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d8c9846f26b74c28bc12117c842fecd3 : SUCCESS in 2m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3134708a532642e492367e19a876a115 : SUCCESS in 2h 33m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e70c2dc44ac7494fb9eb114fded78a99 : SUCCESS in 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"716f9fffdf9d2341b7cfb6cbd919baa9d19d5241","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-23 07:40:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f9b8e1b04f5fa37f3a16c7296aee833a348fadd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 07:59:38.000000000","message":"Abandoned\n\nNo updates since July 2024 - please restore if needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"85f4c83f2f5822fa60c37f84198611349059a853","revisions":{"e5e7d9c8e047abca7d2963c286416b1f5a30acb2":{"kind":"REWORK","_number":1,"created":"2024-03-20 15:19:04.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/95/913795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/913795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/913795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/913795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/913795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/913795/1"}}},"commit":{"parents":[{"commit":"ff798d757bf1fbc684348f32bbf6ebde13ffc3bb","subject":"Merge \"mariadb-clustercheck: Add a script handling signals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff798d757bf1fbc684348f32bbf6ebde13ffc3bb"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2024-03-20 15:18:39.000000000","tz":60},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2024-03-20 15:18:39.000000000","tz":60},"subject":"Fix bug prohibiting images build.","message":"Fix bug prohibiting images build.\n\nCloses-Bug: #2058484\nChange-Id: I7f1ee493cf07909b5ed8fab89133c0eabcef2793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e7d9c8e047abca7d2963c286416b1f5a30acb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e7d9c8e047abca7d2963c286416b1f5a30acb2"}]},"branch":"refs/heads/master"},"85f4c83f2f5822fa60c37f84198611349059a853":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-22 10:19:45.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/95/913795/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/913795/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/913795/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/913795/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/913795/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/913795/2"}}},"commit":{"parents":[{"commit":"f63594bdc1dc429c4601bfe818ad5c7893a93f3e","subject":"Merge \"[docs] remove not working dev setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f63594bdc1dc429c4601bfe818ad5c7893a93f3e"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2024-03-20 15:18:39.000000000","tz":60},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-22 10:19:45.000000000","tz":0},"subject":"Fix bug prohibiting images build.","message":"Fix bug prohibiting images build.\n\nCloses-Bug: #2058484\nChange-Id: I7f1ee493cf07909b5ed8fab89133c0eabcef2793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85f4c83f2f5822fa60c37f84198611349059a853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85f4c83f2f5822fa60c37f84198611349059a853"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
