)]}'
{"id":"openstack%2Fkolla~914529","triplet_id":"openstack%2Fkolla~master~Iccf304a6e8b322b751e28156f4619dd559a979d1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"34213":{"account":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"last_update":"2025-05-06 18:09:16.000000000","reason":"Change was abandoned"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-18 09:06:59.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-06 18:09:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iccf304a6e8b322b751e28156f4619dd559a979d1","subject":"adding rpm/deb support","status":"ABANDONED","created":"2024-03-27 15:35:34.000000000","updated":"2025-05-06 18:09:16.000000000","total_comment_count":31,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"2bc6f45f51643d6cef2a041652f938d2d72c7f24","_number":914529,"virtual_id_number":914529,"owner":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-24 12:30:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-04-24 12:00:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-27 15:37:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-03-27 15:40:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-03-27 16:08:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-27 16:59:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-28 14:10:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"b09b0779f99d80c3e10d9786f84501e7cac0035b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-27 15:35:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62b4d9cd6a76829768f839c6d80102dcbe89a736","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-27 15:37:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16805425f3bd6bf5412fddd7ceccb7f6d93ff8f8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-27 15:40:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"14338a8b9be8594ec8545cc17c99d62be9d18eb2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-27 15:55:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/tasks.py\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"45abaa5d9479f0229f2a035d4f531dd8b7ffa6a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 16:08:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1e29fb9fc3c429eb0f6e179d54dace3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2352aacaeffb496eb3abac9d1720e519 : FAILURE in 3m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a43dc80bab03460d873936d82e444b8c : FAILURE in 4m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/787abc33dfbd432ca8cf53e539efb5b8 : FAILURE in 10m 35s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d437281d42843408acc771a7f0237e2 : FAILURE in 4m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a70309062bd46828bba62bd4b4ad33e : FAILURE in 2m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f72f2925c41b490ca1c86e265611e5df : FAILURE in 5m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6f6c0e9f1b154fe181a96198096f6997 : FAILURE in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"37cddabdbcddceab5d4041bec7f47d6e7f6e631c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-27 16:11:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"522f4397f2cd51be043622c3fe451c8422b28292","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-27 16:19:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"02821cb3c9f1584a9d58da2ff707bcf95e812549","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-27 16:40:04.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1223ab2e23ad642cbcef0531172611b89ce44fce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 16:56:59.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/950ef32e5b3341c2ad1003121eede639\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fd3ca05a1822446c90e95793b213dfd5 : FAILURE in 11m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eb8ce4dd9d55422490ec0db2f0e46b68 : FAILURE in 36m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a45ea98d595d471dba1126fee0a9be34 : FAILURE in 11m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7243bfd3e0af4460ba54059584bb5e90 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/719f59798cac429ba7a91f80635f03b2 : SUCCESS in 2m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9fa98fe7c40d4c81ac196cefa2f24e81 : FAILURE in 31m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/aa79de5ed09845dd9e461f55365d3b5b : FAILURE in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99dc2fda6d880aa1729f3fbc41ec49279774a34e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 16:59:04.000000000","message":"Patch Set 4: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb2b5af4bfe04c2aa4cc6b254762c32d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0c57280669964847bddd1809049e0f76 : FAILURE in 13m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/773a004bbfbd4df6bb3e92670b95be27 : FAILURE in 15m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b162a1c1921e43309f6d142bcc38748a : FAILURE in 8m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d0755ee7409443bfa7a4b355bb4ab3b3 : FAILURE in 11m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/18df73a0607a46af84e126a55469e240 : FAILURE in 17m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/825ec74295c0449c8917f0aae63ece0c : FAILURE in 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d8c2dee2f84831af24d3d44b089a08 : FAILURE in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24e862f99c354ccf813d6bae5dab0081 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9289e76c64e44f5bd8aadef263f4369 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8d1f12ce65b4d9cbe2db798827585da : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fef9cda9ba3e48c2ab3db4bd3007f812 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bae9ad5bc4d407f95399fcf84c88f7e : SUCCESS in 3m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3629da8064d44c59bbc77647e9d98e04 : FAILURE in 12m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8f7b57c82a2a4d2b94a364e6b7568f4e : FAILURE in 17m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d71d31ca1946400d93d14e1c8613de59 : FAILURE in 14m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/151c7325e47744bebf1408c20a0ebae3 : FAILURE in 13m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b1f81a33c46c47998cfc271d4f681b2c : FAILURE in 13m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/abe38037855647c2985bc8526a06f222 : FAILURE in 17m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/40afdbbce98940c49d47e756bee12af8 : FAILURE in 11m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/64309dd5dba642b896d2eed7f4d1c5ab : FAILURE in 10m 54s","accounts_in_message":[],"_revision_number":4},{"id":"2edb0b822a0c0bcc0a14fafd3e7d8ece4ee74a7f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 11:18:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"51fc410e37bdc94b8ac864641004bbb19acebf6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 11:46:02.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d755f08e4b4615937ec157a6768619\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/83fbe7bb0c754429961b171914b73864 : FAILURE in 8m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/69d7a13f1d1344f7a7556152ab15b18b : FAILURE in 17m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/86af69ae121e4d5dbd3499a97451f085 : FAILURE in 24m 30s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/713350741bc4428b8d50cd9a96d7768e : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2b446a29b85744b5b125e0ade1c92b45 : SUCCESS in 4m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/03bcdb41fe0345b9b4c7d010261a0e8a : FAILURE in 16m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fda0658c4b744b8384ae2956b11fb886 : FAILURE in 12m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f9755447aa8a6254d076dba6353c18f9d6527928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 11:47:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e92b071b12664ab19c2de19615b10f6c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6fffc1023f714b0daa432b901948eff4 : FAILURE in 16m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/75d162abc5e74d54b5133fe298ff4e9a : FAILURE in 13m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7d18b1b60ca3427b91853932c856fc36 : FAILURE in 8m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e1fa30de203b4d47937ac182997a6618 : FAILURE in 11m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/22b792bbb11b4cf9a7b5e0a2cd03d2a1 : FAILURE in 9m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7594db636b0c4a568652b639c661bcec : FAILURE in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f02a2f95a2446ea9ac75f7966a0bc7e : FAILURE in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f564da85f5f45079b6d384dfe1b222a : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2db67c3d76fe4f27bedbd6b5445975a7 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/843492cdd6344802b2653a630f7b0ffd : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/420a6aed7b1e4880ad223c1887b67513 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c11b7fd762f47c68cf3d2a69bcef118 : SUCCESS in 6m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fd3ecd92bd82483cad1f08e518061ee9 : FAILURE in 12m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7ecbcc974aef452885c945b96858e72b : FAILURE in 14m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b2dbf00a59b948e2966f8113435c2796 : FAILURE in 17m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6383c57fb8eb45c5b879797887b08259 : FAILURE in 14m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b22a9d48ec9d48bea18d267fcdb7a5a5 : FAILURE in 10m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/74489eb1bfd84e049d3da2cbc7ec149d : FAILURE in 20m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5cb5311e001f462c816cb92d019cc6aa : FAILURE in 10m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/645127fd233745dd9aa8452ea5f8af24 : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":5},{"id":"d7e983673a99825f774f49c178bc6e299e802c72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 12:04:53.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/tasks.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8afbcfabaaeb0da96283258aab4d15213faeea64","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 12:09:46.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"b003b457fb5fa7ca99132cf88b89b7a0a330cb7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 12:10:06.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":7},{"id":"f6f490bbf3522050d645d171a66da1f56a3bee88","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 12:10:07.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/tasks.py\n","accounts_in_message":[],"_revision_number":8},{"id":"48df7f52943aef5ce5a644125c14c1557e452764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 12:10:08.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":7},{"id":"7227eaef4da46b2b501a817c6616a0fb91afd062","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 12:10:41.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"5f1df1795951ac17af3865c1c21f58d16ceb326a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 13:50:25.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"b254188b6cbcc545f4d7133f6254ee40c879f7e3","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:08:41.000000000","message":"Patch Set 10:\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"332fa61bed8a637744164361ecd31ac997a695cb","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:11:36.000000000","message":"Patch Set 10:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":10},{"id":"e70dfa441bac68b5d6de5578ea30b13d453ede4c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 14:13:07.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"af2a377bfca24d6c303180cb2539f9adaff44985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:14:54.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"78d329c03d80147d8cdd9051c0d754db6713cc2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 15:54:20.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/914193 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"fc957ce69e4650c85f88826aa793d5c65c84fc6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 15:54:22.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/914193 is needed.","accounts_in_message":[],"_revision_number":11},{"id":"b35390361006b79dc9112ec7b2db34880f03fb73","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-29 08:45:36.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/tasks.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"17b9ac4d698b726e3526a25fb7717551f649a6e7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-29 08:52:54.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"ea4abe32bb401f3e97baf54d365ae7621409362e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 10:32:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8949847e65d64c0d918d2b34444f09a7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8f4059bc353a40fc8533277bd585f020 : SUCCESS in 1h 20m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79d15402a1ce4dbc9c4eae790e24b43a : SUCCESS in 24m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9b2ab3c78da24b39a6dd2f18942bb74c : SUCCESS in 1h 37m 30s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d6bc9e4e70444c09963f1a3f1c86393 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee543a4c88c44b2ab41d5fb3b28cde76 : SUCCESS in 3m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ed670f654c40486f83558bc2fc7253c1 : SUCCESS in 23m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f87af02bbc9f40ea82cc7d7a51f5b7d9 : SUCCESS in 1h 08m 27s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"62235ec4e64a151ccc9b675445754b569f972928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 10:47:38.000000000","message":"Patch Set 13: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5eb157f1bff4fa69f382d917741b0f3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9fec934fffc94cfcb2ca878af10c7339 : SUCCESS in 26m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/eb3c3379b15b405fb454012627dc8296 : SUCCESS in 1h 04m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c3d204ace00142d6a25a1249ee7b1521 : SUCCESS in 18m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aa15b849abe1464e8940156e51e87eb0 : SUCCESS in 23m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/857a23a306a64db2b8395bcf849cf7ff : SUCCESS in 1h 02m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4bd4ba9dcf70466683500f49998e7116 : SUCCESS in 1h 48m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c3979e710e442ebfb9aa9445167d2a : FAILURE in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4562406fea014de1a1d47d99347cf29f : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6416f471a544da9ba480b0ce0be4ae8 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d5be8f494804e2a8afe280e4692a599 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b77ea658f7e54b28b8124dad2501becf : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d9c11bffc148b48bdeab6dee932d21 : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/377f12c3847d405ba3ee986e056abe8f : SUCCESS in 26m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6ff80bc138914369ba1f21873dcaa925 : SUCCESS in 30m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7b6f763faeca4d44a797998917dd021e : SUCCESS in 1h 12m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4db20497690544b291b15a76459bad49 : SUCCESS in 1h 36m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3133d932d7c84fc5b421bd52c2f64846 : SUCCESS in 27m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/91f4f65d0ca94480b5262a7865497567 : SUCCESS in 1h 00m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e721645fe3d54375885ece5263ae1a6e : SUCCESS in 54m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6dccbfd5d7cd4f84942eca8dee70b1d7 : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":13},{"id":"0758eaf27530fb32c18b20e1261f7b25b36684ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-02 13:24:31.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"451dbe08e1ce214b9d63611a23a990d43e37e8bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 13:47:16.000000000","message":"Patch Set 14: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/347670dd787c40db9792063a92e81e01\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a0c332477dd14e98a24ae0432b0608a6 : FAILURE in 14m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/409f1b72818c4269b9292defc03b61a0 : FAILURE in 13m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4b625647ceef4a09b1802dc59bdf0b1e : FAILURE in 8m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/41aecb9e9b994769867b3fe95ee2364d : FAILURE in 12m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d0655d6d32cb418c80db8bf6e447f828 : FAILURE in 11m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/28680b0c912f46a8bf19f3917cdd77f6 : FAILURE in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e660a6de2294698898bc2fe56d46d16 : FAILURE in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea7c961b378246c1ad83d5f5d0b0a757 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4287dae833444e4699087e638c386ee8 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/827268cecc42464ea5c8f897c354a937 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/145ddfd25aef4dada078497a6a96a8ac : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24596d4421524e6aaf4e47ec3e33d1e4 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/15103f14528e4a0ba2dfcbe778cd160c : FAILURE in 14m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4625aa1b3d5446fa9b18624a6d4ca71e : FAILURE in 17m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/83e9a2e08a7f4d06897383963603d4cb : FAILURE in 13m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/641202bdf4be436dbe14c46603cc23cc : FAILURE in 15m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8a3755584c4c4535a127e3b3b61b86d0 : FAILURE in 13m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f5518cdc69994970b58933bce17fe8d6 : FAILURE in 17m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e3086091dfe348c1bd24ac70f52a1cc9 : FAILURE in 14m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/841a757b90204633b389a3d7401c5db1 : FAILURE in 13m 02s","accounts_in_message":[],"_revision_number":14},{"id":"d28bea14ed31346a1e8cc7e1feb4d2388dcde7a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 14:11:02.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/922e53ccfa0c4cf3aef93947156c60b3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8dd3421dcd9540c9925f979937d3d0f0 : FAILURE in 41m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ee6bb8f7c694954ba01c2fb3586d268 : FAILURE in 44m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f740f06f7e345d89c88cc5aab4c90f9 : FAILURE in 10m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/172f1b899834438190d650fe92b96ca4 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e432f14ac40c4e109106d89512fce8ec : SUCCESS in 2m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/eece5f49c32546bb88bc6e75ccfe0f6f : FAILURE in 39m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f5601e3318ba45819cc69225dd47cb1c : FAILURE in 13m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f8118671e40e45a88f4562151e26540dc497ff66","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-04 10:29:14.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"8e78492d6674b9d4b2e5666f25d5aa86270b20bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 11:57:13.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64de0b4dfd8646ed81a135785672c68b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6186d8ed3b4f4103ac69e4e79ccdb623 : SUCCESS in 35m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6ab8aa6d71d64b8793fe1160e7b54b70 : SUCCESS in 1h 23m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/263ef2dc14fe41528cfbd8a05e3a50de : SUCCESS in 1h 21m 58s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65a236a81a8d49f0bea4c13266a26eb8 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0851185b1bb34a00afce368ac3e97515 : SUCCESS in 4m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2e243783ee294b9387bb7e3087e34a2c : SUCCESS in 1h 23m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ebcc64067314840b83176db07345dd0 : SUCCESS in 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"56171a6263698798ef77aefd1a5c43a13043b165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 12:18:27.000000000","message":"Patch Set 15: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1db5cc47112e42b08843730b19cf7483\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ada35c2f63d64891b2c26269bbc9295c : SUCCESS in 20m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5465aa0577764cddb30b5cca087f49c6 : SUCCESS in 1h 05m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b81ea2390a114afaae9c9f3c1b1c3dd8 : SUCCESS in 20m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/37a90985dd2f42ce8dcd5c6aa6242916 : SUCCESS in 28m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/515c58de26b845f894f630bfe120ade7 : SUCCESS in 1h 05m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/356f4f9bf68948a191f2b6e0f5a6a044 : SUCCESS in 1h 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a3343460d548c5a189dca3839e9400 : FAILURE in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72dc4dbf809e44be9cf7d59f4be05df3 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3cbd921fba94c62bf2c12008b0f84b9 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04e0ffc1e41b4d59aa733f5f7ebbe84b : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/156d5bbf3bc5436b8621f5549f62c0bb : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1ee9f2d4f64ac3a21f05479c14681d : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7a118f24168b4747b0aeffebc386ab85 : SUCCESS in 18m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f3395ddcfb7c4382ac0bfb514baba45d : SUCCESS in 25m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c198cb7b05e843fa801d714d3b45ef15 : SUCCESS in 1h 05m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dd5515b80acf45309e1cf71ee65c09a2 : SUCCESS in 1h 34m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/735b910840aa4d4aa66d493a2cb576ac : SUCCESS in 22m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/90346fce6ad147eaba716eb33b705203 : SUCCESS in 54m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2f04363c9cd3414891171a9fa1628ab0 : SUCCESS in 59m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bf2653f94f264ad9926da2430d3c9b38 : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":15},{"id":"d5a5c385a3d5a3acda226d956fdb7470f9e1a14f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 14:21:19.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nThe following files contain Git conflicts:\n* kolla/image/tasks.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"1438c4ae8bcc9ae505587fb9cd33a6e43d767f2a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 14:22:02.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"04652b859be0d4f3891a37d6c26265d9d3208d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:01:00.000000000","message":"Patch Set 17: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0bbf8f729434965b0b8d223957999ee\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ad32bad80ac64b4cba24e72fa9b37882 : FAILURE in 29m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2cd9c96aa08c444fa6ed8c15034aea1b : SUCCESS in 1h 08m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c4c3e27059774c88b312b6d9e61ecfb1 : FAILURE in 23m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/64fdaea01ee44092b8f7ab6795a2b7cc : FAILURE in 34m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5f9c44dfcfc7440f9e070a002a6ee94a : SUCCESS in 1h 05m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/39a04b76eaf8421083b75d5ebffcb735 : SUCCESS in 1h 29m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7961360c614747b78f088e8a9638e83f : FAILURE in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e2e7c972b96400e864950e269f908d1 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97cb7fe0630a49efa6fcb98afca82d96 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b01b7012c4cb43b2a1c976ab3a81340c : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aca3050d3ded4e02bf17ca393edb50a6 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05461069e01648dd874430a2c8b3daff : SUCCESS in 5m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/66e3fa0a8354483ba303aa74453b3384 : FAILURE in 29m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c1c14bd2fa3c49baa55d8a78c3465f2b : FAILURE in 34m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/47c3cea7729049168f81e7ad8a78065a : SUCCESS in 1h 08m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/df789423b76d44d2aa073226b0824033 : SUCCESS in 1h 32m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ba488021861f428d9f8c2383fd753570 : FAILURE in 20m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/06c1feb59a824036a95ff3fa8c64761b : FAILURE in 53m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ec9f588f41f94f07b3eb5e2d3b27c802 : SUCCESS in 50m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1507baab7f354f32bc1e44d2a3942845 : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":17},{"id":"0d64f363e6d1768fac8905b55bb80cb2a7191049","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 17:45:07.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f5c2956ac4454ab4d309b09060e691\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8a72451c261342f7b65c183f3b35143c : FAILURE in 46m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/af2ad4f1253845bd833c02438ded4bd1 : FAILURE in 43m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5f91ca4dbd1041598b0c92ae5cca747a : TIMED_OUT in 3h 01m 30s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d99cdc81699844889532d7cc009c6b90 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5b3eef37323c4024a2d1ceb3bdf3fede : SUCCESS in 3m 59s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/89a92aa9a3e94078a4fa04ac6ea3e3fc : FAILURE in 38m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/766ba65a49384110a5075a186d0d899a : FAILURE in 37m 27s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"804f996e560720c0945935c1dd2576ba2422bcf6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 12:47:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":17},{"id":"d968d40b60d953e34e752cbecb173beb1f9f5ee2","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-18 08:38:14.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"9902bad6c4b9c3967f3d4e13009c0022473fb9c1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-18 09:06:59.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"1eda58b2d958dd9dc568553b34bfacd55d7c9f7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-18 09:26:37.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"f0371d26371f355b8c45a1ebcae26ce7ccfa4add","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-18 09:32:14.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"e0a5c8c83230b4015cc3ca7d9ab09dbdba234767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 10:51:42.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea97f58eaee5473b8cbd18e32b2299d7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/50395324be4a4ba2819b4060666be06a : SUCCESS in 38m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1190e3ffc34d432cb44d75de2a7ae8fd : SUCCESS in 1h 01m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/53952f86a5844ef8857274f8a2943f42 : SUCCESS in 18m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/51fa25a005bb43489ef3992ee52ca78a : SUCCESS in 25m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3f0bda914ea74b6db728e430a6f92936 : SUCCESS in 54m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8c19634e6abe40909c3a3a406b509028 : SUCCESS in 1h 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c061aa35c3af4b11bb55863dcbfc289e : FAILURE in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a85aedcee4e41449ec3cc476f7f91ff : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5ee5f279cb641fb8c4f0677f61c0792 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45a5d72765354c5093cecca3abe3477c : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f931f3bd020494591963edc138fc8b4 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2abbee65b41447cda73f40822c3db480 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/044a7bfe81a040fa96aeabcc2e4adaf3 : SUCCESS in 23m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/117bc92ac0b34263884536a9db69336f : SUCCESS in 39m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/be37b147f31944c7aa3f6e2b313263ca : SUCCESS in 1h 08m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/75860c90bff94fc6b60a276864716405 : SUCCESS in 1h 19m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5e913e5ec6ae4df2bb9282f4a2ccae98 : SUCCESS in 30m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6ed6d7c7905c4ce688a1c2f7e53672b8 : SUCCESS in 47m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/06840d1d4b64475495232286a1033826 : SUCCESS in 51m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/285fa9ba15b6413cad9ecdddc6c808dc : SUCCESS in 1h 22m 55s","accounts_in_message":[],"_revision_number":18},{"id":"21fddb5d6877eabda75a2fac9210b62d1b1db7a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 11:38:23.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/374c53529d984e31ae96eff4789564c2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bc61502c25a34904b2c4f7e3daa9a29c : SUCCESS in 23m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc2990d55b0a41b68bf381df92e7f094 : SUCCESS in 24m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/63a294cf70ab4cb88ea2d00cdc4a58d1 : SUCCESS in 1h 57m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38932ecb7e594730a8ca9f3fef60b02b : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3fe2303e79894d3a8039ba97cac8652b : SUCCESS in 4m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bc7d15c984e8470787e7cbb7a4a204b1 : SUCCESS in 25m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9d8f3437d9284242ae60f0d8a113190d : SUCCESS in 2h 09m 32s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"bbd9aa24150f5dfcc8c858482f7ab7335abce8a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-24 11:50:11.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"b893e48d75f3769ce9df8df3528c094523ccebb2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-24 12:00:09.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"8bc837eb3e6c494d0a079610b1900ffe5628e55e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-24 12:00:29.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"4fb2fb7b991e25394816a537b3858a4d0a527818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 12:30:54.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8a137fad0344e19ae1e57d4d67373fb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/107e6fb3eacf4ee0bbc9ec2497693c57 : FAILURE in 17m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/eedef34c8d4d4221b5c88befcc3d82ba : FAILURE in 17m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/17c0983718db47e2aefc9dc9448240f1 : FAILURE in 12m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/54c88020d9ce435e84c732a30f647d44 : FAILURE in 16m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/64ba6beb12fa411f804337fdcbe6a1fd : FAILURE in 13m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5287eeb800e44749b6d2f58d8b04b8d6 : FAILURE in 21m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/310d78f029e14000893c1a4d34fd0b36 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fd185a0d51e49c9a21b349eb600fdef : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98bfba26a1824179a577546ec4638088 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cafc20d2c93e434d8b6c78084ee350ce : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb62ac07a15640109a6db7300295c11b : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9c1de17472470082004ddb73342b87 : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d6dfc796b6004c30a89121e7e311b34a : FAILURE in 11m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/812b90c83cc44705a5325f2996276c16 : FAILURE in 18m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f33cd1e76dc245eb9ca83aaf4e38ee20 : FAILURE in 17m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/58d569e848bf49bd90488d8927815dee : FAILURE in 13m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ca24b0a7e88d4d7c90451c30a42d98bf : FAILURE in 7m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/55549534bfaf4910b30c76a44aceaa74 : FAILURE in 17m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f76df91edfea42ab8582dca441058115 : FAILURE in 10m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e5f41526816144708cee607103141fa4 : FAILURE in 11m 24s","accounts_in_message":[],"_revision_number":19},{"id":"269a9882fedd4b86a875cfd4b5e00282fb2915e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:53:59.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd1e02857b0441d1a099b27783bbbd01\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5c8ef8145e8c4c129918d804045fdf63 : FAILURE in 1h 33m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/06cd35944ca44f7d9631b42254309905 : FAILURE in 17m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ebcb281476b14a758e46844723b13796 : FAILURE in 23m 01s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54d71f008b0245b4bd046482947e33ce : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f487c688b39f4ed0849b7e932cd461dd : SUCCESS in 4m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b2f9a83b9ead4129964d2eab49e998d9 : FAILURE in 19m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f31b02a2837d4d17885fbbddb716625a : FAILURE in 2h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"2bc6f45f51643d6cef2a041652f938d2d72c7f24","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-06 18:09:16.000000000","message":"Abandoned\n\nno updates for over a year, please restore if you want to continue","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"d74c8224d0dec99893824520b3b9a0905e2799a6","revisions":{"8ce8772730f61da82b32d4d167c212f53d98251d":{"kind":"REWORK","_number":1,"created":"2024-03-27 15:35:34.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/1"}}},"commit":{"parents":[{"commit":"2576fbde000484502e88b58d372b4c193341ed8b","subject":"Merge \"Bump proxysql to 2.6.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2576fbde000484502e88b58d372b4c193341ed8b"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 15:35:06.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ce8772730f61da82b32d4d167c212f53d98251d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ce8772730f61da82b32d4d167c212f53d98251d"}]},"branch":"refs/heads/master"},"53f6788f26602632e04be6281eaa6fe9309921a7":{"kind":"REWORK","_number":2,"created":"2024-03-27 15:55:22.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/2"}}},"commit":{"parents":[{"commit":"3ba723fe313df6a2b02fb26549d334d32e024351","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ba723fe313df6a2b02fb26549d334d32e024351"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 15:55:22.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53f6788f26602632e04be6281eaa6fe9309921a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53f6788f26602632e04be6281eaa6fe9309921a7"}]},"branch":"refs/heads/master","description":"Rebase"},"34ac1cee0fd463181044e208836bf14c2035729f":{"kind":"REWORK","_number":3,"created":"2024-03-27 16:11:38.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/3"}}},"commit":{"parents":[{"commit":"2576fbde000484502e88b58d372b4c193341ed8b","subject":"Merge \"Bump proxysql to 2.6.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2576fbde000484502e88b58d372b4c193341ed8b"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 16:11:20.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34ac1cee0fd463181044e208836bf14c2035729f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34ac1cee0fd463181044e208836bf14c2035729f"}]},"branch":"refs/heads/master"},"ac251602a6622567162b3856712b4a4638d0be5a":{"kind":"REWORK","_number":4,"created":"2024-03-27 16:19:14.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/4"}}},"commit":{"parents":[{"commit":"2576fbde000484502e88b58d372b4c193341ed8b","subject":"Merge \"Bump proxysql to 2.6.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2576fbde000484502e88b58d372b4c193341ed8b"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 16:18:58.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac251602a6622567162b3856712b4a4638d0be5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac251602a6622567162b3856712b4a4638d0be5a"}]},"branch":"refs/heads/master"},"ddf3d2f48d27beb7b7d26472de303db6a42df4cd":{"kind":"REWORK","_number":5,"created":"2024-03-28 11:18:39.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/5"}}},"commit":{"parents":[{"commit":"2576fbde000484502e88b58d372b4c193341ed8b","subject":"Merge \"Bump proxysql to 2.6.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2576fbde000484502e88b58d372b4c193341ed8b"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 11:17:48.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddf3d2f48d27beb7b7d26472de303db6a42df4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddf3d2f48d27beb7b7d26472de303db6a42df4cd"}]},"branch":"refs/heads/master"},"f8762ba744bba80cd4abce7fa25f3e573843dfdb":{"kind":"REWORK","_number":6,"created":"2024-03-28 12:04:53.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/6"}}},"commit":{"parents":[{"commit":"a9027e0bc62e54d8a42757683489c5a6c2bc528f","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9027e0bc62e54d8a42757683489c5a6c2bc528f"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 12:04:53.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8762ba744bba80cd4abce7fa25f3e573843dfdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8762ba744bba80cd4abce7fa25f3e573843dfdb"}]},"branch":"refs/heads/master","description":"Rebase"},"e70a549895c2cfc3ea870c85a7a072bd24b9b221":{"kind":"REWORK","_number":7,"created":"2024-03-28 12:09:46.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/7"}}},"commit":{"parents":[{"commit":"a9027e0bc62e54d8a42757683489c5a6c2bc528f","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9027e0bc62e54d8a42757683489c5a6c2bc528f"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 12:09:46.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e70a549895c2cfc3ea870c85a7a072bd24b9b221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e70a549895c2cfc3ea870c85a7a072bd24b9b221"}]},"branch":"refs/heads/master"},"556cc610daeee4c2f80ea8fcaf352477f5b1bc3b":{"kind":"REWORK","_number":8,"created":"2024-03-28 12:10:07.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/8"}}},"commit":{"parents":[{"commit":"dedfe791a6754e5056d595891b48fc6daf9a8cc2","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dedfe791a6754e5056d595891b48fc6daf9a8cc2"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 12:10:07.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556cc610daeee4c2f80ea8fcaf352477f5b1bc3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/556cc610daeee4c2f80ea8fcaf352477f5b1bc3b"}]},"branch":"refs/heads/master","description":"Rebase"},"a5d0901626a05d16a186f4bf766817f4d048f7d6":{"kind":"REWORK","_number":9,"created":"2024-03-28 12:10:41.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/9"}}},"commit":{"parents":[{"commit":"dedfe791a6754e5056d595891b48fc6daf9a8cc2","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dedfe791a6754e5056d595891b48fc6daf9a8cc2"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 12:10:41.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5d0901626a05d16a186f4bf766817f4d048f7d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5d0901626a05d16a186f4bf766817f4d048f7d6"}]},"branch":"refs/heads/master"},"5a4a766e86ce3230a2061c9b5e2857f4b80b3c72":{"kind":"REWORK","_number":10,"created":"2024-03-28 13:50:25.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/10"}}},"commit":{"parents":[{"commit":"dedfe791a6754e5056d595891b48fc6daf9a8cc2","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dedfe791a6754e5056d595891b48fc6daf9a8cc2"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:49:59.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a4a766e86ce3230a2061c9b5e2857f4b80b3c72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a4a766e86ce3230a2061c9b5e2857f4b80b3c72"}]},"branch":"refs/heads/master"},"2fac27c754fca4dc02585d41bd139aea27374379":{"kind":"REWORK","_number":11,"created":"2024-03-28 14:14:54.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/11"}}},"commit":{"parents":[{"commit":"dedfe791a6754e5056d595891b48fc6daf9a8cc2","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dedfe791a6754e5056d595891b48fc6daf9a8cc2"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 14:14:53.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fac27c754fca4dc02585d41bd139aea27374379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fac27c754fca4dc02585d41bd139aea27374379"}]},"branch":"refs/heads/master"},"77707440f3ab6eb288fb972af5a574c4d10f3db5":{"kind":"REWORK","_number":12,"created":"2024-03-29 08:45:36.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/12"}}},"commit":{"parents":[{"commit":"b90cdf7c678b454b6ca584c2447f02e2cf440031","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b90cdf7c678b454b6ca584c2447f02e2cf440031"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-29 08:45:36.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77707440f3ab6eb288fb972af5a574c4d10f3db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77707440f3ab6eb288fb972af5a574c4d10f3db5"}]},"branch":"refs/heads/master","description":"Rebase"},"dc5266b6d2455c176fd1ddf9daff0e460be06493":{"kind":"REWORK","_number":13,"created":"2024-03-29 08:52:54.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/13"}}},"commit":{"parents":[{"commit":"b90cdf7c678b454b6ca584c2447f02e2cf440031","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b90cdf7c678b454b6ca584c2447f02e2cf440031"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-29 08:52:54.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc5266b6d2455c176fd1ddf9daff0e460be06493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc5266b6d2455c176fd1ddf9daff0e460be06493"}]},"branch":"refs/heads/master"},"8aff5660700116c8fa1a4f2ec80835339ba880b2":{"kind":"REWORK","_number":14,"created":"2024-04-02 13:24:31.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/14"}}},"commit":{"parents":[{"commit":"b90cdf7c678b454b6ca584c2447f02e2cf440031","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b90cdf7c678b454b6ca584c2447f02e2cf440031"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-02 13:23:57.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aff5660700116c8fa1a4f2ec80835339ba880b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aff5660700116c8fa1a4f2ec80835339ba880b2"}]},"branch":"refs/heads/master"},"ea6700873082ad9f53b7e3876d0ae4af420c0027":{"kind":"REWORK","_number":15,"created":"2024-04-04 10:29:14.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/15"}}},"commit":{"parents":[{"commit":"563c921b6c646ab9fbc845df70fed5809fe8c387","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/563c921b6c646ab9fbc845df70fed5809fe8c387"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-02 13:25:26.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea6700873082ad9f53b7e3876d0ae4af420c0027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea6700873082ad9f53b7e3876d0ae4af420c0027"}]},"branch":"refs/heads/master"},"00d253d62fda7ae3a674de6942a2225e5f5dc9c6":{"kind":"REWORK","_number":16,"created":"2024-04-15 14:21:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/29/914529/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/16"}}},"commit":{"parents":[{"commit":"1df1a25c57cba9aec0e0ef698829d96226ed0519","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1df1a25c57cba9aec0e0ef698829d96226ed0519"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 14:21:19.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00d253d62fda7ae3a674de6942a2225e5f5dc9c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00d253d62fda7ae3a674de6942a2225e5f5dc9c6"}]},"branch":"refs/heads/master","description":"Rebase"},"897c6b51f80e96da38d6ad95c749cac0f0a34d23":{"kind":"REWORK","_number":17,"created":"2024-04-15 14:22:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/29/914529/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/17"}}},"commit":{"parents":[{"commit":"1df1a25c57cba9aec0e0ef698829d96226ed0519","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1df1a25c57cba9aec0e0ef698829d96226ed0519"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 14:22:02.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/897c6b51f80e96da38d6ad95c749cac0f0a34d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/897c6b51f80e96da38d6ad95c749cac0f0a34d23"}]},"branch":"refs/heads/master"},"d5ec0d2ceb113d55b19faa97414c224c46025180":{"kind":"REWORK","_number":18,"created":"2024-04-18 09:26:37.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/18"}}},"commit":{"parents":[{"commit":"1df1a25c57cba9aec0e0ef698829d96226ed0519","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1df1a25c57cba9aec0e0ef698829d96226ed0519"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Kayobe deployment user","email":"grzegorz@stackhpc.com","date":"2024-04-18 09:25:54.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5ec0d2ceb113d55b19faa97414c224c46025180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5ec0d2ceb113d55b19faa97414c224c46025180"}]},"branch":"refs/heads/master"},"d74c8224d0dec99893824520b3b9a0905e2799a6":{"kind":"REWORK","_number":19,"created":"2024-04-24 11:50:11.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/29/914529/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/914529/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/914529/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/914529/19"}}},"commit":{"parents":[{"commit":"92e7a73417b5d0a93c36d830df4f79226163b5a3","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92e7a73417b5d0a93c36d830df4f79226163b5a3"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-27 14:54:22.000000000","tz":0},"committer":{"name":"Kayobe deployment user","email":"grzegorz@stackhpc.com","date":"2024-04-24 11:49:50.000000000","tz":0},"subject":"adding rpm/deb support","message":"adding rpm/deb support\n\nChange-Id: Iccf304a6e8b322b751e28156f4619dd559a979d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d74c8224d0dec99893824520b3b9a0905e2799a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d74c8224d0dec99893824520b3b9a0905e2799a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
