)]}'
{"id":"openstack%2Fkolla~914646","triplet_id":"openstack%2Fkolla~master~I9bfb655252d060a83e73ec850abc9e8edde30f0f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"34213":{"account":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"last_update":"2024-06-20 11:51:34.000000000","reason":"Change was marked work in progress"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-06-20 11:51:34.000000000","reason":"Change was marked work in progress"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2025-06-13 09:53:20.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-06-20 11:51:34.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I9bfb655252d060a83e73ec850abc9e8edde30f0f","subject":"add scaphandre image","status":"ABANDONED","created":"2024-03-28 13:42:16.000000000","updated":"2025-06-13 09:53:20.000000000","total_comment_count":19,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"d2e274e7bd712d395252d6e197926419c14f8073","_number":914646,"virtual_id_number":914646,"owner":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-20 12:43:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"reviewer_updates":[{"updated":"2024-03-28 13:42:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-28 13:42:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-28 13:52:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-03-28 14:05:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-03-28 14:09:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-04-15 14:18:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-06-20 11:53:16.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"84c94e45f2d76071a98e99fd185a2c240339d84a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 13:42:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"621fce2970f3a61badd2538baf1c883ae3f47329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 13:42:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":1},{"id":"fa999496274c9fc1669e261b7f403ca709db37e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 13:42:36.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":1},{"id":"2d82e74a28bc90f936388ef885efbc35329b6ab5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 13:52:35.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"75de809db1d0da723c89bbb8c6e1f8f879143d93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:04:58.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e00f9904dd0d97994df8514595ea8e9ea39b64d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:05:18.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":2},{"id":"62f737386003ae86197058e5bf3f09c9e74b8ae9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:05:18.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":2},{"id":"9ffb07201000db84dbbd7b1f99056456e1cd5c46","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-03-28 14:05:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43c7fedba4130bd3c1d8a4ebe790986a0651a663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:06:01.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b6d82718a8981d6ae5a2eb56782d582281245a4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:06:22.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":3},{"id":"9914127b457497c6529df2326413d4d503bce3db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:06:25.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":3},{"id":"b2cf8b17057373525f9cb6fc712a2ee96a307e3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:07:16.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d7a0d2f9357a80a7e75758df812fd38b72027878","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:07:36.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":4},{"id":"4a3a0c4d210ca73a555675a01fe5e8c184f6a3cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 14:07:38.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914193,6","accounts_in_message":[],"_revision_number":4},{"id":"9d9c325a0a534979e3b16e8b7cbd77382dd9fbc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:09:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9ddc7b555c952b261a82751c7d6553ed8d15b8bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:10:37.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"7b6ac93671c7716109f2d83f66574c945e06a819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-28 14:15:17.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"8e7db06d409a8d88800f765974720dffcbdbb537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 15:54:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4faa6bc3a77b4c92b46b38595f19f86f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e887c2e9e8a24b75a323824ee89d5ffb : FAILURE in 25m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/216378ba98d34e449a688fa1ee8b531c : SUCCESS in 1h 12m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c9861bf60102470384da476b77a43f17 : FAILURE in 19m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/654171cc02084cb0867ab602cdd6be27 : FAILURE in 33m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3339d5cc4fab4cefb108ba24814fc321 : SUCCESS in 1h 08m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e69f7b52b1ed4a3084fcfa60b99df8df : SUCCESS in 1h 25m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76de2cfdb14e4aa48b196aa55d504732 : FAILURE in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c68ee3d531e496cbf5d6b5e5ba993be : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77cc3fc6ece641ec9a0273bb9728fa95 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d1a843c66d94288bdab18b6b45769fc : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00eacc12605b47f9baea80327bbaf3a5 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e89c9f6e9c4f487cad2394b97c74a285 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16d50c1643b84826b75b3285bdbda0fb : SUCCESS in 3m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7cd725ece5414b1da17192ae2fb2bd40 : FAILURE in 23m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9ec7c7f8603b409b9c3533268ad50d82 : FAILURE in 31m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2de3268ad377465db011e429b6a2f51b : SUCCESS in 1h 06m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/35771c522f1f48ce975814889007ca62 : SUCCESS in 1h 38m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2fc57b45ab3e424c9c97411e9bf03993 : FAILURE in 23m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0dd52122821446c4bae5fd8c485fbccd : FAILURE in 54m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f08a3e0dc155469994d1d646b7ce5242 : SUCCESS in 1h 06m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9e87f63390ac4ebda29dfe75887c665c : FAILURE in 49m 32s\n\nWarning:\n  Comments left for invalid file kolla/image/tasks.py","accounts_in_message":[],"_revision_number":7},{"id":"a760e2b070e750d29604ea73afdb52307072fc58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 15:54:23.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/914529, https://review.opendev.org/914193 are needed.","accounts_in_message":[],"_revision_number":7},{"id":"a4eae2258444f2f1d4a707debabe6b5b82c09395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-29 08:53:03.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"165a08a870e8d5b310f1258fa4674087f53934af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-03-29 09:10:35.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"82a911df9e89cb832767f49745f0b4b5cd3ed0ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 10:36:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b0031d1df64707bc95215b298d4c7f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5d5a46a259584fbb8915676a8be98acb : FAILURE in 35m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/24e8e855e8d24ea1888271316013ca5e : FAILURE in 1h 14m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/54d5f1de53fb4884adabd9d476d5fd70 : SUCCESS in 1h 24m 03s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/197724ad7bf0403abc0ec1450a754b97 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3659e814dc3940da8d967dc75d9de4c3 : SUCCESS in 2m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/347d851881984a368cbbbb822ec3d656 : FAILURE in 41m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e9d7fac346ce4f51bd6ec22b0a6fdd13 : FAILURE in 1h 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"01a52b95b7a903534dede920ae77fca33b1b1f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 12:17:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cd9e31c796a4cd79ec5baadd8702e63\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aa42ff3f4a6e47dabad7b6efe94290ef : FAILURE in 29m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4ab946c830a5437995dd9319f81f850d : SUCCESS in 1h 11m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ce0d8bc1dc354fb99ae48e1f38ed5590 : FAILURE in 21m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ee7c677c8aeb4273adf3c895166f57b5 : FAILURE in 30m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f5301f12d90f40788e6cd59d663a8aba : SUCCESS in 1h 07m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/abb40a611f4d43f3b4684ace48081a5d : SUCCESS in 1h 35m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70496a7053c544ab8c44f59e41ac8a42 : FAILURE in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe1f06e10610454d95a2dae1bfb1aed8 : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79a80d49d1134dc8b88ca166754ca4c1 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49666b2512754c008930fd0f25c5baca : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/653e308a9e784754a35c9366fac9fa8f : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2d483183394733846f9b390f76bc1b : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b672b5b79b54d85a0a892f3a2ad5388 : SUCCESS in 3m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6239b5586ee04ad0b4e0f3a6b75fdd80 : FAILURE in 22m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fbdd56d46fbe4c739f1f4d1aef121541 : FAILURE in 41m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/550b074690cf4379a76dbd7777f04f8f : SUCCESS in 1h 04m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8afa7b6a213849aba984ee4a570b17dd : SUCCESS in 1h 29m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b73b6f34a3304ce2bbb1e1f3ae3733a0 : TIMED_OUT in 3h 01m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5230a9bd6b404e2984456cd43171dc0c : FAILURE in 1h 05m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dd51f6604b8645be849c47c104ad2f60 : SUCCESS in 58m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a12e8617d95b4e60afdb02283a8bf4b9 : SUCCESS in 1h 20m 02s\n\nWarning:\n  Comments left for invalid file kolla/image/tasks.py","accounts_in_message":[],"_revision_number":9},{"id":"0c2609be3eac0f95d60b38968e5d67178ecc0d3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-02 13:25:09.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"17fe96bbf053246348b7d6fc0ebe270e826d45ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 13:55:41.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab4d694fc8b946b1936cca0c8d26b680\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b368a6f850df4be4877c1336841819e9 : FAILURE in 13m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0ee54ac6aee742389538299c9743f42e : FAILURE in 16m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d634e4825ce44fbcab6b7cf0e7814d08 : FAILURE in 10m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/931536a191174fa2bc17ee39fadee6a5 : FAILURE in 14m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1bcafe532e674bfeac48e2bb4a779052 : FAILURE in 13m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/983617ba38224cb28d375f27fc720d09 : FAILURE in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6580d17d0d54acf8b5b861052e31a05 : FAILURE in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b4e4dd90b954c72b5ca77b62c55cf43 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/add9463ca27e4fe197b278c7f158cea7 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85bbc07ea0b448848c06bd887eeb9723 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b2df0c9a10d407fbaf4af4905e9f96b : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5fa5318c9a144dbbfcb5090ddf8d3d1 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96ae53d2c36f4b7c8f7b0f2490a0bd34 : SUCCESS in 3m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b48148cc672b4d7f833b5be031a7261e : FAILURE in 9m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8ee83456011e48c6a3ea7efc8a5e6cf3 : FAILURE in 15m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4418c13d48724e03ac876953cb058ffc : FAILURE in 14m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4db0ecd4ead141a1bccf77e2869b7617 : FAILURE in 14m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/148e628dbb2e455d9c577c5deb1f3c0e : FAILURE in 12m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e5edf200948940d68fea71d005907c50 : FAILURE in 15m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9ab5a3aa18d74862835354b89b14d314 : FAILURE in 10m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/21e03b76f1644b8581eb296b6f3eff1b : FAILURE in 11m 03s\n\nWarning:\n  Comments left for invalid file kolla/image/tasks.py","accounts_in_message":[],"_revision_number":10},{"id":"c1d79597847c64d0e39ca465f2c061f8994542f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 14:10:05.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3e5df0c8454ee382e9dea9714ed9ec\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/98c7b83cf0b04ffe8238a8c6dd446ca1 : FAILURE in 42m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/360af42ee79d4b36b664bc5edce7a461 : FAILURE in 40m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d120a1e8a15545ad96f3b7ba2ca5c6bf : FAILURE in 11m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e00c7cea134d40898a012542bc94b8b0 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ab3a6414a4f14cbfbcfde4002d06fa22 : SUCCESS in 2m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0e76e6f79bc3463089eb663d912ca4ff : FAILURE in 35m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b7a53a0b38a8400686e46090160bc2bb : FAILURE in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f976255038095447bcb93336b44b83de90728230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-04 10:29:14.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"9d58219a66af89b42f09578477aff5f235d00c3b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-04 10:55:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2ccbfe63461b49f32c12c57997a5a829d0f4e7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 12:05:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22c1f9737e574e748fcb366be3d89df1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ee6b12f751f64bee81ef0462fbb5788a : FAILURE in 25m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1f42cf6131d9475c9b538ca7f3200f80 : SUCCESS in 1h 11m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8798b281c972466c8e1abf7d9fa62d54 : FAILURE in 19m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aeb93b10d0334829bcf16f7d46751b39 : FAILURE in 26m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4b6797c66bdd4135950b7c6a2bca6f96 : SUCCESS in 1h 05m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/284cdff90bdb40bd8db963e45c217c11 : SUCCESS in 1h 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01e5ee7eb474dd6bd25b0723269fdef : FAILURE in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2686f041d9c340229a6de104116154e0 : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a865d487c104173bd8892fd69b7ebc9 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8f90eff5a0f49718d1a61d8e2c46f43 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/149023be8bc345f5822a4bb3e12be2b1 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab8f2e550c349c9b12f5d3cc01d2c3b : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f96b72adaf4e43ecbc83b52b956fb9b1 : SUCCESS in 3m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/665e34d9ce0a45648ec62a0f1db98574 : FAILURE in 23m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0e934340e1564b8a93b36fa71919b060 : FAILURE in 32m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5d4fcc7534f94b4f81c96ab0546dd5ec : SUCCESS in 1h 02m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/623c73666d3b47e5a4c3975540fd5a38 : SUCCESS in 1h 31m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d8b94ac29f86490099801ca2e0b4cd61 : FAILURE in 17m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/07b9882be12c4b74b9e44f26cd5f92ae : FAILURE in 1h 01m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/06baa471fc094b4d91ea243d98465e5f : SUCCESS in 48m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8e37908221d648609b71b8bd00f34e94 : SUCCESS in 1h 25m 55s\n\nWarning:\n  Comments left for invalid file kolla/image/tasks.py","accounts_in_message":[],"_revision_number":11},{"id":"bb0e4dbc2a5817fedb95b2b7677e7675d0e8ba07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 12:30:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf6e39eeb3334d768e63a5e067138fb1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a8fd0289397e401f8a358fdc5f042237 : FAILURE in 1h 30m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8814d344dd4044de947a8abb845bda2b : FAILURE in 34m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/099be8aa6f13463e8a9fae0b0ccf3c49 : SUCCESS in 1h 59m 14s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75cf359279f64c3fafdae1a7ca58ebb4 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8132d9d2fb9d432ca718c3601916ffc0 : SUCCESS in 5m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5f985e43c3474532830a7b67cd082080 : FAILURE in 35m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/02b9d1d53df14c2a952a23c870c731c3 : FAILURE in 1h 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2a18dd91344c7cc382c2feaabfae9c77fa24bb30","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 14:18:41.000000000","message":"Patch Set 11: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d9e9b9e37d170f03d89f5afd6cf082ff824cb75a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 14:22:46.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_34213\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"}],"_revision_number":12},{"id":"86abd452aa05a67cb6453890077f645a4b004239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:00:20.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5dd5e2f7e4141359b2b75090d38f82c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7cf7158e92d244e08f55cb5a836b7515 : FAILURE in 27m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/30bde10af5b44c7bb17345e31848bf26 : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3d4a2c41bb524499a70b92edcadececa : FAILURE in 22m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/00ae315ca67c4fb49121d30d2d2b2c20 : FAILURE in 27m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/04d04af48274413fb910ad65cbc87b49 : SUCCESS in 1h 11m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/76043da1190346c58efc4a86c3e8e145 : SUCCESS in 1h 20m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44bb85defa6a4e94826facbd2fdd4414 : FAILURE in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc60516c027a4779a4ec479e24166d95 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37ff20539a234cc39399edf9d0458903 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2e90b43676648a0a2252420bdf8f3da : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cedf6ea5510149a686a8dfb75f7e5c73 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ba4db0bcb04bce9318ce5bf8358a90 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9999672aeeda4b6e904b56ac7655b8d3 : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dcf2a09e921447838ad59c045d0a40ca : FAILURE in 34m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c9ff645a2c894cf68e9bbe50c62c30a7 : FAILURE in 33m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ba5a434412dc40cb8d940e049522d5e8 : SUCCESS in 1h 02m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3265ca8095494b70b2d01f3de36a1c09 : SUCCESS in 1h 29m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9c8c7b3abbfb45c39d40155c59acbdbb : FAILURE in 22m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0c139b25a8054a2eb5fe9aeee90caeda : FAILURE in 1h 06m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8f8bf979e8734534bf517c286b8b09e0 : SUCCESS in 55m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8b1dd8d46e97422ca263eaddaab17cf1 : SUCCESS in 1h 19m 51s\n\nWarning:\n  Comments left for invalid file kolla/image/tasks.py","accounts_in_message":[],"_revision_number":12},{"id":"e83174a52ec55249d83496b394c4e70cb7c1ffa7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 17:16:50.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18edc83ddc4b440c912a1bb5efdd3886\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7d7a1fb9f3254a1e961c9528a5960ac6 : FAILURE in 43m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3293cac8d3614f1d99e44f7a92d7e60b : FAILURE in 36m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d2aa087f7c534138a27b958c3ea22a51 : FAILURE in 1h 41m 00s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ec8d6d7971641468d47a08459a72f35 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/92670990392e4275aca270deffc81cea : SUCCESS in 4m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/33202c164b174a598531c140c382ad72 : FAILURE in 34m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f144a69a107f4f81a5717633f088fe0c : FAILURE in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8c01116c5886c942525c28ec9bb9c82a4689a0d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-16 13:57:40.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"61a141d16f72328789549acaa89bdcae629d0288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-16 13:58:20.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"41ffb885dfa0b5d4841766ed3895ee8fa612ae9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-16 14:03:03.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"8f3044272952a8a5d05d32b183b6249a1f1c57ac","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 14:05:07.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4878f2dc7915db7bfdd5e93ca3fb93feb30e62a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-17 11:30:50.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"ca88445563db45ce72a7a606f0775a4c3ada0ec9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 11:31:12.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914646,16","accounts_in_message":[],"_revision_number":16},{"id":"683233fef78abeed6840679889d3963a7ef1826d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 11:31:12.000000000","message":"Patch Set 16: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914646,16","accounts_in_message":[],"_revision_number":16},{"id":"a2d0c56ae07398679a7e5f0712fff413621b2b95","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-04-17 11:47:11.000000000","message":"Patch Set 16:\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"3d26891ccaaeaf5a22f1dccad844891097ffc75c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"real_author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-18 16:15:21.000000000","message":"Patch Set 17: Patch Set 16 was rebased on behalf of \u003cGERRIT_ACCOUNT_34213\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"}],"_revision_number":17},{"id":"11fa0aee0bcbfa3d85906f9fdbd7a750aac44886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 16:15:43.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914646,17","accounts_in_message":[],"_revision_number":17},{"id":"4c9081345814e08fe36ef35bf9f05480f0224b4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 16:15:44.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 914646,17","accounts_in_message":[],"_revision_number":17},{"id":"ac3a0f301121ca43949bb4fb4f8a7172b1db173b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-06-20 11:51:34.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nThe following files contain Git conflicts:\n* kolla/common/sources.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"f5ba7e66bb83f7728d098418aee02ad9fce1de80","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"date":"2024-06-20 11:53:03.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"dd73c315d787e319f59e61fc63895c93d8c87a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 12:43:00.000000000","message":"Patch Set 19: Verified-1\n\n(19 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ded45ee697d249c991091f76b55594a5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/55c3a1200d9349cd8360fc394239866c : FAILURE in 22m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/93e34a72a4a343d290c7145e68a7d74c : FAILURE in 18m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/87b0f648a0f1472f80b66f606fb45b18 : FAILURE in 15m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4895a507158f4270ac372b860ea004ff : FAILURE in 11m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3065972907314fb498cc9f301f74acb2 : FAILURE in 10m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e71d27a400cf4b0f8973cc5c3d42f950 : FAILURE in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/494763806066490d83e4c6eceaecdb81 : FAILURE in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/637495f56168491ca23a1d43ca3cb39a : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1783d137d6274a9db758b6efa8e59809 : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4c338164c434ba0adcd9e13887c5939 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbcb2f03f9954193a7673905d47686ec : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e79efafbfe493192cbce3e34652a8a : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fa4aec3f1ef44b8951ab438b867891b : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e0e1599798b546f98d99cf25102f796e : FAILURE in 20m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/852634f31b7f4bff81a491434d12c3db : FAILURE in 24m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/70a8b09a9ee040eaa07a566109d99692 : FAILURE in 16m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ca7d487fc6984d5fb717d7e839740c3e : FAILURE in 16m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/64f16e1d072e46098bedb458b711c416 : FAILURE in 13m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bf53f53fc608450698a0dcb5753558d3 : FAILURE in 17m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5facd4d6480a4f05aef84bfdb45fed7d : FAILURE in 12m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/131312d533e846a5a8426095884eb0ee : FAILURE in 12m 37s","accounts_in_message":[],"_revision_number":19},{"id":"b40e70555b36824cfd67323bc7ac2e0bdd2a7f56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 14:56:24.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04da814386f94b9d92deaada035169e7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e113180c86b247919e3d3d71f575db85 : FAILURE in 2h 04m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/22d8e42adae34540805cd5cd144f7b0f : FAILURE in 2h 06m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ac6d142d70a8442091225279a67d488b : FAILURE in 12m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59b9c74636624bd79a81f43a0ca0c893 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b66cc4ff82ed497c84a1a0d14c0a800c : SUCCESS in 2m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dbab6208e03f47439c9200ec3c4cc283 : FAILURE in 3h 00m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c8e0ae88763f4c56a6fe621ae0771484 : FAILURE in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d2e274e7bd712d395252d6e197926419c14f8073","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-06-13 09:53:20.000000000","message":"Abandoned\n\nUnfortunately the Scaphandre project is no longer under active development.","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"f8fcbb0f9b72f2828c1aafef8b4519ae15e755e3","revisions":{"29342dd6d06713eaf4374405e036aff9230cb8e4":{"kind":"REWORK","_number":1,"created":"2024-03-28 13:42:16.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/1"}}},"commit":{"parents":[{"commit":"ddf3d2f48d27beb7b7d26472de303db6a42df4cd","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddf3d2f48d27beb7b7d26472de303db6a42df4cd"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"subject":"add scapahndre image","message":"add scapahndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29342dd6d06713eaf4374405e036aff9230cb8e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29342dd6d06713eaf4374405e036aff9230cb8e4"}]},"branch":"refs/heads/master"},"9d0b83e129f9061bf07dc6d9dea6f73628e8ad35":{"kind":"REWORK","_number":2,"created":"2024-03-28 14:04:58.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/2"}}},"commit":{"parents":[{"commit":"ddf3d2f48d27beb7b7d26472de303db6a42df4cd","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddf3d2f48d27beb7b7d26472de303db6a42df4cd"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 14:04:58.000000000","tz":0},"subject":"add scapahndre image","message":"add scapahndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d0b83e129f9061bf07dc6d9dea6f73628e8ad35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d0b83e129f9061bf07dc6d9dea6f73628e8ad35"}]},"branch":"refs/heads/master"},"f4cbd01166c7fa327d8030e74d6022c3f8876821":{"kind":"REWORK","_number":3,"created":"2024-03-28 14:06:01.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/3"}}},"commit":{"parents":[{"commit":"ddf3d2f48d27beb7b7d26472de303db6a42df4cd","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddf3d2f48d27beb7b7d26472de303db6a42df4cd"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 14:06:01.000000000","tz":0},"subject":"add scapahndre image","message":"add scapahndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4cbd01166c7fa327d8030e74d6022c3f8876821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4cbd01166c7fa327d8030e74d6022c3f8876821"}]},"branch":"refs/heads/master"},"8cb958d4f159bb58f81c6b762e762eb9c142856c":{"kind":"REWORK","_number":4,"created":"2024-03-28 14:07:16.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/4"}}},"commit":{"parents":[{"commit":"ddf3d2f48d27beb7b7d26472de303db6a42df4cd","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddf3d2f48d27beb7b7d26472de303db6a42df4cd"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 14:07:16.000000000","tz":0},"subject":"add scapahndre image","message":"add scapahndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cb958d4f159bb58f81c6b762e762eb9c142856c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cb958d4f159bb58f81c6b762e762eb9c142856c"}]},"branch":"refs/heads/master"},"869675d61af9b944578504e8e95cde7c19151df5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-03-28 14:09:03.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/5"}}},"commit":{"parents":[{"commit":"5a4a766e86ce3230a2061c9b5e2857f4b80b3c72","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a4a766e86ce3230a2061c9b5e2857f4b80b3c72"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 14:09:03.000000000","tz":0},"subject":"add scapahndre image","message":"add scapahndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/869675d61af9b944578504e8e95cde7c19151df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/869675d61af9b944578504e8e95cde7c19151df5"}]},"branch":"refs/heads/master","description":"Rebase"},"66b607e43fd2532f5e354ef52c454e85ff35736c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-03-28 14:10:37.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/6"}}},"commit":{"parents":[{"commit":"5a4a766e86ce3230a2061c9b5e2857f4b80b3c72","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a4a766e86ce3230a2061c9b5e2857f4b80b3c72"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 14:10:37.000000000","tz":0},"subject":"add scaphndre image","message":"add scaphndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66b607e43fd2532f5e354ef52c454e85ff35736c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66b607e43fd2532f5e354ef52c454e85ff35736c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a3f95d1b1eab852c9d49a7caea79a532183af5fe":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-03-28 14:15:17.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/7"}}},"commit":{"parents":[{"commit":"2fac27c754fca4dc02585d41bd139aea27374379","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fac27c754fca4dc02585d41bd139aea27374379"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 14:15:17.000000000","tz":0},"subject":"add scaphndre image","message":"add scaphndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3f95d1b1eab852c9d49a7caea79a532183af5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3f95d1b1eab852c9d49a7caea79a532183af5fe"}]},"branch":"refs/heads/master","description":"Rebase"},"bc1e53f980058669981e8470f6f4571ec95c142f":{"kind":"NO_CHANGE","_number":8,"created":"2024-03-29 08:53:03.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/8"}}},"commit":{"parents":[{"commit":"dc5266b6d2455c176fd1ddf9daff0e460be06493","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc5266b6d2455c176fd1ddf9daff0e460be06493"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-29 08:53:03.000000000","tz":0},"subject":"add scaphndre image","message":"add scaphndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc1e53f980058669981e8470f6f4571ec95c142f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc1e53f980058669981e8470f6f4571ec95c142f"}]},"branch":"refs/heads/master","description":"Rebase"},"cff03188e5d59479518a69014a6d1599668b93fd":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-03-29 09:10:35.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/9"}}},"commit":{"parents":[{"commit":"dc5266b6d2455c176fd1ddf9daff0e460be06493","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc5266b6d2455c176fd1ddf9daff0e460be06493"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-29 09:10:35.000000000","tz":0},"subject":"add scaphandre image","message":"add scaphandre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cff03188e5d59479518a69014a6d1599668b93fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cff03188e5d59479518a69014a6d1599668b93fd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0ae130c64082dd54e6e700c0cce5a0247092fa19":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-04-02 13:25:09.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/10"}}},"commit":{"parents":[{"commit":"8aff5660700116c8fa1a4f2ec80835339ba880b2","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aff5660700116c8fa1a4f2ec80835339ba880b2"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-02 13:25:09.000000000","tz":0},"subject":"add scaphandre image","message":"add scaphandre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ae130c64082dd54e6e700c0cce5a0247092fa19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ae130c64082dd54e6e700c0cce5a0247092fa19"}]},"branch":"refs/heads/master","description":"Rebase"},"6e9c4f6c5d24053621e243b5b0dcc82cd370792a":{"kind":"REWORK","_number":11,"created":"2024-04-04 10:29:14.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/11"}}},"commit":{"parents":[{"commit":"ea6700873082ad9f53b7e3876d0ae4af420c0027","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea6700873082ad9f53b7e3876d0ae4af420c0027"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-04 10:28:24.000000000","tz":0},"subject":"add scaphndre image","message":"add scaphndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e9c4f6c5d24053621e243b5b0dcc82cd370792a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e9c4f6c5d24053621e243b5b0dcc82cd370792a"}]},"branch":"refs/heads/master"},"7bdda71aa38215b8365374d36ed04a24c0797580":{"kind":"NO_CHANGE","_number":12,"created":"2024-04-15 14:22:46.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/914646/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/12"}}},"commit":{"parents":[{"commit":"897c6b51f80e96da38d6ad95c749cac0f0a34d23","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/897c6b51f80e96da38d6ad95c749cac0f0a34d23"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-15 14:22:46.000000000","tz":0},"subject":"add scaphndre image","message":"add scaphndre image\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bdda71aa38215b8365374d36ed04a24c0797580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bdda71aa38215b8365374d36ed04a24c0797580"}]},"branch":"refs/heads/master","description":"Rebase"},"e52b77e1c5757445d12cbe460bf4887f0ce2cef0":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2024-04-16 13:57:40.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/13"}}},"commit":{"parents":[{"commit":"897c6b51f80e96da38d6ad95c749cac0f0a34d23","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/897c6b51f80e96da38d6ad95c749cac0f0a34d23"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-16 13:57:40.000000000","tz":0},"subject":"add scaphndre image","message":"add scaphndre image\n\nScaphandre is a tool to measure energy consumption of physical and virtual servers.\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e52b77e1c5757445d12cbe460bf4887f0ce2cef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e52b77e1c5757445d12cbe460bf4887f0ce2cef0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eddb692b2954a93064e11c2156d0b7ebfc8cbda1":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2024-04-16 13:58:20.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/14"}}},"commit":{"parents":[{"commit":"897c6b51f80e96da38d6ad95c749cac0f0a34d23","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/897c6b51f80e96da38d6ad95c749cac0f0a34d23"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-16 13:58:20.000000000","tz":0},"subject":"add scaphandre image","message":"add scaphandre image\n\nScaphandre is a tool to measure energy consumption of physical and virtual servers.\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eddb692b2954a93064e11c2156d0b7ebfc8cbda1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eddb692b2954a93064e11c2156d0b7ebfc8cbda1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"45f9a9d2b03f062363f6b639e4b9d9168bff3e5c":{"kind":"REWORK","_number":15,"created":"2024-04-16 14:03:03.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/15"}}},"commit":{"parents":[{"commit":"8aff5660700116c8fa1a4f2ec80835339ba880b2","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8aff5660700116c8fa1a4f2ec80835339ba880b2"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Kayobe deployment user","email":"grzegorz@stackhpc.com","date":"2024-04-16 14:02:37.000000000","tz":0},"subject":"add scaphandre image","message":"add scaphandre image\n\nScaphandre is a tool to measure energy consumption of physical and\nvirtual servers.\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45f9a9d2b03f062363f6b639e4b9d9168bff3e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45f9a9d2b03f062363f6b639e4b9d9168bff3e5c"}]},"branch":"refs/heads/master"},"b97f4ef4e79bee8afbe372a988b5a223b8fcd3aa":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-04-17 11:30:50.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/16"}}},"commit":{"parents":[{"commit":"897c6b51f80e96da38d6ad95c749cac0f0a34d23","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/897c6b51f80e96da38d6ad95c749cac0f0a34d23"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-17 11:30:50.000000000","tz":0},"subject":"add scaphandre image","message":"add scaphandre image\n\nScaphandre is a tool to measure energy consumption of physical and\nvirtual servers.\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b97f4ef4e79bee8afbe372a988b5a223b8fcd3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b97f4ef4e79bee8afbe372a988b5a223b8fcd3aa"}]},"branch":"refs/heads/master","description":"Rebase"},"e0e8edaa6a718c77a122067eeddb4d169dee57a2":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-04-18 16:15:21.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"real_uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/914646/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/17"}}},"commit":{"parents":[{"commit":"d5ec0d2ceb113d55b19faa97414c224c46025180","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5ec0d2ceb113d55b19faa97414c224c46025180"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-04-18 16:15:21.000000000","tz":0},"subject":"add scaphandre image","message":"add scaphandre image\n\nScaphandre is a tool to measure energy consumption of physical and\nvirtual servers.\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0e8edaa6a718c77a122067eeddb4d169dee57a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0e8edaa6a718c77a122067eeddb4d169dee57a2"}]},"branch":"refs/heads/master","description":"Rebase"},"1a5b4dca1ee693f88a59d1509a177090efba4a1a":{"kind":"REWORK","_number":18,"created":"2024-06-20 11:51:34.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/18"}}},"commit":{"parents":[{"commit":"d74c8224d0dec99893824520b3b9a0905e2799a6","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d74c8224d0dec99893824520b3b9a0905e2799a6"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-06-20 11:51:34.000000000","tz":0},"subject":"add scaphandre image","message":"add scaphandre image\n\nScaphandre is a tool to measure energy consumption of physical and\nvirtual servers.\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a5b4dca1ee693f88a59d1509a177090efba4a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a5b4dca1ee693f88a59d1509a177090efba4a1a"}]},"branch":"refs/heads/master","description":"Rebase"},"f8fcbb0f9b72f2828c1aafef8b4519ae15e755e3":{"kind":"REWORK","_number":19,"created":"2024-06-20 11:53:03.000000000","uploader":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"ref":"refs/changes/46/914646/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/914646/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/914646/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/914646/19"}}},"commit":{"parents":[{"commit":"d74c8224d0dec99893824520b3b9a0905e2799a6","subject":"adding rpm/deb support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d74c8224d0dec99893824520b3b9a0905e2799a6"}]}],"author":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-03-28 13:41:50.000000000","tz":0},"committer":{"name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","date":"2024-06-20 11:53:03.000000000","tz":0},"subject":"add scaphandre image","message":"add scaphandre image\n\nScaphandre is a tool to measure energy consumption of physical and\nvirtual servers.\n\nChange-Id: I9bfb655252d060a83e73ec850abc9e8edde30f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8fcbb0f9b72f2828c1aafef8b4519ae15e755e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8fcbb0f9b72f2828c1aafef8b4519ae15e755e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
