)]}'
{"id":"openstack%2Fkolla~914744","triplet_id":"openstack%2Fkolla~master~Ic8cc29230d82f4d1f07a3c97c3dcbcd14ac76074","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-17 12:25:13.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-04-17 11:13:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-04-17 12:25:13.000000000","reason":"Change was submitted"},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2024-04-17 12:25:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8cc29230d82f4d1f07a3c97c3dcbcd14ac76074","subject":"kolla-toolbox: Bump ansible-core to 2.16","status":"MERGED","created":"2024-03-29 07:24:45.000000000","updated":"2024-04-17 12:37:55.000000000","submitted":"2024-04-17 12:25:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914744","meta_rev_id":"201e8a1d0d07ab2665c60741b57c516afb42af44","_number":914744,"virtual_id_number":914744,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 12:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-17 06:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-17 11:13:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-17 12:37:55.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2024-04-17 07:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2024-04-17 07:47:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-17 06:20:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-29 08:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-29 08:58:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-05 07:58:41.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-07 02:31:13.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-04-08 10:16:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-17 06:48:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-17 07:47:11.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f4ede8de3954cdec499ad5db99b6750e32f1445d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-29 07:24:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99135c98c49ef9b202a4505fb35569d6f65b81e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 08:49:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/606ad04b901046819306285dd9158fb0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0ceff06327034d62a3d3e5b868cd76f2 : FAILURE in 23m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/525655199d0a4f71a89bcda532dcd72a : SUCCESS in 22m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b9b315565610447ca6002d717952be07 : SUCCESS in 1h 22m 35s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a74254c39b74e05963e5b9c89c1e46a : SUCCESS in 2m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c556e658bdb40c4a83d582d7143454c : SUCCESS in 5m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2ff653208f644675aaa6768427030d53 : FAILURE in 22m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7ba9b0010be545e598df3d68c965da7e : SUCCESS in 51m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ed9e2d61f5db553547d48bb9f09b1aca373576b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 08:58:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2241f8339a6140cbad855ddad54d84ce\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/35c1da581f964915a7702dd21e091f6c : FAILURE in 23m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4752119119014901ae1cffff3cde90a9 : FAILURE in 17m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/29e20179aba74a1a9fddc7c28b33bdc0 : SUCCESS in 17m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d86d944f65ce4aef96414f4a66f1e3c8 : SUCCESS in 28m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/000d5eed33074d33bc93704cebbe53c6 : SUCCESS in 1h 00m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a90b0c66a04d48eaa44687b8da15cc64 : SUCCESS in 1h 32m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34abba5816e24eb49a688b103d3646bf : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10ce54aff23e469fb3975e288a5ea1f1 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6042fdcb43ac4bde809efb68e435b716 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a078f569c5e4ea1a6bfdfc35b4db286 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1ba002001d44cdb975aef253c78a925 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156f01003645427fbdf5731719efe980 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bee373f3577c464ead4b55a81832349b : SUCCESS in 3m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fa69fca93e4646b3b06128c82fe1f38c : FAILURE in 22m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b10529dbbb924ba58840c0107ecbc91c : FAILURE in 30m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9ec1f12f44fc4d2eab23ec2720f1b856 : FAILURE in 19m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c98c5f57deee42669e74bfb26789a249 : FAILURE in 18m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eb590fa5de5e430b8ab1ba4798ef5d29 : SUCCESS in 22m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/72bd16a03c804d879bb537c74c07fe4f : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8fa623bfb0184341a15220e68c710bd2 : SUCCESS in 54m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2b6be64299b847f2ae4d77a8e917bf1c : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":1},{"id":"058312ac07ef5059d8bf9efc90557a7697d861fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-29 09:10:25.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"31e6fdefa5a58a330115348b51692faf8fa23d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 10:53:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41816d3c15d442d4bfbc8a86d162155d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ec477428aebe4a49be3bb05fbcdb2844 : FAILURE in 23m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0a05f3cefc0b4695adce10f953d47ad0 : FAILURE in 20m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b8fa6f4baad64df484228a232b4e4324 : SUCCESS in 15m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e49ebaf66f924d4090dad7db963913ff : SUCCESS in 30m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef51ec93682941bab341478622843bd2 : SUCCESS in 1h 05m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7059fc495eff4384a8228f8490c83b3f : SUCCESS in 1h 37m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa8f607e4cfe49e3ae43226b5a4e0c69 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/838e0a1a71ef4ff589c4b4f3cef79499 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/051d523bc8444fefb9f44e185ace9715 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a440158f0774c228b96f7558b0b573e : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04dbd2b0e8e844088397ffbb180c20d8 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86006558bd694081bbc0f435d2c60075 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16fb1502a47c425b87203a321d6d8fe1 : SUCCESS in 4m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/467bb0f19b14486495d744182cbdaa8a : FAILURE in 24m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ac5434625e7e42068c0978387c3ed89d : FAILURE in 33m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4d3f8625b6344b868ed9ab6d64b086bb : FAILURE in 20m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ff428e2d72f44bc0ad324ab5bb80f53e : FAILURE in 21m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a1de632fd37b49b1a0ffcf526adcd118 : SUCCESS in 21m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/63f07710370442c9b567c1d1e2ef536c : SUCCESS in 1h 21m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c899f6a2775c4d3da3ec1e628df00eaf : SUCCESS in 48m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1ba04381ff8e40f286ebb6313bc10432 : SUCCESS in 1h 19m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e169b6a628bcef8bfa36e0bd1b61cbbdbfcb16e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 10:56:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf1b239e1384ed48d5d61ab6e5145e3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ace8adde89b94d0ea5fffb89ae38fa09 : FAILURE in 39m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/904edde3ecd94f67a003cf2f12cbb3ad : SUCCESS in 30m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/af15cce11d214ad9b45407f79ab05998 : SUCCESS in 1h 44m 45s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e49342d6d9c4d60baab08d85a1c966d : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a1518553fd34c9091b07c0e0217d7d3 : SUCCESS in 4m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6a5f039f38644b12b66cb5182645c82d : FAILURE in 34m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/feff738aeee0461a970c1e4f6405c94b : SUCCESS in 1h 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84d227959e798b4e3f38e2c3b4b41844c46f1d00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-29 11:15:45.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"37a0265573110b2e8561a9458303b468b80d6082","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-29 11:16:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"822b841aef9b8c6806af5515148612387464f642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 12:30:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/568e2e103921410cb291de380b1678c9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0c11e4a9fe9e411c9e844d8eff85a928 : SUCCESS in 1h 08m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/71b98868f83343b7848bb50d8d42c3de : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f4e6d1824c847d0b8b8d70aedd78813 : SUCCESS in 1h 07m 32s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aebf84e91c8a4b63a2ce33f5de63f7a4 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9174c551cc9f4615b9629b58987481b6 : SUCCESS in 2m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1c90953ec2784ad5b8eaae14de3d1aed : SUCCESS in 1h 11m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ffc65e864be84aebab43651e78289ff4 : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d9ac37401573401a0076f5f703ede1e10470488e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 13:01:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67614f7d2b244d2cab2d7e61d3a881d4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/412b8a3c89344910a96df6e915beeb92 : SUCCESS in 23m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e5793239a0a14efa898e914ae8037c8e : SUCCESS in 1h 07m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a01459bbb3ea4d028257b19234dd508c : SUCCESS in 18m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bb72e6f5384a404da9b243f6c875bdd8 : SUCCESS in 32m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5b1baddf4a4447fba05c9f3b002e6ac2 : SUCCESS in 1h 07m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a1d18d2a5a7b4d519c8de267082e68c4 : SUCCESS in 1h 28m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1acf5ad4460b47268671614b9e7a9323 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e62ff90a823046ffa57c9410a7d7ab72 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40b5a814cde34c028a2826f31877eb83 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3b7fd9ddbf34ff88c4ea7ed1067dba3 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7df69ed8965c47a7b2affe74f8c7040a : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b2dddc50cc64e6ca7892ed9aa34bb05 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbe1e8fea34d45ada8019f0215dc7fa9 : SUCCESS in 3m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c1041358048b46cdb4f645feaf04128a : SUCCESS in 22m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/be393ae6c6c347dc9cf8a97bd2574939 : SUCCESS in 29m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/57efa9c54b9446ec857b8c731f1ba1fd : SUCCESS in 59m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/35273d7505944d1ab792ce5a9f3c0528 : SUCCESS in 1h 37m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3ee99c103b9642aa861e0de518f37db0 : SUCCESS in 16m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4abf92e2b59b453e8ea3cc26f57eafe4 : SUCCESS in 1h 16m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f1168af541a6488990ba9bb9b2281f27 : SUCCESS in 52m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/45f310a062b842c0b4d56552ef4dfc68 : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":4},{"id":"8a04ccac2654fe338fa33f86dc94cf767f693908","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-05 07:45:06.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"20b74a05ef714a1bcaab2d9fdf77a2ead39be338","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-05 08:02:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"80fef2c380d17648d5c3dd3d066367e2dd6caab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-05 08:06:07.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"54837e86e74070ad3746ee89b50b027a4762407d","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-05 08:12:27.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3991f2f9b04cdded6fe8de8aec84b9e0b6f20c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 09:54:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1c73b91672a48128250fbac1afa606d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6e9aa1106d434da1a102e05c6d241c2e : SUCCESS in 25m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/43a7c13986834f99a87b2a5f51f21e2b : SUCCESS in 56m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4ff6c09e34934c38ae312508c5d5bfa5 : SUCCESS in 17m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6ed8bd95b09943329c5d22b04a1bea73 : SUCCESS in 27m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6928a350219e4fc3a0a49ce1e70d997c : SUCCESS in 50m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/18bd2ef2d7794062a0f285e21ea45ebd : SUCCESS in 1h 32m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3fc6c57b8004d3da886c31335aabae4 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af5477f2e9824b828cb768948e7daa30 : SUCCESS in 2m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/814b12b398794f6d9e9077edb83692ed : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4321c8f34f004992939698745bdd7937 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71617628e91349c8bbc3966a9f61da31 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a018817c254098a31bc524dba3a3b0 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a3285f82dc24569936832aa8155ecfa : SUCCESS in 3m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dc0904ef19a34d8b981e5ae94b89e327 : SUCCESS in 22m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/87dbf4489c07432e97fa05fa065b90fa : SUCCESS in 31m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/658c0593e59944e9ad5ff046810fc095 : SUCCESS in 1h 05m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/477e15a00be045828f45747b5eb7cf3e : SUCCESS in 1h 36m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/803cbbd9871d4c668d628aff5804257d : SUCCESS in 20m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4b444d2949c84c3aaf2e200978327159 : SUCCESS in 45m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1ad378f530274b8fa2f88953588df54c : SUCCESS in 44m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/74eff7a0d3594d6b9c1c15400c70a595 : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":5},{"id":"3fe05af00ccdc349d1ccb645b78a1d28e31df35c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 11:09:25.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848f9103767746c0a053cbef57dfd88e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4b1e527e69054279b29709e499fb5331 : SUCCESS in 24m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bb84dfca28c1471984393f6d6df5730c : SUCCESS in 31m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0c1b83c431cb4f9d99b8944730eea487 : FAILURE in 2h 38m 54s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/411c954c1eaa47298a50f1e908801bc7 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fd47deb02256484088abeadeda10eaa0 : SUCCESS in 5m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f8e102b2f3f34d11bf99444c02452d0a : SUCCESS in 26m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/65a6bd325fcc4209be79569d801af40c : TIMED_OUT in 3h 02m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e62c51d00b7b4fe5704e27c656fe23bc4340a988","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-04-07 02:31:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"aa2daab47cd53fa2a8cc5b63317f46a57561f67e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-08 10:16:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0d9b2b1a5b1c4b6ca5e8487641fe26d8c402f2f9","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-09 09:22:51.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1e98475f25a286448609e320834e9410a13e09dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 06:20:36.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4701337222473e95d1470c724e17c272b1506049","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 06:20:58.000000000","message":"Patch Set 6: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0afb5f44b899233d67e9d9f1aa9b1d8f8cf99185","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-17 06:48:42.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7d5edd9317591b67cb6286de9ad1b85770f0945d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-17 07:47:11.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"caaf7302b1872ae55617b75ff3e7df33a97cd054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 08:08:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dc650de19524526a3885a58b84bdaa5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/87d13d5554aa4b55baede2cfcbe38989 : SUCCESS in 25m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/07f7c406ec5748869a99f7fe9862c06d : SUCCESS in 52m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1095c2bbd9ab48b19a06bb29f5c1b761 : SUCCESS in 20m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b81fc7746e884df48da2c47ab83986ce : SUCCESS in 27m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2eff5fcbc95843178caa5fe901b25afd : SUCCESS in 53m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/572d33dbe0474adaa95065f0e926dcc7 : SUCCESS in 1h 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bbe12a051b84f3b8ef9538e00680792 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2239e8148051489d8ec93788eedaeefd : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86a9595d099a46168b37d67ae47648c9 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0ba3554fc2747ec85ec5a65e5c2de2c : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cd31b0ba9cf49a3834c08dfec9e57dc : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f262caadbbe49e08dbb7cc8af3afb12 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2411bdcbe5674d9b94faad5688a84c7d : SUCCESS in 3m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1933833a10dd4572b63216f4b5f7c5a8 : SUCCESS in 20m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a4d89059f29a4266ab80896544139f30 : SUCCESS in 27m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fcdace9efc6047939b782778c9171eeb : SUCCESS in 1h 08m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/56660ba981a84088aed112dfc55d9e3c : SUCCESS in 1h 39m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b72643ab12ec473fa508d396e8d2ea98 : SUCCESS in 15m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9f734267969f4f278243d98ab1111498 : SUCCESS in 46m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/83f2334d5dba4b45b51cd82118484bcf : SUCCESS in 49m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5b39c438327b476491549ef75de8ee82 : SUCCESS in 1h 20m 17s","accounts_in_message":[],"_revision_number":6},{"id":"a85467e70c17e17b7b52e17622abd2cc3272396a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 08:08:48.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7563656d8b00507afc9c9b13926aefff22321508","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 08:39:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0c6b21b59b943d4846dc3267f04b659\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/59afff8fcc29472cb4b9fd94b30beda1 : SUCCESS in 23m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3bcd980620ff49648ffddfe1f851eac2 : SUCCESS in 2h 16m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2227868625c849fbb88668d3dd3b2d6f : SUCCESS in 1h 10m 29s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec8269fcc3944473a533cc85daa04119 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6b13f92e18ca42a8aff990addd1b4a15 : SUCCESS in 3m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6bb3129511a94625b205acf4c7eb5f1c : SUCCESS in 2h 07m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/db081f164dec450aaa32bc9ab5a5ea33 : SUCCESS in 2h 07m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a55b6dc050037564227dd04df2b0e0d0b1d021df","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-17 11:13:54.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"91604caa8c7ef297a47536279ec1854271c536a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:25:12.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ead6bb5b8d534c7593054889b6ce2102\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/20dafc0520234250aa93b513d777041e : SUCCESS in 20m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bf2ec91158d44129a69abc4bd190b90f : SUCCESS in 31m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/346f866695af4129a5e60a637a69f266 : SUCCESS in 1h 01m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/202968dff97347bdbc3b872d939b5b36 : SUCCESS in 1h 22m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5907747b0c4306b8371c7d6fda2108 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38e3b6e329e64ee690bab2e578aaced9 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc8d3f5b73724ef58e01ecce6d7ba79f : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c34d40c6f4b427796feac726ca4de7d : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f595d23ed6e746f0a6a283decc091d23 : SUCCESS in 3m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ea68cff6c3e74586b6e8a4a4286813f4 : SUCCESS in 25m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c01af68451424af09c46875bd4e7fd47 : SUCCESS in 33m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7872936c91a24ababeb43333e6241e58 : SUCCESS in 1h 12m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/79ae54f575ef40a98c0b997ebf9a0adf : SUCCESS in 1h 30m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d30204a30a51422e899da80c6cc7cc9f : SUCCESS in 17m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fc1f71b90a7044128f94f333a2fc96b3 : SUCCESS in 52m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6af4e9a02da642e6bca925ea6378bf1c : SUCCESS in 51m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3da7eb6d7e0f43c5b5322e459226a061 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":6},{"id":"d195c29c19a873baf0d07a18dc1e375ec9cfd544","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:25:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"4bc4902efb5fa0d16e035e74cdbc74c7486d1fed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:26:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1616617d9f143d98d1b1880c77694a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879f1e2ed54545bbbc312f16987f64c2 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/101b9ca80c7c42b8bf452e9eb99a95b1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6},{"id":"201e8a1d0d07ab2665c60741b57c516afb42af44","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-17 12:37:55.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6efbc371f0b42daf846590bb3017711203d01d5a","revisions":{"840dbcb43fbcb39d32d944607f574eaf72c63fb0":{"kind":"REWORK","_number":1,"created":"2024-03-29 07:24:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/914744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/914744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/914744/1"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 07:23:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 07:24:42.000000000","tz":60},"subject":"kolla-toolbox: Bump ansible-core to 2.16","message":"kolla-toolbox: Bump ansible-core to 2.16\n\n2.15 patch was 2023.2 only - because we missed it last cycle [1]\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/910148\n\nChange-Id: Ic8cc29230d82f4d1f07a3c97c3dcbcd14ac76074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/840dbcb43fbcb39d32d944607f574eaf72c63fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/840dbcb43fbcb39d32d944607f574eaf72c63fb0"}]},"branch":"refs/heads/master"},"abc56fe3b31253449eec87c8cd1299f4432d8ce4":{"kind":"REWORK","_number":2,"created":"2024-03-29 09:10:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/914744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/914744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/914744/2"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 07:23:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 09:10:25.000000000","tz":0},"subject":"kolla-toolbox: Bump ansible-core to 2.16","message":"kolla-toolbox: Bump ansible-core to 2.16\n\n2.15 patch was 2023.2 only - because we missed it last cycle [1]\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/910148\n\nChange-Id: Ic8cc29230d82f4d1f07a3c97c3dcbcd14ac76074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abc56fe3b31253449eec87c8cd1299f4432d8ce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abc56fe3b31253449eec87c8cd1299f4432d8ce4"}]},"branch":"refs/heads/master"},"4126564c60c6d49c7b78a2c394194ea0505dbf78":{"kind":"REWORK","_number":3,"created":"2024-03-29 11:15:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/914744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/914744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/914744/3"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 07:23:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 11:15:45.000000000","tz":0},"subject":"kolla-toolbox: Bump ansible-core to 2.16","message":"kolla-toolbox: Bump ansible-core to 2.16\n\n2.15 patch was 2023.2 only - because we missed it last cycle [1]\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/910148\n\nChange-Id: Ic8cc29230d82f4d1f07a3c97c3dcbcd14ac76074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4126564c60c6d49c7b78a2c394194ea0505dbf78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4126564c60c6d49c7b78a2c394194ea0505dbf78"}]},"branch":"refs/heads/master"},"42ac23709ceb63832311f991299ec7ac534613e8":{"kind":"REWORK","_number":4,"created":"2024-03-29 11:16:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/914744/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/914744/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/914744/4"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 07:23:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 11:16:02.000000000","tz":0},"subject":"kolla-toolbox: Bump ansible-core to 2.16","message":"kolla-toolbox: Bump ansible-core to 2.16\n\n2.15 patch was 2023.2 only - because we missed it last cycle [1]\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/910148\n\nChange-Id: Ic8cc29230d82f4d1f07a3c97c3dcbcd14ac76074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42ac23709ceb63832311f991299ec7ac534613e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42ac23709ceb63832311f991299ec7ac534613e8"}]},"branch":"refs/heads/master"},"f712bbdbb399f2ab96faf421d864969c59f85f7d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-04-05 08:06:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/914744/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/914744/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/914744/5"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 07:23:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-05 08:06:07.000000000","tz":0},"subject":"kolla-toolbox: Bump ansible-core to 2.16","message":"kolla-toolbox: Bump ansible-core to 2.16\n\n2.15 patch was 2023.2 only - because we missed it last cycle [1]\n\nansible-core 2.16 requires python3.10 which is not the default\npython in Rocky Linux 9 - therefore we use 2.15 there for now\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/910148\n\nChange-Id: Ic8cc29230d82f4d1f07a3c97c3dcbcd14ac76074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f712bbdbb399f2ab96faf421d864969c59f85f7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f712bbdbb399f2ab96faf421d864969c59f85f7d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6efbc371f0b42daf846590bb3017711203d01d5a":{"kind":"REWORK","_number":6,"created":"2024-04-17 06:20:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/914744/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/914744/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/914744/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/914744/6"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-03-29 07:23:43.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-17 06:20:36.000000000","tz":0},"subject":"kolla-toolbox: Bump ansible-core to 2.16","message":"kolla-toolbox: Bump ansible-core to 2.16\n\n2.15 patch was 2023.2 only - because we missed it last cycle [1]\n\nansible-core 2.16 requires python3.10 which is not the default\npython in Rocky Linux 9 - therefore we use 2.15 there for now\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/910148\n\nChange-Id: Ic8cc29230d82f4d1f07a3c97c3dcbcd14ac76074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6efbc371f0b42daf846590bb3017711203d01d5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6efbc371f0b42daf846590bb3017711203d01d5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
