)]}'
{"id":"openstack%2Fkolla~915440","triplet_id":"openstack%2Fkolla~master~If91e0330dc149143c82d2183b8ddf6fa9f68d80e","project":"openstack/kolla","branch":"master","topic":"fix-config-json-handling","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-11-06 16:37:46.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-06 16:09:02.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2024-11-06 20:39:17.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-11-06 13:00:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"36201":{"account":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"last_update":"2024-11-06 20:39:17.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-11-06 17:14:33.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-11-06 20:39:17.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-11-06 19:19:50.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-11-06 20:39:17.000000000","reason":"Change was submitted"},"32398":{"account":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"last_update":"2024-11-06 20:39:17.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-11-06 20:39:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If91e0330dc149143c82d2183b8ddf6fa9f68d80e","subject":"Fix handling configs in base image","status":"MERGED","created":"2024-04-10 18:21:22.000000000","updated":"2024-11-07 06:54:48.000000000","submitted":"2024-11-06 20:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":66,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915440-fix-config-json-handling","meta_rev_id":"fe85b1f9c695eb43909178f35bb7bad7479a68be","_number":915440,"virtual_id_number":915440,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-06 20:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-11-06 16:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-11-06 16:37:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-06 17:14:33.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2024-11-06 19:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2024-11-06 19:19:50.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-11-06 16:37:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-11-06 17:14:33.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2024-11-06 19:19:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-10 18:23:34.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-10 18:23:34.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-10 18:23:34.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-10 18:23:34.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-04-10 20:35:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-10 21:34:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-10 22:11:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-15 02:32:47.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-04-30 10:41:31.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-05-02 10:03:54.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2024-09-18 10:40:50.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-09-26 14:05:33.000000000","updated_by":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"reviewer":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"state":"REVIEWER"}],"messages":[{"id":"374c51138549173b1cc16c600592b67ef14963ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-10 18:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"681b0d5f13b3c621383e4fa6911817c7e9b19d0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-10 18:23:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"981f8ecf74e7146ca913c81ff18cf8dde326c848","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-10 18:23:41.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"b6761fe7119746fafa1b06b7ad51e4709c8cb170","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-10 18:25:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35e197f04274a526e85f7ad220acfc5eb06b1991","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-10 18:55:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3efea2e3d68f0ea9f0e3068e8229c120175e797a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 20:35:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b97820819b45f6b9660b0f93812c83\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5088bfe1b3e846fab0fed77f886948b9 : SUCCESS in 2h 08m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f060924d4a2044c3b5282176f8cb13c6 : FAILURE in 2h 07m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ddb23ca09d84fb1b660687fef36bf65 : SUCCESS in 1h 04m 29s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35ccf6426c4c4d9398ba34b167ab11e3 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/496ebb5811374496a20195f5bd00b783 : FAILURE in 2m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d03f028fa644426dbcf8f9fb731a8152 : SUCCESS in 2h 00m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0d13b4e06e154f30ae1ecb2f8887f7ef : FAILURE in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"135859f98c6e8962750eb63b9ae7e87072fc9577","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-10 20:57:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b464a0bb2b38e807ce6efb2b525515861c3078c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 21:34:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d746cfcf8b8f491ab6fe8069689e4715\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2cc8191939fc4ebd947df4f6fb0725dc : SUCCESS in 26m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5906657e5c124a5a90f2ff395973dd30 : SUCCESS in 58m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/811828ef667c4a66b7a6a998bfa80f1c : FAILURE in 17m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5f7bcb16285a4553a5fee3967c88b9f4 : FAILURE in 26m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2b22937b9ae94c4386c4a2a39ac2ef31 : SUCCESS in 54m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d0360a17a69d48c99fd5d9087e9b3f02 : FAILURE in 1h 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b29918bfbe47d5b5bff7d811e71097 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b66fc389d7ba40a5a4f14c025fb432d7 : FAILURE in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbf6eb18b8ea4fcc83da3e798b99a700 : FAILURE in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cdf95026b3a43079219e921beebbfd6 : FAILURE in 5m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b842ce9120e42e396dadcf6b64d2f80 : FAILURE in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f04de50f212421193e94433669d8958 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d97cc3be89844033bad8c983773f5957 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1580cbd64578472493a9fd5eec73c226 : SUCCESS in 18m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f4885273a50346f0b76de82e4ebc15bd : SUCCESS in 26m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dc435995e2004c2bb2823a1d0a5afd7d : SUCCESS in 1h 11m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9f9201817bdc4d84b51518d68ea0de90 : FAILURE in 1h 19m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c8b5bce3999f4f22a4df57cd5d1ea461 : FAILURE in 57m 55s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d457b682c3d74c7da0710653589d69ee : FAILURE in 2h 15m 42s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1df8f34e54394f17bb3b1286b7076b9a : SUCCESS in 1h 17m 14s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/60ff393740b94ced97f668b02eab7238 : SUCCESS in 1h 27m 39s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/a6fa3812224a4d1bab6d951c3a1bfc31 : SUCCESS in 26m 26s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/80a459d0e5d84e48bc0f8585587c6bbf : SUCCESS in 1h 05m 36s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/fa5b699bd52a4fde9e36e7fa60eefd86 : FAILURE in 1h 04m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2a4c604806ea4b6298f647ddfa72be3c : SUCCESS in 1h 23m 27s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f73826c694b64a409ba637ff8d4f917b : FAILURE in 33m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/739bce820b9e45f6acf82836343baf08 : FAILURE in 22m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/04438d71263a40aa8e1ca5f66c7f1ef2 : FAILURE in 56m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5ee38d90216d458bb81e81622ec64cee : SUCCESS in 49m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/313988849ee84578b1213bf9eaa82dd6 : FAILURE in 1h 07m 49s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2219c208fd7f4676a1e407060dba11cc : FAILURE in 45m 21s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d45a755e4e2848c3b492c48aeec5ba99 : TIMED_OUT in 3h 03m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/25e84c4d986042f7aacfc6bd4d048577 : SUCCESS in 1h 07m 52s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/65810ed1664044559c53af460d6af505 : SUCCESS in 1h 04m 16s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f2aed20d2b8c4df39ae3c3d13a05eaa5 : SUCCESS in 24m 25s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a08d1a2bc59e49938d2b6fa61a18eb9c : SUCCESS in 44m 22s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f280e0660f1347a4a3a610779c312336 : FAILURE in 49m 29s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b5a9847b5b114caab28f0cb631a97cef : SUCCESS in 1h 03m 13s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a350b853a7f04d328533cd40fa201ead : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"ded94ee548a071ed82784a37c8b55e7962013f42","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-10 22:11:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d756aaae076d5d38a7d4745a381ffa6d65c0cd92","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-11 07:27:58.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"950d5185edb05e5141f2ad1f841b5ee125c94184","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-11 07:28:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65018484078f2a71b06cfde7a6a65ff185bce175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-11 09:06:51.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4e15f5a595c51be1582ac73c2a58be24570910a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 10:49:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/684da65bdf7a46bebb5754160af72026\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b61fedb059214f53a8f9bdd0e52ceb4f : SUCCESS in 1h 20m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/18dbce0ad496461598598f3a06a510a5 : SUCCESS in 16m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3a1a6ce729e346bdaa5d5f09dd80dce4 : SUCCESS in 1h 40m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c2e9cf731eb47efad4aa82550c75a5d : FAILURE in 4m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9c4f61d2ab3c416bb5d3223bb0f89cf7 : FAILURE in 2m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/07ef12e2cbc44197aa67c4f0103fa1ee : SUCCESS in 1h 08m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/390fa10c7e554cf4973d05b6b5f7a454 : SUCCESS in 18m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d2284e96f8b9e25ca1b494b0827578573b78fb80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 11:28:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53cd7959242b4386851b112e85fd9d80\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f244bbc18d3e4dcbb73bf8fb00533767 : SUCCESS in 25m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a39127d61a2348729aa44c211e4312cd : SUCCESS in 59m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0840db43a84a4cee9eb6fa96b24b59ec : SUCCESS in 17m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2a9a18ba48d34c1b8606781395a91223 : SUCCESS in 23m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/386990d5eec745d3ada703bc9966d9d0 : SUCCESS in 50m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7e6b811d3d5a405c9aaf61bbed295366 : FAILURE in 1h 32m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/561c63c52c85412383dd5d97bf9fc0da : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e86289c85434261a24b6eca03716795 : FAILURE in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef7c2eb53a634df389e86f01ce0d633b : FAILURE in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/214333f638ea492b9b9fb610ff0a98a8 : FAILURE in 5m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8ed2ff95752486eb297642c0c0f7996 : FAILURE in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4eafc6d5d7460896e264a91515e32a : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/544e8170a4c84b4cb84e3f9af5caa30b : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/12e44d9d038045c1b0bae5b811ee4742 : SUCCESS in 23m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b25c9d97e1b54b578ef190e4ceb24b73 : SUCCESS in 29m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/554e0dd87dd244169d6fcc1314086b40 : SUCCESS in 1h 10m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b6af6a435697445fbefde8493a1b6358 : FAILURE in 1h 28m 31s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/624fb43ba52046969b0a5a50fd27e563 : FAILURE in 57m 22s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1133cecd29a042d89fb2cf4855f86d61 : FAILURE in 2h 04m 28s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b544f435128a48bc8a0fada960d691d4 : FAILURE in 32m 39s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/275131d84dfa419483c4d545752ae133 : SUCCESS in 1h 24m 33s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c4a3b4ff910b4c689380d4dea65500c0 : SUCCESS in 28m 28s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/fc8cc651a19c42cd8251cbbcebf95a44 : SUCCESS in 1h 03m 48s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0ac86374ef01412d9f71ff46bd07e4a7 : FAILURE in 56m 58s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4527e3518490436799535d71f14d3fd9 : SUCCESS in 1h 19m 19s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/70b27feaf4ef4da9bc8912257fbbcd41 : FAILURE in 31m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eb1748bc7b6f40ab95517eac7bbd9f51 : SUCCESS in 22m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c0e00a0b4d1b4a17af8d52e534a6f7e0 : SUCCESS in 1h 01m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/48efed6f878d4efbb6188c5955c9b1e5 : SUCCESS in 51m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e0b2928e8b8649fa996cd5b7c24bc504 : FAILURE in 1h 05m 22s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/715c3342dafb423cafc13f0e5cb5f866 : FAILURE in 46m 24s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b2e365d2b727490ebd7ea0f39be5c25a : FAILURE in 1h 56m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c540262446704627af597a28814c31a7 : SUCCESS in 57m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/95a28fc700834284b9886778bb57de67 : SUCCESS in 1h 07m 30s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e773c657f0a343f0857fc3032346bcef : SUCCESS in 20m 20s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/260114c0b0da4ab2a6c3e77a235115d9 : SUCCESS in 50m 20s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6aabb77e210140ed8ce2118b3756aeae : FAILURE in 43m 48s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5b31c5718c574ac8b8af7cc22330eb05 : SUCCESS in 1h 10m 36s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/91479060eed54aca99a7a23c11c82746 : FAILURE in 22m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a91d6e5858b6c9c4ff11d78e0f1d886ec26bc04b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-11 11:33:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b471bc8606746128d11ec9a0e2855f1972ed0894","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-11 19:46:05.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a7c3bf113781bec30531ae612e6bbbb2b7055c7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-12 14:08:47.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1ecb9adc45b1358ce603302c0d723bd35ff32bae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-12 14:28:18.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"75bf8d2e735ff4554f7c8702a627f99cd3e15f12","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-12 14:28:51.000000000","message":"Patch Set 5: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"58385f59a8166625d791b909a30ebdb16d43c5f5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-12 14:29:19.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"604813175c13daaccdc8940b6e71deb9af97eff4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-12 14:30:03.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"4062faa7a4463144d61c4c0951ef939c64c08025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-12 14:40:11.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"16c353da3dddf10587e748850a2f794b3e9354a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-12 14:54:05.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":8},{"id":"f09f7ffc2e5e1ff1bc06c304b1b6c0f62ca7372a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 16:18:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/490e7ab4f9764cd98b58f75c5dc47a3a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6bf878618782417da496b15bf4c09999 : SUCCESS in 1h 20m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/603331abc85c478ab4f0774d5346ed99 : SUCCESS in 20m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0bd9d84f3294430ca9b453966dd88c31 : SUCCESS in 1h 18m 59s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe352c6d26f3424d845b7ec3cfd0d7d1 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c3b17d8292347c38b11a7eac20de9c1 : SUCCESS in 3m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/06c99d5ffe19439791b6c74cb3e8311d : SUCCESS in 26m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3972ce5b551444c9bf4817af61c3ace6 : SUCCESS in 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"74fe48046490059019de5d3fe94a2709648cc563","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 17:33:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9a98d74bad4b7abd3e5394be16797b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9f385e152db740d6a176804aef58c2b4 : SUCCESS in 28m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e1e0dce0dc00425bbcdb42d57f8578bb : SUCCESS in 1h 01m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9db62052f0144afcb44a0a9e29fdac13 : SUCCESS in 20m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fc9e8fcc2bc8438aa2c68501b2697fec : SUCCESS in 27m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5d750c18623b4528b5d2fc763f4287d0 : SUCCESS in 1h 04m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7028642140c8480ba87987af1aabe84f : SUCCESS in 1h 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9affbbbfa4474320a208e124074b6aa9 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6b943d8f95c48e29852bd1ef772cb5a : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfe0aac501b445a0b40410dd6bb260ae : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a2990e0bb53434fa3ac674f5ddcd67f : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55235e1b9cf8453bb08d8b51eeefce01 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33c199ad42e4897b9608701c8d6ac1b : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9e45615326a4f7db8b26910fee4778b : SUCCESS in 6m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a43ecd20f7d645e5aea25ca80e85100f : SUCCESS in 23m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d61f79534f314fb5837bab01abdfb39d : SUCCESS in 31m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d74b5664c25d44c4a1275f2d9cd279d0 : SUCCESS in 1h 10m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/68b0194ddb744532899ad87304b7b90b : SUCCESS in 1h 35m 38s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/56eaece3677b400d9c195f9e3dea1024 : FAILURE in 1h 17m 55s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8169afed29a6409e9252fb27c689d46e : SUCCESS in 2h 39m 10s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/91ef6daded7140fa808c6732e9472759 : SUCCESS in 1h 20m 57s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/0f9a2b2b39d34c0a8e28d4de39509122 : SUCCESS in 1h 31m 17s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/d80892210101446a93147384f1a3ca55 : SUCCESS in 27m 47s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a8253397c7e24731bac5b40b172a303c : SUCCESS in 59m 45s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ef616db448f14b989a855868e023b88a : FAILURE in 58m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/50545fbee94247368188e75c2ba04e6c : SUCCESS in 1h 34m 26s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f1a407d1e94a467480a38136c0196465 : FAILURE in 30m 22s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9f773536d3d14462853673b0ae435a8e : SUCCESS in 25m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a77ae9592f0d4dc4afe0204e4284e562 : SUCCESS in 1h 01m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2af5aad38700422ab7dfbe12ffc9dace : SUCCESS in 52m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8d3cda2b3de24ff0bd484ca0dcc9a3fc : SUCCESS in 1h 33m 30s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/cbd8e407d0d3473d922ef2d01d74e940 : SUCCESS in 1h 36m 12s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/771f7282a5e84cd49b0d07fcc64e5c06 : SUCCESS in 2h 34m 39s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/87fe2f2ec1d946beab9e1baa4147a439 : SUCCESS in 57m 32s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/4a63556646384fcf8e59eecb6c4949f6 : SUCCESS in 1h 12m 04s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f5c6caf33c2e4ed7967e304d413f9650 : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7e59f56e1a5d4d1a9c44128cac661c52 : SUCCESS in 46m 46s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/06c40c7ee53345d6a6b1f841dc66cb5b : FAILURE in 50m 04s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/36e76e9d066547d997b3910800114fbc : SUCCESS in 1h 07m 13s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8768701a7f904c9a9f78d6c82805f3e8 : FAILURE in 25m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"45f5b0687daa5a8dd7309d7d0698ddc74c1447cf","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-04-15 02:32:47.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d2536d34e5ce55996b6411b593f5c2a0f8eede56","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-17 12:12:27.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"165b6efc1dce07c2b5389bf0f4d2a85fac2b8bad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-17 12:25:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"91219819a5ff85fa2f8dfd6e7cf4990ab3bd0d7f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 12:43:05.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e42eb7347ec961ff33eef248d56366b82ed64125","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 12:43:53.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"418cd3a813687ddbd0dc72b9dcc9afc51af35b32","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-30 10:25:47.000000000","message":"Patch Set 8: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"97292f154767771ddaa7ab072b8db5c805f91714","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-30 10:40:15.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"fb6b581993e2f60af558465553504a9d04ebe664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-30 12:03:18.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"32daf5b8852dff27fb9cf8c4446841313854614d","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-30 12:38:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c30d1093de69cb47e2e711caee9ca9f4c0d65bd6","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-30 12:56:42.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"aeaf0a8225df1a57176562ea918f9bcbe128592d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-30 13:22:38.000000000","message":"Patch Set 9: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cc7e5000c9dcca0cffd61ef09705ecd5db45f0c7","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-30 13:34:09.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e6a8455565c578a3da5aff52b60312df051ae0fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 14:16:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8ec3621e9d469da16ce5fc710e1dc8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aee064a7511e4642a2368489b7c21b8c : SUCCESS in 28m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f8d0e8cf7ac74543bb0e891d29c047a2 : SUCCESS in 27m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/51ebcaacac7642608bc97404d0b20aa5 : SUCCESS in 1h 36m 21s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ac91c2184b245a5a8ae040d3ca1b17d : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1ad0318ee11347dabcf28a0736a7eb28 : SUCCESS in 5m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d40b8c232fb44b18817335b6e2d60359 : SUCCESS in 28m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/06a5509766ba4eb1909949c8ad44d2d2 : SUCCESS in 2h 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ab547c1e2916812512c89a1dcc72982b9417bf8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 14:50:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c31396905304e8484c33008469419be\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5153471f912044b79a38cc4e59c8f1ae : SUCCESS in 25m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/680699d65ff846f78ff911166b12495b : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5de5f3fcf2cc4ac2befa6013fa7bba85 : SUCCESS in 21m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d18502abfcb5453086ab95eeb12f60d3 : SUCCESS in 29m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5721f9156a2d4536b7835ad388f18897 : SUCCESS in 1h 04m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4dbbeccf2d96431a89033cf6bac7b4c3 : SUCCESS in 1h 51m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a8eb3dd99184033bf43e6b290337c5c : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8bf260261a74e8c93c3ccfb20adc812 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7cbbcc2d721641cfa08fe57a84616803 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b93fe9731a444968eeed9f3e8d13a7c : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/312ac7a06ac94fd2b6544bf645a137c0 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/947b5ba4975b4f58a9ea9d1fb9180123 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f8bc0a56e4343408aded5bd1cb21e15 : SUCCESS in 3m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/504a25ea240d40b0b6e16ff20e069cb8 : SUCCESS in 21m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f8dc4089468d4b1ba5b1da547090c4ed : SUCCESS in 25m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/959a96ff018d4a19830158e681bd3df6 : SUCCESS in 1h 08m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/24ce5961e01548b2ba99623ddcff1b9b : SUCCESS in 1h 41m 33s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/2d40b09d793b499bb16260823e0eea13 : SUCCESS in 1h 55m 34s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/70da8f2522354dc7ae9706b19abeada5 : SUCCESS in 2h 41m 42s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/80a83ca03b4b4af98a9f0eefb8459cf4 : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/220dc60ce8e14324af838971f17c8feb : SUCCESS in 1h 30m 26s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/bc66bd6184604ef3a97426ca38be51cb : SUCCESS in 27m 27s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9d81cbf43faa4b0aaeafd7ae363f24e9 : SUCCESS in 1h 08m 47s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/92aaf4d4ef4e48b6bedaf75878f4aedc : FAILURE in 57m 28s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f5e1a7665fae4469948cc9c44be1a11d : FAILURE in 58m 07s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9a5c44c427e247af8571c447b016e706 : SUCCESS in 37m 22s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8e531fac1bc2464faaad3c4b4c1277bd : SUCCESS in 19m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4965b8eca0ed4f6ea98b6a7d3bfadb01 : SUCCESS in 50m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/89012c379e6f4dda975dea4a206e128e : SUCCESS in 47m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7357062793174e108965cbc1c6ab4599 : SUCCESS in 1h 20m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/04168f286419448fa66b5713bb86407c : SUCCESS in 1h 29m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f5e99882eab6407b81ed19594300d445 : SUCCESS in 2h 14m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fe8dd521a6614d518800e39000380a73 : SUCCESS in 1h 02m 13s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f23391d71d98404abf684f51f984b78d : SUCCESS in 1h 13m 30s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9b8f21ffd89d4c1c8ab1f973ca6167b5 : SUCCESS in 21m 59s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/cd89dc57aeaa4c8ca552b92f89b145ee : SUCCESS in 55m 59s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c95218304d454ddf812e06ca087f7d66 : FAILURE in 45m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/95ceeca5a3af46b28a002ee672085239 : SUCCESS in 1h 08m 31s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a2bd79c241444580aeac78a48e0e9c5c : SUCCESS in 28m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fcd1e6ef2af9ed149a6c23f8cfe775e8991fa617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-30 15:08:55.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"3e2c382fb3a8581fc3a19f3edf5da049cc26e679","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-30 15:17:24.000000000","message":"Patch Set 9: Review-Priority+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a65c1efa4bd93a1bb21de17e9b5bef2a7da6262c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 17:03:46.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d918224069e341c78b062d66489be501\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/080e914e77fa4532955eb425f807e05d : SUCCESS in 23m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/61904c46416040b19c7c9108cb15e024 : SUCCESS in 24m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6034340e1a224364b5f02b7af39acf04 : SUCCESS in 1h 18m 48s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/968cd61014234946bd6a4ee24fb3cd79 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c39d564c84ea41e69a749fe90f956eee : SUCCESS in 4m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e3e0be0ed3c94525a72689cc0ae464fa : SUCCESS in 1h 52m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f78e4878cb7343b5b2557f4ca3448b19 : SUCCESS in 20m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d8e4870d2810d54dc1bfc3018e6951ddfddf277d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 17:51:48.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/480bbd9eeffa4850863ad9dd0382602e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5493015017fd4f08912ecd719e63078a : SUCCESS in 26m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9f80b3c147584f3994751ef6e920ff7d : SUCCESS in 1h 03m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2a79990b84cc43e0990a37869bdc9e2b : SUCCESS in 19m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c1e1b2a39d9b46689136428db6f08ea9 : SUCCESS in 30m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f567f2dca02f4e41b51e16468a613d36 : SUCCESS in 57m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f96becec04cd4421804964fde1bf776a : SUCCESS in 1h 24m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a922f77cb143b592429c7cdb790ae8 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5138f35a41f54051817e679d14ba23c7 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64f267c97f374859ba5d80f2c4cbb5a5 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a60eb6a9581045fb809723e3ea4eb8a7 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8c54e0eeffd4aa79c1f932a992091fe : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78b98c1f1a84b718f9763039d99b638 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/476a11262da84e67befb6d6f190a830c : SUCCESS in 4m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7d51dd0d50124b24931842f0f3c6565c : SUCCESS in 22m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1659148de43742a6bc7fe3b00b97e32b : SUCCESS in 30m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e16a60dce25f433ca81d94c500f6c31a : SUCCESS in 1h 13m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/be00da891d7b44e18acfde5e01d48cd8 : SUCCESS in 1h 39m 22s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/542099385c8d4675a1c19631a113b569 : SUCCESS in 2h 06m 03s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/603377aa267543ecade0a69ef3a2197c : SUCCESS in 2h 35m 39s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a43f10ecc4fa481094651003ca51a140 : SUCCESS in 1h 29m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/2becd6033caf4bd88c81d83732260742 : SUCCESS in 1h 32m 25s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8f2efea31f854acfbd4ead20764287d4 : SUCCESS in 32m 44s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/990d2f9fc026400aad086127987c1b91 : SUCCESS in 55m 14s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ce2e0b29e33f4d2699962568808c5ba5 : FAILURE in 1h 07m 52s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4ea5c54b794b4fd5b5da0ce14704726e : SUCCESS in 1h 31m 13s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3861f75089484cef8908fa7ad90c9ddc : SUCCESS in 35m 29s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/45e2f2229c1843f2947af0d501a45de4 : SUCCESS in 25m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/10c40a41d65440178e5c4eedbbe40caa : SUCCESS in 48m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b28d8e0c37a84d679c92d57897c947f5 : SUCCESS in 45m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f18e23c662814a85a599fde6423b7889 : SUCCESS in 1h 31m 44s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/dfb191f6752342ce987a9dd4e5036f46 : SUCCESS in 1h 30m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/59cfcdf168aa4383956372703fb14dc7 : FAILURE in 53m 59s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1b1357aaefba4a158c35236839189b30 : SUCCESS in 1h 05m 25s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/035b40bd67de4999bb539e4e00c419a9 : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f89bd48218d44d2e95d1d044c4608434 : SUCCESS in 19m 46s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e17ad67e5c2b40b9be8b20f1a4e85089 : SUCCESS in 50m 43s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/bd18e41580084fb0a64770a1c3c82329 : FAILURE in 50m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4878dc2a54304bba966af6522967b4ce : SUCCESS in 1h 16m 12s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dbbfffc3a48e4f23ae0fc9f95ec808ae : SUCCESS in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"93733359d973bde884e16fec845d4d84e42782e0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-05-02 10:03:54.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"62f863879f07a098f835f358d31c4bcfab79891a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-06 07:06:05.000000000","message":"Patch Set 10: Review-Priority-1","accounts_in_message":[],"_revision_number":10},{"id":"6a7d24aaa22204f7e21d048f7edfbb8fc4a64fbd","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-15 08:13:10.000000000","message":"Hashtag added: test","accounts_in_message":[],"_revision_number":10},{"id":"a59a2ab46fa6d95e41e77ad163f5d93e1ad40b79","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-15 08:13:35.000000000","message":"Hashtag removed: test","accounts_in_message":[],"_revision_number":10},{"id":"c340a7ca24e6c65ef43a7f234172e7c048f5326b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-16 05:24:48.000000000","message":"Patch Set 10: -Review-Priority","accounts_in_message":[],"_revision_number":10},{"id":"c4463306f48d0b4847a2bf140aea0ef66657613a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-27 12:55:10.000000000","message":"Patch Set 10: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"110c68d0ee8f2cd249cc54443fd23fe0d8c09162","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-06-27 16:27:52.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fe5619902f869cd69e7dd542ebc7b2e13e22756f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-06-27 16:28:05.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"48bdf6628499a0fb03c00944dd3fe8e905f393c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-06-27 16:47:20.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"f26c4e299a02db8616fa0bfa98bec08f2be60299","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-06-27 16:47:30.000000000","message":"Patch Set 11: Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"79f07136e120e43f91124dc15c0738c6e467ff17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 19:26:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b424794c7aa2439899b2453afa6b382f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/198dcb3377694e69924f9eda429b6eff : SUCCESS in 24m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5cb2bde8317f4559bcc007dfa435f38c : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/66d19e64cbe64d48b8acacf5251605e6 : SUCCESS in 16m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cc63972b5a8c46f381ecda224b96d6d6 : SUCCESS in 28m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/19edf50735584b909861247881acfaaa : SUCCESS in 57m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/19f4c123ee054133a4507b44087221e1 : SUCCESS in 1h 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f087af179f84e0dae21ce4e02b92b16 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8b0f881f37b4bfdbd030cddf83562cf : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94de9997479944c1955db56294783afa : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3729e013d4d54f35a5939c9e80d0dc29 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e49fa65997ea4e9e9e0b13033e82e834 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e260757bf3141a9b2d941398b59c174 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/030a17cf544f4e7494d4327a8d29a302 : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c53a55b818354dc184b6bb7a682c8503 : SUCCESS in 27m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/568450f568204086bcc5872857df1b11 : SUCCESS in 39m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f597a22fa9ce40f08ec650f6d0f19918 : SUCCESS in 1h 13m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/374f108b0a7a4f21965e71e1cf40a19f : SUCCESS in 1h 36m 17s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8e75f7595b024173a140d4b41bc86adc : SUCCESS in 2h 03m 45s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e47446b710d642ec8af07aa24256f742 : SUCCESS in 2h 31m 11s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/736ac1577d334800acc8e0e6c7aa7b41 : SUCCESS in 1h 05m 17s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/40fec7571fc74588a57b76cfeadbbbca : FAILURE in 46m 07s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/6f7ee9a00263494e820d7f3f357b6dbb : FAILURE in 30m 43s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9e405d29174542ee952d63616e710997 : SUCCESS in 57m 43s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/238c9435315a4b768e94243c7cd958ea : SUCCESS in 1h 10m 55s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/07c3d9f37a2b43d5b661c04f4a49f474 : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/53ef904a1a0c4a57b294b7dced29f490 : SUCCESS in 36m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/625fc0a842954a09a95e72ba050ed343 : SUCCESS in 18m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8ecaabcd2ad144c0ac7bf70dd6065b7f : SUCCESS in 59m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9d4e579f5d1b4a9eb54a768537bb1151 : SUCCESS in 45m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/49990dc0bc494201aaa16d4cb64b2225 : SUCCESS in 1h 23m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/dce61849bd2648f3a70df5eb06214361 : SUCCESS in 1h 30m 28s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1d5b15040ce2447c9205d8ac41e162b0 : FAILURE in 1h 02m 09s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0c003515a57e470bb89e6335e7687100 : SUCCESS in 55m 00s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e4fc9cd3d110473a96e3b7fe602182d9 : SUCCESS in 1h 04m 03s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f637ea39a0834a01815814ea425f09e5 : SUCCESS in 23m 35s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0a3337426ffd4d16a9fcc6592d013057 : SUCCESS in 45m 50s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b295f4b5d3684e6588a158f4cce2af36 : SUCCESS in 50m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/326afe628d9248c8aa15487ad67cea06 : SUCCESS in 59m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7db44dd26bdf45f6bce803cb816968de : SUCCESS in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a04c89d8146198f7f5c168299151acec135cf236","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:05:08.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c4a901aaf34d63910028d1b2576596\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c4f0672a6cb844c4bbfdc5394c99ae2e : FAILURE in 22m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/766132100cbc4e45b3ac78fc98bb67a3 : FAILURE in 15m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cddcc457931d4e9a956196724bfe32ce : FAILURE in 1h 34m 48s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3f6ada62e4894cad84eb199ac03678c9 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b8441e57f9874f5a8c5b7830603ae403 : SUCCESS in 5m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f96b990403534677a3f63d5f5f47e2ad : TIMED_OUT in 3h 09m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/906d92514bd9475ba9068c98fe39302a : TIMED_OUT in 3h 03m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"43518e30790ec7a09d589bf90f9f4bb0d800007d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-10 10:40:04.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"24ade700057a2fe69729c79b4ed5fda3ccb31b9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-17 09:54:14.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"de1ea9ae9278b97411edc25f8057c6409ca1c4c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-17 10:26:21.000000000","message":"Uploaded patch set 13.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":13},{"id":"5911822117b07b30de6a3c06845477868d4b1552","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-17 10:28:53.000000000","message":"Patch Set 13: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"128d066582d179dc61d703c926588269eb65a26a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-17 10:33:42.000000000","message":"Uploaded patch set 14.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":14},{"id":"8b5f81bd5caca233be3e38a3998cd8defa1be639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-17 12:42:20.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":15},{"id":"48c55105fcfc6e9b8161dbd0b7af26c90bb37eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-17 14:29:31.000000000","message":"Uploaded patch set 16.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":16},{"id":"07fa3ac7bd38a07111de94d465b56f92e8366605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 17:05:25.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d2b2eff97e487f92d5ed96b5fa82fe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/df7c1bb42dac4afa8c5cfabe3e009f50 : SUCCESS in 23m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/64ceabe6dacf47fc91af434f47c55f30 : SUCCESS in 1h 01m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/848a56cee5af4f2cb9ee11fb88234591 : SUCCESS in 21m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d0a19294360447a6a77afdb8c23c50e9 : SUCCESS in 30m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9d8aca45a68140b9b5293569224396f5 : SUCCESS in 1h 03m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/40b7bb215c444e1ca4abafc6a4aa27f9 : SUCCESS in 1h 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce52c4cac4744bd0a96dc96466f4a708 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c300b93e07c643ccb90e891bedc48788 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b15348bd812648ec8ad1676ecdc44c71 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f437f45983fd4ec48e328bf5460ee480 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1164368c350b44d2bca6363d68cb7650 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae740528701f4624982d6b6b208e86a8 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba14577ffb44b63a74a13dfc2ee8406 : SUCCESS in 3m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fac11716c43843c087b34c1c9807c650 : SUCCESS in 30m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d873acc32dd496594f2deffbc5ec512 : SUCCESS in 39m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/acd8c44c31ca4ec9a42ea3bb580ce759 : SUCCESS in 59m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7deacfc8916546989dd75bb187617258 : SUCCESS in 1h 37m 52s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4e3a7062fcd64a45987b59b3bd7263df : SUCCESS in 1h 47m 41s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/60166ec74fa240aaa4b75784b5dc0d75 : SUCCESS in 2h 31m 08s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/70d60cb5a210446ab9f5d70b92029f74 : SUCCESS in 1h 11m 41s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/4930c85bf5864bd2a7aac86065abf647 : SUCCESS in 1h 27m 30s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9cb42c5fe95e45418ebe01487ae53149 : SUCCESS in 25m 03s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d55cb127c2fe4d41a026f9cbac90d7ad : SUCCESS in 1h 05m 11s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a26ee4694dc447c680964c56e3d14901 : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ce57d09764fa4ad189227fe14738bceb : SUCCESS in 1h 20m 06s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/703447a671314a5180d443a3c01ba654 : SUCCESS in 33m 23s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/30f210d5bfd945628c97a57438d988db : SUCCESS in 16m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f048031f46d7469da88f7d0f42684893 : SUCCESS in 55m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d1436a1b48fb46b2b001443446b86607 : SUCCESS in 55m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f67102b916ad4b109f090664f1ce1d4e : SUCCESS in 1h 11m 36s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/63a06231d6ba4756a3650dbda9731934 : SUCCESS in 1h 43m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/381c2408e5be4e3e97ad2e759459e741 : SUCCESS in 2h 11m 21s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ac1e788f3c69481986037830453236bc : SUCCESS in 1h 10m 24s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/be90e9ba1d034453b44363eebc4bef0c : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c0b7aa0ce25b4f65924863deebcf836f : SUCCESS in 21m 49s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/80191a8afc53473cad8dd959a5350304 : SUCCESS in 56m 19s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/809262a31f5e4528aa970cb29f80b13e : SUCCESS in 52m 57s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ee8a3a65060b4abd8a847b58ca54f440 : FAILURE in 51m 59s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cbfb041f256b47dca7e47c26a8d9d656 : SUCCESS in 31m 21s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"c0106239c9aa4005225b3351e9ce89296630f452","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 18:33:46.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8820585d367e4bbd81faa914e360b7a5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0b5540c256e64ed688632a86a75200de : SUCCESS in 2h 58m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/167a10d38b554c22bae93b51a5c781cf : SUCCESS in 2h 04m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1a18da0b3c248889980a92e54c4c479 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf903a9052594387aa614e925490fe9d : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"eb0ae6eac1c01f97b9739af33e36344aa38a1538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-17 22:04:23.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"4bbbe283e2badf3c2e33e0114255f5de56d08671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 00:29:38.000000000","message":"Patch Set 17: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edc47650b72f4a1f8e8ec2effc471846\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/528cbfa3647940b68592f0982f24b754 : SUCCESS in 27m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4e3902be41da42498075c74a06ad62a4 : SUCCESS in 57m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/797142b50e8c434eb36fcee564aeb2e7 : SUCCESS in 23m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/834f5fa20cd94f9ea8476dc84269457c : SUCCESS in 22m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2b284fba83f4422ba95c95e5dc4ccbf5 : SUCCESS in 50m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3682f9feca9940aab6ef36c6929dd25f : SUCCESS in 1h 08m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b414384ec643dba6667256606a1bfa : FAILURE in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e79e21e0ac654bbf83f80257f350058d : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee473657d2ec44ba8087ca9ab68cce13 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51d76c601b7b459196e1a47250243c7e : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5edaf1ba9f34ab99d859735834178dc : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c63f089bd94d49dd9e3ac5786a75fb2e : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c85d621c3cda4ddcb0bd997618321e6b : SUCCESS in 3m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/47c0dc5401474263a4a65188f9b76a3f : SUCCESS in 23m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d5710a50a0fe434aad746d6f89b317de : SUCCESS in 33m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7afc682697994f11bfd98435d57b49a8 : SUCCESS in 58m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f489946a63d1441990e12634161b86ca : SUCCESS in 1h 24m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8af8acea389b4dc8b555a2ad0ebf6a91 : SUCCESS in 1h 39m 43s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c69d9285aeb740fe964064ffb3692549 : SUCCESS in 2h 22m 16s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/281d79c1f9aa40b78bcec1d4a92003e2 : SUCCESS in 1h 16m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/54d012c199dd4fd38e3ab1186b03375f : RETRY_LIMIT in 2m 18s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c4278edfb0a94a6396d0f477d6347031 : SUCCESS in 27m 25s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/bbe31359879946368add36df6c9f945b : SUCCESS in 1h 02m 38s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2c8d850104eb471383219203f8256537 : SUCCESS in 1h 12m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/268a741bf4154a0497a74c353bbc9130 : SUCCESS in 1h 19m 48s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b0ec4fc513bb470e8e447abd73d58e39 : SUCCESS in 34m 26s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/94e2a2f0903c42d6a879d67c0b1fb587 : SUCCESS in 16m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/003d24edcbb843329479aa9010a353f7 : SUCCESS in 51m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/139c8e5c5fb8464ba8f0d98f9cc3a447 : SUCCESS in 46m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/90709a7b8370490abaa7116bdf3ee082 : SUCCESS in 1h 11m 55s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1f29a7e097e44f48bfe1781efc69d921 : SUCCESS in 1h 15m 30s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/429cb95e3a20447d8a2f8e991eaf7c94 : SUCCESS in 2h 13m 42s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/3a4d4ab05d32484183a17d4d6f754140 : SUCCESS in 52m 48s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9bf2dd634e0c40bf9dfffa897d20f92a : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4aad6ee4944546738f3578009d9d5b63 : SUCCESS in 20m 41s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4f8bcd03f47445abbe2ff49ce15c2031 : SUCCESS in 40m 22s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/cbc84ad44a7f43df954d870523c29531 : SUCCESS in 45m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/29db8a9f324f4454802e0050bcca038f : SUCCESS in 57m 00s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a70097073ef44ed59869f908e7664f7f : SUCCESS in 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"26df80850c52b6e5ad8cf6436d26abf07a272bc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 01:41:50.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aaacdd823a84a6bb53c767aec413965\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d551480fde3d4a17895e36f8bfc8ebc1 : TIMED_OUT in 3h 04m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c9a7c14f16404419a336d81b0a21fe3e : SUCCESS in 2h 28m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e7e971b8d334fbca99cf81c0a95e477 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b2d39e7dd754ef3998d18a064de326f : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c2256ae3485c77d3cf9905e77a13395cb138ebe5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-18 10:23:07.000000000","message":"Patch Set 17: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7b3c5880f91c5173b5a730d8a60816c37f730277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-18 10:23:22.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"74d052a5252711be8d9e3aef2809a16309452e00","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-18 10:23:54.000000000","message":"Patch Set 17: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8ccea04552edd9b972ed5b370389a2c59da360c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-18 10:35:17.000000000","message":"Uploaded patch set 19.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":19},{"id":"266c1a34ac9060f5711acfd847ac1f41ccd59a5f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-18 10:38:40.000000000","message":"Patch Set 19: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"34372dd3a3bf654f5b24c368ec6018962fe818dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 13:15:56.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f3f3e46348f4608a3f73cf2daadb7c4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9364fc872ee94dd9af329e315c8f03c6 : SUCCESS in 27m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c665976e32604e99aa0a546a3dd66ba5 : SUCCESS in 1h 03m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/29f5a591d9e1484dbb2522329869befd : SUCCESS in 21m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b93c4adb894f48e69131db7050d14fdc : SUCCESS in 32m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8112fd9c8fe2447491ad416299fe603c : SUCCESS in 1h 07m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cecc013daa1f48dda9d5af7a765a28f4 : SUCCESS in 1h 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa6d4aa890e4d0b92a4356d14c8e8db : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92ca07ff76e944ab8f833e044ab3900d : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ff166e9fbca4f36835348752463da9c : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0322a3f2939943548cc84059fe65c45a : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7beb8ad9d0b84064b2e456c4ffeac447 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a59ed1cba5c497b8a1e7604eb9ded41 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/757ec98806924cdea49cfc3e0685083b : SUCCESS in 2m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fc910b986ced4825ac0150d432ed1a76 : SUCCESS in 29m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9dca01c7048b4eb6b22d9d1ed6f90a7d : SUCCESS in 33m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cb7467960b8841dd8e0bdf4efd09719c : SUCCESS in 1h 01m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fa8ae7ded18e4ad39601b4963002c212 : SUCCESS in 1h 25m 10s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/2fb3f409e42247d7aabb398dc305320b : SUCCESS in 1h 41m 55s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0ec26cc6e70d4367beef43d9beb5387f : SUCCESS in 2h 28m 18s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/aea78ff7d95f42739d406f7e513d7bec : SUCCESS in 1h 17m 51s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/310ace75cf824493997ad5c6c1e5ed01 : SUCCESS in 1h 23m 01s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7c4e3f80dbe64b40804c2f3a0226656b : SUCCESS in 25m 16s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c969fd675e3348fdb1aab6db2227f20b : SUCCESS in 54m 42s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e5670df85eb348bbbe58e1ecab50af88 : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1eaf9748da12441a9938da1f36638cd9 : SUCCESS in 1h 26m 03s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3082720390584e5897ee36d8c8c26c91 : SUCCESS in 35m 41s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/361a536ee2734bbe956156017681d4bb : SUCCESS in 19m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cbe97749659e4e7898f1bd5d9f085299 : SUCCESS in 1h 08m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/45bca238774642e7b945b5133abc4848 : SUCCESS in 46m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ac4fb0354d154bdab0ea4a001863d63b : SUCCESS in 1h 22m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ee191684d7e845c68fbb67bdfd08c1ff : SUCCESS in 1h 28m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/87fc5f79969940138125aef8ffe53f2b : SUCCESS in 2h 27m 21s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e5eb839d290d4d988bde32de38cc842e : SUCCESS in 1h 02m 20s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/dac9339ef2954443b2ee78b16c0557b2 : SUCCESS in 1h 00m 13s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2e2a54b5005143559458ea1e22091b3b : SUCCESS in 21m 51s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/26469ac12ac247848044358eb52adca7 : SUCCESS in 42m 29s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6b083ad860564468ba423f0c416219b0 : SUCCESS in 52m 42s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0199c0f7971e4e4eb4e7198478cee30d : SUCCESS in 1h 06m 02s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a7777f349e5f48799ebfa18eb5ff5947 : SUCCESS in 23m 19s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"24760de8718ae82a8e42d7f4a20fe5c77374d2eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-18 13:36:43.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7fb88d408f41caf36efb4c9a721a33cbc09eca6f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-18 13:41:37.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"0633e2593c49c485779dae1dee7d00b9f215be29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:52:44.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18ed2a95fb0746df90cd41529274e5c8\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/392ebde626914b408e07602d078d0351 : TIMED_OUT in 3h 03m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/362747bf78924cc38e65e83a6cd2a898 : SUCCESS in 2h 10m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b88ad7d46d0486481a7f6bb1f344181 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6aaed1d4ec7b4f3482b835ac14d08118 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"a3ea13ecc06af59ec8d82a60aef4c913eeeafa3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-21 20:18:03.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"4397364b8254627600491c32446885370fa4ef74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 22:07:19.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f671991a61846b4bc5f584663dfd30a\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a0864c78a388475185f877d64c682bfc : SUCCESS in 1h 47m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d7f5701dcdd34db58215830a413736f8 : SUCCESS in 1h 17m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3c874e2a72442f1a9c7ed981abd7604 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f3eca826a76b48f8b52b3ea6bcf12368 : SUCCESS in 3m 07s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"3c0eec8c6d9a70f26ededdb005c860f00ae650ce","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-21 22:33:53.000000000","message":"Patch Set 20: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"007990572ebd89f02e1634af20cbb092b551a7b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-22 01:19:18.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/219c7fbe266f44b580048f2c543054dd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/58a99b95205c425c941a295f79ad3c92 : SUCCESS in 27m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/98fd8d48af184348b5fa930b515b14f6 : SUCCESS in 57m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/de9b2d2d34ab4715b5e703279b62476c : SUCCESS in 18m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a2ca8218f251470a9591d26b825a5c60 : SUCCESS in 32m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fb60365001114995b2a815e10e3829c9 : SUCCESS in 52m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/936c7dfff65c4d47b5e1859cabbc2f56 : SUCCESS in 1h 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76f74977330415692389ec3bc47cdb2 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56bb274113dc4cd48fad329d0e8cbae2 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3efd77207d3d49f397a7b11107632396 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1c712b293914c64b039922ddc88bf7f : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c82a89b86974c6e832c60c1c56100f8 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2c43556c96438086bcf2b9e924b3aa : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5869e3a72e314c41a7364eb248ae75df : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8e05e12e54bb4ecfb8265b6a888c6977 : SUCCESS in 32m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/45df5aed554d4ed2a26701f74dce9cd7 : SUCCESS in 36m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bdf4479b0a0243fbb744fc1e0daec1e1 : SUCCESS in 1h 02m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8f18120cc297474888250e8e4c7685de : SUCCESS in 1h 25m 56s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e82a4d548cde4dc8a44cb6d4bc7b0552 : SUCCESS in 1h 34m 05s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fd878f6c0afc4bff909fd2c23fd94597 : SUCCESS in 2h 00m 18s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/497176aceda84478a90703ea0ded78e4 : SUCCESS in 1h 12m 11s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c0c75f29784d4e138b7a883e6de676a5 : SUCCESS in 1h 18m 14s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/020a52500f0446f3a0edbee059d8b769 : SUCCESS in 27m 05s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/86f606d770a2457a816307e2ab8baf32 : SUCCESS in 54m 52s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/149ef7dfa97542618e3d21aed76c023e : SUCCESS in 1h 02m 45s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5c681716afe243afa97a739e42cd2e58 : SUCCESS in 1h 21m 10s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c458579be2684374aa31083963caa81e : SUCCESS in 39m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ab5488d6873f41ef8fb5a641a6655483 : SUCCESS in 14m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ec49e25e9dee4012b2d141b0c1f42177 : SUCCESS in 50m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b546c47ea16b4c75aef8f778f8cfbd3e : SUCCESS in 43m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/80633f71bd6946afb4595b2f6f1443b1 : SUCCESS in 1h 07m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/468d645fb6504a7e8fd037c1d7d26782 : SUCCESS in 1h 34m 07s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/059f2864a19e4b99952466c530d15b39 : SUCCESS in 2h 16m 46s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/332f6f12819f406a8c8d3c0e00f64b54 : SUCCESS in 56m 57s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/ad0834a356ca48e49971d74d9d91785d : SUCCESS in 1h 02m 39s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7499f0c650124e4897425e21037daf52 : SUCCESS in 20m 59s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/94787baca51340b1a5e237872e742612 : SUCCESS in 44m 35s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ef63934857b040cfa3d62655fdbd9d32 : SUCCESS in 51m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/87680674c9754c74b380909a0b104df5 : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7c4807ef8a754b20b11bfcaa8c49620b : SUCCESS in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"5e9f67af0cb2b66a6b9a21fa0b689c69a0edd714","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-09-26 14:05:33.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"c7ff5642fa62bdd828b2d92a65f21a1825d52198","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-03 08:33:22.000000000","message":"Uploaded patch set 21.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"ea78453ba501038767a350b2d1e408dea048344a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 11:11:58.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03251657c2644a1e820f0e83cd85fbfe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/52118c46b4f44a4c9c363d268dfcab5c : SUCCESS in 36m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6983f1d3db6d404ca371169458e43d3c : FAILURE in 5m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aa5722e2086240faa22e29d1c1ce0c4c : SUCCESS in 31m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c5df01b940184e5c820ded4b5b85cd50 : SUCCESS in 27m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5ea0be7a2c354458bafbab98e389d3bf : SUCCESS in 54m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e00dc95873124edc8d3dc98e1b9b8bc2 : SUCCESS in 1h 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd32754f7d3c4965937111db33241d6c : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/500710e891b24a77ab060a03dda41265 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a78fd29691a247deb101fa1bc66b43a1 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b12f0c3d004b43509c1526b11e37d377 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1545e1b22fc4efbb58a3251c263938a : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f89e7b727c694752b96179fc7258a162 : SUCCESS in 3m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5143019354d74518bd2c64e63a55b115 : SUCCESS in 23m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e1be2d6c44794540978aec44973098d7 : SUCCESS in 41m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/70766119dd5043388c0fd3c8bab4380e : SUCCESS in 1h 02m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/32dda27766dd4569be534d5d52e71389 : SUCCESS in 1h 26m 59s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1233013ad00b46d7ba69eead66fd54f0 : FAILURE in 1h 18m 47s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ab5f034ddef342ffbaa9326d42f823b4 : SUCCESS in 2h 26m 27s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6142dc99e3844e68a4011dbc8dea594b : SUCCESS in 1h 11m 25s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/dae8e7a45b0849b18bb5589dd892999c : SUCCESS in 1h 28m 31s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2ccd6130f55946238a48140f77b41cd5 : SUCCESS in 27m 20s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/47bb1daa0e8d43d28765f5a0f40b65e6 : SUCCESS in 58m 19s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1914133f69da42739c8012734c7db60d : SUCCESS in 1h 21m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a1a1978948064f9f885e24d7fac54bd3 : SUCCESS in 1h 24m 32s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3bad0f9d5a344c34a3d393b91f22837b : SUCCESS in 29m 37s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c34541a3f9dd49af83829a782ff66eef : SUCCESS in 22m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a6c359cd798e41cdad4eba20f64ba00b : SUCCESS in 25m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/78d1ab09c53e445bb5de529a79bcdacb : SUCCESS in 56m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ee79db7c2cea4e3592c267a1f787bec6 : SUCCESS in 1h 11m 52s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/97832c1127914e5dbe4e8b15016cd7a9 : SUCCESS in 1h 32m 42s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5ce0a1be03cd43e699f278f6871f9ab1 : SUCCESS in 2h 15m 54s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ca15faa681114cb4a14b107a9eb1879d : SUCCESS in 1h 04m 07s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/93f385d2240d48ed8e0e01ab84890f98 : SUCCESS in 1h 14m 02s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/09f0d234655045bcad00e972a285e773 : SUCCESS in 24m 49s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/11403a1579004235bd367b1e216d2bf3 : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1c7e87c798194a6ca46e210a517b2ff1 : SUCCESS in 1h 11m 51s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9021593fa12c4f59a744972230bdc8ff : SUCCESS in 1h 25m 50s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f41c6c8a13744341994e61164dbaa956 : SUCCESS in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"34ab6b1df91d13df0ec4f606a58305fa76b07eba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 18:15:56.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d713b2a48646efa1c97710ef86dffc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/058914dccd06412daa619b75c9169495 : SUCCESS in 36m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/df1e688865f54f88b1f8fb470d02c78c : SUCCESS in 26m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a1ad7f2552440a8bbd523d54764ce51 : SUCCESS in 2h 11m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7836b1b386ed42348739dfc766ba8b8b : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f64cd74a1edc4f21bbe5c6f87d21a42e : SUCCESS in 5m 03s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/24dd3c8e874a46f5a1f3d240e5a826fc : SUCCESS in 30m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c158c074b5f548638762656050710054 : SUCCESS in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"cb97841d1157412827d99793bfade0c8b7f2037a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-10-09 19:36:16.000000000","message":"Patch Set 21: -Backport-Candidate Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"db37f27c0c3ce233c970e01a88410c97422b2566","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-09 21:24:47.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4a19568b3f1e4da121cb0429c3167db7ee9ac3b7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-09 21:36:52.000000000","message":"Patch Set 21: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"d54e0cc189faf560306a1d8905f3f510a1e8bfa1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-09 22:13:57.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"a28d9af1db4771c565ea35a9cf1c8176ec2b3075","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-29 09:16:03.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"ac8ec94a8a8538145403d225235f31577ba9892d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-03 11:26:11.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"315a491e66506ddebe14e9c39a600859ff9cb6cb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-04 08:06:26.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"67ca50e96823eccf0d51b3b59988b379f4d0f0d5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-05 17:42:50.000000000","message":"Patch Set 21: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"37cd2f65788cc9b88d06843e9118884dcb5a31f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 17:43:12.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"1eead2bd51714d3e75cd4b73306dd700e8a81a30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 19:11:14.000000000","message":"Patch Set 21: Verified-2\n\n(3 comments)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8daafdc187ee46adaad3cad6e72faf70\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/66ca7362453d487285cd68f199a687b7 : SUCCESS in 21m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/60f5660f7bce4e498274b251221cbe86 : SUCCESS in 26m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5bc1cc0c78264625996349fe75416c15 : SUCCESS in 1h 04m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/056d2f76149e4883bfb0b56769fce79b : SUCCESS in 1h 27m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e07ea7c42fa48c89b6e04301b177244 : FAILURE in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a22568d5a03f4aaf962b1523b85dbab7 : SUCCESS in 2m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40224cd17e2b47c09fcbbc9b7517ebb0 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14273b379d043778929e3a6d059282b : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/638a858fcfa0442183a7b1f927956c30 : SUCCESS in 3m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7c511aafc728461f96f8be49caceac09 : SUCCESS in 27m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d3aa7f0b092a4097937d0a2e0c983c63 : SUCCESS in 35m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/470d7a0271c94ddaad53af9d4236d642 : SUCCESS in 1h 02m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b80278c6b0604c56ac4bdc8825201058 : SUCCESS in 1h 27m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fc2c0baa291c45b295c6ca13b8469485 : SUCCESS in 25m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c187518c36314ca9b1f587057a7444d4 : SUCCESS in 36m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/11ffb9f8d49740c3b611e4d51d53aaa9 : SUCCESS in 1h 09m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c08ffde374544803aa19c3bf188ccd99 : SUCCESS in 1h 18m 07s","accounts_in_message":[],"_revision_number":21},{"id":"5385d844494311f0e167e058f50c0b2b51ab881b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-05 21:43:26.000000000","message":"Uploaded patch set 22.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":22},{"id":"cc970a21aa5af586368a0dc77a345dfe60b74b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:14:16.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd4a2308b5654e6c9fa99c079d123fc1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1dc4818edcef47d2bed2df089a4ef2fd : SUCCESS in 24m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5825f5256a534c8f9f3ab18d56e7f5cd : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8f287ac98e2e45c3a1a65c8bf35425ac : SUCCESS in 19m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/316885e79ce44e27ae52f50a5f554384 : SUCCESS in 30m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/028458af804149bbb68bec02ac05cb7a : SUCCESS in 1h 09m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e30083355853465d9fbf1c21dbc29eae : SUCCESS in 1h 33m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdba169a9764cb5bd2de45aadcda225 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f345d9bca8e8490fa335805188df101b : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b639afc67d64407a43d4dffc9146120 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/819b1cb118204862bbbaf8951b4f31d7 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497b48d647cf488abf4992cc6a8739c6 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a704a8314df40528109a10f8e5ccb85 : SUCCESS in 3m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/275df2f085f6438b99a8f7bce49ee63e : SUCCESS in 20m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7d42ff4c25d84c509894c6e286425587 : SUCCESS in 30m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dcb6d2cd5f214063bdbb4a45bd5a53a3 : SUCCESS in 1h 08m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0f57efe3a62f47b49aa1302234f1ed1a : SUCCESS in 1h 14m 11s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/58f8e9efeb0c4bf586828d660fcc05b4 : SUCCESS in 1h 50m 35s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/81c1c99abb4548e6a9b25e49d3f1236f : SUCCESS in 2h 12m 17s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/59fe348ab5244366a807be297c703611 : SUCCESS in 1h 13m 49s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/fb6bc015a2b049f48029700c6b4878ac : SUCCESS in 1h 12m 47s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/31425aa754df44f58c4e93533da6acbf : SUCCESS in 23m 16s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/2271a1f6ad2947f186115559c1f062ad : SUCCESS in 57m 32s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6c8a07e1a1ab4627a11490ec1a1ceb21 : SUCCESS in 1h 10m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/30c56a6efbf649eaac97283fe287fe61 : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/58d4ea20735247be99ffd9b14cfaf814 : SUCCESS in 26m 43s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/27baf0cf6c014984a74ca8be48382af2 : FAILURE in 22m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cb60d72cf4f74445881da4ee16c1c4f5 : SUCCESS in 23m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/86f0f9d13f254343985c8873ba0a3208 : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0e981940a31b4bd2839ae139d5cdb349 : SUCCESS in 1h 24m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/57bef6d205e940039fd6c7bae1c29dfc : SUCCESS in 1h 46m 04s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2788e7ad92364d5481c90e4ffcb0d24a : SUCCESS in 2h 24m 54s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/63d39eed36d44a4086894b2c50cfa4f3 : FAILURE in 1h 06m 31s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/20f5e5707dcf44d8a94d986b14be84f5 : SUCCESS in 1h 18m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ec84fd2e24af44d680c29bcb98585eba : SUCCESS in 28m 06s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0b04ceafd2a342cfb84de881855d2b4d : SUCCESS in 57m 03s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/da65b776de8d4c7d9e9bbcf40a44f795 : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4378a61a7d7a423fb1e88fe69fbb8d03 : SUCCESS in 1h 25m 55s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d3a4bc2c2df84d6181d509856678fc43 : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"46daf5a147366cf773a23ba9963cff1719c89145","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:52:46.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f823a3ac2a49f584c53207af9139a0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b1d5fc2acc98462cb197a012c6ea793a : SUCCESS in 2h 34m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/50eff0e7c2b947e3a3733cde38e2dbc3 : TIMED_OUT in 3h 01m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b2ca05f6418547f1a9b6921e883c6c63 : SUCCESS in 1h 47m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cd98e855dc44b0288ff0227cba2bffe : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6ec11be6ca0748f993cec892f99733a6 : SUCCESS in 4m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a5d85203189042aa83cb6fa31d6aa61c : SUCCESS in 2h 37m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3a3dd25b88f0418eabce036dc5d56ca8 : TIMED_OUT in 3h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"15ff09f5e358c2c68e9c4c0128c632abb910d712","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 03:58:20.000000000","message":"Patch Set 22: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a2adf38b786e40bafe8669655de797b1615f15da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 07:07:53.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b45907e0a89142629111578318b65acc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9bff6b76563841f5a3b0a0c112e4fed0 : SUCCESS in 19m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/68b06401f4544973978b5780b7b49b03 : SUCCESS in 52m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d59804e7c01e42f7850e7e7ea8d3c4fa : SUCCESS in 21m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9e65b810b30e473ebab29fe7ecf2688d : SUCCESS in 30m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3063e180221f467fbc9011b7160337b4 : SUCCESS in 1h 05m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1a9985ab6442463d8aa1568315b6d30d : SUCCESS in 1h 29m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c43a3ba27143eb8d80758dcf8b4412 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61566a0313174b0b922bce16dcc01a5b : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f69fda73d8454232b774eafabb222096 : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ef1b01ace144a828a88de6db3f56038 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9246979f77884b5f8aae76f841ea7a58 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e14456bec84a56b0ee0f21641d8520 : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3614f0d353df4bb7971c37b0a03cd07f : SUCCESS in 22m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e180228d3c964fcda58ab4c5bc4a739c : SUCCESS in 2h 38m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/69d54e32bc8543b9be455ed2de3d4867 : SUCCESS in 55m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2558e9e6e10b42de86e8c7e0cf31bbc6 : SUCCESS in 1h 27m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8bb54d1d06b243159b19629f9a0f6752 : FAILURE in 2h 36m 54s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3814da0d929943f0b4e69277ce77c457 : FAILURE in 2h 16m 53s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6d5523d173e44e1e89f07424df1e50ff : SUCCESS in 58m 26s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/f86a88843ac74c2dafb2ca90d0849a3b : SUCCESS in 1h 08m 03s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/69b189f21bb8401dba0cde4ef9cf1671 : SUCCESS in 23m 01s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/30c05d9c8d6a4061a1817905a60eb34f : SUCCESS in 1h 03m 27s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/48674cddad084d53b52f4c93abb6ca9d : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/bebca5cdc6a74353add89e367f526d46 : SUCCESS in 1h 08m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/08c2aa815a2944c4aeb4dc00d3468782 : TIMED_OUT in 2h 02m 02s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aa3d873647ef4e8f879cd67482e4cf7a : SUCCESS in 20m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a602a3cf2ada4ffdae9510cf19b22141 : SUCCESS in 28m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/06642cecf7a041508dec1ea3d6d39295 : SUCCESS in 1h 03m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bbe321c07ad542c484392e03eda916bd : TIMED_OUT in 3h 02m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8ea4695cb3c7407fa66955054f801ce1 : SUCCESS in 1h 26m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/857348fac0fe45b78d70f3fcbcdd7050 : SUCCESS in 2h 43m 04s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e61397531df34681b7ef4d043f8d5f28 : SUCCESS in 1h 13m 36s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/57801b3b413c4fe8bcb284d73fa2c1f7 : SUCCESS in 1h 20m 49s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6f4fd34aaeac4656ad5db61242937c71 : SUCCESS in 26m 17s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/d4dfbf7d3d6443aeb50349a0c56a3f8f : TIMED_OUT in 2h 01m 59s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2f34a733b2d84a8baa7d06b1000518e1 : SUCCESS in 1h 10m 14s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/59cb0acee21945e1a16392e455a9733f : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6843cb4cc4b44a08aedfb2cd0e3fe977 : SUCCESS in 25m 35s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"143988a829bfbd4019269811f31dbd84af2d2e96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 07:28:08.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcad2c99a7d54d4b96b030b992a9a9df\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a9dda476bfbc450f87618437ee2aea49 : TIMED_OUT in 3h 01m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c113df6aea5b4a8894a202a9274b1dba : TIMED_OUT in 3h 01m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4fd4ab377db048989d5fefdfd1c525c7 : SUCCESS in 1h 32m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/773acd0bfce24f1ba965cac4c68365a2 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80ea131b204948089a1cadae8d880f73 : SUCCESS in 5m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5ddaf76fe15444e58484109e0b5a8734 : TIMED_OUT in 3h 02m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0c613a98b78e4b4496fe5b7b1946ede1 : FAILURE in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"0253907531c66664c99889a9d2551cc8ddbb5a3e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 07:48:15.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"fea0448bb6b098590cf1fe2e58520220db261b58","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 07:48:23.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"31d2e19623cd5615fe0745118e22d111551dcd72","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 08:36:31.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"b17ab08248df9fdf2df538a2bc67b13b4c686cdc","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-06 08:50:27.000000000","message":"Patch Set 22: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":22},{"id":"5a682a02c5010e25473589d61df4cc5351f6ff11","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 09:04:10.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"575a1cd7f8b505abfa1e050ef35b5178e96f7404","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-06 09:12:50.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"5db7512c5f3ff61ab1c1f621165cc786e16ecf7e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 09:36:41.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"7b15ab78967040dd3f70a5c816cb4472a29d9c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 10:11:44.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/899c8c0ec2234563972b3b6626fa3bd5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bde07c2b52704cfbb54d4e41cc10e516 : SUCCESS in 22m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4859524afe0b40078b83afc5644d30f1 : SUCCESS in 59m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aed42c98d59347faad1abd2b710cc0de : SUCCESS in 18m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6cd4f8975307462186a3a49065e9ede7 : SUCCESS in 29m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0ba9323fe5b64597af1a70294292c53b : SUCCESS in 1h 01m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2fd61f09236b4465a825bfb7e84dfb56 : SUCCESS in 1h 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046ba1cc77494062a6d2af8351117aba : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f90cbefb26554f92918d3e7e79edeb57 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44f4f697ef194ea09120522663d6196f : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7797bcd169814d749c1a37d49dfba89f : SUCCESS in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e788b47a9bb54532997033937160423b : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32e3f2c62fb44463a667ae21cebd6534 : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/96ccb1c6bb2e4d3fb0dd55ead781506f : SUCCESS in 31m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2e8447f30a954236a37e78e82d885010 : SUCCESS in 29m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/35012307f3c6415db60978932b9f8688 : SUCCESS in 1h 16m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1c200da42da64caab91d3a4a84bdac91 : SUCCESS in 1h 31m 22s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/21e94d24dbd14b7f987068dbb9326b92 : SUCCESS in 1h 47m 39s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a093dcdd811d49df8d1b4b0e7e331d58 : FAILURE in 2h 20m 00s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e12414ebf71e44ccba6ab2e1eca730b1 : SUCCESS in 1h 01m 24s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c1c68b419c8c4280945b333a0f4f4883 : SUCCESS in 1h 16m 02s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/75554aab7800424fa6ae919e8c5f15fb : SUCCESS in 23m 57s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a5da36c80b0b48f7bc81513e4ec32552 : SUCCESS in 50m 00s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/cdab28baf2204d5d9c405a571769091e : SUCCESS in 57m 57s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/928a7045857a4404a08b441f878c0001 : SUCCESS in 1h 15m 56s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3d548eb33cc54f618670c2449ffb4bce : SUCCESS in 26m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6fe1f645c257409b9d00708acf924ede : SUCCESS in 24m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/54744effa1894c95aee3b74158d63f19 : SUCCESS in 35m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3a42f2704b7e45248a75999ebb8b7335 : SUCCESS in 1h 05m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4b6b96447c6a40ffb65530bcaeaabbe6 : FAILURE in 7m 09s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/13809d8163094b06a29f40ad091e8eeb : SUCCESS in 1h 41m 08s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dee42492507f4962a4303f7a902648f0 : FAILURE in 1h 51m 46s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/431721acfcbf48f9a2ce36ac5f3f476b : SUCCESS in 1h 10m 21s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b85ffbb3b08045d78a627b5b32f537bc : SUCCESS in 1h 18m 58s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8c8e39b94d264f9ea6d6ecb2efb8de4b : SUCCESS in 22m 37s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c31052f0a1634e61a6b0912b159efd5c : SUCCESS in 58m 39s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/86e8a82208f1470eb26ac201e8d2e59f : SUCCESS in 1h 13m 31s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5aaf9b66fa974b729165c92dfb93a971 : SUCCESS in 1h 24m 40s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/19b956ae8cef4a10b05ab2e9203ceedc : SUCCESS in 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"fc0f8665bcfe96afad3bf62858e821813b582005","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 10:57:53.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97818b543d7c4402aa040830fdf4e33c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/473f85e300bf48bfa89860e5cac18f40 : SUCCESS in 2h 02m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4e12ec54888e47df98b7b0343d7e1d92 : SUCCESS in 2h 26m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4aa7d8babb40405fad0ca208e2d0c0cd : SUCCESS in 2h 05m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0715a3c97c474dd99941de42d3041b92 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/294e261c52214151ac1079fee677d594 : SUCCESS in 5m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b01cd74311534f5b9c38c7f4d542c19c : SUCCESS in 2h 07m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b30ab260844a4186b8305105bbcb1f8f : TIMED_OUT in 3h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"256bb269d6b88efb42aa1fcb714827b81ebc7a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 12:55:36.000000000","message":"Uploaded patch set 23: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"deda4b12e2a7515dac931e7040c879895811f87f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 12:55:49.000000000","message":"Patch Set 23: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"ea071f596f2f50d4485d6effa9e1c5748ad60584","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-06 13:00:13.000000000","message":"Patch Set 23: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"aff3b4215267c7c29f68431096f11c44bb9d0ab2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 16:07:00.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f04aa2a9ffa1412fbb47e7e5df9bcbed\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/67ec5b64a39441e783f0957a9566fceb : TIMED_OUT in 3h 02m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc7801e5a4ef4ea29652b79f09044f5e : SUCCESS in 2h 25m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e7cda9ca0ef45918b01fba205157e7f : FAILURE in 2h 17m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22edc6b2598a44398c22aa947ca93a35 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f9cccc05e8d43028073595d78f82d96 : SUCCESS in 6m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0a10cc50459e4a188a1ddbf0f8588cc1 : SUCCESS in 1h 56m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/25fa9be80fbe48f184dc0faaa6d9a3a9 : FAILURE in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"fc4e637833a5a9e3d80d613d0a3e131557d720cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-06 16:09:02.000000000","message":"Patch Set 23: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"9a78add78a680244f3c92ad1ba5974c736700f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 16:19:04.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bcc5d983cdb4464808c4fc50b7a68ef\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/98b4339f65544d0280f5ba04eb0eb258 : SUCCESS in 23m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/44a9b81e932047548593ae2493d7c37c : SUCCESS in 1h 07m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c53d9797150041e4bb6ea3df7832f1f7 : SUCCESS in 21m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/70bae33ef3874b3f8fa2788dff2220d6 : SUCCESS in 34m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8b1e5890e051468a872cd9feb159449d : SUCCESS in 1h 00m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ac06e9d482354e11b3c6327a8f10b9b0 : SUCCESS in 1h 25m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b3fd8da3cc47e9ae35a766a9bb61de : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8da8bd2f18d24105a5e3244352774554 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3529a1ed9b6d4127b38ba40e4f9d695d : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e43ae7c744004cd58b25216d52240446 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fddb46e8695a43e7996f4d0f30881b2f : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1d7ffe8f83f4682a5d2d66df6ce9174 : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5dd34802e07245df9656c9455dafd52a : SUCCESS in 20m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/55bf01bc8b0a43cab5892c24bc73b3b2 : SUCCESS in 40m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/225e154259b94f18a1848ad0163063f3 : SUCCESS in 1h 00m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/eef31bb8315e4887b43dd8cf2ad9ac92 : SUCCESS in 1h 24m 47s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/54963359fe2a4868ad6ba5c0d7d8b411 : FAILURE in 1h 16m 47s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ba6402b6973c4942b6c0c5e5b0d71511 : FAILURE in 1h 51m 51s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6f4162b701ae4af7a202ad0587c0ba96 : SUCCESS in 1h 09m 21s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/3f550e0820a843f98b44bb0b54cc2bb2 : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7d1277a276ab457fb1b5b652a402708c : SUCCESS in 24m 02s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9b910a0916bc43589be3415f7b94321b : SUCCESS in 53m 48s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1da145bb741647f9b4533848b89cb47c : SUCCESS in 1h 15m 21s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c1ba7c74fcda44098495e126656530e8 : SUCCESS in 1h 17m 17s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f187c7bf879143b1b6ceea326be2bb09 : SUCCESS in 36m 24s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1167dd93362545a28a7852aeff77b08c : SUCCESS in 21m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d7799fc6805f47349954f01e781171c0 : SUCCESS in 27m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/53570ef1f9cd48ef82063aa0b309cf75 : SUCCESS in 1h 03m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8ded3a16609d477881482625764c13fc : SUCCESS in 1h 23m 41s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7754a9ece97d4768a2fe41dde063da0d : SUCCESS in 1h 45m 51s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cf03b5f3a0f74bee9bae70e934324d44 : SUCCESS in 2h 36m 06s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/165bb401cc43494b918ff5150528301f : RETRY_LIMIT in 1m 34s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/88e75746cd6b4a3594d94dc3bb32cec7 : SUCCESS in 1h 12m 01s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/551c2eaa782e47579cd6ee605ba98222 : SUCCESS in 25m 43s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/50413387e6d24fa7acabcc1173da6324 : SUCCESS in 1h 00m 48s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2a4b93f4e3094e4fb3e3b29b68ad54fc : SUCCESS in 1h 02m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/094a5f712960403296d1b96d51c02533 : FAILURE in 1h 05m 44s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1609203997904636b6e9d1b3dca32407 : SUCCESS in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"523f6ce04ce704106c4a5798c6ccd05bc78f5097","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 16:22:10.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"9f8a888b6369d3201ce35d30b4b9ddc714340e7e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 16:22:23.000000000","message":"Patch Set 23: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":23},{"id":"2f1b1921586d9ba00e14dfdc3652decdd7d46ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 16:22:34.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"63c9bd552a756c3a4dc094de7f594a99a190f77f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 16:22:36.000000000","message":"Patch Set 23: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":23},{"id":"74844c363dd6aa76ee205b76c3a5f7f4f839ec3d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 16:22:40.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"481cb6d788c7e6120f95bb23214fe86bb9436222","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 16:23:13.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"b96b6b74d450ca28ffd337be973545c9e3fb387b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-06 16:37:46.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"b7d17e0ba7a6105e84c4146f17f2d376794a0b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 16:37:48.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":24},{"id":"19fbc681f7605ff56332ef55ef06993ccc816377","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-06 17:14:33.000000000","message":"Patch Set 24: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":24},{"id":"1342a57434bd42ceec4e8f5a25a366c1b699fc99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 19:13:02.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee7f21e0416249958af7e221cbee2afe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b60a569bd80c4835a91fd223b7b17406 : SUCCESS in 17m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9352b58a717c46fbaffe34104be67305 : SUCCESS in 58m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e467b44d9ca543eda09b5fca62878708 : SUCCESS in 22m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/556a4ebc2cc943a4a6538e265a9cf085 : SUCCESS in 25m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3736cd5840954c4b8032f8d2bb9d812b : SUCCESS in 1h 00m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b82b906b9fd543a3ac922ae8c5623f5c : SUCCESS in 1h 22m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62eb138d38448bbbd71ba3b5e770b1b : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15ceae7c99e0435998392bf577903469 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cea22275ffc477c8d78dffe2d0fa50d : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64824ac5bd844f16a588e5cc032716ab : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c95c59f83da24ffdb58a02254cb96290 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d99d4d2c4b584546a1cf3e4924fc5c2b : SUCCESS in 3m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/25be71e95eb643e48004676d64a4006c : SUCCESS in 24m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/96b23687e8cf4cf0a305f81a038b5ad2 : SUCCESS in 35m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9ef763cdbb0840bd9967ca767b4dae55 : SUCCESS in 1h 09m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c981c6af11654d3c960ee511b6c6c4f1 : SUCCESS in 1h 33m 10s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/46cbdb3aac5647d4b8640fa86c3ed8f4 : FAILURE in 1h 32m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/723a33ef4cf8429a8272c2d915ea60bc : SUCCESS in 2h 14m 00s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/907b9d4b09424d3285044e45e874a56a : SUCCESS in 1h 06m 12s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e0941a3490204d51ba5548044a4b1aa1 : SUCCESS in 1h 17m 23s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9b6abfb602de44088ebc6464cc3bc791 : SUCCESS in 24m 13s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/975a897258c84a07aac1b6a654066695 : SUCCESS in 59m 09s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b129c5dc9ec04ff2a33176f452fdeec3 : SUCCESS in 1h 11m 06s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/18e25089aece42bda369a8baca9b5767 : SUCCESS in 1h 18m 12s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c64a996a98a54c048463a6abbd7cde4d : SUCCESS in 24m 46s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/02c5ca57039445b184d8864d0a694529 : SUCCESS in 23m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5e354fd746bd4aceb4bafe52ea0bed92 : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e408111dbd554527bafa53c03ae7e56b : SUCCESS in 1h 11m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e2cc90ae9bec4d79a06fd77bfe08a646 : SUCCESS in 1h 21m 18s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/13d7d7f9c1ae4c0e9b08c0b267bcf65e : FAILURE in 1h 27m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d1964dc363b44afc8770c3141904567c : SUCCESS in 2h 25m 56s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/552b346c51684b959ed6dd9975cfcf26 : SUCCESS in 1h 15m 35s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0306f5e12d2a4644971e0a45733e4327 : SUCCESS in 1h 12m 30s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7eedb71ff1f44f6fb9d1bc20815e2777 : SUCCESS in 28m 27s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/267c4243c0274093bb374a05eb63d069 : SUCCESS in 51m 17s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f95bde2930164675877bb607f3444d93 : SUCCESS in 1h 00m 01s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/27fff2fa5a07482abd0df607d17f169a : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/077705f5a72c448bbf2d98c7ea40dcfe : SUCCESS in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"be6f6b18a9d58ac4a8ff6e492412a5f6774a77a5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-06 19:19:50.000000000","message":"Patch Set 24: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"5022e02d09b3bb2ae15bec1d5ee13a8ef1b6846d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 19:20:13.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"d87aad3ef5be9b39ec7061cb142c3d048b8a1082","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 19:57:31.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ae16255da2c4a58ac6abccc7740d423\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/54d09e5382514701abf397342d8ff1df : TIMED_OUT in 3h 01m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/805366a2a8d84d8ba13feb18ea61d155 : TIMED_OUT in 3h 01m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25494e2386114c35985fec6d179e850d : SUCCESS in 2h 33m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3a4cf2d33964aabb49e6f6288053078 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53bb6e6bfc07468581fdb9a4c121dca2 : SUCCESS in 6m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9ffb7eab257742218c3ea5aaf86fad10 : SUCCESS in 2h 55m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7946bb87ce0746968649d43aa2cac125 : TIMED_OUT in 3h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"a1e7651879d79a9396211eec507d255f74493dac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:39:16.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5feeb6489db54ce3b51fe8fea6e45572\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bf4537143d81482cac13995d64025dee : SUCCESS in 16m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/413b1cd3c7b94b77b87b6a73c41e6064 : SUCCESS in 34m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3d3a4b2f44d44bc8a6c5c919ff2fcb97 : SUCCESS in 53m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c62f2c1f69e14befaebdb564f3f754aa : SUCCESS in 1h 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e3f02ac108417eae0396d4e6f84709 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23d5de2181564a368fef6092051a04e9 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ccabac57b1746d7b7fd381aa38c1d63 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/715bd697a8b647f3bccff759544f020d : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03121080f088404593d50586be8dfed2 : SUCCESS in 2m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b8c4d0763f544a9bede19ba69b6f365 : SUCCESS in 32m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/13822584219643e4a5451d15ade0fb04 : SUCCESS in 32m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dd442fb52724455f8ec619dabc7fd579 : SUCCESS in 1h 04m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f2b207b9b78a42768deb74752059795f : SUCCESS in 1h 18m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b2c40a2619c84319b8d5973939617c72 : SUCCESS in 23m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d870145033934b689166d0c3846f9173 : SUCCESS in 25m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/547c85088ae64451bf10c7c11f0b3863 : SUCCESS in 54m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7c0dbbb69c054b869c211518fe2ff1b4 : SUCCESS in 1h 15m 02s","accounts_in_message":[],"_revision_number":24},{"id":"269b2e029cf356a071b8b96896943d5a577694ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:39:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":24},{"id":"50fc288cdd769a089ccb76202801fa746fbf0a87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 20:40:22.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b230fb22f82e458c830ae05049aa9510\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db452ea0ab904f0382bb888db72ca975 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c03ef24c4364c20b72cdbc3abe02600 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":24},{"id":"fe85b1f9c695eb43909178f35bb7bad7479a68be","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-07 06:54:48.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"36c12676ff185f524a7f4d2f2f4c70b5d60462b4","revisions":{"4e4752adc3d7e7ee689e79367f313bd571dd4d82":{"kind":"REWORK","_number":1,"created":"2024-04-10 18:21:22.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/1"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:21:11.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e4752adc3d7e7ee689e79367f313bd571dd4d82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e4752adc3d7e7ee689e79367f313bd571dd4d82"}]},"branch":"refs/heads/master"},"20c7f7983c490f8db202b8837a88c9b7846a9ac6":{"kind":"REWORK","_number":2,"created":"2024-04-10 18:23:08.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/2"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:22:46.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20c7f7983c490f8db202b8837a88c9b7846a9ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20c7f7983c490f8db202b8837a88c9b7846a9ac6"}]},"branch":"refs/heads/master"},"0ea34f8f08dd6e3936683b601419c7e97c9dc251":{"kind":"NO_CHANGE","_number":3,"created":"2024-04-11 09:06:51.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/3"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-11 09:06:15.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ea34f8f08dd6e3936683b601419c7e97c9dc251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ea34f8f08dd6e3936683b601419c7e97c9dc251"}]},"branch":"refs/heads/master"},"4808fbd539cfab4d2ffab52dda7133fe27ee573d":{"kind":"REWORK","_number":4,"created":"2024-04-12 14:08:47.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/4"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-12 14:02:42.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4808fbd539cfab4d2ffab52dda7133fe27ee573d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4808fbd539cfab4d2ffab52dda7133fe27ee573d"}]},"branch":"refs/heads/master"},"4b5220b29ccd7bd0d223f59430fe52f1866a9084":{"kind":"REWORK","_number":5,"created":"2024-04-12 14:28:18.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/5"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-12 14:28:10.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b5220b29ccd7bd0d223f59430fe52f1866a9084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b5220b29ccd7bd0d223f59430fe52f1866a9084"}]},"branch":"refs/heads/master"},"d2ccb1aebfbdcc134d7217441dc1227ac60a7337":{"kind":"REWORK","_number":6,"created":"2024-04-12 14:30:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/6"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-12 14:29:54.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2ccb1aebfbdcc134d7217441dc1227ac60a7337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2ccb1aebfbdcc134d7217441dc1227ac60a7337"}]},"branch":"refs/heads/master"},"1b79d90efa647d5beffbfc44984427c226f96e79":{"kind":"REWORK","_number":7,"created":"2024-04-12 14:40:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/7"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-12 14:39:54.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b79d90efa647d5beffbfc44984427c226f96e79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b79d90efa647d5beffbfc44984427c226f96e79"}]},"branch":"refs/heads/master"},"cb233d1ed0f38043978322ec697f07f18b73611c":{"kind":"REWORK","_number":8,"created":"2024-04-12 14:54:05.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/8"}}},"commit":{"parents":[{"commit":"3e502dc34dd4bd2b643c8131839d2853d855171d","subject":"Merge \"Update prometheus-mysqld-exporter to version 0.15.1.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e502dc34dd4bd2b643c8131839d2853d855171d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-12 14:53:51.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb233d1ed0f38043978322ec697f07f18b73611c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb233d1ed0f38043978322ec697f07f18b73611c"}]},"branch":"refs/heads/master"},"a84ce7436e571dc7a059f41d869d01f0e2d6a0b3":{"kind":"REWORK","_number":9,"created":"2024-04-30 12:03:18.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/9"}}},"commit":{"parents":[{"commit":"26d95c182c479f5ab12386289420e1688b3c7986","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26d95c182c479f5ab12386289420e1688b3c7986"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-30 12:03:04.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a84ce7436e571dc7a059f41d869d01f0e2d6a0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a84ce7436e571dc7a059f41d869d01f0e2d6a0b3"}]},"branch":"refs/heads/master"},"5a68f863e6b21e13bbb25dcfee8b1dfcf9fdc392":{"kind":"REWORK","_number":10,"created":"2024-04-30 15:08:55.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/10"}}},"commit":{"parents":[{"commit":"26d95c182c479f5ab12386289420e1688b3c7986","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26d95c182c479f5ab12386289420e1688b3c7986"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-30 15:08:40.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a68f863e6b21e13bbb25dcfee8b1dfcf9fdc392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a68f863e6b21e13bbb25dcfee8b1dfcf9fdc392"}]},"branch":"refs/heads/master"},"c546580d23a17cf87db6f7875f2f06cbda40f84f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-06-27 16:47:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/11"}}},"commit":{"parents":[{"commit":"f36e45382ed6ef5767531c2da30a1aa3e72a15ec","subject":"add kayobe to release management doc\u0027s Launchpad admin section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f36e45382ed6ef5767531c2da30a1aa3e72a15ec"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-06-27 16:46:59.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c546580d23a17cf87db6f7875f2f06cbda40f84f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c546580d23a17cf87db6f7875f2f06cbda40f84f"}]},"branch":"refs/heads/master"},"431a3e715da77761486b8ebc28ed5971ad3f0321":{"kind":"REWORK","_number":12,"created":"2024-09-17 09:54:14.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/12"}}},"commit":{"parents":[{"commit":"0b390c35aecde1ef862f760eb5f1af36283ee51a","subject":"Merge \"Gnocchi: Bump to 4.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b390c35aecde1ef862f760eb5f1af36283ee51a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-17 09:53:28.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/431a3e715da77761486b8ebc28ed5971ad3f0321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/431a3e715da77761486b8ebc28ed5971ad3f0321"}]},"branch":"refs/heads/master"},"121b483c78fa783156520a606377279329a709a3":{"kind":"REWORK","_number":13,"created":"2024-09-17 10:26:21.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/13"}}},"commit":{"parents":[{"commit":"0b390c35aecde1ef862f760eb5f1af36283ee51a","subject":"Merge \"Gnocchi: Bump to 4.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b390c35aecde1ef862f760eb5f1af36283ee51a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-17 10:26:12.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/121b483c78fa783156520a606377279329a709a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/121b483c78fa783156520a606377279329a709a3"}]},"branch":"refs/heads/master"},"32fa2dfa04423d720fbec81fbc20909caf53aee8":{"kind":"REWORK","_number":14,"created":"2024-09-17 10:33:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/14"}}},"commit":{"parents":[{"commit":"0b390c35aecde1ef862f760eb5f1af36283ee51a","subject":"Merge \"Gnocchi: Bump to 4.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b390c35aecde1ef862f760eb5f1af36283ee51a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-17 10:33:35.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32fa2dfa04423d720fbec81fbc20909caf53aee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32fa2dfa04423d720fbec81fbc20909caf53aee8"}]},"branch":"refs/heads/master"},"e01f9538eb860d9c754353faec4a521f353e9be3":{"kind":"REWORK","_number":15,"created":"2024-09-17 12:42:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/15"}}},"commit":{"parents":[{"commit":"0b390c35aecde1ef862f760eb5f1af36283ee51a","subject":"Merge \"Gnocchi: Bump to 4.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b390c35aecde1ef862f760eb5f1af36283ee51a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-17 12:42:11.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e01f9538eb860d9c754353faec4a521f353e9be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e01f9538eb860d9c754353faec4a521f353e9be3"}]},"branch":"refs/heads/master"},"9a4862fed2df4cdd76f3b0e0f9b7216835accd32":{"kind":"REWORK","_number":16,"created":"2024-09-17 14:29:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/16"}}},"commit":{"parents":[{"commit":"0b390c35aecde1ef862f760eb5f1af36283ee51a","subject":"Merge \"Gnocchi: Bump to 4.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b390c35aecde1ef862f760eb5f1af36283ee51a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-17 14:29:24.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a4862fed2df4cdd76f3b0e0f9b7216835accd32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a4862fed2df4cdd76f3b0e0f9b7216835accd32"}]},"branch":"refs/heads/master"},"285ad6d11ae9d06c37a8bcccf9454ec7b79882f4":{"kind":"REWORK","_number":17,"created":"2024-09-17 22:04:23.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/17"}}},"commit":{"parents":[{"commit":"0b390c35aecde1ef862f760eb5f1af36283ee51a","subject":"Merge \"Gnocchi: Bump to 4.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b390c35aecde1ef862f760eb5f1af36283ee51a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-17 22:04:14.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/285ad6d11ae9d06c37a8bcccf9454ec7b79882f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/285ad6d11ae9d06c37a8bcccf9454ec7b79882f4"}]},"branch":"refs/heads/master"},"85012ddd2d87c348410293b285869b97e93e3397":{"kind":"REWORK","_number":18,"created":"2024-09-18 10:23:22.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/18"}}},"commit":{"parents":[{"commit":"0b390c35aecde1ef862f760eb5f1af36283ee51a","subject":"Merge \"Gnocchi: Bump to 4.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b390c35aecde1ef862f760eb5f1af36283ee51a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-18 10:09:28.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85012ddd2d87c348410293b285869b97e93e3397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85012ddd2d87c348410293b285869b97e93e3397"}]},"branch":"refs/heads/master"},"d8d281b822a398fe13e27c30563c58bfe63029d9":{"kind":"REWORK","_number":19,"created":"2024-09-18 10:35:17.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/19"}}},"commit":{"parents":[{"commit":"0b390c35aecde1ef862f760eb5f1af36283ee51a","subject":"Merge \"Gnocchi: Bump to 4.6.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b390c35aecde1ef862f760eb5f1af36283ee51a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-18 10:35:02.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8d281b822a398fe13e27c30563c58bfe63029d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8d281b822a398fe13e27c30563c58bfe63029d9"}]},"branch":"refs/heads/master"},"0d9aa2b311595743782eb49d4d185df75ae38f0c":{"kind":"REWORK","_number":20,"created":"2024-09-21 20:18:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/20"}}},"commit":{"parents":[{"commit":"a7135c9e15b4ec919f971afa0821752c692d6247","subject":"Merge \"Fix state checking via kolla_set_configs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7135c9e15b4ec919f971afa0821752c692d6247"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-21 19:46:15.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d9aa2b311595743782eb49d4d185df75ae38f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d9aa2b311595743782eb49d4d185df75ae38f0c"}]},"branch":"refs/heads/master"},"1796e6d0f26d7899fb05f19d04978271390d44e4":{"kind":"REWORK","_number":21,"created":"2024-10-03 08:33:22.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/21"}}},"commit":{"parents":[{"commit":"01515cdff135a02aca5df4962220d216ac61a68d","subject":"Merge \"erlang/aarch64: Move to openstack-kolla owned COPR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01515cdff135a02aca5df4962220d216ac61a68d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-10-03 08:33:02.000000000","tz":120},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1796e6d0f26d7899fb05f19d04978271390d44e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1796e6d0f26d7899fb05f19d04978271390d44e4"}]},"branch":"refs/heads/master"},"39a337399c6deca83b98303bc67098792b546095":{"kind":"REWORK","_number":22,"created":"2024-11-05 21:43:26.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/22"}}},"commit":{"parents":[{"commit":"ee7fe59f7ba366bb3f203eb072ae2c9babc2175c","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee7fe59f7ba366bb3f203eb072ae2c9babc2175c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-05 21:43:17.000000000","tz":60},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nIt also updates the logging configuration to use a more\ndetailed format.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a337399c6deca83b98303bc67098792b546095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a337399c6deca83b98303bc67098792b546095"}]},"branch":"refs/heads/master"},"1935d23ba11b1c70dd4079398a14f7e18b63f5a4":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2024-11-06 12:55:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/23"}}},"commit":{"parents":[{"commit":"ee7fe59f7ba366bb3f203eb072ae2c9babc2175c","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee7fe59f7ba366bb3f203eb072ae2c9babc2175c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-06 12:55:01.000000000","tz":60},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1935d23ba11b1c70dd4079398a14f7e18b63f5a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1935d23ba11b1c70dd4079398a14f7e18b63f5a4"}]},"branch":"refs/heads/master"},"36c12676ff185f524a7f4d2f2f4c70b5d60462b4":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2024-11-06 16:37:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/40/915440/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/915440/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/915440/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/915440/24"}}},"commit":{"parents":[{"commit":"8eaf4149b1f9a9bf56fda7dbe41049f25d35334a","subject":"Enhance logging format for better readability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8eaf4149b1f9a9bf56fda7dbe41049f25d35334a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-04-10 18:11:23.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-11-06 16:37:31.000000000","tz":60},"subject":"Fix handling configs in base image","message":"Fix handling configs in base image\n\nThis commit restructures the handling of configuration\nfiles in set_configs.py, introducing functions for managing\ndefault configuration files first.\n\nCloses-Bug: #2060855\nChange-Id: If91e0330dc149143c82d2183b8ddf6fa9f68d80e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36c12676ff185f524a7f4d2f2f4c70b5d60462b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36c12676ff185f524a7f4d2f2f4c70b5d60462b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
