)]}'
{"id":"openstack%2Fkolla~915847","triplet_id":"openstack%2Fkolla~master~Ib973325bb3ad04e1646b71595fa43903604d9b20","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-16 16:45:38.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ib973325bb3ad04e1646b71595fa43903604d9b20","subject":"[release] Use Caracal sources by default","status":"MERGED","created":"2024-04-15 11:18:47.000000000","updated":"2024-04-16 18:22:31.000000000","submitted":"2024-04-16 18:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915847","meta_rev_id":"3b39fc1241630caa05eaaee3477a984052f66863","_number":915847,"virtual_id_number":915847,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 18:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-16 14:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-16 14:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-16 16:45:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-04-16 14:47:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-04-16 14:47:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 12:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 14:37:18.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-16 14:47:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"fad7fa943c54e5efe4f3e0c72de8f814f06dd1f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 11:18:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3ae0d5d7f403339d7f69475688ea13180e0cb29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 12:21:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f4c123f514a46f1aa9903d499292297\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2caad3cdac344e71b913f33f6247b93a : FAILURE in 21m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f64af8da62444fa0ad2bdb5141268187 : FAILURE in 18m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3a210d5423484cc190c134923537b0aa : FAILURE in 18m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/78e108466c5b418490757f63e7ef9e96 : FAILURE in 24m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/73cf7a8507814f92a0a8764a4c325c1a : FAILURE in 15m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0fca0c9a83f74a14bd469d8f3f93313f : FAILURE in 22m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8168783a534743a5b254ce316c380a56 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cd00390bfb94bbcb41f19baa809729b : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c326e9aed0244c1804b250605cb975c : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdb81c40c6214518a46989d0ce7a74fa : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eec6587601bb4866842e6e4c9d033ad6 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da89fe2410a4b6daa431284d6b0e62e : SUCCESS in 4m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fb8b573e923648f8b8cce1fab704b1e4 : FAILURE in 23m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/65a12851b6c0481e804879f53c5bcf61 : FAILURE in 33m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5cc8e72b96b749809428b6efe4bd39fb : FAILURE in 19m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/954c57c666bb435cab1ede2e1467f7d7 : FAILURE in 18m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b176432d1413476d95b1472c8f2ffdb6 : FAILURE in 21m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d3cbb8cda0d84edabf4acd5bb0328c33 : FAILURE in 57m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/29e933c9efa74bad8b59f15264f68a19 : FAILURE in 15m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/55996707b15b4b1ba1eee45825123e9a : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"cb3888307b13caa5abd5db2d9c2053c1c7c5352c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 12:54:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5dcd6631c414a9b80fc51086dc5451a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/262c18170c714c7fae7c87d3d5e65856 : FAILURE in 23m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d57b31c9b8b94def9786ba0f7af9710f : FAILURE in 1h 26m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/35b24eedf73c4462b7d1d651f60c51e6 : FAILURE in 45m 45s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42e26656e9714be4bbac34322de7837c : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44c0b6ea7cdc483bb72386a4feacd4ff : SUCCESS in 4m 03s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/76df05e125d245e69a17f720afbe317d : FAILURE in 1h 32m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/35b084b6548b4f6398da52cd4b29ce7c : FAILURE in 1h 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f75c776051f8d2d6616af2ed0ccc05d2907f936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:27:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a06c361120cdfb9208467b22dd1053e5904c2fbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:44:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bad521d79df1c76bdbf5ec302ff90c39284920f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:38:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01682c06ec6c402caaea5935f4dfb437\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5bad33266b9b4befa498fa56e6ca7c12 : SUCCESS in 25m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c4afd9ba5a4e4c609d84a645824e8987 : FAILURE in 41m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0c2ee1f4e5a249e9bd6ae10f8f3bce00 : SUCCESS in 23m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/036aa749c2174b3cb5ccb6e96e9e2327 : SUCCESS in 30m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/db49f7516967477ab0393dc923cfe616 : FAILURE in 45m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d4cfcfcde9dd49a78e462c3b99674262 : FAILURE in 1h 37m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d3036b0f9743b98b75adb8acf5f48b : FAILURE in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2e8e5670fe441fe8bfcde29a08af689 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ddd9750001e4e25b1f9e19f9362d113 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4ba061ecaee41c688b7d8716ba1d04b : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5798e32b5da4757948806a4a400c322 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375fe27c567541d0a8ddb7102a80f041 : SUCCESS in 5m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/573c8154b0ad472b8040a42518cf86e6 : SUCCESS in 26m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7d3fc42e4bb342a4bf771623d23ba32b : SUCCESS in 29m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d782b5c134524a669d7d65e10bd6b84c : FAILURE in 45m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/de6d17d2dff24e5c8593e55f52034caa : FAILURE in 1h 35m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/00e8a305916d4f37a5fa2bba7baecfd1 : SUCCESS in 22m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bd9b4d1ad989483b92a0d422787172d1 : SUCCESS in 1h 13m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/76c2a409a6d24069821b7aae1daed4a8 : FAILURE in 39m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/aeee12fbac474038905e2f1f8c944bee : FAILURE in 1h 31m 13s\n\nWarning:\n  Comments left for invalid file docker/horizon/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"127b2e239f32be29316caa6779783b0673fcd342","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 17:52:33.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8c7863a98384c01a7afc795679caf8c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/00f40204465a41f184e84490cf50736a : SUCCESS in 2h 21m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5bbd2cbd2cb84959bad704bae8a35da1 : SUCCESS in 2h 15m 10s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b20d27d539a9464ba78d073cc43c886d : FAILURE in 2h 07m 06s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5fae92a2f4f467a964656ab541345ff : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/241c96736bc64b2b9e18a41355dc3bfd : SUCCESS in 3m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/adf3b8514edd4fb1afed46b20001c7c4 : SUCCESS in 42m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/daf90e6ab6984241a44b5041fa315f32 : SUCCESS in 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66570aa39464c30a036ee80fcb0b66e216dd80c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 12:22:47.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3ad9936b9cd5084c2cd5306b28ba8e5453da41b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:04:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03e462fc5da14677ba056c6fe4ee668d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3f11fde602694a94b6fac8e329b3125d : SUCCESS in 1h 37m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ca4b595e2bd5407b95a3b025616d1a99 : SUCCESS in 1h 39m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/86a91223a0c841739b7ccee656de8fa0 : SUCCESS in 1h 13m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cbd5d162bdea435b8a56dc078fcebe2a : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f407d37016e14488be52bbd8e23b59cf : SUCCESS in 3m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/87b6ade9a8054d728ac21eca471f4852 : SUCCESS in 24m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1d8f7acf9b3a42e5a5e4c47d27eadaf3 : SUCCESS in 22m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d743f115a52fa59fca49e713e34d0d331b1b1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:32:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9be0ce17f44607b67922226d8a477e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dfaf4dd33cbc45559241a6d16a4059ca : SUCCESS in 28m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/df635e8fd885434481f7fd3534bae47d : SUCCESS in 1h 02m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f356fa4ceb104e5789dd84fdc534be0d : SUCCESS in 21m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/704f5474babf4930827c3609c5143cef : SUCCESS in 26m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a8ac18ad785441f9ae4f69bc54e70071 : SUCCESS in 1h 09m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c8aeb7c6953643b9943cf5569afcbf32 : SUCCESS in 1h 34m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032aa1d86b664475a0a2b97b79c07902 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cde2374420143d19355fc873293f4ba : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e6c57b99eb64ec398f59d7464b75d05 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c76cb398d2fc4a0a80280f5e52db5541 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dcf5f62ef5c4ab694a905afd277b436 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9c60cd180744f1ad4f7eefc66ad082 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8a5ff7ab340b4d1aa3616c857d8e6e42 : SUCCESS in 17m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/46a42927c8bd47508665097df634c07c : SUCCESS in 31m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/849ede6e5a644697aceaae672c7e04c5 : SUCCESS in 1h 09m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/76caa880840b4165a9c5ac61ca4bcfab : SUCCESS in 1h 36m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/20055b05d0fb40778ecda86978b57559 : SUCCESS in 25m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3ca259f5d5a34ba39368031606d0e603 : SUCCESS in 1h 11m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9a064dfc6c82482d9e5276e532735690 : SUCCESS in 59m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a7b4c1e6a1304ad5aab084e3ddc91a1d : SUCCESS in 1h 20m 41s","accounts_in_message":[],"_revision_number":4},{"id":"6cfebeb3d382149bb473ea0e75e78213c9b46827","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-16 14:37:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a747f94b212d288dfb18331372648823dc9b3d46","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-16 14:47:26.000000000","message":"Patch Set 4: Backport-Candidate-2 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"8e052c1d158d4398fb06f0fab633c17c1b7f6f72","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 16:45:38.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fe4792564dc1d7084f6b43ac5cf0f2b0f0f90284","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:46:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"724961dc2cb4f96fc3ef171828207d0c713f9a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 18:21:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee4c23aa7964666a5b49696ee055fc7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4c3e4435a7d8457cb1cae6dd30a7c1f6 : SUCCESS in 20m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/511e8a8f4ba0470fb371551d50a189e6 : SUCCESS in 28m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2b94f9d440d146ac9034b7f8b84d4711 : SUCCESS in 50m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e4df47b33c4047f98362f8ed6d0d1532 : SUCCESS in 1h 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8017cba4f4284ff0a130e75a72c0e01b : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9822bb481a7443ac94e36d0213caabbe : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41962d405cfa47368c63580af06658d9 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c490ae84fbc4444a55bb146349e428b : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/20862c866f164e978ee83c1b8ac1f7a2 : SUCCESS in 21m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ee9e57470730462692b25a548ac74dfc : SUCCESS in 27m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fefaeed7f3024078a232a1717fa06204 : SUCCESS in 1h 09m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/23a38de9f3fd43f693c9e2a10db543e3 : SUCCESS in 1h 32m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/18f3f41b40d942238cdfb9787bf408ed : SUCCESS in 15m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/eec1fa61c3694146a87d60022a07776b : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f34d0069ee7a4f29a6e37699715f6361 : SUCCESS in 54m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0374a28efb8d49e6a950717d55b83f6b : SUCCESS in 1h 30m 07s","accounts_in_message":[],"_revision_number":4},{"id":"73ffbfc89a802049cd3f3cac8cc670de75704cc7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 18:21:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3b39fc1241630caa05eaaee3477a984052f66863","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 18:22:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b35de6d01df4d89a242d8a23af1157d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2839b2c588c84aa896e4f82946989c19 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0fb24ceefce742a374d725ffa5cc3221ca18dfda","revisions":{"6f21a1ede04f90132fe1c5de62489cd37fc603a6":{"kind":"REWORK","_number":1,"created":"2024-04-15 11:18:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/915847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/915847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/915847/1"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 11:18:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 11:18:44.000000000","tz":120},"subject":"[release] Use Caracal sources by default","message":"[release] Use Caracal sources by default\n\nChange-Id: Ib973325bb3ad04e1646b71595fa43903604d9b20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f21a1ede04f90132fe1c5de62489cd37fc603a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f21a1ede04f90132fe1c5de62489cd37fc603a6"}]},"branch":"refs/heads/master"},"4c0cea86b9da59d28091f424c2bf314b84c319c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-15 13:27:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/915847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/915847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/915847/2"}}},"commit":{"parents":[{"commit":"9f8ffbe8777e66981d282d85542beb64d9ebbd1b","subject":"Drop Freezer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f8ffbe8777e66981d282d85542beb64d9ebbd1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 11:18:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:27:55.000000000","tz":0},"subject":"[release] Use Caracal sources by default","message":"[release] Use Caracal sources by default\n\nChange-Id: Ib973325bb3ad04e1646b71595fa43903604d9b20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c0cea86b9da59d28091f424c2bf314b84c319c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c0cea86b9da59d28091f424c2bf314b84c319c9"}]},"branch":"refs/heads/master","description":"Rebase"},"8fa6126da7bc364e8fada34fc4924a3f8f08931e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-15 13:44:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/915847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/915847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/915847/3"}}},"commit":{"parents":[{"commit":"ec737505978402292a2e5de97ea62196782933f7","subject":"Drop Freezer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec737505978402292a2e5de97ea62196782933f7"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 11:18:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:44:45.000000000","tz":0},"subject":"[release] Use Caracal sources by default","message":"[release] Use Caracal sources by default\n\nChange-Id: Ib973325bb3ad04e1646b71595fa43903604d9b20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fa6126da7bc364e8fada34fc4924a3f8f08931e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8fa6126da7bc364e8fada34fc4924a3f8f08931e"}]},"branch":"refs/heads/master","description":"Rebase"},"0fb24ceefce742a374d725ffa5cc3221ca18dfda":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-16 12:22:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/47/915847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/915847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/915847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/915847/4"}}},"commit":{"parents":[{"commit":"c69284fd8b6a44a8c1f24771d28d8d510a0855b4","subject":"Drop Freezer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c69284fd8b6a44a8c1f24771d28d8d510a0855b4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 11:18:30.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 12:22:47.000000000","tz":0},"subject":"[release] Use Caracal sources by default","message":"[release] Use Caracal sources by default\n\nChange-Id: Ib973325bb3ad04e1646b71595fa43903604d9b20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fb24ceefce742a374d725ffa5cc3221ca18dfda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fb24ceefce742a374d725ffa5cc3221ca18dfda"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
