)]}'
{"id":"openstack%2Fkolla~915855","triplet_id":"openstack%2Fkolla~master~Ie715eac24247c38b21382c735eb00a9b3b56f145","project":"openstack/kolla","branch":"master","topic":"kolla_caracal_drops","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-16 12:03:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie715eac24247c38b21382c735eb00a9b3b56f145","subject":"Drop Murano","status":"MERGED","created":"2024-04-15 12:48:48.000000000","updated":"2024-04-16 12:04:22.000000000","submitted":"2024-04-16 12:03:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915855-kolla_caracal_drops","meta_rev_id":"9d51763585fc1ec99735ead2c4f647d8d0beb910","_number":915855,"virtual_id_number":915855,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 12:03:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-15 15:20:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-04-16 10:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-16 10:21:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-15 13:44:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 15:20:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-15 16:25:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-15 16:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 10:21:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"abbd14a9b778dc8796abfd74975f4a5836ad4369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 12:48:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e4b24779f2450aa729746702dacdc02bebc9808","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 12:51:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e591fafe3319733e2c7899378d489dadd70ca8ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 12:58:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e8d4ca818fcc7e252b83512979d41f46796296c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 12:59:02.000000000","message":"Topic caracal_drops removed","accounts_in_message":[],"_revision_number":3},{"id":"91aa6244d4d979ae1a1fe7783e55efb926ce28d8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 12:59:06.000000000","message":"Topic set to kolla_caracal_drops","accounts_in_message":[],"_revision_number":3},{"id":"baa7f801433efb8793c09bfd5af9ef1e225c4a93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:06:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d03b6bab896c113f19f7c2fbd61f37494f79c1e7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:21:50.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"295da310902af7dfbe404b7867429ee152d8c36a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:41:16.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":5},{"id":"62afefd94ab7f672342c812dafe9c49bf8822a97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:44:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"928a28b188ce05cf171e3d7aa2f70c37825dc11b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-15 15:20:18.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5ccb553cac540b10cea26629711a7ead7b72f972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:25:10.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a538699955449f7af6600362c56ae9e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0f4ae45a81704db688cf1de9fa30f70e : SUCCESS in 2h 27m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/831bdc370be5482d97be8f97bdef09a7 : SUCCESS in 2h 37m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3bfefe56985341c397cda54e90402d7b : FAILURE in 1h 55m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7985fdb87374fa186d0ea76452646b3 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3276a386a3340b5adad44a3f5316998 : SUCCESS in 3m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ef898e62d2bb49d1a65599482ccbfb19 : SUCCESS in 49m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ef9d320b7ba4bd29bff02fbe547b104 : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"971b27efe2c0412ac698051464bbe08121753459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:36:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a11f944fac45a6b67849ffb30c6e14\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2c371f3128a949709cfdc0b5fa2316f9 : SUCCESS in 1m 39s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/da8f3155977f4c168c45d6fca86f8dd5 : SUCCESS in 29m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2af979be753b4b319e458a06957775c3 : SUCCESS in 1h 08m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ad5bb3dde7144455b0e29efbebfe4488 : SUCCESS in 24m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/63a46a31c0ac49839bbc85bf1596fd9a : SUCCESS in 33m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ac6d02cac87a42de903bab8b5e0866d1 : SUCCESS in 1h 07m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/56614645d363468da6637cba0bbe0a0c : SUCCESS in 1h 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98903b1cd24c4e4a89f94a3f629c9405 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9697ea63d549464b976805df40c6826c : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/263d32c3c573412b83120d8817f05fd7 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c165b46ff0c8420381659667bc5b743b : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fba272f25d71473c9610258a7c5468e6 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9604a4424d754584b7360f0ff4e848cc : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c8906cb08c74fbfaf2b0edd77aa223c : SUCCESS in 3m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8381cac975cf49ae814ccec5665ca571 : SUCCESS in 29m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/00bc2fc4790d48bcb4add0739e9a4939 : SUCCESS in 33m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bc4fd03105954ae6ac3cf92e48410386 : SUCCESS in 1h 08m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d43ac53a260f43a5ac68c15bcc8863ec : SUCCESS in 1h 42m 01s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1bdf26759681488d8a4d48852470e22c : SUCCESS in 1h 47m 32s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/35192efa411a4ff486012d5b14aa7545 : SUCCESS in 2h 46m 27s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/4a7420aa9b1047dab459aa9ea25a87cb : SUCCESS in 1h 25m 53s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ac18d837766041b981242e7a29f6215b : SUCCESS in 1h 17m 39s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a468fdfcf52542dcb2f476566f0cd4e2 : SUCCESS in 1h 13m 45s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/42067ce456004944a8f70758f3feaf53 : FAILURE in 1h 10m 56s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7694b56c7a344d20a494b17304b07068 : SUCCESS in 1h 26m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7ecf1156c29247b69a42099ef2b72b6c : SUCCESS in 23m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/077e88678a264c30af77a6042cb54116 : SUCCESS in 50m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dab7f161711d48c1aa6b7c8b15409f5e : SUCCESS in 46m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/13a64953ab924016a5cfd30252854858 : SUCCESS in 1h 09m 55s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/68923b6b827f48e6bc049ae237a7ab63 : SUCCESS in 1h 21m 24s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ae1b9da9d2454c829b9051a74ba2cda4 : SUCCESS in 2h 21m 26s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/87f64e291942476a8b817573ea72c9c0 : SUCCESS in 1h 13m 45s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/ae083b198a494036aebff08c84311568 : SUCCESS in 1h 13m 35s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/cea8acde1c9440e18abeb9c775f41738 : SUCCESS in 48m 34s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/279a0ff3291c48f69ef27ce997b7160e : FAILURE in 50m 09s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a4912dd285c547089de5d2bb2427054b : SUCCESS in 1h 06m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bbd93e9e72e73678f9d52f7f4bb096cd1f16c01c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-16 10:21:16.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c7b9c10ac6bc3f81f83f4c911cd68d7541fc4008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 10:21:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e676600a6c8aa1710d439d03caa031c62feef753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 12:03:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d648aa76744f4681ca5c5632a25e7f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/99f34273d7c4492f8cd96c7c3bfb9fbd : SUCCESS in 1m 50s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/428503fed70e4d1698c2b9854168ba9c : SUCCESS in 26m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/322ffb3ad0174d8f9877a879739116d5 : SUCCESS in 30m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/37176187f99d4fd49199aec730c6850c : SUCCESS in 1h 05m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bebf41d95dad45e2b0c3a922e8d70468 : SUCCESS in 1h 26m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/608b6ca269e148c389ecc0b7c26073b4 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee0ac68cb15b44f5a8f60c506f6a6d8b : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1769a94aeeb24f2f84d41ae423950bf7 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f79370dfcb40d6a0375d14d7781789 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce19f8d8c3564588b6fe64527ebd3c0b : SUCCESS in 3m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7d2e771566f941de96b8fda626f4e3b1 : SUCCESS in 24m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/09339492c1894290a3615dd210182ab8 : SUCCESS in 36m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c01b79fd7b064caaab4ba7d4d49eefb5 : SUCCESS in 1h 07m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/74c5e74d10474a499f9d4d73eb2d79f2 : SUCCESS in 1h 37m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d1c7321a0fb64e518cb2fd410d8699f1 : SUCCESS in 25m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ed4bd69d8be049d5b979ca9fdcd0f6e7 : SUCCESS in 58m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0aa3085dcee64da9be3f97ed261ee118 : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9cac7a5dd6e64238ba4c5c03c10d3311 : SUCCESS in 1h 22m 28s","accounts_in_message":[],"_revision_number":6},{"id":"c9bb55e22184789720c81427f6f2c951676cd1cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 12:03:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"9d51763585fc1ec99735ead2c4f647d8d0beb910","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 12:04:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e63bff1519a4b2496d9969fec231a81\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21279f2019794b1c898ded62cde0bc46 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38502882553244c795b4ffebbd693f64 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"813035ba8fcc2a92e78412d8e4a96740c55e0c2e","revisions":{"9c78cf70e2836a632eea13f19eac71616f3a7af6":{"kind":"REWORK","_number":1,"created":"2024-04-15 12:48:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/915855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/915855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/915855/1"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:47:31.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:48:46.000000000","tz":120},"subject":"Drop Murano","message":"Drop Murano\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nIn addition to that, there\u0027s a CVE that hasn\u0027t really been patched [2].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n[2]: https://lists.openstack.org/archives/list/openstack-announce@lists.openstack.org/thread/4FYM6GSIM5WZSJQIG4TT5Q3UBKQIHLWX/\n\nChange-Id: Ie715eac24247c38b21382c735eb00a9b3b56f145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c78cf70e2836a632eea13f19eac71616f3a7af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c78cf70e2836a632eea13f19eac71616f3a7af6"}]},"branch":"refs/heads/master"},"3a1ed83307402a4b82dc76a710f94988023e9264":{"kind":"REWORK","_number":2,"created":"2024-04-15 12:51:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/915855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/915855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/915855/2"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:47:31.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:51:22.000000000","tz":120},"subject":"Drop Murano","message":"Drop Murano\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nIn addition to that, there\u0027s a CVE that hasn\u0027t really been patched [2].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n[2]: https://lists.openstack.org/archives/list/openstack-announce@lists.openstack.org/thread/4FYM6GSIM5WZSJQIG4TT5Q3UBKQIHLWX/\n\nChange-Id: Ie715eac24247c38b21382c735eb00a9b3b56f145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a1ed83307402a4b82dc76a710f94988023e9264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a1ed83307402a4b82dc76a710f94988023e9264"}]},"branch":"refs/heads/master"},"5390337df1a154d4f70f45c3410afa269e71a197":{"kind":"REWORK","_number":3,"created":"2024-04-15 12:58:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/915855/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/915855/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/915855/3"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:47:31.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:58:20.000000000","tz":120},"subject":"Drop Murano","message":"Drop Murano\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nIn addition to that, there\u0027s a CVE that hasn\u0027t really been patched [2].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n[2]: https://lists.openstack.org/archives/list/openstack-announce@lists.openstack.org/thread/4FYM6GSIM5WZSJQIG4TT5Q3UBKQIHLWX/\n\nChange-Id: Ie715eac24247c38b21382c735eb00a9b3b56f145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5390337df1a154d4f70f45c3410afa269e71a197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5390337df1a154d4f70f45c3410afa269e71a197"}]},"branch":"refs/heads/master"},"4d06a98bb14580da4129e07e3d672bd8d7296681":{"kind":"REWORK","_number":4,"created":"2024-04-15 13:06:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/915855/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/915855/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/915855/4"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:47:31.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:06:42.000000000","tz":120},"subject":"Drop Murano","message":"Drop Murano\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nIn addition to that, there\u0027s a CVE that hasn\u0027t really been patched [2].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n[2]: https://lists.openstack.org/archives/list/openstack-announce@lists.openstack.org/thread/4FYM6GSIM5WZSJQIG4TT5Q3UBKQIHLWX/\n\nChange-Id: Ie715eac24247c38b21382c735eb00a9b3b56f145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d06a98bb14580da4129e07e3d672bd8d7296681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d06a98bb14580da4129e07e3d672bd8d7296681"}]},"branch":"refs/heads/master"},"df3db06e7433a1c35f3c73678cd4d421b623ae11":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-15 13:41:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/915855/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/915855/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/915855/5"}}},"commit":{"parents":[{"commit":"57161c011f84893234062bba8cace3d05d2b5ca0","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57161c011f84893234062bba8cace3d05d2b5ca0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:47:31.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:41:16.000000000","tz":0},"subject":"Drop Murano","message":"Drop Murano\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nIn addition to that, there\u0027s a CVE that hasn\u0027t really been patched [2].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n[2]: https://lists.openstack.org/archives/list/openstack-announce@lists.openstack.org/thread/4FYM6GSIM5WZSJQIG4TT5Q3UBKQIHLWX/\n\nChange-Id: Ie715eac24247c38b21382c735eb00a9b3b56f145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df3db06e7433a1c35f3c73678cd4d421b623ae11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df3db06e7433a1c35f3c73678cd4d421b623ae11"}]},"branch":"refs/heads/master","description":"Rebase"},"813035ba8fcc2a92e78412d8e4a96740c55e0c2e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-15 13:44:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/915855/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/915855/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/915855/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/915855/6"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:47:31.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:44:06.000000000","tz":0},"subject":"Drop Murano","message":"Drop Murano\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nIn addition to that, there\u0027s a CVE that hasn\u0027t really been patched [2].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n[2]: https://lists.openstack.org/archives/list/openstack-announce@lists.openstack.org/thread/4FYM6GSIM5WZSJQIG4TT5Q3UBKQIHLWX/\n\nChange-Id: Ie715eac24247c38b21382c735eb00a9b3b56f145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/813035ba8fcc2a92e78412d8e4a96740c55e0c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/813035ba8fcc2a92e78412d8e4a96740c55e0c2e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
