)]}'
{"id":"openstack%2Fkolla~915860","triplet_id":"openstack%2Fkolla~master~Iad4d850ba8f8b8446591cab883d2ad6243576926","project":"openstack/kolla","branch":"master","topic":"kolla_caracal_drops","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-16 14:14:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad4d850ba8f8b8446591cab883d2ad6243576926","subject":"Drop Senlin","status":"MERGED","created":"2024-04-15 12:57:04.000000000","updated":"2024-04-16 14:15:02.000000000","submitted":"2024-04-16 14:14:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915860-kolla_caracal_drops","meta_rev_id":"673c00eefa289eb4b445c9866ef260d543244ac4","_number":915860,"virtual_id_number":915860,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 14:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-15 15:37:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-04-16 12:23:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-16 12:23:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-15 13:44:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 15:37:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-15 15:44:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 12:23:06.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"bd2f77873b67e042da3349b4fd8ccbaff2b227e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 12:57:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83c42b2d8d001eaadae8dfdf7c461e731e800422","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 12:58:57.000000000","message":"Topic set to kolla_caracal_drops","accounts_in_message":[],"_revision_number":1},{"id":"e8066b6a00ab49392a001d4cd51bb08448d8fb62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:08:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5632752329676ab9432365a09ecb4cda9e8a512c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:12:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"596061428355a03913fa4d934198f8e27179d291","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:13:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2c9e9d6a14bac8375de19a41b93a5aa85f4e31df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:16:44.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"5937c404de24d29b27a4a0e71283760ccd6526be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:21:55.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"97473daaad8e9db6506176e4b968a4c9c918c846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:44:15.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"4ce5dff05ab9ace27f20adec5e24d6af89a1df8d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-15 15:37:12.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9815b3e079c60d9791353c8ca0410a504884f851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:44:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423609ceefa54bbfa9cdd4c0a816d5b8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a9de19dafa18445c8bd8c866c22a7cf8 : SUCCESS in 1m 54s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e96bd129f8d1411f99befec30470e9ad : SUCCESS in 29m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/30368cf464ac438289f4a3352981314f : SUCCESS in 1h 07m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7d53e2811c384e2b8bed2863389f09db : SUCCESS in 31m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b631ddbb67374cb2b5bb3d3d40a0c3cd : SUCCESS in 35m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3652c93ab3014f83a042e0f57fe35eb6 : SUCCESS in 51m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/35abe2b2e5b3470d97a12ce8142df2bb : SUCCESS in 1h 29m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd06f79ad7d4b0e8b0f63990757ea54 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/354a624c0e4f4f84978a9eeec06e150d : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31059b210a1346bba8e50ce4a7d1e119 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e66ef0fbbaf4a5d86fe34a67c46cfe8 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c75accc232ce487bb2909c6e65150394 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12bb487b47b4f8e911157698e8fe3f6 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea2144d7c96e41ccbceb78e1bbb29f35 : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f92534da3a8c4f0885873a0ea52d7b06 : SUCCESS in 39m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d49c943e371c41afa0b312c3bdfbe00e : SUCCESS in 33m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8bba3db2eb7d4e3cb8dfe24791e8ce87 : SUCCESS in 1h 09m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4e17369f6b67425eb49fc4081f0b81ac : SUCCESS in 1h 34m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ec5cffc4780645e18b3e0a6e6a8edd9b : SUCCESS in 21m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2e288217e77448fe879158f8bf49fe40 : SUCCESS in 1h 01m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/abb5b87c80214d708b8a029e71d2d3dd : SUCCESS in 50m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7802f515b6594b8ea610210ebdaa80f4 : SUCCESS in 1h 52m 49s","accounts_in_message":[],"_revision_number":6},{"id":"9182ee5f9070505fbcb656c6f986960fbcda2209","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:17:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6253b7aeb4b4552a17b1a6ed948c154\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/55b3c324af6847a7bcb7190038e994d6 : SUCCESS in 45m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b31884ec4e9b4bd5acafde220d953d4b : SUCCESS in 2h 31m 10s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/da16baaf1cf243f4a0ce6458c93ab741 : SUCCESS in 2h 22m 25s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a51403eb5b284fd6a51ec860d8d9eb98 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/34574c65945b442aa91ec6b780564fa9 : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/51a059895c8343b181286af0d2fb15cd : SUCCESS in 2h 21m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/634e9b91e3a24241878ca3b7706e0367 : SUCCESS in 2h 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"97a318128cc6dc6ac5727cd5512738b66fe31a1f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-16 12:23:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"455f693018dfd46c3f3c35baadffe8907bd6694d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 12:23:29.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4853b0337442491106058ba3f2aa82e685a5d85e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:14:01.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd0ccb774d374168ad6c44e352fcc7e9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/228951699dba4f61abb4ece9abd05c65 : SUCCESS in 2m 38s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6631790157274037a0028a7efa9a4409 : SUCCESS in 23m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bc999da6b7eb4eea85f4a892e1f5ced5 : SUCCESS in 26m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f938780a82fb4d9c9100d4b336ecf00a : SUCCESS in 1h 05m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ee7b2b51ba94455fac6b8aa79230d1f3 : SUCCESS in 1h 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb804c6a0b94f409ad5d64318b307c8 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/655ab91c499248eabf572eed849ec974 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1ccf867511d45bba8fcbd1da9d72147 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fbae19eb43940748b8a66b7f8a077e1 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e43130e412cc420dacc036e22bf66473 : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/43d4a66885934e3fbd3667d6e762a47a : SUCCESS in 28m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/287807e3bfab443f83b273e5cd5f059c : SUCCESS in 35m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5e9ab077598c48f2bdc0aafb1007fa6d : SUCCESS in 1h 04m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/51b8063f3042409d878eb4c0c5b6b166 : SUCCESS in 1h 38m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/297230f7075942fdb6f615c6102a11f3 : SUCCESS in 19m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d706d035b65a43fa96d25472647c1530 : SUCCESS in 50m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/480c9386276d427091895d0627b26bb8 : SUCCESS in 45m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2333709f28874e14a9b0c66229dca9bd : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":6},{"id":"d48ad82398f9eddb9c52cb7aae6b88aed9538983","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:14:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"673c00eefa289eb4b445c9866ef260d543244ac4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:15:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63117c03aecd4c1691c13831fdfff8b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf8c8e7a760453294d36471fb9be3de : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e340ff3a2b74c138f4685e8b908af6f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"14a2e37f00a8197de97f88eebe88b2ff385b4ccb","revisions":{"cab7197820ec81072abee2900d10957bcc893ab7":{"kind":"REWORK","_number":1,"created":"2024-04-15 12:57:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/915860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/915860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/915860/1"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:57:01.000000000","tz":120},"subject":"Drop Senlin","message":"Drop Senlin\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: Iad4d850ba8f8b8446591cab883d2ad6243576926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cab7197820ec81072abee2900d10957bcc893ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cab7197820ec81072abee2900d10957bcc893ab7"}]},"branch":"refs/heads/master"},"e722d163e3a16e6ed3542e69a14d928344eaaf70":{"kind":"REWORK","_number":2,"created":"2024-04-15 13:08:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/915860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/915860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/915860/2"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:08:41.000000000","tz":120},"subject":"Drop Senlin","message":"Drop Senlin\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: Iad4d850ba8f8b8446591cab883d2ad6243576926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e722d163e3a16e6ed3542e69a14d928344eaaf70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e722d163e3a16e6ed3542e69a14d928344eaaf70"}]},"branch":"refs/heads/master"},"85baa1e37addee0cdbeb62ceab7d0643b34a7b6d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-15 13:12:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/915860/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/915860/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/915860/3"}}},"commit":{"parents":[{"commit":"4d06a98bb14580da4129e07e3d672bd8d7296681","subject":"Drop Murano","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d06a98bb14580da4129e07e3d672bd8d7296681"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:12:44.000000000","tz":0},"subject":"Drop Senlin","message":"Drop Senlin\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: Iad4d850ba8f8b8446591cab883d2ad6243576926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85baa1e37addee0cdbeb62ceab7d0643b34a7b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85baa1e37addee0cdbeb62ceab7d0643b34a7b6d"}]},"branch":"refs/heads/master","description":"Rebase"},"a481e9eeb1e5c63b217707af3e93ad1496ecb5a6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-15 13:13:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/915860/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/915860/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/915860/4"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:13:19.000000000","tz":0},"subject":"Drop Senlin","message":"Drop Senlin\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: Iad4d850ba8f8b8446591cab883d2ad6243576926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a481e9eeb1e5c63b217707af3e93ad1496ecb5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a481e9eeb1e5c63b217707af3e93ad1496ecb5a6"}]},"branch":"refs/heads/master","description":"Rebase"},"ba2aba8cf384156e01fd6ca40d95b0f62fc89b69":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-15 13:16:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/915860/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/915860/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/915860/5"}}},"commit":{"parents":[{"commit":"4d06a98bb14580da4129e07e3d672bd8d7296681","subject":"Drop Murano","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d06a98bb14580da4129e07e3d672bd8d7296681"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:16:44.000000000","tz":0},"subject":"Drop Senlin","message":"Drop Senlin\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: Iad4d850ba8f8b8446591cab883d2ad6243576926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba2aba8cf384156e01fd6ca40d95b0f62fc89b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba2aba8cf384156e01fd6ca40d95b0f62fc89b69"}]},"branch":"refs/heads/master","description":"Rebase"},"14a2e37f00a8197de97f88eebe88b2ff385b4ccb":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-15 13:44:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/60/915860/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/915860/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/915860/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/915860/6"}}},"commit":{"parents":[{"commit":"813035ba8fcc2a92e78412d8e4a96740c55e0c2e","subject":"Drop Murano","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/813035ba8fcc2a92e78412d8e4a96740c55e0c2e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 12:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:44:15.000000000","tz":0},"subject":"Drop Senlin","message":"Drop Senlin\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: Iad4d850ba8f8b8446591cab883d2ad6243576926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14a2e37f00a8197de97f88eebe88b2ff385b4ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14a2e37f00a8197de97f88eebe88b2ff385b4ccb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
