)]}'
{"id":"openstack%2Fkolla~915863","triplet_id":"openstack%2Fkolla~master~I387e5c345cd5ea45e9231b9ce5baf58a8d74caf9","project":"openstack/kolla","branch":"master","topic":"kolla_caracal_drops","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-16 16:16:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I387e5c345cd5ea45e9231b9ce5baf58a8d74caf9","subject":"Drop Solum","status":"MERGED","created":"2024-04-15 13:12:30.000000000","updated":"2024-04-16 16:17:47.000000000","submitted":"2024-04-16 16:16:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915863-kolla_caracal_drops","meta_rev_id":"b93d67bd9e3347919a3cbb6f91a5f4ddcadd368f","_number":915863,"virtual_id_number":915863,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 16:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-16 14:27:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-04-16 14:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-16 14:37:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-16 12:22:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 15:58:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 14:27:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-16 14:37:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"7856531272776aa0a30c8a6f8f6a44929ef1d9e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"708cd984c23bb0ca6478f6a58eeff100daf7a7af","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:19:57.000000000","message":"Topic set to kolla_caracal_drops","accounts_in_message":[],"_revision_number":1},{"id":"59be38df03cb7f914d09b0f3711bf4d08c028976","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:20:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* docker/horizon/extend_start.sh\n* kolla/common/config.py\n* kolla/common/sources.py\n* roles/kolla-build-config/defaults/main.yml\n","accounts_in_message":[],"_revision_number":2},{"id":"b73f5d3f7dd8a2440db932493f0a5a44c66424df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:21:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fd52052b76fc0492d70a402c21c0d0dc9015f725","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:21:39.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"2cbd55a1f239b9494908406cbc0938e741001826","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:22:03.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"2f05cfb646f88e6409d5cc428dead592988bfd55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:44:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"0be8e1bd88b015debb76a4e706bb7e002f2a2604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:58:27.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30509d4f572b496bb53b46359b97e55c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/570f671f2fde40158275388adfe7f1c1 : SUCCESS in 1m 53s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b645430db9cf49e0bee1c62fbc6106bc : SUCCESS in 23m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ebc82e88be1841ef9acc755def484ad9 : FAILURE in 52m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5167f68195534445b27a39fc65db2883 : SUCCESS in 29m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/28bf0613f8904d4983ccdebf09b4f2ff : SUCCESS in 26m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9e867037930c4622bb89da8919184769 : FAILURE in 38m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0eb39b924d294952af9d4f8d17036aa8 : FAILURE in 1h 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8d390c4e5344d2b8b3cabef21c6e46 : FAILURE in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22df7bf759f0403db922cfcd26ed8ad9 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8efe918a99544f3bcae7aa4b123cd4e : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e861698abac3453eaa36cf483a7512e1 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/059071d46b7a4dd89709fed6bbc07fac : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477d57f4b6ce4b1a94eb1ccfc6b07755 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e801d9195f134f71be20c46d4eb44ce6 : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/eb0943d8d5674d8babc0a595e3ca38fc : SUCCESS in 38m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2e3a21c0df9446ef8f59bdc012c3dadd : SUCCESS in 30m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ac6e0b757bac4fbea5a9c3ea0d4dd5d3 : FAILURE in 45m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0892faf949bd413dae62db42aa08092b : FAILURE in 2h 06m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6d81f302a87c4808b2c37b7133120a65 : SUCCESS in 18m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e3ddddebf6354559851d09fc6e9c8a19 : SUCCESS in 58m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/55ba074f8a2a45b192f8160b7b66262c : FAILURE in 45m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ac9c504c9a3f4f9cb5269a515fbc2925 : FAILURE in 1h 33m 01s","accounts_in_message":[],"_revision_number":4},{"id":"c9dee789fd38b3b1c11168be8194955617128e56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:25:10.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bf41ef18bf340bea3291a1ea62c7948\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b5f20a4a03a8411fa1318ba4dfc724c0 : SUCCESS in 49m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9626f8f0c0154c1789d3e3ca7172412b : SUCCESS in 45m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/14acd35dc9d14278859052623c9695c4 : FAILURE in 2h 38m 32s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f89d62bec70d4f0eb7dbc60506a2654c : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cbbcc56fa53c44138e1e5faa4e9b5320 : SUCCESS in 3m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/21932cb6f2ba4654a6e073830051e6ce : SUCCESS in 50m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/11658970ab21400d81799a43665c0ede : SUCCESS in 46m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1b3c16e536aa586ae13a2a41819930cd2cc0b560","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 12:22:12.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nThe following files contain Git conflicts:\n* docker/horizon/extend_start.sh\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a041b3cada271525c79a9d70e72fc2fd3ddff885","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 12:22:36.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"c9ae19bc42237fa43c40f6cc84b4958eaf6f331c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:04:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56c9e572fd944434bd9a36e0abfa3aef\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0f2e5826cad94a0c9d80e7739e630194 : SUCCESS in 1h 39m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/276346034f3d4e5a90dc27389761f12f : SUCCESS in 1h 37m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/73824a8047c24c6ca8063da9051b08aa : SUCCESS in 1h 07m 42s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0abc2f78258e4f4692315f7af5ecd481 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a99ec86f0a104bd9bb19596cfc769fe6 : SUCCESS in 3m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/567e58687a474661ae5c0dc551f91074 : SUCCESS in 1h 31m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7a5c06b367314b0fbb5e82fb8de624e1 : SUCCESS in 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6668ed28e2ca169872937e6b7dc152ee7ba662a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:07:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16577a8364164cf59269640f4148a6e7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/86b79805cce0483a8d77fa455f6fb317 : SUCCESS in 2m 42s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/22d53e14874e44b19d5677ecfa298fce : SUCCESS in 23m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/37db03f2a9b0457f8f90b251232f3b57 : SUCCESS in 56m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/beb981ea4e94445b8d334f6504c8f0d1 : SUCCESS in 25m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/09bc98da515849729c0b7a6c4dd6531b : SUCCESS in 29m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a317f33669c8469d83856e7f0897ef5d : SUCCESS in 54m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/29438bf062f74756bf67a119c4ff1d7c : SUCCESS in 1h 34m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3280a4514ae94076afa4f1cf4d4b7af3 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/002af0547d6f4ee8b29534b1667df0ef : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8342b7155092411bae6344d3ca79bdd1 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5684a8368ed4075a44b0afd0b1a2dce : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ae7e7678c7a4a3e8addaca8236bc79e : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c334ee38ac4331b222c7ce67689e6e : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3408029c740409fb3c522f8267f44ab : SUCCESS in 3m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fcbd3b2e337d4fc3846b3bf23d1d5b7d : SUCCESS in 24m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/22825abdf69648d6af7724244f8e296b : SUCCESS in 33m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f2d1b74e31da4d5db1ece5590dc8f27e : SUCCESS in 1h 00m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/55d0010da37a4ce4bf8bc60b3aa6d9c4 : SUCCESS in 1h 36m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e7136146fdb45c78f0020daf0e1892a : SUCCESS in 25m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f60a5652a96544bb960f7ae99ee2ee7c : SUCCESS in 55m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/458d8c3f1a0343e39293668d483d2bf9 : SUCCESS in 53m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3c9040bb3a044f3c93966ae90fdaa606 : SUCCESS in 1h 20m 43s","accounts_in_message":[],"_revision_number":6},{"id":"5119bb8fcd9f803bd5fbf1b0825868bc2a22e22a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 14:23:35.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"c6b67295189a53794a5f2ed7ea46c39e21fbe678","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-16 14:27:20.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"dc11f235ca978710da0389a7aa78e6a3acf2efec","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-16 14:37:04.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"269dcc0b7de17811f794c8c8b790cca718e9a25c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:37:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d3fda718c774bce302c24ed441960998a58f36aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:16:40.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/874d9255b7e9428881c933432a3ea8c3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cf826c83a9be42feb3242d484d5e3a16 : SUCCESS in 2m 06s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5ba307cfd95d4bf09ebf450c28ff0b5e : SUCCESS in 26m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6a3cfdd733934abdab762bfa750cea81 : SUCCESS in 29m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/60810d9b229741958ac77be1f0acfd63 : SUCCESS in 1h 10m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/363a2cc329834cc593e8b5e5eb9dd652 : SUCCESS in 1h 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca394ed6a05e49138c2cee332de5b9c9 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7be1ec63f164d0ca155f55ac0bdd922 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da3ab3b9996e4f268523127d26371c62 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cba2d137f924f2da81e7692a6853472 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61eaa277945446e3b251cabd15909530 : SUCCESS in 4m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0fc37871bfb1487aa8b611c5a3182d2d : SUCCESS in 30m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e20f8c7e8aa14a4f9be70e0a07d02693 : SUCCESS in 33m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1d74ae61001b4586bbe10b9c3c53854b : SUCCESS in 1h 08m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9832f87fd0854424af99e04c38cdc479 : SUCCESS in 1h 37m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c4af5f4f1d2f47129d5be64abdea868d : SUCCESS in 18m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a32292c492fe47ab9292d6aa9e01b316 : SUCCESS in 50m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/214fa9f103de4074a507b3ed7c68f808 : SUCCESS in 50m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/32849754f3b040b88b373b0cefdddcc4 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":6},{"id":"f91ffeba13f061cfafd43f14c2a3ae714e5d7d57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:16:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b93d67bd9e3347919a3cbb6f91a5f4ddcadd368f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:17:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c8d78f6a8004c4486f5fd204f7f81e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81cbb25d52b4563b5a2defb041dbb34 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99ebc65343df4c81b59e8e5eda76139f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a9bde9b1ac5651f2859b4bcc00f273bf772b8198","revisions":{"c4e2c09a4f6311fb100a189cf220d159ac37f760":{"kind":"REWORK","_number":1,"created":"2024-04-15 13:12:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/915863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/915863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/915863/1"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:12:27.000000000","tz":120},"subject":"Drop Solum","message":"Drop Solum\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: I387e5c345cd5ea45e9231b9ce5baf58a8d74caf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4e2c09a4f6311fb100a189cf220d159ac37f760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4e2c09a4f6311fb100a189cf220d159ac37f760"}]},"branch":"refs/heads/master"},"7a9b66c26531862c7c6414caafdceb840a44dfa4":{"kind":"REWORK","_number":2,"created":"2024-04-15 13:20:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/915863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/915863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/915863/2"}}},"commit":{"parents":[{"commit":"1b52ce0f37e45d73b24701352dad5a5f25c7345b","subject":"Drop Sahara","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b52ce0f37e45d73b24701352dad5a5f25c7345b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:20:10.000000000","tz":0},"subject":"Drop Solum","message":"Drop Solum\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: I387e5c345cd5ea45e9231b9ce5baf58a8d74caf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a9b66c26531862c7c6414caafdceb840a44dfa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a9b66c26531862c7c6414caafdceb840a44dfa4"}]},"branch":"refs/heads/master","description":"Rebase"},"976b513fbe23dc16aa2d177d6c0882b626e29adb":{"kind":"REWORK","_number":3,"created":"2024-04-15 13:21:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/915863/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/915863/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/915863/3"}}},"commit":{"parents":[{"commit":"1b52ce0f37e45d73b24701352dad5a5f25c7345b","subject":"Drop Sahara","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b52ce0f37e45d73b24701352dad5a5f25c7345b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:21:20.000000000","tz":0},"subject":"Drop Solum","message":"Drop Solum\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: I387e5c345cd5ea45e9231b9ce5baf58a8d74caf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/976b513fbe23dc16aa2d177d6c0882b626e29adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/976b513fbe23dc16aa2d177d6c0882b626e29adb"}]},"branch":"refs/heads/master"},"d6d00277803e3ee4f21fec402bba42f5affdaea0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-15 13:44:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/915863/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/915863/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/915863/4"}}},"commit":{"parents":[{"commit":"67dab4c5ecb288044ba72eded9d6166d48049dd2","subject":"Drop Sahara","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67dab4c5ecb288044ba72eded9d6166d48049dd2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:44:31.000000000","tz":0},"subject":"Drop Solum","message":"Drop Solum\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: I387e5c345cd5ea45e9231b9ce5baf58a8d74caf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6d00277803e3ee4f21fec402bba42f5affdaea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6d00277803e3ee4f21fec402bba42f5affdaea0"}]},"branch":"refs/heads/master","description":"Rebase"},"41a53a72bc714144ee6f9a890965d7055a0328e0":{"kind":"REWORK","_number":5,"created":"2024-04-16 12:22:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/915863/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/915863/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/915863/5"}}},"commit":{"parents":[{"commit":"4d9379ecc8d3c66902763aca0ae14643350a8976","subject":"Drop Sahara","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d9379ecc8d3c66902763aca0ae14643350a8976"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 12:22:12.000000000","tz":0},"subject":"Drop Solum","message":"Drop Solum\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: I387e5c345cd5ea45e9231b9ce5baf58a8d74caf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41a53a72bc714144ee6f9a890965d7055a0328e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41a53a72bc714144ee6f9a890965d7055a0328e0"}]},"branch":"refs/heads/master","description":"Rebase"},"a9bde9b1ac5651f2859b4bcc00f273bf772b8198":{"kind":"REWORK","_number":6,"created":"2024-04-16 12:22:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/915863/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/915863/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/915863/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/915863/6"}}},"commit":{"parents":[{"commit":"4d9379ecc8d3c66902763aca0ae14643350a8976","subject":"Drop Sahara","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d9379ecc8d3c66902763aca0ae14643350a8976"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:11:56.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 12:22:36.000000000","tz":0},"subject":"Drop Solum","message":"Drop Solum\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: I387e5c345cd5ea45e9231b9ce5baf58a8d74caf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bde9b1ac5651f2859b4bcc00f273bf772b8198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bde9b1ac5651f2859b4bcc00f273bf772b8198"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
