)]}'
{"id":"openstack%2Fkolla~915874","triplet_id":"openstack%2Fkolla~master~If2d034789d8c1128597c9072b4e82d14f72492e0","project":"openstack/kolla","branch":"master","topic":"kolla_caracal_drops","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-16 16:31:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If2d034789d8c1128597c9072b4e82d14f72492e0","subject":"Drop Freezer","status":"MERGED","created":"2024-04-15 13:27:35.000000000","updated":"2024-04-16 16:32:06.000000000","submitted":"2024-04-16 16:31:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915874-kolla_caracal_drops","meta_rev_id":"7d4afb979c0aae594c3440dbc85438dfc1e0cdd4","_number":915874,"virtual_id_number":915874,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 16:30:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-16 14:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-16 14:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-04-16 14:44:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 16:33:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-15 16:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 14:37:09.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-16 14:44:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4dc18da1223ef37d5a55e33cce0cadf7a7c845f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:27:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5da6f0df46b83b5d7fda324b68f22616b3b10015","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-15 13:44:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"741382c179a109f5c3c74fd863cb31d5724cc6f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:33:36.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcab73d8140449189185b982b96be5ef\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7d73d0db697142b8840bf2a51d0b5e35 : SUCCESS in 44m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/890bb6c5d7d24d2aa37d19c3393bd1f7 : SUCCESS in 2h 40m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d9c3ad2d55a94a5db3b91ac3d3e974a5 : FAILURE in 2h 24m 04s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a50d30640c94fbb9581481ae9924c8a : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bee4c94577a4d579a5ab701a3fdd241 : SUCCESS in 5m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d7fe8b5156024afa80faf058a592d33a : SUCCESS in 39m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4220ea56f5c44038afe1d47ff4c6f12e : SUCCESS in 2h 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d40cef2fcba0cdee0dece3ff44e6829fd21172a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 16:52:25.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95fe63d3b9a741d8bcf8a5dd542b9415\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a59e446d6614fb89d45b02b7044fea5 : SUCCESS in 3m 02s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/40027fa679f0462fbea4331dd47b054d : SUCCESS in 28m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fa7df53fa6df4eadbd19fa09f35f9570 : FAILURE in 46m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8d750130a9d0453488bb51bbe1cb0636 : SUCCESS in 24m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bf843cae47724601a52cc1b70b44c3d1 : SUCCESS in 29m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fd7aef5a482c44a9b327a407ec3eeb97 : FAILURE in 43m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2447d05b08364b4e8ac349ba6130f3fc : FAILURE in 1h 24m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09791e4c776d49379d912e4c938ff043 : FAILURE in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe3d5555ab0048fda1031d9ae0c71339 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28ef84133dba44e1b72973dd5f1d81ca : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cd29d8105f9401ea74905615ae9a2e1 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd37de50c5d44cea199e8dff8a41b6d : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d248f5ba7cb64754af1e9f9f53e6cbde : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7e587e239a942ee9a4ac585affdfbb5 : SUCCESS in 3m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0bfb3147411245b6b56775db69f2722e : SUCCESS in 31m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1dbac298ea014d0c857d45d5b3e9978d : SUCCESS in 33m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/99152d5749b4449dad86db759d1991af : FAILURE in 46m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0deddf8cff434a50a7e25af512d3cc67 : FAILURE in 1h 26m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2fdc79dae81a48d2bcd03f0f1ca28816 : TIMED_OUT in 3h 01m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b47ddc2c679e48788c579b3c502ebd77 : SUCCESS in 49m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5997c7cae78b4b2ebfb2fceada6f4071 : FAILURE in 41m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c65a6ace1049498a994363de60cbb20d : FAILURE in 1h 24m 26s","accounts_in_message":[],"_revision_number":2},{"id":"792e0f6379dfcfe94406f2dc010d2f728f4f7834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 12:22:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8643c7ad6479c285cf7c4b36b978542fc0732baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:07:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7597e0bb92664cba99a4744d69d47d9d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f0090f22c9a4c0ab0b4b46b2085ffc9 : SUCCESS in 1m 59s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ff0cc3b090874f5287ec0195ac969864 : SUCCESS in 26m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cee3e975cc7d4c36847f325a8579812a : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8d62ed8a3ae74a2da5c36061bf806b13 : SUCCESS in 24m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9b14ae9aa62647f9b3dd02f07486bcc7 : SUCCESS in 32m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ffb117c881f3444bb6aaf6040a19be3e : SUCCESS in 1h 04m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6081a2c45e824fdeb45038d00735d2bf : SUCCESS in 1h 25m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b7a53246604422b4c8df144fa6f049 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67ae0c43ac9646df81168e8f894bf6ec : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2abc82b9792f4af6973d2bf3e01d7ab4 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e19921fe13e4d8c8613d683cd8869a5 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e2f88e0b7ee48a9a1f5610b915cd4ed : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282ddc0beaba4feea4d534a9cd8cfec5 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1726ca4f31ae4cc1922d1bcef77dcc72 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/31089c0bc81847559f4b23458ec5d66a : SUCCESS in 25m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b7426c60277c4f47a712caf02dbcf121 : SUCCESS in 33m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1419e2722eb44186957264d6ffd1fc4c : SUCCESS in 1h 06m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/36a276b252584e76a4490ae928267611 : SUCCESS in 1h 39m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/55d1459bef7b4649b1bf3322b7f6c379 : SUCCESS in 25m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/96bd7b8684184a84be4ada5c8c9bd3b0 : SUCCESS in 49m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cd7b112bf9d6423fb2c13bdbb75184b1 : SUCCESS in 53m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/03a4b5b6428e4aa9a0b6514243951d64 : SUCCESS in 1h 27m 00s","accounts_in_message":[],"_revision_number":3},{"id":"8e2abf0b37f98b62d70c5a2853d36382697b2856","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:19:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe7ae7e39e5469ab2c828ea83c4cebe\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4a2f80a39ec8465f8c909a836945c304 : SUCCESS in 1h 41m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ce7861d2e9af4e29975dc292b5f2c65d : SUCCESS in 21m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/53d1fbd76f7a402e9b877d6154c8591a : SUCCESS in 1h 54m 07s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d165219b49944e299e8a8a42272d1de7 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d65bdbeea1cf49709a4d5cf3a2973290 : SUCCESS in 3m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d0c2f5a3d5d5413e864b7d67c4360049 : SUCCESS in 1h 35m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3cab9e12b9144752af09848eeb76fe54 : SUCCESS in 20m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"01a1ac4359b6487b8029ac7b900b0072b75497f2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-16 14:37:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"89f60485a3c07f1aa090ef929ba065e34a866163","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-16 14:44:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5b482c0b6cba2a86b5355932d300010b298a7213","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:44:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"587c2234639912cdfb518d39425bc0e3261b9695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:30:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c22d499420364b62b9e8e2de4105de83\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eb63ebc9856f437fa87391a7892a835e : SUCCESS in 1m 55s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8063f156fb504e0596cc6614f5f0fa45 : SUCCESS in 28m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1f8a4c9066da4b09a315318cd858c6f1 : SUCCESS in 32m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dc7767ac125f455dad4170c2b5637ef8 : SUCCESS in 1h 02m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6a396643e6de47a1882fd5422545139d : SUCCESS in 1h 41m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3a022da81d14ceab24e9f68101d55fd : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e519b4a9c244cc2b49c6501e54d8eef : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/718021667c864cb8a4c0918fd19fc385 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1fed59166a426695a8d0552b310b56 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7b66788476f4f20a1be34c8fe6f652e : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b170c4e88fe64b67ae94242328850da1 : SUCCESS in 23m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1c60a862595a49f593917f3ffcaddd16 : SUCCESS in 34m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2bc922a6d86c4a7db074972ea4adfea3 : SUCCESS in 55m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0bb992a9b09a4aaa841dbfaf181019b0 : SUCCESS in 1h 35m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d0ac03028d724ae79d675a8503bac78a : SUCCESS in 18m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7cd1d90cb0e54084914e6c7b46f4a997 : SUCCESS in 44m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b14c5c543ede4f4f8cfeed25f23dd79b : SUCCESS in 45m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bf36ba69ed8b4cd28b39f74b2462484e : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":3},{"id":"d8b7cf27d27bbb8873933244a83e45dedc1a3f67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:31:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7d4afb979c0aae594c3440dbc85438dfc1e0cdd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:32:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72d70651b7a44b2baf4bc1bb06921e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a721cbaa2c48f1b18dbbc02ef5f9b2 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1371ebb20514e80956412c8b572d156 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c69284fd8b6a44a8c1f24771d28d8d510a0855b4","revisions":{"9f8ffbe8777e66981d282d85542beb64d9ebbd1b":{"kind":"REWORK","_number":1,"created":"2024-04-15 13:27:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/915874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/915874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/915874/1"}}},"commit":{"parents":[{"commit":"976b513fbe23dc16aa2d177d6c0882b626e29adb","subject":"Drop Solum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/976b513fbe23dc16aa2d177d6c0882b626e29adb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:25:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:27:25.000000000","tz":120},"subject":"Drop Freezer","message":"Drop Freezer\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nThere are some efforts to restore Freezer, but let\u0027s remove it for now.\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: If2d034789d8c1128597c9072b4e82d14f72492e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f8ffbe8777e66981d282d85542beb64d9ebbd1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f8ffbe8777e66981d282d85542beb64d9ebbd1b"}]},"branch":"refs/heads/master"},"ec737505978402292a2e5de97ea62196782933f7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-15 13:44:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/915874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/915874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/915874/2"}}},"commit":{"parents":[{"commit":"d6d00277803e3ee4f21fec402bba42f5affdaea0","subject":"Drop Solum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6d00277803e3ee4f21fec402bba42f5affdaea0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:25:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:44:38.000000000","tz":0},"subject":"Drop Freezer","message":"Drop Freezer\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nThere are some efforts to restore Freezer, but let\u0027s remove it for now.\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: If2d034789d8c1128597c9072b4e82d14f72492e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec737505978402292a2e5de97ea62196782933f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec737505978402292a2e5de97ea62196782933f7"}]},"branch":"refs/heads/master","description":"Rebase"},"c69284fd8b6a44a8c1f24771d28d8d510a0855b4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-16 12:22:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/915874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/915874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/915874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/915874/3"}}},"commit":{"parents":[{"commit":"a9bde9b1ac5651f2859b4bcc00f273bf772b8198","subject":"Drop Solum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9bde9b1ac5651f2859b4bcc00f273bf772b8198"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-15 13:25:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 12:22:41.000000000","tz":0},"subject":"Drop Freezer","message":"Drop Freezer\n\nIt\u0027s inactive and hasn\u0027t produced a 2024.1 release [1].\nThere are some efforts to restore Freezer, but let\u0027s remove it for now.\n\n[1]: https://governance.openstack.org/tc/reference/emerging-technology-and-inactive-projects.html#current-inactive-projects\n\nChange-Id: If2d034789d8c1128597c9072b4e82d14f72492e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c69284fd8b6a44a8c1f24771d28d8d510a0855b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c69284fd8b6a44a8c1f24771d28d8d510a0855b4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
