)]}'
{"id":"openstack%2Fkolla~915952","triplet_id":"openstack%2Fkolla~master~I5491adef46fe0cec013a597cb05018d8d7e862f0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-04-17 06:36:10.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-17 15:39:56.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2024-04-17 15:39:56.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-04-16 18:41:07.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-04-17 15:39:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5491adef46fe0cec013a597cb05018d8d7e862f0","subject":"Revert \"CI/Master only: pin opensearch{-dashboards}\"","status":"MERGED","created":"2024-04-16 18:33:43.000000000","updated":"2024-04-17 15:40:59.000000000","submitted":"2024-04-17 15:39:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":915322,"submission_id":"915952","meta_rev_id":"ff239f111f4562855039f53ef248768d830f4d50","_number":915952,"virtual_id_number":915952,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 15:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-17 06:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-16 18:41:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-04-17 07:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-17 07:45:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-16 18:34:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-16 18:33:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2024-04-16 18:33:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2024-04-16 18:33:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-16 18:33:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-16 18:33:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-16 18:33:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 07:45:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"83dd798c21fdd273a8c6ab0cb5b03963601a2b48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:33:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b0a1ac76da90a357d7e3c3cf9ea2ad26a1a48fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:33:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"28b7e69f321ca26d53a9091545cbacf448e68bf5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:34:04.000000000","message":"Topic CI_pin_opensearch_212 removed","accounts_in_message":[],"_revision_number":2},{"id":"bf2bb2c629815326c78dbe4e2c01de2b956afd04","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:34:26.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"6c29b612f8a21bcb6249364dec807f8073e3f179","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-16 18:41:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b101a38644ce6791b27dd4d244d3fe90ffda42d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 20:54:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f95e28972d44c2ea96050fbac04acac\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c587304267be42c4b7b487b7971312c1 : SUCCESS in 22m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c054824727f14fb296765aef5f426f57 : SUCCESS in 21m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f4a221d6b81c4778a28e3749a98bf977 : SUCCESS in 2h 14m 49s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3566d8a98187410bbfedc0d0febde132 : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d86dc76e1bb7465ca4392a894fce3ff9 : SUCCESS in 5m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c7745caaa63c471fa06cfb650929f5cf : SUCCESS in 24m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f6802fd08fe545fcbfd81ddc6977e2d6 : SUCCESS in 2h 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee158ad0199d53ba001c4bd4d803f45cf90883a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 21:01:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d47314a5238041e68f66d55a6a095a79\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9d6cd3a347d445cbb7ec5f6eca6c7656 : SUCCESS in 21m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d4ae27bf009b41178bf2eb5694a03a3b : SUCCESS in 52m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/da99c5645f194eeb9d8c6390d5855e82 : SUCCESS in 19m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0d306780698a450ba2f4b1650f531949 : SUCCESS in 28m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3b8aa4253ae140e7ab4d689ef8f5c59d : SUCCESS in 59m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0320c0a6a62b4d628afca8e821c1cc06 : SUCCESS in 1h 23m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868a81b7de6e4bb9b6412b0fd377ac76 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8e3566c01a04ca8b70925cd79614878 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13f50f07d2f34a689716b6c650a5e82e : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da9ac18659d1423f8f3f8b8004d0aa01 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1df3796c386a4fa3942600142c2c4715 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09b928bf0c174a3b97cd07980de6cc5d : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc9fc27d9340440fa29eb558450d2d1c : SUCCESS in 25m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3ec98b505e054d899e5bdff38be23676 : SUCCESS in 29m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8f788e0bc4104c96bd5117b51ea77f18 : SUCCESS in 1h 08m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b0db34716d164f48b423134baf1bfe21 : SUCCESS in 1h 34m 21s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/f5e400f0bfb44d9e92e34fda46eff333 : SUCCESS in 1h 30m 28s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e80aa316c8d7450684a059a215b16d7a : SUCCESS in 2h 26m 38s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/4be0d0f444f841a2bf6009b8841ce2d8 : SUCCESS in 1h 12m 05s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/f722248f690e4e1da689de9df059fd3b : SUCCESS in 1h 19m 31s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/50b6de30546846c58694be7f04aabc65 : SUCCESS in 31m 13s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/bab9450d51af4fe59c07bd4461cf0b2f : SUCCESS in 54m 16s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6c2d3c9690d446868180bfbb99b3f132 : FAILURE in 1h 00m 36s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a2c0c9f11c3a4654a69fc035c9bd7bb8 : SUCCESS in 1h 27m 38s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/30461a6554e340a59ad671698cacdcb0 : SUCCESS in 38m 20s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c2eb41d2509842cfbac014b4992217cc : SUCCESS in 19m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4c4d4079def9497ab5d49647a0fda5cc : SUCCESS in 43m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9e713469b17244008adf754347550b77 : SUCCESS in 50m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/530489ca7a9c48af88b84d9330b00e35 : SUCCESS in 1h 21m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/63916d7061c64b65b97f5dfa8ff12be3 : SUCCESS in 1h 33m 07s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d546f02469034314aefc6817bc62a123 : SUCCESS in 1h 57m 42s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e4f0b62b48ff4c26b50bbca7083e00b6 : SUCCESS in 58m 34s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/bd2eca9707d84b438251a421fb9e5988 : SUCCESS in 1h 08m 40s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2053b75bb5ae49f1be30cb1437e654a9 : SUCCESS in 21m 41s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/fa5e6e82ff524540be79b1019ca7292a : SUCCESS in 53m 20s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d1e3ba3661a6420db9df08e33386b2a0 : FAILURE in 46m 09s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/92c7a1a762db4c9495e91595fd34df58 : SUCCESS in 1h 09m 15s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ba04465bc2044b47ad398bf78728b04b : SUCCESS in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11b7b3a2680c8bb587bb5715014ea0e93684d539","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-17 06:36:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"01e9f02db6f2c8c3f663b1f3ef6fdee7ab02b091","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-17 07:45:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01ab903b618d79bf0307779c49c1eb80036fd749","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 07:45:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"872ea8a311c95583f2c7224018b1c992e62512ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 10:53:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0bc573b3d994a31be02964cf3866e85\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4360a590035f44f794e9b662a976f666 : SUCCESS in 19m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ffbe118e81e34ded8f5d96599cac1dbc : SUCCESS in 26m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4bb3eb6b8ec943ce98d05af321e059a9 : SUCCESS in 1h 05m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/516a06a206bd4fa9a9cc308572627cd1 : SUCCESS in 1h 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b343af8b73341839d4943446752def8 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a238dabe709e43f19ae17e6bcb441587 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56ccb05aa0f7427ab328b24636ec428e : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed3d393eb6a4446bec1d5b5934ec685 : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2c910da966be4a2ebb122f72f268c401 : SUCCESS in 26m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f390528ed47c4d2ca5b3168e39d69daf : SUCCESS in 29m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e8660f6f3ff940e5938fda92409d6cb9 : SUCCESS in 1h 05m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/258f92bb3ca24976bbe8df74021c3dac : SUCCESS in 1h 45m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3baf407b3bf24c598b53cd8e22e08644 : TIMED_OUT in 3h 02m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/28b5f64f61a248ddad75921f0f592625 : SUCCESS in 1h 06m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9b1c9343091d4f3ab4e4c17506c4aaf3 : SUCCESS in 57m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c53659e5d0f04088ad4cacf9a0435ec2 : SUCCESS in 1h 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"62c97d14ac6c1728b2e7aa175dba9e36b107db87","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-17 11:16:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0acfed37ee1902d0920a964532abd8981ee46116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 13:57:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb023630af06411a96a8ed48df91c796\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c25dff08eaf849d68628fe15e912616f : SUCCESS in 30m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4070c937237a4c4fa0a0cd7fa0f778c1 : SUCCESS in 1h 01m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b877fbcfc30b4b01aaf0163444dde526 : SUCCESS in 22m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/45a46687a8d047388a10c34918f32512 : SUCCESS in 39m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fdc7b34e15ca4c6c81477c358851fd3e : SUCCESS in 1h 01m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/24eda9b78ec04500a30c43010026bcdb : SUCCESS in 1h 42m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08035fec338646d983a08a6b39c7f672 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/155e41f031c1491eb15f65a0e2934343 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5c480b43e6a47448ae35dc0ff97712a : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6444fab5038412ebe6ffc5e2ce8e97f : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/038ec897f6fc49918473e6fec482e2af : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b323b13020b64a8cbcfc8f378f361de5 : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3e282f7e91a34283bbf08b8829ed63e1 : SUCCESS in 23m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f69ed6e3fb354f618894914c33274283 : SUCCESS in 36m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/61a79634d1a6454d890e7aac7f3877d9 : SUCCESS in 1h 02m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/169a4c2ca1cf4bb59a56c189f07c837f : SUCCESS in 1h 37m 30s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1fed540cf2a24cf2a08953034f4bbdd8 : SUCCESS in 1h 54m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d3ce8347d778465e81d944c91dcf8f58 : FAILURE in 1h 14m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e0a64605546148f79a4d510d72a9e1fd : SUCCESS in 1h 24m 16s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/3d083fdc1cff48f3941ca2135d4597a7 : SUCCESS in 1h 34m 55s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/4b0a8673ee8943cd8681e0d505728f6b : SUCCESS in 28m 56s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/b9195b56bf3041b6be2ba34f952addf7 : SUCCESS in 1h 03m 40s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/361e03f751b94513a17831b7e2c958d9 : FAILURE in 1h 03m 06s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ed126d6a0c0f42b8b455593e5e993dd5 : SUCCESS in 1h 31m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5161f4fae48d4d55a61aa5bb1aacce6a : SUCCESS in 40m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f524996f73f345dc8eac9ea1b8ebc702 : SUCCESS in 19m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b869793bfcf54df49c667d7b17ee69d2 : SUCCESS in 56m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6189df33d83e4a1da62341e22dfe4664 : SUCCESS in 54m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/071be2c1175e45259231ce77e33c23ff : SUCCESS in 1h 13m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/375be779c79a4b78a429588a026520d6 : SUCCESS in 1h 34m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/19434f7cc4874d4a9af0e4f576e02236 : SUCCESS in 2h 33m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/00c0e0ee2b164311b2c3d107b7f25169 : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/ecc30502e6c845e5b3766ed4c01c9743 : SUCCESS in 1h 07m 35s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7732748463494e90b8d4aa69f6d28aec : SUCCESS in 21m 39s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/eaf5fe9220684b19bdb46533a67bce88 : SUCCESS in 53m 16s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/eab6023324fc4a91b4af57c58e8969ce : FAILURE in 50m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cd56cd3644474c4ab52eddcf7cd020b7 : SUCCESS in 1h 06m 30s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a102f04652614489aec1f46881ffee31 : SUCCESS in 30m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"495c4fceb849f919d0f6bb2920cbbe611bf7b9d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 13:58:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c8bfc274f5428c0677e006160d435fd24b11036","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 14:12:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331f652ae63c44b7a92db0138026bc66\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e9f95c28185842f6867d2af17896acb3 : SUCCESS in 24m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec0cb91671e140659e501478da4ba2d9 : SUCCESS in 17m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f1652f227e1447e3bab3cdf3d6871d88 : SUCCESS in 2h 03m 49s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/565d7b8ca1d440929088bbf86e5b0fee : SUCCESS in 2m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1b939ed66014391b6cf74cb1a7b8ef7 : SUCCESS in 3m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/df991cceebf74274bab7368a8a86caa3 : SUCCESS in 21m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dc0b2e9d21654541acba811e487ffd26 : SUCCESS in 2h 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3aee60bddb25844d1ea75af2585bd24a2892ded","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 15:39:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/332de68780664bf8907b158a7c7f99e4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/17d40893a40743858dae3842da6f7331 : SUCCESS in 23m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/699a6204147449e7a1bd1d418437eb81 : SUCCESS in 25m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d46b8dc6253041038cc395f872a022d8 : SUCCESS in 47m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d788a35a0e5545c88770eafbd154b9a2 : SUCCESS in 1h 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ce56d58ee7248ee9c6a8d06ca93a091 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a82bfec719f744949c141a60b2a500cd : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/abade639f44d4729af8a18f1bdd52480 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986718c45346460cb1c7b2d8c0badccc : SUCCESS in 6m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/da4879c7fae049a8a1a3a0d36e573654 : SUCCESS in 23m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0d73f26052e6486bb6b3079b2db54ac3 : SUCCESS in 30m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/525c2b5d43d442f492f71390c91ec0c0 : SUCCESS in 1h 06m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/67b5681c63da4c0d8160bb4bb04a4a4a : SUCCESS in 1h 39m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f7e6d0a9cd844fc7bde45d959a4090e8 : SUCCESS in 16m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9f1644fde9cd40488ba3e0d4bee4bf4b : SUCCESS in 49m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6a8e7b02ccbf4c81b3220b2827562e4e : SUCCESS in 57m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fe9faea4a8374557b473878d46a9d9aa : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fccd2c4b0c377daeee751f57c75bad02d000574f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 15:39:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ff239f111f4562855039f53ef248768d830f4d50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 15:40:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cecbda52dd4be8a3e95938b41df0be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cf68b82e134605b72ab4aa054cd985 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"340350a387cbbe417eff9385156d665c3edd683d","revisions":{"acc45af2b5d3d623069d8be9517660d67d97730a":{"kind":"REWORK","_number":1,"created":"2024-04-16 18:33:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/915952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/915952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/915952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/915952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/915952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/915952/1"}}},"commit":{"parents":[{"commit":"19a004e0a500ea4839fa1e089e1c50c953cec3b3","subject":"CI/Master only: pin opensearch{-dashboards}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a004e0a500ea4839fa1e089e1c50c953cec3b3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:33:43.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-04-16 18:33:43.000000000","tz":0},"subject":"Revert \"CI/Master only: pin opensearch{-dashboards}\"","message":"Revert \"CI/Master only: pin opensearch{-dashboards}\"\n\nThis reverts commit 19a004e0a500ea4839fa1e089e1c50c953cec3b3.\n\nReason for revert: Fixed\n\nChange-Id: I5491adef46fe0cec013a597cb05018d8d7e862f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acc45af2b5d3d623069d8be9517660d67d97730a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acc45af2b5d3d623069d8be9517660d67d97730a"}]},"branch":"refs/heads/master"},"340350a387cbbe417eff9385156d665c3edd683d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-16 18:33:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/915952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/915952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/915952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/915952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/915952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/915952/2"}}},"commit":{"parents":[{"commit":"7c098bdd3cac73505851d4c97d7f39e966f00e1b","subject":"Fix OpenSearch file ownership issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c098bdd3cac73505851d4c97d7f39e966f00e1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:33:43.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:33:55.000000000","tz":0},"subject":"Revert \"CI/Master only: pin opensearch{-dashboards}\"","message":"Revert \"CI/Master only: pin opensearch{-dashboards}\"\n\nThis reverts commit 19a004e0a500ea4839fa1e089e1c50c953cec3b3.\n\nReason for revert: Fixed\n\nChange-Id: I5491adef46fe0cec013a597cb05018d8d7e862f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/340350a387cbbe417eff9385156d665c3edd683d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/340350a387cbbe417eff9385156d665c3edd683d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
