)]}'
{"id":"openstack%2Fkolla~915981","triplet_id":"openstack%2Fkolla~master~I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-17 09:16:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8","subject":"Fix OpenSearch file ownership issues","status":"MERGED","created":"2024-04-16 11:11:03.000000000","updated":"2024-04-17 12:14:28.000000000","submitted":"2024-04-17 09:16:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915981","meta_rev_id":"b3760c85ac32a694a0ffc51d8ef9aba144aff5d3","_number":915981,"virtual_id_number":915981,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 09:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-04-17 06:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-17 12:14:28.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-17 07:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-17 07:30:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-04-16 18:34:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-16 11:20:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-16 11:57:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 06:35:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-17 07:30:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-17 12:14:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"77b8248e7bfef6958e4c6e6dae1ef5ff99e94ca6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 11:11:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1489bcd2bdcae2e9c63a858e2a421ddf1a771849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 11:13:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37c97ee2df8ca3d11575a7e1bd11f0fdcf93a431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 11:20:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b216a2a354844e895e31124c61ac8bf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7ccf9b07c504b49a8df13f247f40e88 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e605ab07234b42aaa4ebd9818e0ed85d : SUCCESS in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c4ebe97493ba75f969d4accf55f746c2e4423d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 11:57:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2738c09ad914b2d8e2867bb1f8f4855\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6b88db289f4e86ac738f7dcb28044d : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c143bdb9f2a648a7a19438333dcb3471 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41af1e81c3034c00b9cd3839a7a4cbb0 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c82a0ed3b8564339b2a2db847c304d7e : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc3f390c28634537a2d3531709d4ab8b : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f157a67007d24ed799fc17c6f998dadf : SUCCESS in 4m 28s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/78601a37f0c44e48bc4c4165caa97c9b : SUCCESS in 37m 57s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/aa1a4818b238492dae77369bd8bc48f8 : SUCCESS in 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71db355623073c060c7ab6c63342ce4ada714034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 12:18:56.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b90d12fc13952a86409b0a57c113c4eb58edc644","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 12:22:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/813c7420ede74dc79c03c5729bd98645\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d2a4b8a644c44ed9994303ba9961cd2 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb18a66dd99048b0b93127ce8c7ed1a4 : SUCCESS in 2m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"82fb82141af242ad3df560b99954cedf33479014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 13:02:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4721668424384750be99abe8be300a38\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c95a1f427a4cd59ece064f09bfadaf : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ace42663674244da8e71639e01a61d2e : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b343a63b26db4f389b17a610ac9d72b2 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8bdb5cb7a6464129940cbee0d664cfa5 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aef7ea4ef4ad40d085ea0b312fa95021 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16214a2ae1364e47aac00ff069ee096a : SUCCESS in 4m 08s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d9d5f2aa044747ec8a4731e02b1a41b2 : FAILURE in 27m 27s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/80bfa6a4797548169d7ddbcadc79218f : SUCCESS in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"31a77421a04f388f6ffad21fab2e5ac2f0553b7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 14:35:17.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"15cd823a188ec4ca5c1108c6a4d16dcd36d8082b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 14:39:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17f4a1666f87411984c02da4e3733433\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9385f38d801242b688adb4a1a9ee6241 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc5d71542e1a4b4ca13f0b55b4d76ce8 : SUCCESS in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"57deaffd8a972d387be781e683cf0f8eb069d273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 15:12:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9acaa1dde2d1437da512195a3bb56518\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3ff3f4e5f9f43a5941de3766cc2ee3f : FAILURE in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03c18fa84dfc4c76b551fd71e06a639a : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bcf6b3b0799f43c790c3f4cc4323b16a : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/716771a152824f91a72d9e957c0d801f : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9a52e5e208b4c618cf2bdbc3f5df3f2 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc25e8f5b2a4eeaa1165505997fe6b2 : SUCCESS in 4m 41s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/03d4b393f59d4f3187da16dcd6466476 : FAILURE in 32m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/122a584e2f674d3d813ab2cc16a63014 : FAILURE in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3711260602b30d2ef43051eabc6fca94e4ce264b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 15:18:28.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fe2146926d871a774824e7c8f74f05e669fba9fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 15:22:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30715a2393ee4412a4641244a6360c16\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ada1eaea89a4960a214c738c9349f5e : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c72c41e53cb4708beeb4d53fb00c18a : SUCCESS in 2m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7141f156a8b89789b4a19ca0373716f55f5f6caf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 15:52:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08bc118eef2d4395aa4b392a02c7ffbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94d7061aae34654908570445ada97cd : FAILURE in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7098717dfc1d4c618e38f2b2b5d0aa67 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03b0ffbb772747a4a1e3d158abdff5de : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9073eef0b4ac4ffaadcd2fa4e3711483 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4be60405d35545b1af5fd60ce3b6605d : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153224e1c8af41e0abdab1eced17cc30 : SUCCESS in 5m 08s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7d00c04bad9843078d7ba05d922d45ee : FAILURE in 28m 46s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4416a88ea1414c9daded7ae6fb94f18e : SUCCESS in 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"18ab56e4006e0f7687f8a49a2511cd172fb88345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 16:35:17.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a6c2bd1abb6148ba366f3dbfe56ee7fde9ad55c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:39:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dfb7cdc0f584ba0a22136d529e3bd80\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f5a78553fae4592ae208fb564c56aa7 : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1935af784de348e290d172d747993674 : SUCCESS in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"67d1843f155454aa59d30235a324852849979712","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 16:44:02.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"aaf49d4fd626d831c8944e887a60dee81f0a85e5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 16:44:16.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"2c3a800868142ee765ab295fe7d9c4d54d794d12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 16:49:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b964fdae28d14dd28dbd78656dac0316\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de9bcef18d2a4176b3e853f4c1cee9f6 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad54eaf4a78e4a23aefb20d71bafba88 : SUCCESS in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"efb249d3d48595a45a784049388d52475bf3b28d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 17:35:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/993fd6da127a40a08995d1a608339d8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd8bfc1985a4572b45e4ccc7d0fc294 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0204c7595b994b839c1a0eab1757675c : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c322580eebe4ba6b6f564be96ceab0f : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89dd7b33ed4943f7a4cac401406c2f43 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea21419ab09c40409af073034b6417ca : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e136a18d8546ed9a697122a98937e4 : SUCCESS in 4m 12s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7b6efadfaf8b4dea81c5f8fbc3ab9cff : SUCCESS in 35m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/894158fd043c4940ac0d31117ff3b55a : SUCCESS in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ce11b13311a7d9da1719142f49f1fbda576139fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:02:49.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"6239cf584367b15954a02a0a0f3e8faf646fb6fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:03:13.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"35fb782c86c3625fd3474f14b2c5a38043d47ca5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:03:30.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"75ba8ad1af4dca8b7ef878aef2deb0a957ae80e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:03:47.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"7a7ebadddf67057e66c5e7e70bc0df7041e9b9be","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:04:50.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"7badf3954a492543a069602a50fc3e2d86bc8e0d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:05:03.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"4ac5f65de24c04d1c6e1d9c3442c4d4b26b10ded","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:05:09.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":13},{"id":"5e00ee71c8dfefe0f1a1f5feb15dcdd7685308be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-16 18:34:22.000000000","message":"Patch Set 13: Review-Priority+1","accounts_in_message":[],"_revision_number":13},{"id":"618d47631a99de97744eb9ed662b7d5ad2ff4eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 20:01:14.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f563b00921e4faf90618a7fca5a3be9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c7ceefe244044a879c0e7d1267cf74ab : SUCCESS in 30m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c6c0499b3c23495d97c822fca0e2fa79 : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6a6a1f3838a7415daca602571c7724a9 : SUCCESS in 19m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/40704dd176e14d2a92d82abb32193598 : SUCCESS in 31m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9f6d51ee1d454c0eb6032c5302daf41c : SUCCESS in 1h 00m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f029ea3310714dba8440aaf93d195e06 : SUCCESS in 1h 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc15a5249134eabbdbdd78aa29b23e6 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88dc7af5ec3e49eeb9ade068d4466233 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e96bc3a88ff449c82e382962dcf373b : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73dd5a267dab45b7be88c6839a5e0cf6 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e03d62c6318a46a6a1eca3541a7ac7e9 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a7ab7e228e4022bc30059daf4e0d6c : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fa508e1703524911bdbd96ffa1227e13 : SUCCESS in 25m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5f566f2f2b9541d9ab2710fa51f9dda6 : SUCCESS in 38m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5042de8fd12f48eeac87d703edd56112 : SUCCESS in 1h 05m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/977e82e5404e434aa869106a743e3881 : SUCCESS in 1h 46m 20s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d0c8ae207315440d80712057d3c73af6 : SUCCESS in 37m 49s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/077d2139b4e54135ae67506c5396a335 : SUCCESS in 20m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/58fb181ea2104e7b97612e630d084e7d : SUCCESS in 47m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/58724d8ae79048349a959baa472e5248 : SUCCESS in 47m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/68f52d533cde43f4845e08a441789814 : SUCCESS in 1h 09m 01s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/129bdc9b599d441aaff21fd70dfaf59d : SUCCESS in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"dea0563bbf8a7c4b3cc5422f73bb3de9b744cdaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 20:34:26.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b164b1028445d49e76f0ecd6e23bf1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7d6f74fffd02492089736e13f22859c8 : SUCCESS in 2h 26m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/80f3e6ac83dc452d8da542b45b19bdc1 : SUCCESS in 2h 28m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a6688a7d3774b67b1adbe02d64125c4 : SUCCESS in 2h 00m 24s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98f3e00c50f746eb921423b462cade93 : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e32399fe4bc44197b511ce5f11895165 : SUCCESS in 3m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5366885c2a6445d5b707a608dc7bd86a : SUCCESS in 22m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5a20473652ff4acf93e085c4f382b4e9 : SUCCESS in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4b9b0440a1076ba442c70b2b570e739e1429a004","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-17 06:35:23.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"ef202c2df0869c350a5db846a6c68191faedefb9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-17 07:30:08.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"b44ea2e6e6b46bb5c0110af4df8cba8db6f47cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 07:30:30.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"d396d0f56d29bb7fc4dd9a80da345968c80d0f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 09:16:23.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/392e195e7e214836a9bc3385851612e1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4f9fac03eef041348dec340e7ceb4480 : SUCCESS in 20m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/59b8c6d30ce440f6a530d1f57b2bba71 : SUCCESS in 25m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f19f59deedc94bc49f20c8b8b0cd7450 : SUCCESS in 1h 03m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f1e437ce0e2d4d5aaf0e280b5b3acee0 : SUCCESS in 1h 28m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7444e01ea3164048b97bb68a337baedb : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f361fe786984e7abf42db697cffe0f4 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/176b88bc695d45708ab44490e814c5ed : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516005c6751e4878bf0801bbfd4424b6 : SUCCESS in 4m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a3288e5fb3d491dbdc288d0f15fc0e5 : SUCCESS in 23m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9143c935cd3f423693b50b334c3698f8 : SUCCESS in 33m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5fe17f5767844e90a569482e6cc957fc : SUCCESS in 1h 08m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6096863ecf864300ba1a0f279eaf8cdf : SUCCESS in 1h 36m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ba6f3c51ff43413ebd3062c974fe0e3b : SUCCESS in 26m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e781962510cc4cd3a2d59571e4a9cb61 : SUCCESS in 44m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f5107fb2a6da4293a9d26622ddf2e0c7 : SUCCESS in 51m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/202180f3deec44f684e63690b04883fb : SUCCESS in 1h 23m 45s","accounts_in_message":[],"_revision_number":13},{"id":"d6ad8ae94589c2e5062fe39f7d40e55e2eb885ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 09:16:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"2c3d1c3b78eb3b11257fcde5d3bb7ada820b986d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 09:17:21.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9325d2790273487e91a5099523fe0b2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abea60436bb64512b78417878855c9a6 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":13},{"id":"b3760c85ac32a694a0ffc51d8ef9aba144aff5d3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-17 12:14:28.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"7c098bdd3cac73505851d4c97d7f39e966f00e1b","revisions":{"cb772a36fe3e72eb51ca23d135c00384d33a1f17":{"kind":"REWORK","_number":1,"created":"2024-04-16 11:11:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/1"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:11:01.000000000","tz":120},"subject":"WIP: Debug OpenSearch Dashboards sec plugin removal","message":"WIP: Debug OpenSearch Dashboards sec plugin removal\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb772a36fe3e72eb51ca23d135c00384d33a1f17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb772a36fe3e72eb51ca23d135c00384d33a1f17"}]},"branch":"refs/heads/master"},"8de0767bccdce6ea53b9a327eec071613607eaa1":{"kind":"REWORK","_number":2,"created":"2024-04-16 11:13:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/2"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:13:25.000000000","tz":120},"subject":"WIP: Debug OpenSearch Dashboards sec plugin removal","message":"WIP: Debug OpenSearch Dashboards sec plugin removal\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8de0767bccdce6ea53b9a327eec071613607eaa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8de0767bccdce6ea53b9a327eec071613607eaa1"}]},"branch":"refs/heads/master"},"6b7e4a510f05065519d2ed12ec77aba068d66003":{"kind":"REWORK","_number":3,"created":"2024-04-16 12:18:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/3"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 12:18:56.000000000","tz":0},"subject":"WIP: Debug OpenSearch Dashboards sec plugin removal","message":"WIP: Debug OpenSearch Dashboards sec plugin removal\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b7e4a510f05065519d2ed12ec77aba068d66003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b7e4a510f05065519d2ed12ec77aba068d66003"}]},"branch":"refs/heads/master"},"f1dcb98baf9830f65e15f9985f0a0e537cad2845":{"kind":"REWORK","_number":4,"created":"2024-04-16 14:35:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/4"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 14:35:17.000000000","tz":0},"subject":"WIP: Debug OpenSearch Dashboards sec plugin removal","message":"WIP: Debug OpenSearch Dashboards sec plugin removal\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1dcb98baf9830f65e15f9985f0a0e537cad2845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1dcb98baf9830f65e15f9985f0a0e537cad2845"}]},"branch":"refs/heads/master"},"95f9453d916ee677f601c6c0391024e0dab7d950":{"kind":"REWORK","_number":5,"created":"2024-04-16 15:18:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/5"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 15:18:28.000000000","tz":0},"subject":"WIP: Debug OpenSearch Dashboards sec plugin removal","message":"WIP: Debug OpenSearch Dashboards sec plugin removal\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95f9453d916ee677f601c6c0391024e0dab7d950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95f9453d916ee677f601c6c0391024e0dab7d950"}]},"branch":"refs/heads/master"},"8bbca8ac69e5fb3be22b9fa7f79e006594082c1f":{"kind":"REWORK","_number":6,"created":"2024-04-16 16:35:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/6"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 16:35:17.000000000","tz":0},"subject":"WIP: Debug OpenSearch Dashboards sec plugin removal","message":"WIP: Debug OpenSearch Dashboards sec plugin removal\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bbca8ac69e5fb3be22b9fa7f79e006594082c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bbca8ac69e5fb3be22b9fa7f79e006594082c1f"}]},"branch":"refs/heads/master"},"aa20ae92e2c887cb343f6a58202de265b20fc77b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-04-16 16:44:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/7"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 16:44:02.000000000","tz":0},"subject":"Fix OpenSearch file ownership issues","message":"Fix OpenSearch file ownership issues\n\nSince 2.13 OpenSearch and OpenSearch-Dashboards packages\ninstall files in /usr/share/opensearch and\n/usr/share/opensearch-dashboards with root:root ownership\ninstead of using opensearch/opensearch-dashboards user.\n\nCloses-Bug: #2061877\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa20ae92e2c887cb343f6a58202de265b20fc77b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa20ae92e2c887cb343f6a58202de265b20fc77b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c0712de9abd6e9ffc0f5ee31380e9bc9c81b998d":{"kind":"REWORK","_number":8,"created":"2024-04-16 18:02:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/8"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:02:46.000000000","tz":0},"subject":"Fix OpenSearch file ownership issues","message":"Fix OpenSearch file ownership issues\n\nSince 2.13 OpenSearch and OpenSearch-Dashboards packages\ninstall files in /usr/share/opensearch and\n/usr/share/opensearch-dashboards with root:root ownership\ninstead of using opensearch/opensearch-dashboards user.\n\nCloses-Bug: #2061877\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0712de9abd6e9ffc0f5ee31380e9bc9c81b998d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0712de9abd6e9ffc0f5ee31380e9bc9c81b998d"}]},"branch":"refs/heads/master"},"0b47ee7555996f6b08114b1eba11ee434f2d4c7f":{"kind":"REWORK","_number":9,"created":"2024-04-16 18:03:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/9"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:03:13.000000000","tz":0},"subject":"Fix OpenSearch file ownership issues","message":"Fix OpenSearch file ownership issues\n\nSince 2.13 OpenSearch and OpenSearch-Dashboards packages\ninstall files in /usr/share/opensearch and\n/usr/share/opensearch-dashboards with root:root ownership\ninstead of using opensearch/opensearch-dashboards user.\n\nCloses-Bug: #2061877\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b47ee7555996f6b08114b1eba11ee434f2d4c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b47ee7555996f6b08114b1eba11ee434f2d4c7f"}]},"branch":"refs/heads/master"},"848778c7b8e75d6c3e1c4f0ee06215af4cb619db":{"kind":"REWORK","_number":10,"created":"2024-04-16 18:03:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/10"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:03:30.000000000","tz":0},"subject":"Fix OpenSearch file ownership issues","message":"Fix OpenSearch file ownership issues\n\nSince 2.13 OpenSearch and OpenSearch-Dashboards packages\ninstall files in /usr/share/opensearch and\n/usr/share/opensearch-dashboards with root:root ownership\ninstead of using opensearch/opensearch-dashboards user.\n\nCloses-Bug: #2061877\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/848778c7b8e75d6c3e1c4f0ee06215af4cb619db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/848778c7b8e75d6c3e1c4f0ee06215af4cb619db"}]},"branch":"refs/heads/master"},"b1395e9640872ec4ec4d53c2ffd6ae831aae59ec":{"kind":"REWORK","_number":11,"created":"2024-04-16 18:03:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/11"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:03:47.000000000","tz":0},"subject":"Fix OpenSearch file ownership issues","message":"Fix OpenSearch file ownership issues\n\nSince 2.13 OpenSearch and OpenSearch-Dashboards packages\ninstall files in /usr/share/opensearch and\n/usr/share/opensearch-dashboards with root:root ownership\ninstead of using opensearch/opensearch-dashboards user.\n\nCloses-Bug: #2061877\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1395e9640872ec4ec4d53c2ffd6ae831aae59ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1395e9640872ec4ec4d53c2ffd6ae831aae59ec"}]},"branch":"refs/heads/master"},"2e98534bedc69025629f64ec36fc6e8b55b97ea4":{"kind":"REWORK","_number":12,"created":"2024-04-16 18:04:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/12"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:04:50.000000000","tz":0},"subject":"Fix OpenSearch file ownership issues","message":"Fix OpenSearch file ownership issues\n\nSince 2.13 OpenSearch and OpenSearch-Dashboards packages\ninstall files in /usr/share/opensearch and\n/usr/share/opensearch-dashboards with root:root ownership\ninstead of using opensearch/opensearch-dashboards user.\n\nCloses-Bug: #2061877\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e98534bedc69025629f64ec36fc6e8b55b97ea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e98534bedc69025629f64ec36fc6e8b55b97ea4"}]},"branch":"refs/heads/master"},"7c098bdd3cac73505851d4c97d7f39e966f00e1b":{"kind":"REWORK","_number":13,"created":"2024-04-16 18:05:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/915981/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/915981/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/915981/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/915981/13"}}},"commit":{"parents":[{"commit":"7e95ffb09815d1918b3b262e074d325b4fb35d70","subject":"gnocchi: Bump to 4.6.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e95ffb09815d1918b3b262e074d325b4fb35d70"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 18:05:03.000000000","tz":0},"subject":"Fix OpenSearch file ownership issues","message":"Fix OpenSearch file ownership issues\n\nSince 2.13 OpenSearch and OpenSearch-Dashboards packages\ninstall files in /usr/share/opensearch and\n/usr/share/opensearch-dashboards with root:root ownership\ninstead of using opensearch/opensearch-dashboards user.\n\nCloses-Bug: #2061877\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c098bdd3cac73505851d4c97d7f39e966f00e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c098bdd3cac73505851d4c97d7f39e966f00e1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
