)]}'
{"id":"openstack%2Fkolla~916081","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-17 13:37:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8","subject":"Fix OpenSearch file ownership issues","status":"MERGED","created":"2024-04-17 09:23:56.000000000","updated":"2024-04-17 13:38:45.000000000","submitted":"2024-04-17 13:37:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916081","cherry_pick_of_change":915981,"cherry_pick_of_patch_set":13,"meta_rev_id":"762cdfde0c475de07b172e1e1628994f5d150159","_number":916081,"virtual_id_number":916081,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 13:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-04-17 11:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-17 12:13:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-17 12:28:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-04-17 11:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-04-17 11:52:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 11:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-17 11:52:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-17 12:13:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-17 12:28:15.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"60476dcb85eed6b508d95c6cde13038eff7fd3a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 09:23:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a4e4c9a4433c942c1f24dfd18787bb1d98961f0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 11:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89b490738f374a3ba39e2f471d1ed27a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/87fc1736a8b2444cbeceac2c03a15d46 : SUCCESS in 24m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/11782b0414c649748db419042c4f2cce : SUCCESS in 1h 06m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ed0167073ed042508e8e0acfb66038b0 : SUCCESS in 22m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6ca31ec44a4d4e83a6bd3179335a9c96 : SUCCESS in 32m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dc26089dd23749649964dbdc3b67f3c5 : SUCCESS in 1h 03m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6c374abbc8784511a1f1b348a710fc5c : SUCCESS in 1h 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b51a81c0ee3b4743b8195985f5221ded : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/097dcafc1359497e8fb31de22792eed8 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7682c5648f7421a86084c3361624555 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/821151fd1a914712b9735440178d60e7 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e4f4ec5847841de86735e44aa2cc58f : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12a6263e72f149c18508602f6c748c67 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30072c0282fe4894a43369d8a5be1dc6 : SUCCESS in 4m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/80a8573f45c54a348c044f559188b56e : SUCCESS in 25m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d0ca95ab174f44b89db74490e63e61d1 : SUCCESS in 31m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f842b6001d5f491b96e6687823f2810e : SUCCESS in 1h 06m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a6795ed8535f4f43bc6669e226296a8e : SUCCESS in 1h 19m 54s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/42b84ba0b4a64940913e707b7ce158e7 : SUCCESS in 35m 10s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/df698350e7154aa0ac5f25172f0754f9 : SUCCESS in 18m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/93ce6f8f767f434daac622151f942a43 : SUCCESS in 59m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/59798a906b804752be41d50648b4e70c : SUCCESS in 51m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9ab1a30786884d5bb8f732c36d280ba9 : SUCCESS in 1h 28m 29s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0ee780c96cf34775bff9cdb5b21a4013 : SUCCESS in 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93b42018b293e28f767a77e7aa78684c66a49089","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-17 11:51:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"58885a789957fffe5f43ef89f0997d9094422755","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-17 11:52:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13191e0d5b375ec88d50d6061a552c5848b35667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 11:53:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0300f2109edba96c24927a79ab74a09e7d45a4c7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-17 12:13:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0fe31235331d1104f3f1090fdc3c0ada68f10e0f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-17 12:28:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf712db51d6fe0da8f4532ad35755f00f84f9ece","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 12:30:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e62bcb4b6ddd4f4a9a2a157656777b2e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/baf4f676c18741e1952bd96b7e4ddf5b : SUCCESS in 33m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c72ef25433404bb9a57dbd60e32787a4 : SUCCESS in 30m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/008ea9a26b884968bc708af517b7a3b3 : SUCCESS in 1h 05m 11s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d92265838cd042cebcee98c51f4487a3 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb8ddace0fc24817984082e69d7258b0 : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e088c294af9648d0b4177f74d74938ea : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/360c91cad25c4bd8b44794ef49f5344c : SUCCESS in 3m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fc938dc9b7294fb29467bfe899079a3c : TIMED_OUT in 3h 04m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c3dfcdb2b0334b38b7c7454da71ed886 : SUCCESS in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfcddf866aafd5b6d6a162de3ce45d76982b40a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 13:37:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b932f5293143472c96ae4cd82c9b35e5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b4fcc0886a5e40519bfb429cf37ab506 : SUCCESS in 23m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a1307bf7a8ac41f8af7b1a31cfa47faa : SUCCESS in 28m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/696ddd763649442a9818b4b925e8c835 : SUCCESS in 1h 03m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/66669b7470d340b69646bec825d4804c : SUCCESS in 1h 38m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c0059c93b646a3807d006ec803bffc : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12ecaadff8b64bb2ba16079d7db4f058 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03e4861f9863410ba0ff34ab6613949c : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09048cacc0f84623b80b45181bd129b5 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cbcc927094243ccb7a7bd99640237a7 : SUCCESS in 5m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/391ba5c9305248b0b8b969b533006314 : SUCCESS in 21m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/36d9edb850d94ff2a6d10abad645e4dc : SUCCESS in 31m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9047aaa037e44820ab78b7a876af68d0 : SUCCESS in 1h 13m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/28650c1236924c358fdc30647f2335c6 : SUCCESS in 1h 39m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6d9e5cf29b404f859c539bafbd46cd2a : SUCCESS in 22m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/887d7c10909842d280d10a53c5046d05 : SUCCESS in 59m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/81dbe9521cb44db28f8c5f572cf84958 : SUCCESS in 48m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b9b16eb157aa4c1284d363293780cb0c : SUCCESS in 1h 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a1f6c6acd8e8ce296a0f16d1f7450298889d4fd6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 13:37:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"762cdfde0c475de07b172e1e1628994f5d150159","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 13:38:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c4e1eba65924f2394af696631fa95c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c970bb65a8042d382f72460e0e38748 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cdfa5c00e205635e28f401b8f8c88e560dd133a","revisions":{"8cdfa5c00e205635e28f401b8f8c88e560dd133a":{"kind":"REWORK","_number":1,"created":"2024-04-17 09:23:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/916081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/916081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/916081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/916081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/916081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/916081/1"}}},"commit":{"parents":[{"commit":"66c7dec630d01450d31fe13119fa7df7dc049d76","subject":"Merge \"toolbox: Bump ansible-core to 2.15\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66c7dec630d01450d31fe13119fa7df7dc049d76"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-16 11:10:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-17 09:23:56.000000000","tz":0},"subject":"Fix OpenSearch file ownership issues","message":"Fix OpenSearch file ownership issues\n\nSince 2.13 OpenSearch and OpenSearch-Dashboards packages\ninstall files in /usr/share/opensearch and\n/usr/share/opensearch-dashboards with root:root ownership\ninstead of using opensearch/opensearch-dashboards user.\n\nCloses-Bug: #2061877\n\nChange-Id: I3831d8ab76c8c7136ce7869db3d0d1e05b4ba8f8\n(cherry picked from commit 7c098bdd3cac73505851d4c97d7f39e966f00e1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cdfa5c00e205635e28f401b8f8c88e560dd133a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cdfa5c00e205635e28f401b8f8c88e560dd133a"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
