)]}'
{"id":"openstack%2Fkolla~916489","triplet_id":"openstack%2Fkolla~master~Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b","project":"openstack/kolla","branch":"master","topic":"fix_2062072","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-24 13:47:57.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-04-24 11:10:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-04-24 13:47:57.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-04-24 13:47:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b","subject":"swift-base: fix rsync missing for rpm installs","status":"MERGED","created":"2024-04-19 15:05:24.000000000","updated":"2024-04-24 13:49:21.000000000","submitted":"2024-04-24 13:47:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916489-fix_2062072","meta_rev_id":"716441ff16a398e75c23235146a830e3a0fe690c","_number":916489,"virtual_id_number":916489,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-24 13:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-04-24 09:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-24 11:10:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-04-24 09:41:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 16:36:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2024-04-19 17:27:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-22 10:11:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-04-23 10:33:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-04-24 09:41:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"75906af617ace2368e071015e7a9cdb39a6fbc80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-19 15:05:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4834149c2beab81603890ec4b7c77ce66986d801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-19 15:32:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7377e418908b9c8aa1e8d7bfc63a8fb3231f01c5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-19 16:36:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e3baa617b4f429af0cac62aee7f590e7a40bbe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:27:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3554ac9c45514d60ac128918f27740a9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a75e927cde574aa4bfe58b9ee626728a : SUCCESS in 28m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/15b080b6884447e780794d39aebd4cc6 : SUCCESS in 56m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7962f8b8d0114b18b0102a6d07d3deba : SUCCESS in 18m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1c1026d8521d44fcaa7216e1f3600460 : SUCCESS in 38m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ce8562022c784591ac2262c7084e5ae5 : SUCCESS in 1h 04m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/eb675b109c8c49ca95bf672fe6c8ea29 : SUCCESS in 1h 32m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1691ea2d5f44174aa743cde4f036fbf : FAILURE in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4248676fda9b48308da66c52e4b0ff8a : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e21fd48d8d541e4a5bb7598af9284e4 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a18354cd40df4cceb3364aa2a1584716 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a8d7263e6ea4ca1bafd48cb0c69ac9c : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2859fb459ba4da68657625b80290b0d : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a25a5fade8ea40fab020651119f69b7a : SUCCESS in 5m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6c8d5c2669a94c02b90991c8da67f2ed : SUCCESS in 25m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0b61f7083b1d4dddac940958f93af37e : SUCCESS in 36m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b53b624725814fcbabe9c536727ec8fe : SUCCESS in 1h 12m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/af534adca2ba4ad182bbbae25ed2815e : SUCCESS in 1h 37m 09s\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c583e256518b4d589a675dd356bdf437 : SUCCESS in 1h 46m 01s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e5d93ec594a24575897a61d79e35066e : SUCCESS in 18m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e93d61c938e9429399523d599c5974ce : SUCCESS in 1h 00m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9f30c8e84c504615b4962411cc27f217 : SUCCESS in 48m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ed1147ccbe194194aaa733b55c0cee23 : SUCCESS in 1h 24m 55s\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2baac3527a1d46c68a1f8667b395a8e7 : SUCCESS in 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1181d5b9bcb56cd9cedcf3c14e1a462d9a65f5fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:39:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dbd500614b14911864652b28eb27ea3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/97282886c2b5496ebce7696fd1527642 : SUCCESS in 24m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e99bf4492b2a43849e670ffc9f0d4558 : SUCCESS in 22m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8d39e69b0c94501a824b0f63a933e13 : SUCCESS in 1h 42m 27s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15c158be65ef4ff0b65580535bf0c575 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/19d90031d90946a1befa7cfbdf01232f : SUCCESS in 4m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/61030b4068c343afaf0f7277140e9c46 : SUCCESS in 22m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/28046e2ea384425386488d65deaf8e0b : SUCCESS in 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"561e7d3efcea20f5a9410179c9c34942dd54ca89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-22 07:34:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"81c26d2bda7975f5fc08423a73e3a72ff116096c","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-22 08:33:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5ed29c42d836cb7749f5383f2250f61158da6d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 09:26:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98ed78aab0cc44fb8ddac29828788e74\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a1028f7d8caf4d4db77374fbe82694b9 : SUCCESS in 23m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c07f724cd5984878a66d0e9b7bc250d6 : SUCCESS in 1h 05m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7beebcbc0fad42be9af0f953dab1bfe9 : SUCCESS in 18m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3059d31d345943b1b8f6a2d316c58cf5 : SUCCESS in 22m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bd736d01c4474319b2ebd50eaa278dd8 : SUCCESS in 58m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f8f8c9efaad5469192e997fa0d1e5c2a : SUCCESS in 1h 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a3f58f22cf34f4590952e0c6c6bdaba : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50e41c7dbb2e456093babbd65f7719a3 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/456706171d984f668e1f9ca43640a223 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/891815f5799444759f9db21652325022 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e11bcbefcc924ae086a7ef0962a0a0e7 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82fdf8a75a0d4ae28e53b8dc2d7af10f : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5fe3689a30b4fb5909abf07fe0b5c06 : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e3431c81aaec432d87d635172a4d6ea0 : SUCCESS in 25m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3dece6111ab14eb6aeab2b72b271a0ee : SUCCESS in 32m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ca06decdae6d4df3811b2107b59342a0 : SUCCESS in 1h 11m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5089b7873c054ca8892e11c383404d4f : SUCCESS in 1h 44m 15s\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/fa4d97465249406b89cc7a3546060f20 : SUCCESS in 1h 23m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a70b54e9faab4fcb96fdc0d306fd5d16 : SUCCESS in 21m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/59efdc573c8e4dafa2bc5ed55b2418d1 : SUCCESS in 40m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a05e5b4c76894f7a992d66f7e81c3878 : SUCCESS in 42m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0b2657d20ae444ba8381e771a63e67d3 : SUCCESS in 1h 05m 17s\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/db8228b7be1846f9a3a91545b9b96d77 : SUCCESS in 53m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"39c43564354afa6592738003264a44666a8b24f8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-22 10:11:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1ab8cdc006e8f3fb1cb00ed7420612721dd329e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 10:42:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10cf2883e25e44c69c12ab8af9632fcc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cfb63afb49da41a5a8e69954efe7d5e0 : SUCCESS in 25m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1c52ebd7a2f14c47a699d7d25b56f3e2 : SUCCESS in 19m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/583a4145749e46798664efef4c567a8c : SUCCESS in 1h 45m 04s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4abc40b01e1e452aa3da18604b55f180 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/907a89df0cbb41d898ddfe1a649c9672 : SUCCESS in 5m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5cb8713f3a584220af0fc6c8db17edab : SUCCESS in 23m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/db4a289b3d3e4d3bb4543442f100f2bf : TIMED_OUT in 3h 06m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ea232355eef560b7bad80d713b84cbe0623cc67","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-04-23 10:33:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e89bcba94d81760cd53a52cc873eac3e160be13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-24 09:41:39.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"fdf35ee27a128cc7af80f1cad098c75e1ed52dca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-24 09:41:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e8f9ad87f890b6f480ab29ca89e58b36175efdeb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-24 11:10:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8d138531248ca77e6b358281b53424392d71edbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 11:50:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c29d9ecea1d6471aa47f0bae7033b2c5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ac58f5a7a1d14099a298ed4d748fcbb8 : SUCCESS in 23m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e0361f4e80f547f0b9af882411168c75 : SUCCESS in 1h 07m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d97b4f4968f949b6b7b93bf853cd794e : SUCCESS in 20m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/045671c35e78430fb526d0ed7b290194 : SUCCESS in 30m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9393d1f3a2ca4ae8a7553d4e823a9b49 : SUCCESS in 1h 07m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bd759a53361649b588747a3aef49c429 : SUCCESS in 1h 43m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e7954e38a3c45ec82d1a7db7985ac17 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2984656b51b440fe9093fe53f2f3aa7e : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63e63c13dac04e47a7239b96c97c5498 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c097b86144c34460aa1af41bf0d38441 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8965b246e24c4cf98e59b5c7da0f1e37 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae9eaa48e124fd08558e9c51df4a828 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd38c2d776a54e95beb187270259ac08 : SUCCESS in 3m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/10215a3fac9541f1a7c1ea0281fb2b21 : SUCCESS in 23m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/79e8643337ff4ad1908620701030f7a8 : SUCCESS in 29m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7a7d10afca0f4b678c7e7ec106c59b2d : SUCCESS in 1h 03m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/128db5700136419892842ba334d8cd46 : SUCCESS in 1h 26m 03s\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/0ada4483a1f142ed982d1be5ad59a14e : SUCCESS in 1h 31m 52s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a1740e0834e74199b50e69cced4516e4 : SUCCESS in 18m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9b88156d135b40c0b3bc494b1fbc5b5c : SUCCESS in 58m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/06327458db654c9b9e32b99e028ca12c : SUCCESS in 57m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ac6930e683f54feb9f83db838f2cceeb : SUCCESS in 1h 21m 46s\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/092e699b0e4149acb37b69d354e7a12e : SUCCESS in 1h 08m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a6a09776958b7bb7b60d4ec4b758257a6b763363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 11:50:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d93534856e5fa5ada6483f0b626b60200ed998d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 11:57:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f014900c864906918518559d1bc37d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c2e3e8c62f9c4401a5f61180400e6cd7 : SUCCESS in 25m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7fe6b0ae68484ef782dcf51b95d4d4fb : SUCCESS in 2h 14m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c1c0dbec9786477ba8e758b34af6a2a8 : SUCCESS in 1h 18m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e983d03796a144fca2f198b85024396b : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/676edca118204858afd4678c05a9132b : SUCCESS in 2m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/009c71d51823480b88691be9137e2157 : SUCCESS in 2h 11m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dfd74d524d974a118c037bf22cd5bc83 : SUCCESS in 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d6d95cfe832c7c5350e422ecd68f97b550d6339e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:47:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b066931fabc045f984f2c3edcdb33567\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/976ba93d3b774c1c98f27a54da6f718f : SUCCESS in 15m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e0727b6f94fb4506aae732e4f1614aa9 : SUCCESS in 24m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b0d09dbe630c4b1683c133c8bcd3a165 : SUCCESS in 53m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c128b9c6a3c641979abbcd6f32af99ab : SUCCESS in 1h 27m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e289c3275bc44ba0a3bb770912339fd5 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d10e55bd5e75448182319eef0f0fe094 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d78e5dd52ec64550aade206ee51198d4 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23124a3faa74617b0432578c50fc0fe : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7a49dde98bd4b18ac30e2b3eedf7b52 : SUCCESS in 3m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4ec0b220f7c343b49930c0c56510eb1f : SUCCESS in 22m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/addd9217c883468097d4f41b29a2c84b : SUCCESS in 33m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e3dc3a93010c460ba9467b8865c128ce : SUCCESS in 1h 03m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fefe99e255c1417b910c9a92091c3615 : SUCCESS in 1h 45m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/920d6f2d364d4a2e8fdff2909d4d9cac : SUCCESS in 27m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4dbf6b8614e0411ab2caa7d5634c0c8f : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/caa920834d5a426c96e0fa1556935186 : SUCCESS in 49m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3861ce0bd7b34bacad8d1e22d86c19b8 : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":4},{"id":"09bb676b4a6ed312de30519c705d7ccd5a189edf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:47:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"716441ff16a398e75c23235146a830e3a0fe690c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:49:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2a3eb86e1ed4638b1fce88e9e88e8f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe2d265376b4c24898cef173691e298 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/755c7371e56f4d5f80da42d1d5f8be35 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7766abb4398d13355a3f76bc1b91f6c92d36111b","revisions":{"d249dbc82055fcd6cd6b21b3524281ed44a0ab17":{"kind":"REWORK","_number":1,"created":"2024-04-19 15:05:24.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/89/916489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/916489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/916489/1"}}},"commit":{"parents":[{"commit":"52183ab561a4098a016292a18d5a9a3243549141","subject":"Merge \"Revert \"CI/Master only: pin opensearch{-dashboards}\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52183ab561a4098a016292a18d5a9a3243549141"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-19 15:04:22.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-19 15:05:23.000000000","tz":120},"subject":"swift-base: fix rsync missing for rpm installs","message":"swift-base: fix rsync missing for rpm installs\n\nCloses-Bug: 2062072\n\nChange-Id: Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d249dbc82055fcd6cd6b21b3524281ed44a0ab17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d249dbc82055fcd6cd6b21b3524281ed44a0ab17"}]},"branch":"refs/heads/master"},"c5591dbdfbd13c3ee6f997dfc0a3576d2119ca0f":{"kind":"REWORK","_number":2,"created":"2024-04-19 15:32:47.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/89/916489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/916489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/916489/2"}}},"commit":{"parents":[{"commit":"52183ab561a4098a016292a18d5a9a3243549141","subject":"Merge \"Revert \"CI/Master only: pin opensearch{-dashboards}\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52183ab561a4098a016292a18d5a9a3243549141"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-19 15:04:22.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-19 15:31:42.000000000","tz":120},"subject":"swift-base: fix rsync missing for rpm installs","message":"swift-base: fix rsync missing for rpm installs\n\nCloses-Bug: 2062072\n\nChange-Id: Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5591dbdfbd13c3ee6f997dfc0a3576d2119ca0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5591dbdfbd13c3ee6f997dfc0a3576d2119ca0f"}]},"branch":"refs/heads/master"},"621c790a09babd0fb667f6030c01dcc38f6c102d":{"kind":"REWORK","_number":3,"created":"2024-04-22 07:34:00.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/89/916489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/916489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/916489/3"}}},"commit":{"parents":[{"commit":"52183ab561a4098a016292a18d5a9a3243549141","subject":"Merge \"Revert \"CI/Master only: pin opensearch{-dashboards}\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52183ab561a4098a016292a18d5a9a3243549141"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-19 15:04:22.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-22 07:33:49.000000000","tz":120},"subject":"swift-base: fix rsync missing for rpm installs","message":"swift-base: fix rsync missing for rpm installs\n\nCloses-Bug: 2062072\n\nChange-Id: Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/621c790a09babd0fb667f6030c01dcc38f6c102d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/621c790a09babd0fb667f6030c01dcc38f6c102d"}]},"branch":"refs/heads/master"},"7766abb4398d13355a3f76bc1b91f6c92d36111b":{"kind":"REWORK","_number":4,"created":"2024-04-24 09:41:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/916489/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/916489/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/916489/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/916489/4"}}},"commit":{"parents":[{"commit":"52183ab561a4098a016292a18d5a9a3243549141","subject":"Merge \"Revert \"CI/Master only: pin opensearch{-dashboards}\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52183ab561a4098a016292a18d5a9a3243549141"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-19 15:04:22.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-24 09:41:38.000000000","tz":0},"subject":"swift-base: fix rsync missing for rpm installs","message":"swift-base: fix rsync missing for rpm installs\n\nCloses-Bug: 2062072\n\nChange-Id: Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7766abb4398d13355a3f76bc1b91f6c92d36111b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7766abb4398d13355a3f76bc1b91f6c92d36111b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
