)]}'
{"id":"openstack%2Fkolla~916642","triplet_id":"openstack%2Fkolla~master~I799d6bf7a381c377dee565c13b1732f7313520f4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-11-08 12:04:11.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-08 13:48:45.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-11-08 13:48:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-11-08 13:24:46.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"I799d6bf7a381c377dee565c13b1732f7313520f4","subject":"CI: Add reno linting","status":"MERGED","created":"2024-04-22 14:35:20.000000000","updated":"2024-11-08 17:12:29.000000000","submitted":"2024-11-08 13:53:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916642","meta_rev_id":"e30923d8717761e20248c0ec3be1234bb2f9ad23","_number":916642,"virtual_id_number":916642,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-08 13:53:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-08 12:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-11-08 13:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-11-08 13:48:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-11-08 13:24:46.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-11-08 13:48:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-11-08 12:04:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2024-11-08 13:48:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"date":"2024-11-08 13:24:46.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-11-08 13:24:46.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-22 14:37:05.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-22 14:43:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-22 14:50:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-07 19:40:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-11-08 10:47:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5a9c6b6ddc0ab6b7b35375e182e97c31d4b1159c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-22 14:35:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5441517ef0f5500fe1b191b6f9d5e1452eb7c72","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-22 14:37:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"81777088af54cc17c7c0c55e44ca9e31f7480898","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 14:43:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21621e8bb2b34119837a5c4193291b8e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4dafaf3684944a79874db80685684bdd : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e751f9b903f9404fbae4a993a735bcae : SUCCESS in 7m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b7b3091a633c4efe88af3aa8b8fc6e3dfdb0142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 14:50:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51adc54c07bd4c68b35b217724884f4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22e4713f40c40d7a0d44ccaac7a0887 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab86d336cc2c482983c149902bb0dc06 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a55e0be255941c7863b9ac161d587f5 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8091f313a39493e9040c84036aceb15 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbe0778d729d4eda947bd182b8fddc9c : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d8e74e3ce04d04bcf0c99dc9ae5fdd : FAILURE in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8020cef09de4a67a35095786f970226 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4a4d530400325562aa48fd84f2ced4f15dfac8c0","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-22 14:58:48.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"d6cb2fb5acb80bbee6d68881c5243acc9911c4d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-26 07:18:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b96defe10184f1d84769e1ab9a217e3add461cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-26 07:22:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db6c0c4c75cc4526a55e3b3359b70519\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd3a0f3bc35344bfaea2e712b4ed5229 : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/206d3df2522e48e6a583a3f66ec4b26c : SUCCESS in 2m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7cc61cf13596fa6aecfdb5df29d3eac54224ff99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-26 07:50:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18d6a956c6614129824b3a6dca9e9d5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5674333590f54883b0a7e0692c0394bf : FAILURE in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e86d89542a14ed0aa5f8a16abb304b4 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/175e99a9ff534ba7b297f366d87aad27 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fcc3dc21c8844da481e95acce00f4460 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8afdc358c9f4f2893092133b95d195d : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0358fe3c176346768264250fec3c0964 : FAILURE in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4585f7f22de2401f81e40c22faf19a28 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"cc58358b5cc0f874d4b3dc6410ddc57ff329cc47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 20:21:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c0b142020550bf5aeabeeb61647c1ce9254b18a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 20:37:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97612dd9dbec4c6cb6d2431a8b2e2347\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edbbfcaa0ed24981b4439345193668ca : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e776170803224ac884d67ae3101a5378 : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8329d4566e7944319310f8506a6993c4 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a750312da8d4718bf8dd6f7a4563def : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/890413e3bd2541208bb2acd000e65aa5 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4b4b9e6ecfd4699b7f025f5bb35d190 : FAILURE in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82ad6785832041d49695e3d5f9f5f0f6 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"80614731a467b9733430f9799c4568fe7524fcbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 20:57:56.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4ffe5a4969d7bb6dac6ef71edd7544d8a566c65e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:07:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51859a5138ba4246b69d45467084cee6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93ef7e79428e49bba86a88994c388fef : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab3bb74b86164d67b8a1a1e3fe70b60c : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5d45b82bb8a4099b4c0bc8571f400af : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ae4183da8ec49c5844d8eea3d4cbd07 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4eadf067d0a4672ac39db63a23504c2 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f3a589394c415883b8e6030a883b84 : FAILURE in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ab7e6ffaee340f28d692cf46928b380 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":4},{"id":"2a77c8437f013b5dae347bd05a595e3a0cc2aefc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 01:08:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a2bd1cd350a45ed868028c6d2a9e9e9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac67e768c47f4fcc9f7fd5ad7aca0486 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01136c64d7b944d1bcaa79966e8fe605 : SUCCESS in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"44f89c108eca702e1bd7657b6dded2b6cd5215be","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-28 09:09:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"57c7a10a2766ba28fcbf33acf159431405f4b80f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-30 14:22:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"12756d25dcdce95d3ce849095a26f030b5b5409a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-30 14:25:00.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71b184e3df98b44114c12a31d82b7115e8ad03e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:29:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f0d2e499f446208aa82e0625c6e012\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/740b8f4222a04edca0aaa83763d99765 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a941368d698a4ce4b38171b01774915d : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8412aca2006ed83f1a2164fc514f9873e6146a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:36:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175ad4ea1fcf4db8b3e57c5e90a615d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0cebdf4d2c47619f72cac51a977e9a : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e5683f5516648598490cd352ad6608a : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6140f0685f44a499ffabc224494fd88 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bf94db5b57a4daebb35ce68117a5384 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3c8c4084f9642fa8ef8bc3ea9e1480b : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8243f348241d45a384ca5065f2778ee1 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f24d3c0c5f1433293bcfab5ec1dff7e : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":5},{"id":"5a360a9fa6ec349cdadddb0fd11fc9c54a9b0991","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-07 16:46:12.000000000","message":"Patch Set 5: -Code-Review Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6f4bde490afa7d2caca302a98c6b13d56e88b9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 17:45:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68b7f9f32a1d43109c20fa4b3350de60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2711c574c20240f687d2b4dd18f0a968 : SUCCESS in 2m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eb7e7d886da4ea09dcfa2014fb99ada : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a2b89fb5ed8425793af928463f1b7e4 : POST_FAILURE in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f651203659d42e3b3e7f1d7e60c5de7 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0681988309344038603ab38bfe6578a : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b7e6ab3946406a90cc2a5816d12f76 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":5},{"id":"b9009c2821a465d2d1cb4cfe4079649eb2e15609","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 23:07:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a80c7475384e4f0c9d1cbc63061ea70f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0cb8d6be9b844c2bcb478b1fd66a6e5 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/177a326f47cd480184dad5444d98d0ab : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"53221e76c5940293baadb8c65432bb06b2a09ff4","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-08 09:06:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9be468c46c1abc435e29c0d31072bea6df301b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 10:00:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f27fbf118da4e88bc173c1c0bc1b153\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4c03b7d53504af2953db8b0eb87a386 : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2f96dcb987b4fa7b3d7108416b72634 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5f020f304fe434092ba44be11b6e825 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2059afacaa8f4c6caae107cd27b7edf2 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d769f9b99f34f40a5dfe6f71066a281 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbb0c279d3e14f9985a572e535309565 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":5},{"id":"be96a4a0ed11bc9e0da0afd94faf216d7497f379","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-08 10:33:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7406fb517de87af894025a762e3b351ebc7468b7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-08 10:47:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1fba635cec8fd7df93fe434e8673cd9c35568319","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 11:24:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f71862aedbf4608a0bffb3b06c4f281\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26165fed2ffc423eb03995beaef702c8 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6453f467c85a4e8999e8e5dc1f6ad3ee : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8fea529a2280de976e94ff78a8ab52c44069c108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-08 11:28:22.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"dd56e74263d4203ddc4f205022f683099b95d72a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 11:32:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b0f1e24e7b4d27890e2c2b42073b9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a3f2cfbc92f4db6b97b507792039200 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47e2ec6f5c754fdbabcd5ae6332f37c9 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d6daa94ec1c40c295e1c9511c360260 : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042ae2cbe7e04b1a85bab8476bd386cb : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4100e1767664b87b3b159c9d9f20a0a : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a00290a7d1e94d98a5bbcdcb906f1002 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":6},{"id":"95a3a710b91cff14b9af1a1bdd92eb182ec3a818","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-08 12:04:11.000000000","message":"Patch Set 6: Backport-Candidate-1 Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7ab3e4a239d9241dc8a0deca52b12140cf614c0b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-08 13:24:46.000000000","message":"Patch Set 6: -Backport-Candidate Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"c73cab3c905364c631a6044093fda6a5a57f53f7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-08 13:48:13.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b7878fe9e81a778b7ff450ae56d64a8ea380958a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-08 13:48:45.000000000","message":"Patch Set 6: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9d473bf4f46ea287e9d67e4be8c5acb4da16bbc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 13:49:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d75b8583575b710fbc90e26e2a6b8d13a1beb857","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 13:53:16.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2956b586c2e4ea8aac19294805da839\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47bee42773d94f89a8010e9ba95d1bb5 : SUCCESS in 2m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7f1d1b9c3624c3b9db91945affd3754 : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e334ccc0500f45889caf43e60db2a2c7 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4229941244124c22ac9df8500beb9734 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f01e1afc9e40426687b04852b4665ec4 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":6},{"id":"0bae46fb0876a3a6f7e6ad23417ffc9109b19be2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 13:53:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1f7a8d83e559a16c989664e9c316cddcd0717f3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 13:54:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb2560a4d0bc4bd2bed109a35ebfa1c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a6b855b6144edf9615231844ac11de : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8db63bb9da7d4e61bd440831ed2726d8 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6},{"id":"e30923d8717761e20248c0ec3be1234bb2f9ad23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 17:12:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0fad66668fb483c885dc74e02da9463\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9099f62ccc5b45a4b898d7f18599e57d : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de5b50ca278f47d0b2444581f3b7ddf2 : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d930fca08c735b848ca92ac0d553eeebf3eaa028","revisions":{"d25f4ab6583376225b6a5d19130cdb46aa375010":{"kind":"REWORK","_number":1,"created":"2024-04-22 14:35:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/916642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/916642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/916642/1"}}},"commit":{"parents":[{"commit":"52183ab561a4098a016292a18d5a9a3243549141","subject":"Merge \"Revert \"CI/Master only: pin opensearch{-dashboards}\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52183ab561a4098a016292a18d5a9a3243549141"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-22 14:35:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-22 14:35:17.000000000","tz":120},"subject":"CI: Add reno linting","message":"CI: Add reno linting\n\nChange-Id: I799d6bf7a381c377dee565c13b1732f7313520f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d25f4ab6583376225b6a5d19130cdb46aa375010"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d25f4ab6583376225b6a5d19130cdb46aa375010"}]},"branch":"refs/heads/master"},"1b77c9d0c88d3753ca4500cfe0eb14c775d9486e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-26 07:18:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/916642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/916642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/916642/2"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-22 14:35:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-26 07:18:49.000000000","tz":0},"subject":"CI: Add reno linting","message":"CI: Add reno linting\n\nChange-Id: I799d6bf7a381c377dee565c13b1732f7313520f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b77c9d0c88d3753ca4500cfe0eb14c775d9486e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b77c9d0c88d3753ca4500cfe0eb14c775d9486e"}]},"branch":"refs/heads/master","description":"Rebase"},"f8af7d7881ae169b203f6c3946cbfb70ac3bab76":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-27 20:21:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/916642/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/916642/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/916642/3"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-22 14:35:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 20:21:15.000000000","tz":0},"subject":"CI: Add reno linting","message":"CI: Add reno linting\n\nChange-Id: I799d6bf7a381c377dee565c13b1732f7313520f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8af7d7881ae169b203f6c3946cbfb70ac3bab76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8af7d7881ae169b203f6c3946cbfb70ac3bab76"}]},"branch":"refs/heads/master","description":"Rebase"},"967d9691f31ce2fe8b13fd07d1eea0e9616156bc":{"kind":"REWORK","_number":4,"created":"2024-08-27 20:57:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/916642/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/916642/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/916642/4"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-22 14:35:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 20:57:56.000000000","tz":0},"subject":"CI: Add reno linting","message":"CI: Add reno linting\n\nChange-Id: I799d6bf7a381c377dee565c13b1732f7313520f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/967d9691f31ce2fe8b13fd07d1eea0e9616156bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/967d9691f31ce2fe8b13fd07d1eea0e9616156bc"}]},"branch":"refs/heads/master"},"840f627901ee58fe8427d4cb79f68066e377b255":{"kind":"REWORK","_number":5,"created":"2024-08-30 14:22:09.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/42/916642/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/916642/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/916642/5"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-22 14:35:04.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-30 14:21:41.000000000","tz":120},"subject":"CI: Add reno linting","message":"CI: Add reno linting\n\nChange-Id: I799d6bf7a381c377dee565c13b1732f7313520f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/840f627901ee58fe8427d4cb79f68066e377b255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/840f627901ee58fe8427d4cb79f68066e377b255"}]},"branch":"refs/heads/master"},"d930fca08c735b848ca92ac0d553eeebf3eaa028":{"kind":"REWORK","_number":6,"created":"2024-11-08 11:28:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/42/916642/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/916642/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/916642/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/916642/6"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-22 14:35:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-08 11:28:21.000000000","tz":0},"subject":"CI: Add reno linting","message":"CI: Add reno linting\n\nChange-Id: I799d6bf7a381c377dee565c13b1732f7313520f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d930fca08c735b848ca92ac0d553eeebf3eaa028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d930fca08c735b848ca92ac0d553eeebf3eaa028"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
