)]}'
{"id":"openstack%2Fkolla~916767","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I456a5b8f66aa88a82fb54938e8df7195d127d9cd","project":"openstack/kolla","branch":"stable/2023.2","topic":"fix_2062572-stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-02 11:37:15.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-05-02 11:37:15.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-05-02 11:37:15.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-05-02 11:37:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I456a5b8f66aa88a82fb54938e8df7195d127d9cd","subject":"Nova: fix swtpm and swtpm-tools missing from deb installs","status":"MERGED","created":"2024-04-24 12:59:55.000000000","updated":"2024-05-02 11:38:30.000000000","submitted":"2024-05-02 11:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916767-fix_2062572-stable/2023.2","cherry_pick_of_change":916615,"cherry_pick_of_patch_set":1,"meta_rev_id":"d7757762b02af31dcc3c21e059e63d3ceae6ccf8","_number":916767,"virtual_id_number":916767,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 11:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":2,"date":"2024-05-02 09:57:56.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2024-04-24 13:00:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-25 12:08:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-05-02 10:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-02 10:36:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-05-02 10:03:35.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":36804,"name":"Michal Wyszkowski","display_name":"micwyszk","email":"michal.wyszkowski99@gmail.com","username":"micwyszk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 12:59:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":36804,"name":"Michal Wyszkowski","display_name":"micwyszk","email":"michal.wyszkowski99@gmail.com","username":"micwyszk"},"state":"CC"},{"updated":"2024-04-24 14:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-25 12:08:49.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-26 11:08:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-04-26 11:08:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-04-26 11:08:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-05-02 09:57:56.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2024-05-02 10:03:35.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-05-02 10:36:42.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"state":"REVIEWER"}],"messages":[{"id":"fe9f2fa8e68bc6e8518678de477fae9007212f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-24 12:59:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c31310bd80e2341165d46cc57396dc0d29a04fd8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-24 13:00:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8b2eb0740a7607d8dc4afe27987de52d2ca5119e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 14:42:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aef27e37db84af79fb54a8bb287ff1f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b485e751cf664868a99cf076a97d153a : SUCCESS in 24m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3a1172f14e0b41c6809b20e6ee9095cd : SUCCESS in 1h 04m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/738a8c8a7bc54282b75c9d2007f369e9 : SUCCESS in 19m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7770f9437be74627b94899749937aedf : SUCCESS in 31m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/913a55de64334176bbea345a73fe3ece : SUCCESS in 1h 03m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d062b403b5f746e7b841c4bb76469860 : SUCCESS in 1h 37m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cc22cb87dae41cdb00b45212ccb498a : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db5e55889d874023ad9aabd0ac489468 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20a8b2f12a8b466db420d9afb9f5d11c : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8f1762958444df0839ec76dfd951e57 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f03c5841659453db48cb6ca6f49ce9b : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46b877b182f24bf9be1568cfd919a68f : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/549502a3ba0149a79892e6998f8fc185 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a53e6f9ed8641809f20c359a31a1ed4 : SUCCESS in 3m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fec60eb991c4457cb336542298301b25 : SUCCESS in 22m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c22850aaa5fb4456a268c16ccf4c27f8 : SUCCESS in 32m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cbad1f0dc04d489a8c171964c96fc7be : SUCCESS in 1h 14m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/792b9faefcb9436188f16de2e2748e86 : SUCCESS in 1h 35m 13s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/2a85abfef97e44008584dbde42aef0e6 : SUCCESS in 1h 02m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82bd2e6b35c94228b1a808a36f04d6ca : SUCCESS in 25m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a9da8b467a1f4c2cb0725e7c3623cfac : SUCCESS in 1h 02m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f0f3e3589f83401490129123a92c2a32 : SUCCESS in 1h 01m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c22dac929f7f4c03b9ab1abd908899e8 : SUCCESS in 1h 25m 17s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/32a23fb13e0e4801bec539754a5f54dc : SUCCESS in 52m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"797b979eb5290636b441cc24b7390d292b37577c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:05:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd2552f5da14996862d9db2bb590ec7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/58e55c2ee792465d82ae6cd37b8c30c5 : SUCCESS in 3h 03m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f3d0cc46ff324d6d83798a531af02d55 : SUCCESS in 34m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9e4e6fa55d6744578bcb7d6be680f937 : SUCCESS in 2h 03m 17s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb916b2954954f8ab82f289c04a4d8fa : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/034cbea7a846460889eef9dedf25a12f : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b26589d353a7492a8a7ea6b572d45114 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/68b76bd4577c47eebf3abdea66b1a923 : SUCCESS in 4m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b8f3d994a2fe4668888f93c7e4690002 : SUCCESS in 36m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4f2d2b4a12214928b8d055956000835e : TIMED_OUT in 3h 03m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cb250a01e601a44f47d7080ab2f844665510114","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-25 12:08:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5d607e484d3e281fe88d2c09139710fc35af91c0","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2024-05-02 09:57:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0439a3ac011d2d1903f4aafa17136a42f02d5ee","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-05-02 10:03:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1de4099531532b89c1db8fa4866dae7f09dfde76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 10:03:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d80379f42a5519ee0819681a04e3d2c485ca73ad","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2024-05-02 10:36:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5fa6917b99ef2adeaa93efdcf8c7fdf256f96c66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:37:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b15124272a074ce68862bd731f126a74\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/48e2985d4d4f497081e5d68d548581a9 : SUCCESS in 18m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e1d42159c4b146019506486a884cb330 : SUCCESS in 28m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f12fac6565e04b35b4d5d68e5d3099e2 : SUCCESS in 49m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/39955276c1b2409984bdb9219cec0589 : SUCCESS in 1h 31m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c1f65c82bc0487b8eacd40d4d0eba94 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e51a1456175407e85490170567bf341 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f57329784a30467ea1a694b172bc57aa : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4599e40d18244eac8356454ca67e00aa : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e37339ae2e9340cb9f38b49d7d9bc834 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9139ca5a9d534b34b74bb552c0d0204f : SUCCESS in 3m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c0b922d105ae41de8f104a7f361c0c6c : SUCCESS in 20m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5fff8df311734949a6477dd6a0446bd1 : SUCCESS in 28m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d190cee2f97e4b08be4c831e4655a0e7 : SUCCESS in 1h 01m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8928ea69a206462ab2a6918d32b1f037 : SUCCESS in 1h 26m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d463651ad7b5404d8be0ad5a1a098f52 : SUCCESS in 17m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bddb1c87d25b49d18634a4bc600d217b : SUCCESS in 1h 01m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fe04cc931cb74dceae6e917b3b38124c : SUCCESS in 49m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b2df25197a4a42e7bbd2840e41cf72a7 : SUCCESS in 1h 11m 19s","accounts_in_message":[],"_revision_number":1},{"id":"51c57ce0acaaeee2507718f80d953ae85e9ae608","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:37:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d7757762b02af31dcc3c21e059e63d3ceae6ccf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:38:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0218b087400a4ac08997423fc142aff3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa1effcc7464c84a126e20f8e65a2d8 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a82acac3b0214069944581624ed5634c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c9a41d6f9bd3a6f273eee7d3a361233be903087","revisions":{"6c9a41d6f9bd3a6f273eee7d3a361233be903087":{"kind":"REWORK","_number":1,"created":"2024-04-24 12:59:55.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/916767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/916767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/916767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/916767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/916767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/916767/1"}}},"commit":{"parents":[{"commit":"8cdfa5c00e205635e28f401b8f8c88e560dd133a","subject":"Fix OpenSearch file ownership issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cdfa5c00e205635e28f401b8f8c88e560dd133a"}]}],"author":{"name":"Michal Wyszkowski","email":"michal.wyszkowski99@gmail.com","date":"2024-04-22 11:03:38.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-04-24 12:59:55.000000000","tz":0},"subject":"Nova: fix swtpm and swtpm-tools missing from deb installs","message":"Nova: fix swtpm and swtpm-tools missing from deb installs\n\nCloses-Bug: 2062572\nChange-Id: I456a5b8f66aa88a82fb54938e8df7195d127d9cd\n(cherry picked from commit abb969502f75517844bd44c25e0484c7495284d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c9a41d6f9bd3a6f273eee7d3a361233be903087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c9a41d6f9bd3a6f273eee7d3a361233be903087"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
