)]}'
{"id":"openstack%2Fkolla~916768","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I456a5b8f66aa88a82fb54938e8df7195d127d9cd","project":"openstack/kolla","branch":"stable/2023.1","topic":"fix_2062572-stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-06 12:08:22.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-05-06 12:08:22.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-05-06 12:08:22.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-05-06 12:08:22.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-05-06 12:08:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I456a5b8f66aa88a82fb54938e8df7195d127d9cd","subject":"Nova: fix swtpm and swtpm-tools missing from deb installs","status":"MERGED","created":"2024-04-24 13:00:26.000000000","updated":"2024-05-06 12:09:26.000000000","submitted":"2024-05-06 12:08:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916768-fix_2062572-stable/2023.1","cherry_pick_of_change":916615,"cherry_pick_of_patch_set":1,"meta_rev_id":"3b0ac73a2fe3de78872b6b80d5d2ab1eb0c24d5f","_number":916768,"virtual_id_number":916768,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-06 12:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-05-06 10:37:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-05-02 09:58:05.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2024-04-24 14:08:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-25 12:09:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-02 10:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-05-06 10:37:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":36804,"name":"Michal Wyszkowski","display_name":"micwyszk","email":"michal.wyszkowski99@gmail.com","username":"micwyszk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 13:00:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":36804,"name":"Michal Wyszkowski","display_name":"micwyszk","email":"michal.wyszkowski99@gmail.com","username":"micwyszk"},"state":"CC"},{"updated":"2024-04-24 14:42:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-25 12:09:06.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-26 11:07:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-04-26 11:07:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-04-26 11:07:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-05-02 09:58:05.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2024-05-02 10:41:21.000000000","updated_by":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"reviewer":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"state":"REVIEWER"},{"updated":"2024-05-02 13:17:10.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-05-06 10:37:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"404e9b2bf3f6505547bd675cd1f9c0c7b8b31489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-24 13:00:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8749bf4fdc411b081fd1f7777e91e47413a94fb0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-24 14:08:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad27d3e04bb79ccba94d808dde016b8ddcf09423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 14:42:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2d61d77e87343d790c0ce5d128d478a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a2be0b2335a4455bace165e8e176b431 : SUCCESS in 24m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c515c7f3b15d48f08e61595ab3376ea1 : SUCCESS in 1h 00m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5378a6ecfaff4f70bc468fadfdc00fd0 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5d588567d94d1ca0c3fb651aa5a954 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6491be96a30e4d0abeca66436526091b : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df2bd3c71e8e4b578aecd2a3cf2878f3 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd2a51bcc50461f87f172adf482b0d9 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a79e3c6b6e6541519dccddf28880e737 : SUCCESS in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/757c85a1475f4ecdb2b030ee0a88dc40 : SUCCESS in 23m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2dddbbc9f9ab4bc4b32894cdfdfcd43e : SUCCESS in 1h 03m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/57f94b8ca3414e4fb358e5145edcddf4 : SUCCESS in 1h 28m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3476ad555c9048789a7be77cb36e82a5 : SUCCESS in 22m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bb20a6590b3341e3b83f33aad2b622bc : SUCCESS in 50m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0f06cb9d01424a4cb1b07b7a1d861b20 : SUCCESS in 1h 17m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2dc45fc96b7a4aec24129f82925a397dee186443","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:05:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/215dbff08d3f4960b66523c022a05658\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1e28beeec8ba404e8fbe430b3cab22a7 : SUCCESS in 33m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f5ba072511bc40d68d9e04aad6a08fe2 : SUCCESS in 34m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4561c85e797427089b0109e91471b69 : SUCCESS in 4m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3fd2c18cd5bb443fa5bb1865b2a6dcf3 : SUCCESS in 35m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f001c7f167474184b8644604bc8be205 : TIMED_OUT in 3h 03m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a30b9dd916166e6b0b9e0f5beca3c7c73afe967","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-25 12:09:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"83ce4b246a2b7576e72469f3804914fffdb14b13","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2024-05-02 09:58:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e435d962524871063353cdbcfd5fbf3f49ade45","author":{"_account_id":26319,"name":"Michel Raabe","display_name":"Michel Raabe (rmichel)","email":"rmichel@devnu11.net","username":"rmichel","status":"openstack 💚\nceph 💚\n\nworking for B1 Systems GmbH"},"date":"2024-05-02 10:41:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b0062f6e1c56e31200b1c9e26294b39cc2420db0","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2024-05-02 13:16:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae62301b982fa49c7b08504911106bba6d70ceb4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-06 10:37:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e69f68d75cf6b5ad47c304a0d06581825bcdf7ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 10:37:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b582b53afaf140b5f49fdf135e1137a8887b09f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 12:08:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab3c74f0939b4102b3edf6d82c687a8b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fa7a97c5b4c745a48f1dbdbec246a807 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a2c2357cef40b0ac236f5bb53410f8 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02608a78001a4f0fbb217b95854ae2a5 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b8d3de173b4b2180d3aa44f814fc0f : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79080e7645934cb09edcf0986ebe51a1 : SUCCESS in 4m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6435037310b14c10813f7fde12afcd81 : SUCCESS in 24m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4a4ce97d18154ed8bd88b7bbafe4e995 : SUCCESS in 1h 05m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5062fba2bff44116a20bd6196ce143b7 : SUCCESS in 1h 29m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/29362b36691a4986a2f1e33b7d9ee160 : SUCCESS in 20m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e4b8d0b5b3e2447288052fe349f28802 : SUCCESS in 54m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9bdd692cad5c4a32870e11a9e1f11f81 : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0fe7a00b212ab8ceae5c047f9f581855e3e3d752","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 12:08:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3b0ac73a2fe3de78872b6b80d5d2ab1eb0c24d5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 12:09:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f91e1a8161b4ec3bab7db29a95c38a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b9bc4157906477d88e9414640684a06 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5a148b0b13e42e7af444231a5438043 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64eed98cf111567bb418b5c92024526981a8dd88","revisions":{"64eed98cf111567bb418b5c92024526981a8dd88":{"kind":"REWORK","_number":1,"created":"2024-04-24 13:00:26.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/68/916768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/916768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/916768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/916768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/916768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/916768/1"}}},"commit":{"parents":[{"commit":"7bd57039e3f2370954112b523af0dfadc047a389","subject":"Fix OpenSearch file ownership issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bd57039e3f2370954112b523af0dfadc047a389"}]}],"author":{"name":"Michal Wyszkowski","email":"michal.wyszkowski99@gmail.com","date":"2024-04-22 11:03:38.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-04-24 13:00:26.000000000","tz":0},"subject":"Nova: fix swtpm and swtpm-tools missing from deb installs","message":"Nova: fix swtpm and swtpm-tools missing from deb installs\n\nCloses-Bug: 2062572\nChange-Id: I456a5b8f66aa88a82fb54938e8df7195d127d9cd\n(cherry picked from commit abb969502f75517844bd44c25e0484c7495284d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64eed98cf111567bb418b5c92024526981a8dd88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64eed98cf111567bb418b5c92024526981a8dd88"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
