)]}'
{"id":"openstack%2Fkolla~916771","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b","project":"openstack/kolla","branch":"stable/2023.1","topic":"fix_2062072-stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-29 12:03:45.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-06-05 14:31:20.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-06-05 14:31:20.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-06-05 13:00:00.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b","subject":"swift-base: fix rsync missing for rpm installs","status":"MERGED","created":"2024-04-24 13:54:20.000000000","updated":"2024-06-05 14:32:16.000000000","submitted":"2024-06-05 14:31:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916771-fix_2062072-stable/2023.1","cherry_pick_of_change":916489,"cherry_pick_of_patch_set":4,"meta_rev_id":"8c77f7f577bf9b65d966032b277c6f24c3f28784","_number":916771,"virtual_id_number":916771,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-05 14:31:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-06-05 10:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-06-05 10:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-06-05 10:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-06-05 10:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2024-06-05 13:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2024-06-05 13:00:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2024-06-05 13:00:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 13:54:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2024-04-24 14:02:05.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-24 14:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 15:37:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-26 11:08:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-04-26 11:08:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-04-26 11:08:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-29 01:53:25.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"b018457d6c79e69ac64a178b088f73d6067c93ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-24 13:54:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"65aad023086949e4d7d4573119bdd1189763ef8e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-24 13:54:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a3c3cb2d66b38639aba6d449e442ff5c2f19986","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-24 14:02:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f0a1ae0fa4dc24f5dbdb3c1d66b01113cea436ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 14:49:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba569cae4d75484c9440d077e9b245bb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6915993cb76749d58be126928176af37 : SUCCESS in 42m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a92742efaa2643f4b43bd25c279cb168 : SUCCESS in 41m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dbcf1bcb6da477cb73c4c2d442c12e1 : SUCCESS in 3m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/95c7ec591bdf4f1d80a39473875a50e9 : SUCCESS in 42m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cda348e03f814a6fb67d302236393f89 : SUCCESS in 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e3f4e3d20ad3488b30215bcf2bc6baa8284b919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 15:37:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c95ac5c98629408a95e5e3b0e620a32e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7105341563d34a798aaf206d6c429458 : SUCCESS in 28m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c26368870ea24469a13a23181f918def : SUCCESS in 48m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/719d8c980c3d4dd4bfe3fd3d0c349fbc : SUCCESS in 22m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a01ddba7fd1e41c1978b694370c54d22 : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25eeb7b4dd3f4e7fb1354b9f08a20064 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/149ece62b2f64355abe342d2ecf1f3b3 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ada4e921a3e454a8f13895c5a5eaa2f : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98329175823c4a079ecf852be30e7a33 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/62f8f9923c774a89a0e55d52d51cf656 : SUCCESS in 20m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fce85b5feeb34a2fa6a5b1bcb1d855ff : SUCCESS in 1h 00m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a35c65929a5d46bbb83498db30726f3f : SUCCESS in 1h 16m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/88cd34fefd014fe3883ad3c411b7c3d9 : SUCCESS in 19m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b77c780cfe794015a41d629dd3bc15c6 : SUCCESS in 53m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3c81349eca314818b05a8e6cf5b019ee : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"069b0b7afa8a2d19d22c2b8fdf275f0a5528ed02","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-04-29 01:53:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1f300f4e55ef94b206c1ffb4b949ece8b2833482","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-29 12:03:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe89658fe95aea2c3cfcb76cda57079a88ccd158","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-05-29 14:57:09.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40305879799bfc0c4fa025c2d500d6d5f9249c28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 14:57:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e43ed0f8aedb0795b1873dc270381cc6e0e430c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 16:20:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4263afa24ca243b49fdaa3aacc70d8cc\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a89842165ee042cb939360fd05599f7f : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc30465d44964f3ea4fc6201a64aa3eb : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8972ff334a404d7aa5c696d845d1670d : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7e71c65f1f41448483a94267c2aac8 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e18c67e3ca84e1c83fc3554ef2214eb : SUCCESS in 3m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7330aa0f773b4e62b80a2e61bf6af8fd : FAILURE in 4m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1d91956f35e247c58732b5894d4f027a : SUCCESS in 1h 02m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bae9a8c017bb41c08a0237092e0fb034 : SUCCESS in 1h 05m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0e385fa5a25455ba4abc47f310f6cda : FAILURE in 3m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/729eee26823e47b590997e20cffefdf0 : SUCCESS in 42m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/77a7a57256c1451f8507b36b18726041 : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b0ede0ba4333f1481a86ae829f78ff2cfb13c68e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-29 19:55:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"498d94095e99b4cdffdc9e2b7209e8235aa596c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 20:05:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779da5b3f134443984ea20a0856c23a7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b6d30d5506c6466586b58a1eecfcf56d : FAILURE in 6m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/338131cb85364b6aa4e3a260b3526909 : FAILURE in 5m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbaf0339387240e9ab1a599db9049c77 : RETRY_LIMIT in 1m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4da4c0cd33c84f1da4da60d04be254fa : FAILURE in 7m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7d9f49b70ef249acb3b5a6ca6e45cac8 : FAILURE in 2m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98f6ee2af0f05ac59634a281e772f67860a65da1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 21:20:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e03cec648f740feb508435af183aaa1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/166e7f461be9451f992d5e8682a7d049 : FAILURE in 4m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/17669d8e06534148a3f8e15d4e7232ed : SUCCESS in 1h 01m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/453535716a354daaa5731b280017d4b6 : FAILURE in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136502e42fda49fabe0899e23bc46d6d : RETRY_LIMIT in 1m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d537a7dd7154e47bb3e7dfaea25bcb7 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72d81ffec8ff4b89a90843792e26af21 : RETRY_LIMIT in 1m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f563cd837c49c0adffcbfadf078edd : RETRY_LIMIT in 1m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2df34ed96b6640fa95450f05f8079c04 : RETRY_LIMIT in 1m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8a56a1f7fa3e4268a3bbbc864fe52b1a : FAILURE in 3m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fbfb2db196624686b994795def88fa09 : SUCCESS in 1h 02m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b5419868bdf640d789dbcc437d1ef0f0 : SUCCESS in 1h 20m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2dc6081fe82b4b7da047f01a2d7345b8 : RETRY_LIMIT in 1m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d6d03174df5448189098dc24cb7b033b : RETRY_LIMIT in 1m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fd7c3f8c5ed74a619661e10c1f8129c7 : RETRY_LIMIT in 1m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b3fd3dda714c0c7bb73c6ce38f02b38d1705d795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-05 10:52:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_14200\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"_revision_number":2},{"id":"56601857d93f378925e42179d7f36d2b24b8d0a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:29:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/626ddd4a909347f5b22cc9ded6dfd109\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/48aef0bba1524f45b762f1429f634ac0 : SUCCESS in 23m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9225d6d101ed4c8b88b527d962bcc215 : SUCCESS in 1h 05m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/06bd171d94af47cb8f4c0f473a3d2865 : SUCCESS in 26m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec0bcd0ad6fe493bb5cff7c1108bd56a : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db7fd02d5ea14dad80d97e552ac7974a : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4205930e6864412a73c9a2a93cabf48 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16393dc41fc345a4ade06d1fb46680c0 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f118aee6d1b4e629e060ccf8747ad1d : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ff93a983c79f4f219402d5565b802bfe : SUCCESS in 24m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c3c654463f7243e6bd7d6e8624c07a29 : SUCCESS in 57m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8ef4d41e96d84660a56fe76271da3bd6 : SUCCESS in 1h 27m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d64cee2459bd49f1a7caddffd4d73cba : SUCCESS in 24m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2bf67078c6f74e689107a0c0f5763457 : SUCCESS in 58m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fa7cbe9171ea412287b87dfecc972367 : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3103ea357856536b76891bba20657e32e3f0b454","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-05 13:00:00.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2d80d4eaaf313dae41bb53783c4508c72246b0e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 13:00:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"52df14ee25b941b54484cebc460bc53a5758d886","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:00:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fdf786d7a141e5adada8922f4207f9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5286bdc6d93f4f54ad32624bb2097654 : TIMED_OUT in 3h 04m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6cc3db9be9434c95a9149cc57d073787 : TIMED_OUT in 3h 03m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0753894c855442f2a5d2f869c2fa37ae : SUCCESS in 4m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d1053db2113a4d4c99c312b9715f5f3b : TIMED_OUT in 3h 05m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9b234467786d4bba9b3051000e936c4b : TIMED_OUT in 3h 03m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b8bcf7bc3cce837e0d30e35daf8d3d1c9a9baaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:31:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e195040b604f4e17acdbc4487e9036f7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6d23c022c351409eadef7ad7fdda88fb : SUCCESS in 21m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ec8fbfa28048d981b355d5fb415cf5 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95efca8ec2054b1e866b068c4fa677cc : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c15fab21486e45e88608d878afeb5620 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/441c208824fb4de2bf9c2b5bddc54b88 : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dbbe401da5364c6791ea5d9aed010698 : SUCCESS in 21m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8fd00729dec94fa58063e28266119a43 : SUCCESS in 56m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2d14c0ad10354c649de60b2410411794 : SUCCESS in 1h 23m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9897add0a71d42c18047b726a0de7431 : SUCCESS in 21m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d9acd15c1da5431da86105087cc24935 : SUCCESS in 53m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/644a110820f046de80f128fff0e53ad4 : SUCCESS in 1h 16m 03s","accounts_in_message":[],"_revision_number":2},{"id":"099a86733fedc58c95ddb8ea11e2577aa820c1c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:31:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8c77f7f577bf9b65d966032b277c6f24c3f28784","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:32:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c7191facad42009ecf6a4bf7532e99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74fc98f0ff14a37be587134f79a4494 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be3a301486e149d4a81e18d00d6fe671 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c0f6339a4ac85814dfd16fa623f898526342f77d","revisions":{"c5386180d12eb38c328dbb125ae10a137db9c135":{"kind":"REWORK","_number":1,"created":"2024-04-24 13:54:20.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/71/916771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/916771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/916771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/916771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/916771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/916771/1"}}},"commit":{"parents":[{"commit":"7bd57039e3f2370954112b523af0dfadc047a389","subject":"Fix OpenSearch file ownership issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bd57039e3f2370954112b523af0dfadc047a389"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-19 15:04:22.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-04-24 13:54:20.000000000","tz":0},"subject":"swift-base: fix rsync missing for rpm installs","message":"swift-base: fix rsync missing for rpm installs\n\nCloses-Bug: 2062072\n\nChange-Id: Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n(cherry picked from commit 7766abb4398d13355a3f76bc1b91f6c92d36111b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5386180d12eb38c328dbb125ae10a137db9c135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5386180d12eb38c328dbb125ae10a137db9c135"}]},"branch":"refs/heads/stable/2023.1"},"c0f6339a4ac85814dfd16fa623f898526342f77d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-05 10:52:18.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/71/916771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/916771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/916771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/916771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/916771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/916771/2"}}},"commit":{"parents":[{"commit":"b5c936a3a564e10b35b0e378bd6d5034f0db93b4","subject":"Pin requests to \u003c2.32 for docker sdk (2023.1 and older releases)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5c936a3a564e10b35b0e378bd6d5034f0db93b4"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-04-19 15:04:22.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-06-05 10:52:18.000000000","tz":0},"subject":"swift-base: fix rsync missing for rpm installs","message":"swift-base: fix rsync missing for rpm installs\n\nCloses-Bug: 2062072\n\nChange-Id: Id6e4bbe0aab2360c4e7e5f74fff6170bcc71080b\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n(cherry picked from commit 7766abb4398d13355a3f76bc1b91f6c92d36111b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0f6339a4ac85814dfd16fa623f898526342f77d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0f6339a4ac85814dfd16fa623f898526342f77d"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
