)]}'
{"id":"openstack%2Fkolla~917098","triplet_id":"openstack%2Fkolla~stable%2F2023.2~If39e045357563111219a008fa7a81a82ef906627","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-30 18:36:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If39e045357563111219a008fa7a81a82ef906627","subject":"nova-compute: Remove duplicate openvswitch package","status":"MERGED","created":"2024-04-29 13:06:49.000000000","updated":"2024-04-30 18:37:23.000000000","submitted":"2024-04-30 18:36:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917098","cherry_pick_of_change":917299,"cherry_pick_of_patch_set":1,"meta_rev_id":"35289501273e9fda348f9685decdb6c625ccfece","_number":917098,"virtual_id_number":917098,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-30 18:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-04-29 13:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-30 18:03:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-04-29 14:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-29 14:15:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-29 14:15:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-29 14:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 18:03:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"05b4bbade5f97807f65929db3dfa2209364b8ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-29 13:06:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"75cc595956df03c22dbd62897b417f1099db6d4f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-29 13:07:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4b2805e49e6f3fec25e5033b51ef576cd231c54","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-29 14:15:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0fac0200830fdcfed8de6f42cf80aec635dbaef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 14:47:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db7e548ff8754c6ba76b6e39de12b1d3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a62b139bcf814d7dabca4cd8965c1e2f : SUCCESS in 23m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d8afd1d92b77439b82f9fa11fd1b5320 : SUCCESS in 57m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/460d8d12f20444f2b287e38ef37bef15 : SUCCESS in 15m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/55a57aab23e641048d7e6517b1c3bb0f : SUCCESS in 26m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b35f0670de1f457fbb4f7dcfb68a0818 : SUCCESS in 1h 04m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/37eb2f6ac56b418c9755f9f1d5f36051 : SUCCESS in 1h 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4229fd37a694f8b9a93e56137cc58bf : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/134fd18d6f014f168fb6ce6d54000554 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faabdff5bdbd4484929c55b6881a5679 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eafdd3eeb6f444bca98d4682f0d2aea0 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa9a338157d1495f834ab1b2581e5fa4 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df6a70bd4fe348cdb1f842093a1b810b : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ef07480c9b431e9b0d640988b5630c : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/97bca7e0410f4058ae09d0a3ae82ff9a : SUCCESS in 21m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bfb29eb74f634384bb57d7059e7e9c6f : SUCCESS in 27m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c5b92ff822c44122b987abe20ac47392 : SUCCESS in 1h 01m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/49fb7fc2480b43cfa6ded1052f9fa5d4 : SUCCESS in 1h 35m 36s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/bd7972300a1e40c28a70942870fd2b76 : SUCCESS in 1h 02m 02s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bf52a61c4cfe45e4bdbb91afccc9d17d : SUCCESS in 18m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b1a62a3e6b34eb2aa609eca16d48160 : SUCCESS in 1h 16m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e630e4b0dba44a77964723ab2edf47c6 : SUCCESS in 47m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/860ae8a65276414fa795002f68b8511b : SUCCESS in 1h 09m 00s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/9a6c1123b4b8495690bc80f6f7df8abf : SUCCESS in 53m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86f89296fe7c6151bae9668bca1b63b89af7a8ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 14:47:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c83a86ce41ac454ae10d715f4eb6aff13b301dfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 16:11:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab7e642d19c945859acbd56f73c842be\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2e7cd92fca45473dbf58472cd0d588cf : SUCCESS in 32m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ad627b4c9b354c6db957017dcfa96e5d : SUCCESS in 3h 02m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7712862c5aa84996bbc04f6c9bffce27 : SUCCESS in 1h 06m 59s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de342cd8c807496888d8e4ff7ad92555 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a126c9dc33914daea86cd73aa5572f91 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97a3e4d561864df784c556332e851544 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/311a6bc0ccee42a9b81e0c2baf69d12b : SUCCESS in 4m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3e61f6bb0b9f43d5bf3527ec83b201bc : SUCCESS in 2h 43m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0086c6e04eb14daa802add9bc85fb935 : SUCCESS in 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7758f4a8370efba7550115eb8d1a96a9398b269c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 16:26:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c47abe1286ec4dc1885485b3afab201e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/864545c672ba4f62a0db3a4d8b41e5a2 : SUCCESS in 22m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c5a7190df581477fa8542e8961caa582 : SUCCESS in 29m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dce401f680584ec49d9e35e8a1ab63d8 : SUCCESS in 1h 01m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f51fcd87bd184264bb8a75ea93db820c : SUCCESS in 1h 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc80124397fb4e08850991a1501358f8 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bef972500194f6988614102f5ce9fa5 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1deb2db73687436d8fe3b7e3b6b04033 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94a6aea3fc7246ad888da33bf50fcceb : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa42f2a6cad84322b3f12dca30ed18be : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1deb636841624e4ca21836f719942efe : SUCCESS in 21m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e5c0be2411954749aa5720641c6f535b : SUCCESS in 28m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e659b77904aa4271a2feac8da2376a6e : SUCCESS in 1h 05m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5402866f73dd44619f8fb80a8bcb0f2d : SUCCESS in 1h 34m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/53c705d9e6754764962ec6e7eee8a23b : SUCCESS in 26m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6c931e983f91495e84a7c40819a29920 : SUCCESS in 47m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/30c5ecf86276493dafe6785c8b12ab53 : FAILURE in 15m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/93094dd4b4d741f995020fcb5cc8a9d2 : SUCCESS in 1h 28m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d941b98b90c13492adc9f43f78bc14b214a03be0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-30 14:53:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ccbdd7630d35311a6540de5eb58fe6958aacb7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 16:32:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd2b41aede644814b2e15fc4009a1c4e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/efc5a9da47e446afbe8ea3117330d8ac : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a51098ffcf684b8ca33e8235d41e9bba : SUCCESS in 1h 01m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/53c5c48c47df4fb4a923916912a710f2 : SUCCESS in 21m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d8ea44ff1f804060ae9837f4095cf980 : SUCCESS in 32m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1dd9cf7a437847239663ef9aaa708b53 : SUCCESS in 49m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f5e92176c7974d87a998061f35c03f76 : SUCCESS in 1h 31m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baaed4ddc2f74915972ac862bf64f843 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bd07b58f64e4cd88cd73754c1319963 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c780166a4b0f4566bcd8c9e8efc72dab : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/927d229188bb4c0b9c12024afd9d07c9 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd8d22368d0f499cadb8bd77f5f51972 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/121d19aa8d0640cdacbc6a5fd81ba013 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5942998bfd42bfbd353e0f46be9388 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/119d7ccc8db448708877eee85a673d11 : SUCCESS in 21m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9568338c3c714eb483a8708de3c21cbd : SUCCESS in 35m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/df233499f0984961998ea19222242fa7 : SUCCESS in 1h 10m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/78aaad4db21a44ab90bcdff531f1fa41 : SUCCESS in 1h 34m 13s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/29fb1aef22574b60ada2dafe4e2abfb8 : SUCCESS in 49m 40s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/51bba4dae7e14f8099708025a640bf18 : SUCCESS in 37m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d3cede425d2d4c50bcf749681d44b2cd : SUCCESS in 54m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4eca41710c544e2b89983cd11bcea1f0 : SUCCESS in 53m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/86de055220034d909a0d7ebd909d4f6c : SUCCESS in 1h 25m 40s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/1cebd56ef07b40989dedbc5d333fce1e : SUCCESS in 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbfbfa1ff66e000572e0f87a32b46672937abf65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 16:32:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"376ed01ad2d01b148a5a59056df6d519c5408313","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 16:45:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77bb6a8d8b43479483fc9ae28e606003\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c96186a78f034d43b648ecc09c6d6d10 : SUCCESS in 28m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/558b393cd2a94255b9c9294f135e5f70 : SUCCESS in 20m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc0798c6cfc7419dbe4d4f1f9ad854bf : SUCCESS in 1h 48m 19s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/713874a5963142948db460b5f349c81f : SUCCESS in 2m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a16f6dc12ed464b90a67f8fbccdcadb : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d83a13861cbd4d238a6ee8268b1077f5 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a293b896c28640af93c40d514d2e9872 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/39ab9c711ecd4902bd1c7bde857cfcc1 : SUCCESS in 26m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eec0f55400434decb3f5518bf29231f1 : SUCCESS in 1h 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cefa0bb1d05161a38b37e92e952e831cc665d3f7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-30 18:03:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8f5dfa46de4be7ecca36a82504fd47887ac28cc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 18:36:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c06e4d8eb046b38a382e124b03cafb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fd1e1740f0014fb193dfd23fcb4d31ec : SUCCESS in 22m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e046adfc2cc647ab855036374248e874 : SUCCESS in 33m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b3fd639a297b413cb01f46a2bc8de767 : SUCCESS in 1h 07m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0558a0ce84f749c7a65d46715eaea53f : SUCCESS in 1h 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805e7e0c042043bab76fff3de81e7ab0 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/701c97b31cd84185a8468e23828adf0e : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b4727289c134696914e5912c97a0581 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/346bca46af2a42dd927b70fe00b707dd : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02b5231bcd2b43ed8afbce4552d40422 : SUCCESS in 4m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/47c6180f502849e0bac94d242d514e91 : SUCCESS in 24m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2a22ae6e09ce40188219642beffdf19b : SUCCESS in 32m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/372da1c3806f42feba042ab6192045db : SUCCESS in 1h 08m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d0a29701a75940e6ae5bafc11526ef74 : SUCCESS in 1h 32m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e32ab8f03de4eeb8f39cd3936223e5b : SUCCESS in 30m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/427527f720ac471a892fbd74dc23efbf : SUCCESS in 49m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d864c4dce338452baac962360bf29a7e : SUCCESS in 1h 01m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/92599dbc7c0c4f98872f04f6a4261c46 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"82759183246104378f2cbdab2dfdf8062d38b8a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 18:36:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"35289501273e9fda348f9685decdb6c625ccfece","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 18:37:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5089fd1f03f24897b77056fb945ef6d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb32b9867fe748689d74d61a92a09443 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e38ef1251a09a28a027ac739f451882ef46e98d","revisions":{"9e38ef1251a09a28a027ac739f451882ef46e98d":{"kind":"REWORK","_number":1,"created":"2024-04-29 13:06:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/917098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/917098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/917098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/917098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/917098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/917098/1"}}},"commit":{"parents":[{"commit":"0bdc01d64083892df9d0c56251872364de6ce4b5","subject":"Revert Deprecation of Masakari","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0bdc01d64083892df9d0c56251872364de6ce4b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-29 09:41:43.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-29 13:06:49.000000000","tz":0},"subject":"nova-compute: Remove duplicate openvswitch package","message":"nova-compute: Remove duplicate openvswitch package\n\nopenvswitch package is already installed in nova-base from\nCentOS NFV SIG repos - and in nova-compute we don\u0027t enable this\nrepo - so we get openvswitch installed from RDO packages.\n\nCloses-Bug: #2064098\nChange-Id: If39e045357563111219a008fa7a81a82ef906627\n(cherry picked from commit de7939a9754cffafb77b518f9efa8be5af91d67f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e38ef1251a09a28a027ac739f451882ef46e98d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e38ef1251a09a28a027ac739f451882ef46e98d"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
