)]}'
{"id":"openstack%2Fkolla~917099","triplet_id":"openstack%2Fkolla~stable%2F2023.1~If39e045357563111219a008fa7a81a82ef906627","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-29 16:29:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If39e045357563111219a008fa7a81a82ef906627","subject":"nova-compute: Remove duplicate openvswitch package","status":"MERGED","created":"2024-04-29 13:06:59.000000000","updated":"2024-04-30 18:03:54.000000000","submitted":"2024-04-29 16:29:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917099","cherry_pick_of_change":917299,"cherry_pick_of_patch_set":1,"meta_rev_id":"e57781abac59aee88fa3870a26f516dcce7e72e6","_number":917099,"virtual_id_number":917099,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-29 16:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-04-29 13:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-04-30 18:03:54.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-04-29 14:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-29 14:15:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-29 14:15:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-29 14:47:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 18:03:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"0602fb2c474c05305941fc2a40278e558a7fe6bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-29 13:06:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eea0747faca0b9f0d06a8fe3f8c4fcac46b5ecce","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-29 13:07:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"520900bb969531f6a2795e8e54251aae09796579","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-29 14:15:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44964b0faa67d688399b2efb73203556a719c0aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 14:47:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d65398fc1144783bd8663ed470f56ed\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2217786c7cb740e19cec99d032d4acaa : SUCCESS in 26m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/02b35068b12c4afdbf599cefb01e6821 : SUCCESS in 50m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3b92547a3fd74cb895e0a4ca7130d54d : SUCCESS in 23m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869ee6551ef5413d88703cba60c75e78 : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c3fdffd070043cc86e499f1ba28ce72 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55c1605d0e354b56bd5a869f69a23ad8 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c6cfa014b9f43b2a8e2597748922744 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc8946d6735a47a18d96f9112bff605e : SUCCESS in 20m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/597b23bf7e194b1486ae99b4afebd02f : SUCCESS in 1h 05m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1de5c89fa4b84d3e9b6c878ee9ed1c7f : SUCCESS in 1h 23m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e9b1d55dcc64a24b1a772361146f59b : SUCCESS in 18m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/00a832f54fee4935b3059dd2c3ae6d61 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f1cf94af33194b9c9a309a0094ad4226 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5bd64fc234fc065d43a1b6cc2966c4a9927e30c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 14:48:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"232ec69ac0211237a80a856f915fd59043800b56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 15:51:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414a9b91fd5a4d4db7a705450ec50620\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d0d1ff81e5714c68a76d38c6b5e24115 : SUCCESS in 2h 42m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aa0d50fb2f4d4a21b87fcf6dd4ca60fc : SUCCESS in 29m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bed36d98c4044c594d694b7b63abb80 : SUCCESS in 5m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5645c0b4d1b34cc49d3564bdd28c62a0 : SUCCESS in 31m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7e560b3b229e4bd898dc8dd9c6161683 : SUCCESS in 2h 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d22d34d3daa7dcac5ef18e8890b872962da504b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 16:29:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11a4772d959242daadbc851e45b67c2f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c949f89a3b324ffab59f2c4e7e634375 : SUCCESS in 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b35e89bc8494200a504c5f427fa1f1b : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/162c364786e04693be726859fbdb5ed8 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8de8c71cf84344917074449c350c61 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/22599ce25d4847b6aae5bdc0f2041e88 : SUCCESS in 18m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d415bea061254ceb91e7aa5808f12b7c : SUCCESS in 59m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e48f4567c9f343f4903c69ee1259cc2b : SUCCESS in 1h 16m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b440de8f2e65480daca92632d051efb8 : SUCCESS in 19m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6ae0e489482c4ef9bc5f8d8144b9283a : SUCCESS in 42m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e32eb3ac7f804ce2a94a588f6f87e192 : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e2fb7a6ac60eecef54b95c74fea77a8c0dd64d88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 16:29:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff038b22e09f0f538349ec0c8679fe571945e8b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 16:30:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bbd27f8a3ee49bc8667330c20e74e8a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8882fe3b476b45b1b802ab10f46490a0 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e57781abac59aee88fa3870a26f516dcce7e72e6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-30 18:03:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef5472a2031611d7e3b2d4f20f24b3a68c658fd5","revisions":{"ef5472a2031611d7e3b2d4f20f24b3a68c658fd5":{"kind":"REWORK","_number":1,"created":"2024-04-29 13:06:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/917099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/917099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/917099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/917099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/917099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/917099/1"}}},"commit":{"parents":[{"commit":"7bd57039e3f2370954112b523af0dfadc047a389","subject":"Fix OpenSearch file ownership issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bd57039e3f2370954112b523af0dfadc047a389"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-29 09:41:43.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-29 13:06:59.000000000","tz":0},"subject":"nova-compute: Remove duplicate openvswitch package","message":"nova-compute: Remove duplicate openvswitch package\n\nopenvswitch package is already installed in nova-base from\nCentOS NFV SIG repos - and in nova-compute we don\u0027t enable this\nrepo - so we get openvswitch installed from RDO packages.\n\nCloses-Bug: #2064098\nChange-Id: If39e045357563111219a008fa7a81a82ef906627\n(cherry picked from commit de7939a9754cffafb77b518f9efa8be5af91d67f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef5472a2031611d7e3b2d4f20f24b3a68c658fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef5472a2031611d7e3b2d4f20f24b3a68c658fd5"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
