)]}'
{"id":"openstack%2Fkolla~917299","triplet_id":"openstack%2Fkolla~master~If39e045357563111219a008fa7a81a82ef906627","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-04-29 13:06:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If39e045357563111219a008fa7a81a82ef906627","subject":"nova-compute: Remove duplicate openvswitch package","status":"MERGED","created":"2024-04-29 09:44:27.000000000","updated":"2024-04-30 18:04:01.000000000","submitted":"2024-04-29 13:06:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917299","meta_rev_id":"573396a94268ead27e1c10dc6075c604927aaa39","_number":917299,"virtual_id_number":917299,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-29 13:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":1,"date":"2024-04-30 18:04:01.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-04-29 10:46:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-04-29 11:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2024-04-29 09:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-04-29 11:08:58.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-29 09:56:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-04-29 10:46:19.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-04-29 11:08:58.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-04-29 11:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 18:04:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"31abb45ada0278d14ffdd45159fecb92f155eed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-04-29 09:44:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"466bdd0fc21b89b85803ecf667dc2eb220703339","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-04-29 09:56:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be482166b73c335c99660133ca70b569eeaf826f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-04-29 10:46:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6e55e2785ab21b62478c751b72879109cd658d10","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-04-29 11:08:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51772e65bb287bb8e547da189ebbb19b3bab2e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 11:25:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419d01c2b4fb442995c40d9fcd8126d9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/de368d830fac4883aca2556173b83696 : SUCCESS in 23m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/248f06724c714e36afa3d5f559b4e4b3 : SUCCESS in 1h 08m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/547373bf14ae414fbd56a8add8fff204 : SUCCESS in 19m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2105690b093e499b97bcb8b3b5ba7fd2 : SUCCESS in 27m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e8a46ae1130143d5a5b99b9d85b83ed7 : SUCCESS in 1h 04m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2b16da623aea410c913d7eb7e74b88b8 : SUCCESS in 1h 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b94a9ff08ac420780f90dfbfc17996d : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13e51ea695f54a80a8ee05cc3b1b4faf : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/068406e1de744b9e97d27c17daf036c4 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/659d327dfe6648648343964c5782c680 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8e141ed134441a0bcb27b43563ffa26 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28fdf525e25e49e1b16e6bc02a318e79 : SUCCESS in 5m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/879d8640e49f49c1b7e0f547069caca6 : SUCCESS in 24m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e21866332be8422c8f0446bedfc998ed : SUCCESS in 30m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a9a7c096172c41fc83353c07e73c46c5 : SUCCESS in 1h 10m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3c3afacceb3f42ebadbf086590875279 : SUCCESS in 1h 17m 38s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/e6f00f428f5a4fd4b28722c1c07ea706 : SUCCESS in 59m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/91708fef5c9e4a3d913945807f4d0c9f : SUCCESS in 25m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5506b8ccc1bd4182a80fee5921ecab7c : SUCCESS in 53m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/86fa14e73b174cb09d576e906499efd7 : SUCCESS in 51m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5583db75ec6f419d91499def50922015 : SUCCESS in 1h 12m 36s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/3637564235564859a6bee20227702d38 : SUCCESS in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0438a1f1557371896ec634fde4eacbb72ab12a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 11:25:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4b052425fc79492a6019c3ce3237e87ab8c0ac0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 12:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d734ee876274016ba6e5161d3425fd3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ae4f6abe5a1a42e1927211a977be8c80 : SUCCESS in 2h 23m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/60b78aa2ea78401db3fc7aa14b44b6f1 : SUCCESS in 2h 18m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8947f36f69d04a4c99b1d782af64e430 : SUCCESS in 1h 06m 40s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9616ead471dd4cb2acc4e01d9517ed83 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c9f261618f5f4282a29b7fd7ad2d5e13 : SUCCESS in 3m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/48a9a7d7f868430cab5f4a1d3c1970b7 : SUCCESS in 2h 29m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f201b70d0e6c4aa1a05e6e84932f77ca : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"caaaa9768c94417881efa07400577580b43f544c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:06:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab16cd57bad4f50a0ed2628ecde3c20\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/00a7ef532bac456fb1ee0e10fdbe1d1c : SUCCESS in 18m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/02341346db654b1890d4683e0904c619 : SUCCESS in 33m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f64fdc5143594f379c1abf5ea45873e2 : SUCCESS in 1h 11m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/188fbcbe5951402196e644383941adb8 : SUCCESS in 1h 31m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36fa0d401c3241bcbfd2f36845e79beb : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b47a7258372f4668b98dfc230fb50a15 : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e067786574b24d75b87c9c5cc1a43a53 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafb40129f22428b9e632170c5c53fb7 : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f55c5fec46b24499a4fb94619fd2860b : SUCCESS in 22m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d28fa8695114654b9946eb719cde973 : SUCCESS in 27m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d2cbefc565f044a885e196865a55acb1 : SUCCESS in 55m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8c6eebedc10f49b3bf3d906a3bd336d3 : SUCCESS in 1h 34m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/014bc542df0546e8a022c0787b75a722 : SUCCESS in 24m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5a49df872f19495e84ad7f2d8136f3b6 : SUCCESS in 45m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c45d792241fd4d379540e3febfaae494 : SUCCESS in 42m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/191f88f16f054b06a0084851061746a9 : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2ef974f2bff139f4c8d34b69f42534ec444c6a95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:06:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"822482c699c0d626248cf788990ea32d95ead1da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 13:07:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d890156a1210452099f3878b805577b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c25c560c86f4497bac78bfdb3da3a8c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"573396a94268ead27e1c10dc6075c604927aaa39","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-30 18:04:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de7939a9754cffafb77b518f9efa8be5af91d67f","revisions":{"de7939a9754cffafb77b518f9efa8be5af91d67f":{"kind":"REWORK","_number":1,"created":"2024-04-29 09:44:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/917299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/917299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/917299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/917299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/917299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/917299/1"}}},"commit":{"parents":[{"commit":"c4225f35eef2f9ff590dfb1726579bfce2b1c1a0","subject":"Merge \"Move apt_prefs pins to respective images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4225f35eef2f9ff590dfb1726579bfce2b1c1a0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-29 09:41:43.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-04-29 09:44:25.000000000","tz":120},"subject":"nova-compute: Remove duplicate openvswitch package","message":"nova-compute: Remove duplicate openvswitch package\n\nopenvswitch package is already installed in nova-base from\nCentOS NFV SIG repos - and in nova-compute we don\u0027t enable this\nrepo - so we get openvswitch installed from RDO packages.\n\nCloses-Bug: #2064098\nChange-Id: If39e045357563111219a008fa7a81a82ef906627\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de7939a9754cffafb77b518f9efa8be5af91d67f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de7939a9754cffafb77b518f9efa8be5af91d67f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
